)]}'
{"id":"openstack%2Fpatrole~701055","triplet_id":"openstack%2Fpatrole~master~I80cff7328c47081f20abaf25396f340d1482ff20","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"I80cff7328c47081f20abaf25396f340d1482ff20","subject":"Do not skip nova API extensions policy test for rocky","status":"MERGED","created":"2020-01-03 19:09:36.000000000","updated":"2020-02-26 19:39:15.000000000","submitted":"2020-02-26 19:37:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701055-1582745868846-d1a6a7d1","meta_rev_id":"b52faca924392e425b4442cd04f335c0312e33de","_number":701055,"virtual_id_number":701055,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 19:37:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2020-01-28 02:59:25.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":2,"date":"2020-02-26 18:22:55.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":1,"date":"2020-02-26 18:22:55.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 02:59:25.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-02-26 19:37:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b949d4ceda775cf495f4791d48899281356059f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-03 19:09:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c7a86301276bc4487a4b561cd8f7a741c155126","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-03 19:26:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c57854b6d1d4ad4431fdcc73f6af1a3b46c78cb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 20:54:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e11f8b6b389347a89869cdccc59ae699 : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c998230115ce47d68ab080022d993770 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0730a2a73de945d191af5b67c1b7c5d4 : SUCCESS in 2m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20a5c15b58184d2ea2231fbbfac4ecb1 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ac84d440a4b40699ed9d72aa9e6d861 : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bdf0d290af414245829a992667f19040 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f47caa143483400da8a59d9fbc012cab : SUCCESS in 5m 26s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/481614a675ce48c092c457a173116ef5 : FAILURE in 1h 08m 57s\n- patrole-member https://zuul.opendev.org/t/openstack/build/609446c05cbe4a6aaf33bd941495245a : FAILURE in 1h 05m 29s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/2004a87f8f87439d8fe28cb7e7c9eb71 : FAILURE in 1h 21m 32s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/30f5dfb2e24a4af2a30d8661cc7bbf56 : SUCCESS in 1h 04m 59s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/81f0847cd11c4a1dab226337ae9a3d60 : SUCCESS in 1h 11m 21s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/fe46678641104feaa71ea91a51394443 : FAILURE in 1h 09m 53s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/eb0372d840c34f368af683847f7394ae : FAILURE in 1h 04m 18s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/24025e81dd4e4b25b208318025462619 : SUCCESS in 1h 07m 33s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/8b8e52fddf584a1eb5a8b6e2c7a5b3ac : FAILURE in 38m 40s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/b838796ec6be4b0e8815dad28864a5a6 : FAILURE in 41m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"988d9127ad81f1711d0a42070f4a7851b1b45a09","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-01-26 00:20:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4610bbd5ba362ef6e7775d212ac03caebd42b118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 01:41:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/698077ed9080476c85e0b48a59d31b59 : SUCCESS in 3m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60bac97bd6cf4c0db9b3c510aa0bbbfc : SUCCESS in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2219e52cd9f949489de66155e30ccef1 : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fb7e6fc125141ffaaa5feff7591adbb : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b441638f6ce246fd88fdb2a8f0479040 : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/000bdc83c91e4c6fa2fbcfea12fff05d : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4356d67b56134ecdad65aabb0b35bed9 : SUCCESS in 4m 56s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/ebc1243a54d94133a4cc23891e73e7ec : SUCCESS in 52m 19s\n- patrole-member https://zuul.opendev.org/t/openstack/build/d07beac0339a4916a384dd8327ae50ae : SUCCESS in 1h 07m 48s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/ce08eba93be2494183f582739346a019 : SUCCESS in 1h 06m 31s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/66c18651ed7541b0b3ee732a855107b8 : SUCCESS in 1h 08m 22s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/d354295d411f4fecbf3fadd0c178e12d : SUCCESS in 1h 11m 41s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/d5d99e4d5c944e98a10b4b7553ed396c : SUCCESS in 1h 15m 30s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/1a3c499865564249974198e1c6723119 : SUCCESS in 1h 01m 06s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/7c196496dc1246559fbbdd7aaf2800a2 : FAILURE in 1h 02m 44s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/a29cb6ac56b64e63a40e370fedc733b5 : FAILURE in 33m 23s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/38815746eeeb40388102f51a83f7d909 : FAILURE in 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c51d8aaf6df5a58a38eb94bba8d2d3eb6875997b","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-01-28 02:59:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ace8cde834131f987e69ba13c4c011c0156ce17e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-26 18:22:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nlet\u0027s get this in for new release with no py2 support.","accounts_in_message":[],"_revision_number":2},{"id":"f2bdb6d3a33b827021ae410a0be78ce13cf634d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 18:23:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"93a5ad2650426518083dd4f0a8ac46efb0f0b025","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 19:37:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fe69350eee84b3b9405c5abdf82c33d : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1682dcbd3b924cd9b56f1e7115d2fac4 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b41743ace8440498d3d3b716bb2d735 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a08680badd148f995fbe18baae59f47 : SUCCESS in 6m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ccfddf4e9a604303896f4c7273de61a5 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d0627e5153f4cebb0e3b0ee33705b13 : SUCCESS in 6m 56s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/979d13242976492190ead0aaad822f09 : SUCCESS in 56m 47s\n- patrole-member https://zuul.opendev.org/t/openstack/build/de81429e5cd948e7830214fdf9b4341d : SUCCESS in 1h 08m 50s","accounts_in_message":[],"_revision_number":2},{"id":"524762d4902c6672e6b6fd9cd37eaa9968c161e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 19:37:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b52faca924392e425b4442cd04f335c0312e33de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 19:39:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e5a7007b464aaba50bb4842adc8fe8 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7085622dbad260c8c78c594f805f1ed342ac8d1d","revisions":{"00904b4b6236da1145d05ab3434981d2af343fd7":{"kind":"REWORK","_number":1,"created":"2020-01-03 19:09:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/55/701055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/55/701055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/55/701055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/55/701055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/55/701055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/55/701055/1"}}},"commit":{"parents":[{"commit":"45f01b741754f91023f7038851fc018c8fbd538d","subject":"Merge \"Fix missing \"raise\" before cls.skipException\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/45f01b741754f91023f7038851fc018c8fbd538d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-03 19:08:08.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-03 19:08:08.000000000","tz":0},"subject":"Do not skip nova API extensions policy test for rocky","message":"Do not skip nova API extensions policy test for rocky\n\nNova API extensions policies were removed in stein but\nflag to skip the tests for those policies is not correctly\nset to False for Rocky.\n\nChange-Id: I80cff7328c47081f20abaf25396f340d1482ff20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/00904b4b6236da1145d05ab3434981d2af343fd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/00904b4b6236da1145d05ab3434981d2af343fd7"}]},"branch":"refs/heads/master"},"7085622dbad260c8c78c594f805f1ed342ac8d1d":{"kind":"REWORK","_number":2,"created":"2020-01-03 19:26:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/55/701055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/55/701055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/55/701055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/55/701055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/55/701055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/55/701055/2"}}},"commit":{"parents":[{"commit":"45f01b741754f91023f7038851fc018c8fbd538d","subject":"Merge \"Fix missing \"raise\" before cls.skipException\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/45f01b741754f91023f7038851fc018c8fbd538d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-03 19:08:08.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-03 19:26:35.000000000","tz":0},"subject":"Do not skip nova API extensions policy test for rocky","message":"Do not skip nova API extensions policy test for rocky\n\nNova API extensions policies were removed in stein but\nflag to skip the tests for those policies is not correctly\nset to False for Rocky.\n\nChange-Id: I80cff7328c47081f20abaf25396f340d1482ff20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7085622dbad260c8c78c594f805f1ed342ac8d1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7085622dbad260c8c78c594f805f1ed342ac8d1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
