)]}'
{"id":"openstack%2Fpatrole~701271","triplet_id":"openstack%2Fpatrole~master~I3655cf176f12685dd87c52d8e4979d8f8e000a38","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"I3655cf176f12685dd87c52d8e4979d8f8e000a38","subject":"Add Patrole gate job for stable/train","status":"MERGED","created":"2020-01-06 19:59:46.000000000","updated":"2020-05-08 18:15:32.000000000","submitted":"2020-05-08 18:10:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701271-1588961457863-d95910ec","meta_rev_id":"1411b22348274ff0a1a8a30616377d97f3563fde","_number":701271,"virtual_id_number":701271,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 18:10:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"recommended":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":1,"date":"2020-05-05 09:25:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":2,"date":"2020-05-05 00:53:01.000000000","_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2020-05-08 16:34:10.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2020-05-08 16:34:10.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 21:38:30.000000000","updated_by":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"reviewer":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"state":"REVIEWER"},{"updated":"2020-01-14 14:02:49.000000000","updated_by":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"reviewer":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"state":"REVIEWER"},{"updated":"2020-05-04 14:36:22.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-05-05 00:53:01.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-05-05 09:25:54.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-05-08 18:10:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14f95e5f9fc06997a767ee5ba59d26ed17cb88f9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-06 19:59:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07b0acaaba0a0226036aa16b5a95a54247dbb309","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-06 20:00:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7f235d9890b8b79e62d8daabd1bbe2d2e62de04e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-06 20:02:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2928f1bb8f439a3e07b21296b13502a26c67d8cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 21:28:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b23ef18b069b4305ab0c2f61e2956d46 : SUCCESS in 3m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/227a195600774e35b626d111982e451e : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c5575f92aa84acb88f9c175a4d6f9e3 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f5d9deb5cef41ed906ba350764c3698 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e9aba82bd23470680d96bf1a88d90b1 : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ced98726734a49ec9937bdc815713744 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d5f145807724c57b4f2627ed73f513e : SUCCESS in 5m 43s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/6c6cc96d64b74db1a1c77ea5359d8b5e : SUCCESS in 1h 01m 33s\n- patrole-member https://zuul.opendev.org/t/openstack/build/f18a9827a20747e2ba28b5da23d34a02 : SUCCESS in 1h 08m 43s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/8c0ddaaf4c2047658ddca63bb0afc7fa : SUCCESS in 1h 04m 44s (non-voting)\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/cd12f7717f9745acb5e82a29c2453958 : FAILURE in 1h 13m 57s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/c0360273b8194cb6ab32d317d820886c : SUCCESS in 1h 19m 59s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/19707295489846d7aca9829ca4e636fd : SUCCESS in 1h 14m 53s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/53454bbe17a44f739f5f0a44c894b1e6 : FAILURE in 1h 04m 57s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/dffb6cfc94fa46e3bdd0d775bc8a5962 : SUCCESS in 1h 10m 35s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/ca82cd2106624e269223e659316fa0d3 : SUCCESS in 1h 05m 35s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/78d075f98f964b598c7cdc7216966a00 : FAILURE in 39m 46s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/cf829d80901c4c63882e50a1eb8605a9 : FAILURE in 38m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"80686c73aabc30c6f0d2bd8a16a5fa8e8cae677b","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2020-01-06 21:38:30.000000000","message":"Patch Set 3:\n\nThe patrole-member-queens job is even failing the periodic checks on master:\nhttps://zuul.opendev.org/t/openstack/builds?job_name\u003dpatrole-member-queens\n\nAny ideas on how to fix these particular failures?","accounts_in_message":[],"_revision_number":3},{"id":"41881957a1f50028c5fbd62e5162679d8c180b14","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-07 18:14:41.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a66141255f5c15afc18db06db68e5052e5ea1f5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 19:43:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/502c756944b547cab7a5aa846b21a330 : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6e73179f1ba4eafbbcba8f0be5ca4b4 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05c8a23134b140f582863aa3337fc11c : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77cb99b085634a67abb8fe37b1555b9e : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01276966ce4947c89dad589b137bbef3 : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf437938e177455dad67270b07cbf39f : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b177d789b9834e288d97dccf08bd2fee : SUCCESS in 8m 29s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/caf1c182644b4f159e52b55082fe57f3 : SUCCESS in 1h 13m 56s\n- patrole-member https://zuul.opendev.org/t/openstack/build/1a6b293c408c426dad935ba694129278 : SUCCESS in 1h 11m 12s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/bebd70dc76864614a499d4b4d328669e : SUCCESS in 1h 08m 13s (non-voting)\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/3422cbc74ef646d4a5ccb8772cb3b295 : FAILURE in 1h 26m 25s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/106770bdfea641b88c5d4a2e9ef2fab8 : SUCCESS in 1h 15m 29s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/7bf32336c88840d9ad5a6eba4ecebbf2 : SUCCESS in 1h 17m 20s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/7da518ee606d4c9fa77bd511a78e7f8c : SUCCESS in 1h 04m 50s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/6518e5049c974f6aaa48d88a0e467ebc : SUCCESS in 1h 07m 26s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/85db7462329e4293879fcb147362c719 : FAILURE in 1h 20m 28s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/f70fb6f03dd24085a0e35bc1aaa0d9ce : FAILURE in 38m 03s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/bb4f4af608894a1291026d40372bddeb : FAILURE in 36m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7266a6185fd7ddff6dc22263fbceb0aed3869da0","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-01-12 20:11:22.000000000","message":"Patch Set 4:\n\nShould work after https://review.opendev.org/#/c/700949/8","accounts_in_message":[],"_revision_number":4},{"id":"53408b329f6ecd131a0277dbc944d5c715a9fd71","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-01-12 20:11:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ef9b0185a7dccae5cc91b3df72cb70ba69caeebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 20:29:14.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5dc71a4d7ab4d30b1a2834699b58f46 : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f684a2092eb4e85b630b011fe30b19e : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac1c29b373c64a48bd18ff96281ad943 : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8096f23e85cb457ab6ccea967d71200c : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16832edca2ef4a16a1c3d7290f198784 : SUCCESS in 2m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6f72681c1164f01ab95dfe12085a8ae : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c72c6e32b8347288beda1aeb38ee68a : SUCCESS in 6m 37s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/993ac3eef6ac46ffb21f290ba6588ec7 : FAILURE in 8m 17s\n- patrole-member https://zuul.opendev.org/t/openstack/build/497f2d81cccf4ddda9b678798c93d45e : FAILURE in 7m 59s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/ca75740c647d4156a179cf0c655c9a8e : FAILURE in 9m 12s (non-voting)\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/bf9698a042bb4e2d9c77c752ee0407f2 : FAILURE in 9m 00s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/c3b65ce5fac64d72a479a5554b6b6c35 : FAILURE in 12m 35s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/1fc86a6db93c4fbca441e58814bcdfca : FAILURE in 7m 48s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/38c1496832c54f1fa626161ab20e7303 : FAILURE in 10m 54s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/b420943596f44948828ec152594fa7c6 : FAILURE in 12m 54s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/26509fc81eb54612a5072521ba8d1971 : FAILURE in 11m 15s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/f20857fdd2f64792b87a544a62042afc : FAILURE in 8m 59s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/8cb7fc480803467dbd382e5480239370 : FAILURE in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"856c0d8e584dd128056e103340355733032f8308","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-14 14:51:45.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9193c90ec5c15f7607ef66e9d770cbd53c178b87","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-14 14:51:56.000000000","message":"Patch Set 4:\n\nlet\u0027s get the fresh resutls","accounts_in_message":[],"_revision_number":4},{"id":"5224a96eb6bf6fbf6e4a5fcaad941bd6a26150b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 16:37:54.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64b07f37d10f4d128e64caea8b89fa24 : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11abd82347cf49eeb44a82f737c4aca0 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82caa22488344c80a3b77b888ba84643 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0108836a5651437b9d57b149a4cbb986 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f41fde6b1954fe5809db441b9b7435b : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ecf2a7e161c24128a997458d27cec3d2 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/248d96323b3344c6a183d7310a906991 : SUCCESS in 7m 51s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/734ac9b69a474dd5859bec5e9a2b5bc1 : SUCCESS in 56m 32s\n- patrole-member https://zuul.opendev.org/t/openstack/build/9cabe2ba525c44f4b59d19d32bc13d24 : SUCCESS in 1h 12m 48s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/2d80a6d993084785b74eb16320a43164 : SUCCESS in 1h 17m 24s (non-voting)\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/91183c4a5f274ceca261e0ef02345460 : FAILURE in 1h 13m 19s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/32225abcda7647bdb7ab7d53d7a1e66d : SUCCESS in 1h 08m 13s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/7e984bcc8e4e4612af5425ca603ae603 : SUCCESS in 1h 07m 44s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/1bb0ef2290d54754b70b0285a04727df : SUCCESS in 1h 04m 00s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/0dbc847f4bc24c56a0fb616fa605e549 : SUCCESS in 1h 07m 37s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/73dd0b8d89e54670bb54cf010ac51911 : SUCCESS in 1h 06m 04s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/f6d7e5111f1f43eb87b55627ba1d1f42 : FAILURE in 46m 52s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/6f477b1e27ac40b2b3ac3b54d1a7ca61 : FAILURE in 38m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6117c1c4679217c5c8da0adf708cbda9533f25ea","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-01-26 00:23:31.000000000","message":"Patch Set 4:\n\nI\u0027m guessing that the following error is getting thrown:\n\nDetails: Policy action \"os_compute_api:os-services:list\" not found in policy files: [] or among registered policy in code defaults for nova service.\n\nBecause Nova uses \u0027system\u0027 scope type for listing services: https://github.com/openstack/nova/blob/3e7e2530f1d0470323916349e01a1f5b11686421/nova/policies/services.py#L49\n\nBut Patrole doesn\u0027et yet support this scope type, which is likely why this is failing. Besides that, since it is scope type, this won\u0027t work with usual admin user workflow upon which Patrole has relied.","accounts_in_message":[],"_revision_number":4},{"id":"a39f335fc5d4a9b0b57cc5926dc012d78bcedc57","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 01:27:33.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"675a694b7a8e0a14d24d1cbbf10baea4ea6d765c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 01:29:45.000000000","message":"Patch Set 5:\n\n\u003e I\u0027m guessing that the following error is getting thrown:\n \u003e \n \u003e Details: Policy action \"os_compute_api:os-services:list\" not found\n \u003e in policy files: [] or among registered policy in code defaults for\n \u003e nova service.\n \u003e \n \u003e Because Nova uses \u0027system\u0027 scope type for listing services:\n \u003e https://github.com/openstack/nova/blob/3e7e2530f1d0470323916349e01a1f5b11686421/nova/policies/services.py#L49\n \u003e \n \u003e But Patrole doesn\u0027et yet support this scope type, which is likely\n \u003e why this is failing. Besides that, since it is scope type, this\n \u003e won\u0027t work with usual admin user workflow upon which Patrole has\n \u003e relied.\n\nNova changes are in ussuri and train policy are unchanged and should keep working.","accounts_in_message":[],"_revision_number":5},{"id":"2fe079294d6e21faa7e5df9e27ab8987b8e9e434","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 01:33:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d2cf773fe4c87ad39e6f06a58eaffd61061e8e83","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 01:33:48.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"33f772fe6c7b73656e6c873dcb257094cdc222e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 02:52:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e06271fe18f44233a17cc5b5e295c0c3 : SUCCESS in 3m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ce956a5e231482e8690cf794e101dcd : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0c32ae3ca3e4200a89db7b59444d870 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c067b98a19ce46a4b95e084da7b3f570 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9bb451085864fc3bcccff039de9a2f8 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e726d77da35140fe9757336b30047ba3 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c9144ce395744aca68e3c0c04b6f5ee : SUCCESS in 5m 15s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/5ec43e31d9eb4d68bb5665dced48390f : SUCCESS in 1h 01m 21s\n- patrole-member https://zuul.opendev.org/t/openstack/build/8ecbfc4838a44911baf9fbb0ec078b13 : SUCCESS in 1h 13m 07s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/cd9ee28d1b224addbc76a2b13f12df97 : SUCCESS in 1h 05m 33s (non-voting)\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/e5196ac729c448d3a2cc942bb1e730c9 : FAILURE in 30m 53s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/d181ce4d6f674d15bac697155bff67ec : SUCCESS in 1h 07m 06s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/9ce1f99aaf144664aefea422007fb27a : SUCCESS in 58m 12s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/1b593594bb5e4fec9b5ddfa208368975 : FAILURE in 59m 10s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/dabcac4e1d5d4a7dac9aee674e89fad1 : FAILURE in 34m 02s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/da2982e24d7241b7814051ebd51adb8a : FAILURE in 36m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b89c1c0547a715270a8d05bb097a2c58eb06a6c3","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-05-04 14:36:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"07c9c763ff5f20e94ffd16f6c64d06b75fbfd740","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 15:57:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"da5a09a6398ebbfc6353f90cfe01fded9c202620","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 15:57:49.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"02d8ea34ed0e49d4c0690a7c5e8df1bd8eca29e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 17:28:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cbc4dd946564ca2a48adbcbdce06a19 : SUCCESS in 7m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68719c9ef9664c4b975a0b0c44672a74 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f20d1e087f3c4aac9dd463630cbbecc5 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3bb66749a074e32bb9c963f561644e0 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80379b86a378447ea369ccbf585da34c : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1d64fdc854b45e8b6c56aa53a6916b2 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5409113516064a9a903aaebb474aaf9b : SUCCESS in 6m 22s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/50682685d0ee447cb717476668f5e593 : SUCCESS in 1h 00m 20s\n- patrole-member https://zuul.opendev.org/t/openstack/build/109d0c904532422097e5489292e99596 : SUCCESS in 1h 13m 48s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/2369a14410e748a487228e497b80f1c0 : SUCCESS in 1h 07m 35s (non-voting)\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/38817120e16e47a78eeb5b58398f0ad0 : SUCCESS in 1h 09m 42s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/7914cad6574148c2aaf206dc6a83b93a : SUCCESS in 1h 11m 58s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/98de5e1b839c484fade3e26671fea504 : SUCCESS in 1h 07m 41s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/ad5b09ebf8634e7180cd789ef4341200 : FAILURE in 1h 03m 56s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/c3894cb534ae4d029bab29d0f7f29636 : FAILURE in 44m 59s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/622fcad8b16e401380597c37307c8d45 : FAILURE in 45m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"589bae2bcfdbc71b2f06445ae9680e51511fd2b5","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-05-05 00:53:01.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"5b4033d65e3f580f64a2d58da7d12fec6f9eaff8","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-05 09:25:54.000000000","message":"Patch Set 7: Code-Review+1\n\nThanks Gmann.","accounts_in_message":[],"_revision_number":7},{"id":"05704b328cb130b97f66e983c2e978e9c4579568","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-08 16:34:10.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nmerging this as it is green now.","accounts_in_message":[],"_revision_number":7},{"id":"c9a95441e26b6012d2a5a8c5602e89e4bdb6b73a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 16:34:31.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f24943dadaf78a93a2f5a10865ff38f41eeadad3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 18:10:57.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2be1f5834dde4a4f932b26f5126ea655 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9840f37121db4e178739203c34597b38 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d43638d229346a7b7bc515b303b232c : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71a26080958247deab0ad12e7eacb9a9 : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6459004d323449228d9f931f5874ea98 : SUCCESS in 14m 09s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/f9e6a6715e4049aba1aee65a9991e7e1 : SUCCESS in 57m 32s\n- patrole-member https://zuul.opendev.org/t/openstack/build/daa9052c6d1947e7a0eccebd81aba91c : SUCCESS in 1h 35m 16s","accounts_in_message":[],"_revision_number":7},{"id":"fb66491fcc5018eb25e1dc3e4cd459ea1a176b9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 18:10:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"1411b22348274ff0a1a8a30616377d97f3563fde","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 18:15:32.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b7a23440d74ab48510d8bd2e257297 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6d97d0f8cb55fc89588e5d340ff4d7a5af00655a","revisions":{"f622cc56517ea02f655a20ca7159c0b78ee1882b":{"kind":"REWORK","_number":1,"created":"2020-01-06 19:59:46.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/701271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/71/701271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/71/701271/1"}}},"commit":{"parents":[{"commit":"45f01b741754f91023f7038851fc018c8fbd538d","subject":"Merge \"Fix missing \"raise\" before cls.skipException\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/45f01b741754f91023f7038851fc018c8fbd538d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-06 19:58:59.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-06 19:58:59.000000000","tz":0},"subject":"Add Patrole gate job for stable/train","message":"Add Patrole gate job for stable/train\n\nNow we have stable/train branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Patrole testing for stable/train\nby adding new jobs running on stable/train version of\nopenstack.\n\nChange-Id: I3655cf176f12685dd87c52d8e4979d8f8e000a38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f622cc56517ea02f655a20ca7159c0b78ee1882b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f622cc56517ea02f655a20ca7159c0b78ee1882b"}]},"branch":"refs/heads/master"},"84bbe49a109213802700b99f1efcc63a66ca9738":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-06 20:00:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/701271/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/71/701271/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/71/701271/2"}}},"commit":{"parents":[{"commit":"f8923d1ddfd2608a5f402c7014f9e95321c43caa","subject":"fix: admin and member gates are broken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f8923d1ddfd2608a5f402c7014f9e95321c43caa"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-06 19:58:59.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-06 20:00:23.000000000","tz":0},"subject":"Add Patrole gate job for stable/train","message":"Add Patrole gate job for stable/train\n\nNow we have stable/train branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Patrole testing for stable/train\nby adding new jobs running on stable/train version of\nopenstack.\n\nChange-Id: I3655cf176f12685dd87c52d8e4979d8f8e000a38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/84bbe49a109213802700b99f1efcc63a66ca9738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/84bbe49a109213802700b99f1efcc63a66ca9738"}]},"branch":"refs/heads/master"},"58975edf863d01f9c1fc88453dfd00dcb2254630":{"kind":"REWORK","_number":3,"created":"2020-01-06 20:02:17.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/701271/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/71/701271/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/71/701271/3"}}},"commit":{"parents":[{"commit":"f8923d1ddfd2608a5f402c7014f9e95321c43caa","subject":"fix: admin and member gates are broken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f8923d1ddfd2608a5f402c7014f9e95321c43caa"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-06 19:58:59.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-06 20:02:03.000000000","tz":0},"subject":"Add Patrole gate job for stable/train","message":"Add Patrole gate job for stable/train\n\nNow we have stable/train branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Patrole testing for stable/train\nby adding new jobs running on stable/train version of\nopenstack.\n\nChange-Id: I3655cf176f12685dd87c52d8e4979d8f8e000a38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/58975edf863d01f9c1fc88453dfd00dcb2254630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/58975edf863d01f9c1fc88453dfd00dcb2254630"}]},"branch":"refs/heads/master"},"6a4673caedcbc6d41ff4eff7ec3989de1c027c30":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-01-07 18:14:41.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/701271/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/71/701271/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/71/701271/4"}}},"commit":{"parents":[{"commit":"f8923d1ddfd2608a5f402c7014f9e95321c43caa","subject":"fix: admin and member gates are broken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f8923d1ddfd2608a5f402c7014f9e95321c43caa"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-06 19:58:59.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-07 18:14:39.000000000","tz":0},"subject":"Add Patrole gate job for stable/train","message":"Add Patrole gate job for stable/train\n\nNow we have stable/train branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Patrole testing for stable/train\nby adding new jobs running on stable/train version of\nopenstack.\n\nDepends-On: https://review.opendev.org/#/c/701404/\n\nChange-Id: I3655cf176f12685dd87c52d8e4979d8f8e000a38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6a4673caedcbc6d41ff4eff7ec3989de1c027c30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6a4673caedcbc6d41ff4eff7ec3989de1c027c30"}]},"branch":"refs/heads/master"},"aefd8529dc85768fdfc47e90f4e62151314ed29b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-05-04 01:27:33.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/701271/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/71/701271/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/71/701271/5"}}},"commit":{"parents":[{"commit":"1fcf4bbd05b529d7d329c5720ac316f6ae8e4da0","subject":"Remove the stable rocky job from Patrole master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/1fcf4bbd05b529d7d329c5720ac316f6ae8e4da0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-06 19:58:59.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-04 01:27:33.000000000","tz":0},"subject":"Add Patrole gate job for stable/train","message":"Add Patrole gate job for stable/train\n\nNow we have stable/train branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Patrole testing for stable/train\nby adding new jobs running on stable/train version of\nopenstack.\n\nDepends-On: https://review.opendev.org/#/c/701404/\n\nChange-Id: I3655cf176f12685dd87c52d8e4979d8f8e000a38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/aefd8529dc85768fdfc47e90f4e62151314ed29b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/aefd8529dc85768fdfc47e90f4e62151314ed29b"}]},"branch":"refs/heads/master"},"18d392a0b5f31a63377c61a750ae6ec583e6e16e":{"kind":"REWORK","_number":6,"created":"2020-05-04 01:33:48.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/701271/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/71/701271/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/71/701271/6"}}},"commit":{"parents":[{"commit":"1fcf4bbd05b529d7d329c5720ac316f6ae8e4da0","subject":"Remove the stable rocky job from Patrole master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/1fcf4bbd05b529d7d329c5720ac316f6ae8e4da0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-06 19:58:59.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-04 01:33:44.000000000","tz":0},"subject":"Add Patrole gate job for stable/train","message":"Add Patrole gate job for stable/train\n\nNow we have stable/train branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Patrole testing for stable/train\nby adding new jobs running on stable/train version of\nopenstack.\n\nDepends-On: https://review.opendev.org/#/c/701404/\n\nChange-Id: I3655cf176f12685dd87c52d8e4979d8f8e000a38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/18d392a0b5f31a63377c61a750ae6ec583e6e16e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/18d392a0b5f31a63377c61a750ae6ec583e6e16e"}]},"branch":"refs/heads/master"},"6d97d0f8cb55fc89588e5d340ff4d7a5af00655a":{"kind":"REWORK","_number":7,"created":"2020-05-04 15:57:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/701271/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/71/701271/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/701271/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/71/701271/7"}}},"commit":{"parents":[{"commit":"1fcf4bbd05b529d7d329c5720ac316f6ae8e4da0","subject":"Remove the stable rocky job from Patrole master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/1fcf4bbd05b529d7d329c5720ac316f6ae8e4da0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-06 19:58:59.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-04 15:57:47.000000000","tz":0},"subject":"Add Patrole gate job for stable/train","message":"Add Patrole gate job for stable/train\n\nNow we have stable/train branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Patrole testing for stable/train\nby adding new jobs running on stable/train version of\nopenstack.\n\nDepends-On: https://review.opendev.org/#/c/701404/\n\nChange-Id: I3655cf176f12685dd87c52d8e4979d8f8e000a38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6d97d0f8cb55fc89588e5d340ff4d7a5af00655a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6d97d0f8cb55fc89588e5d340ff4d7a5af00655a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
