)]}'
{"id":"openstack%2Fpatrole~716765","triplet_id":"openstack%2Fpatrole~master~Ic540a42be0b05fc7c53c7ca78f6ff8e5725340e1","project":"openstack/patrole","branch":"master","topic":"bp/policy-defaults-refresh","hashtags":[],"change_id":"Ic540a42be0b05fc7c53c7ca78f6ff8e5725340e1","subject":"Update compute hypervisor tests to adopt new policies","status":"MERGED","created":"2020-04-01 20:23:29.000000000","updated":"2020-04-02 01:52:32.000000000","submitted":"2020-04-02 01:51:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716765-1585792263198-d25266a9","meta_rev_id":"b7f2391539949ca2522d65fe14383167feba4c89","_number":716765,"virtual_id_number":716765,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 01:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2020-04-02 00:28:53.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-02 00:42:14.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-02 00:42:14.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 00:28:53.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-04-02 01:51:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9eb8473755cf4fd3003990591281e906d02a2de5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-01 20:23:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49a2fdbc59b4f809b46093edebf777da1bff2662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 21:41:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/030236d8b5ba4f78b4906853c6885a67 : SUCCESS in 3m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/718135aea8e0441186edb01023281222 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1af1fea2f2a4b6e87a010864e2e09b6 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfb7c7d30bc64695851e71601da7d239 : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34347709340b4e4db026db0bdfe03a63 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5624432894e415f9de12fabbfa80fe8 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89a365544bfd4109b612a1879bfbe53a : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b21666e20f1141f9997e4592ccb0f426 : SUCCESS in 3m 42s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/87f42014b79f4d3284e26279ac4dfba1 : FAILURE in 53m 58s\n- patrole-member https://zuul.opendev.org/t/openstack/build/701d5e5388404398bc4981eb785b039e : FAILURE in 1h 03m 19s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/979f643aa4ed4e1ea69e924df005e198 : FAILURE in 1h 06m 04s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/45955893c6c74873b85911240ed84a8a : FAILURE in 1h 12m 54s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/8ebadb65b34e42b6a2b97944b061e908 : SUCCESS in 1h 12m 57s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/de726a792a9543508f4447fb6bf4f548 : FAILURE in 59m 06s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/271f67c214fb49829ac963c0f1f8b709 : SUCCESS in 1h 07m 53s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/73e61ccdbbd24873af1b2f387a95a2ee : FAILURE in 34m 19s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/5eb4773d46374899b4ad711efc652d3e : FAILURE in 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf87e21d9c3544803c3d7cc3a7034b9742ee47ff","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-01 22:59:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d187f89f7cdd15d20bf312647ff6598d77cde938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 00:24:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8af593a37ad54f27b3902b11458ff0a8 : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b1a9cb6e11e4764a2faf217244da9fc : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27e3826ebcb14272b5c3c20bb94010bf : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0d8d0c720cc499e95d1b656a805cc9c : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c6e45dff0a04263aea4f7757c7c2773 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66e509d6dd7f472a9bedc40b5ed85d87 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc5127bfaff1453fb2e57b2d529aac0f : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2d217d3f4974203a891c8346378b5f5 : SUCCESS in 4m 27s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/6b2df9f3b9a34d639daf2cbddde57f65 : SUCCESS in 52m 48s\n- patrole-member https://zuul.opendev.org/t/openstack/build/5ee333b4fd024fd2ad092f900c611762 : SUCCESS in 1h 14m 52s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/1363527f030b4c87932dc47fda974b7c : SUCCESS in 1h 16m 44s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/315398c17b774605a4599879b99e81b9 : SUCCESS in 1h 20m 55s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/5ba4f6b923584b94a360a927b89bb52a : SUCCESS in 1h 19m 26s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/a205c05a6d2c4f1b906126188019c9ce : SUCCESS in 1h 09m 16s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/2ecc138445684659b7e1c2ec93aac37c : SUCCESS in 1h 09m 49s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/9ae48e04dba6414aa788b653c006f26e : FAILURE in 33m 01s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/5122860b1f9d4afea3658d9a3aebca9c : FAILURE in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7920a216907285fce95506c22036ad5044b696c","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-04-02 00:28:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f4213926bf6379785225cd9a3c9e1eca875a7046","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-02 00:42:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n\u003e Patch Set 2: Code-Review+2\n\nyou can +A as per new policy :). I am just doing by myself for this before other failue happens.","accounts_in_message":[],"_revision_number":2},{"id":"236c465683f398c129b818ae0f7486ba4637cb7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 00:42:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"622e2ca9bc585548011572bb2c68e9b36ccb742e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 01:51:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad54ed3f52304cc7a1a993240cc60d50 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfe88bb0112449a491bc3a71fc3f19b8 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe44a22d4a9e481fa73037f7727502df : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e3f54b594b049c584f2146caf1f9cd2 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4452840f3cbb4829b7608e57c791d4f7 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/435572791d6145049dc4bbe0a4286ace : SUCCESS in 3m 46s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/15dfab6e4c4f432ea94996af84cb6af4 : SUCCESS in 51m 00s\n- patrole-member https://zuul.opendev.org/t/openstack/build/bb114863e257488fb9cbe29ba4f2518f : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":2},{"id":"764682c43afc02f2ffc3db60a076c290f96ef151","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 01:51:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"79ecde884bcbf152c13b068293e8ef05e5215ed5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 01:52:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed1718197019469ab50bb1f395936c69 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d30cda9e94b4011966bd087f46d3f72 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"84cb426c1649f888a2a9b75f4efc4b1141d6eed7","revisions":{"d0c2a655867a0a6ea043195d849440e6ea6ce4fe":{"kind":"REWORK","_number":1,"created":"2020-04-01 20:23:29.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/65/716765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/65/716765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/65/716765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/65/716765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/65/716765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/65/716765/1"}}},"commit":{"parents":[{"commit":"be0154c17d1b88c867f3d837fd425139b44b9787","subject":"Update compute instance-usage and agents tests to adopt new policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/be0154c17d1b88c867f3d837fd425139b44b9787"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-01 15:21:23.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-01 15:21:23.000000000","tz":-300},"subject":"Update compute hypervisor tests to adopt new policies","message":"Update compute hypervisor tests to adopt new policies\n\nNova is moving to new policy defaults in ussuri[1] where\nfew polciies are made more granular to adopt the new defaults.\n\nWith granularity in few policies make change in policy name so\nwe have update the patrole tests to start checking against the\nnew policy names from ussuri onwards.\n\nThis commit updates the hypervisors policy tests\nto move to new policies from ussuri onwards.\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html\n\nChange-Id: Ic540a42be0b05fc7c53c7ca78f6ff8e5725340e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/d0c2a655867a0a6ea043195d849440e6ea6ce4fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/d0c2a655867a0a6ea043195d849440e6ea6ce4fe"}]},"branch":"refs/heads/master"},"84cb426c1649f888a2a9b75f4efc4b1141d6eed7":{"kind":"REWORK","_number":2,"created":"2020-04-01 22:59:24.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/65/716765/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/65/716765/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/65/716765/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/65/716765/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/65/716765/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/65/716765/2"}}},"commit":{"parents":[{"commit":"be0154c17d1b88c867f3d837fd425139b44b9787","subject":"Update compute instance-usage and agents tests to adopt new policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/be0154c17d1b88c867f3d837fd425139b44b9787"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-01 15:21:23.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-01 22:59:21.000000000","tz":0},"subject":"Update compute hypervisor tests to adopt new policies","message":"Update compute hypervisor tests to adopt new policies\n\nNova is moving to new policy defaults in ussuri[1] where\nfew polciies are made more granular to adopt the new defaults.\n\nWith granularity in few policies make change in policy name so\nwe have update the patrole tests to start checking against the\nnew policy names from ussuri onwards.\n\nThis commit updates the hypervisors policy tests\nto move to new policies from ussuri onwards.\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html\n\nChange-Id: Ic540a42be0b05fc7c53c7ca78f6ff8e5725340e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/84cb426c1649f888a2a9b75f4efc4b1141d6eed7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/84cb426c1649f888a2a9b75f4efc4b1141d6eed7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
