)]}'
{"id":"openstack%2Fpatrole~720601","triplet_id":"openstack%2Fpatrole~master~If661299231d548ce40a2e340b1ddb9ebe8d3f964","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"If661299231d548ce40a2e340b1ddb9ebe8d3f964","subject":"Gate fix and update compute tests to adopt new policies","status":"MERGED","created":"2020-04-16 21:55:52.000000000","updated":"2020-04-23 02:47:19.000000000","submitted":"2020-04-23 02:45:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720601-1587609928367-fbe8d8ff","meta_rev_id":"cbf241b8d064e93245da8c04325b63ba08f4888d","_number":720601,"virtual_id_number":720601,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 02:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":2,"date":"2020-04-22 23:19:21.000000000","_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-23 01:33:41.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"date":"2020-04-22 04:03:58.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-23 01:33:41.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 04:03:58.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-04-22 23:19:21.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-04-23 02:45:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"33e2719e2cfcde6cd9240a4d8ca3e5e5778cce67","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-16 21:55:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3c5ab93cf416e4fd5ca889ad21816dd6ed899d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 00:25:58.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fee0e80abd34748a18f779c94aea930 : SUCCESS in 3m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f47000275c7429ebf5255c3869afebf : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e6eb633341844eaab062f8a33215e4f : FAILURE in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcd0616a323e4d7ebe3ba14c7759c67a : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1307cd1f34fc47b48eb8392f07b32ea5 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2569ee364500441d97bf85393dec5b67 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92bd7d566f004397abea88e893e2f34b : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/feecec75c1e141c5b8663598bb665cc0 : SUCCESS in 4m 55s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/aa4a224cab324ceb8f09ad2f76a73149 : FAILURE in 1h 04m 32s\n- patrole-member https://zuul.opendev.org/t/openstack/build/20f79af85de548418928b256bf9230fc : FAILURE in 1h 14m 36s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/87f87a26ecd34a9781337329ad5d2d28 : FAILURE in 1h 09m 05s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/fb40294b25944c6582cbb6404aa8bc88 : SUCCESS in 1h 08m 20s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/a89e35629f874932ae24f326b1c2b48a : SUCCESS in 1h 07m 38s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/b60768ce41ae4d2cb437a6b92ece9268 : SUCCESS in 56m 39s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/9222e47a81d049e2959bda11fef87c2f : FAILURE in 58m 24s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/5286062065474a0d8351612be20af394 : FAILURE in 34m 24s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/ce842dd0839044f4bcfccc876532fa81 : FAILURE in 34m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e1305c1dafb2f39bfd2c333b980a40fb4aed685","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-21 17:11:58.000000000","message":"Patch Set 1: Workflow-1\n\nneed to fix the glance failure aslo","accounts_in_message":[],"_revision_number":1},{"id":"a37405628e2dcf8a9e9ad342d195f2d53da5d03f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-21 21:41:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d21c7d3f55082d66fc539de8442ec3b55db26254","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-21 22:53:07.000000000","message":"Patch Set 2: Workflow-1\n\nthis is still failing","accounts_in_message":[],"_revision_number":2},{"id":"6bf2aa0c5be1997c15b0428720b53a8e2ae1988f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 23:04:51.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1353ec1f945c442481d73953ec68513d : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8f4ba5fa6044469a66d9457a58ea8e3 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc6fe4ea05d9461e85535a14d6d8d8f3 : FAILURE in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/445e010a30084759b71c8da00295295f : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16c289f39f074c90bd8a38d035abc7c5 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b1f70d3e40e4f7ba5aa36c53196e151 : SUCCESS in 4m 01s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/70ced1fd98314f8d9810db696784bbfb : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f94a9cf1cf34c6e935ba8f87797ec9d : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5b9aef00145475096302b62096afe55 : SUCCESS in 4m 43s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/de96e1e100404c959ccf7b5e54a3a03c : FAILURE in 57m 08s\n- patrole-member https://zuul.opendev.org/t/openstack/build/fd649a8e7b7a4310827af0947c6d0da5 : FAILURE in 1h 10m 23s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/819b54965f504966860519d9e3446065 : FAILURE in 1h 06m 01s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/b05de236797f4d7ea14e405543709ca2 : SUCCESS in 1h 17m 16s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/da971626c2c74a85b3411d6ed1ba8978 : SUCCESS in 1h 14m 24s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/daf67684aeb84c8eaf2ac0074c112e89 : SUCCESS in 1h 03m 54s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/da63eba213c94e948de6db1e16f60d06 : FAILURE in 59m 15s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/c0b3aeced60541c589dbecab7117acfb : FAILURE in 37m 13s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/f6ab4be82e2b4e6bb5d89cdc8492a6b0 : FAILURE in 35m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47e0243678b49868c7cc184e1d052a09c7ad7ef8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-21 23:34:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"260a43e11bfadc4eb5470de5ea458b0555b27abe","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-22 00:28:25.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"8a1850d491a84e4272eae8d93acf2de84bb49584","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-22 00:43:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c36cec77d6f4777a5f3bb092d845c8aa9b429472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 02:04:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8339b62a714f4fdf91a0fc894c3f6009 : SUCCESS in 3m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0dc02c0d48d40f4beb2bf01ba270082 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29eaee811d444d58a5daec8c9b3a2816 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3c95a8892aa4bc98f12b6f6dc809b29 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0eb92090eb3449979f48b60def2bf287 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/561c601129b74c22bee49c36894a5e5f : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cecb7eec1214297a3f547ef91034fe4 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2eacd1c31604bbe87eb046398b1e657 : SUCCESS in 3m 26s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/a908eca39fba44a9a3e8ee205f457650 : FAILURE in 46m 17s\n- patrole-member https://zuul.opendev.org/t/openstack/build/b5fd4309f6d44659979d0f1fc2dd95dd : FAILURE in 51m 55s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/a70b8f449a9a4415b247bfa28f02fece : FAILURE in 53m 46s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/a9bb81b7f0a94c688f051f8bcdc17075 : FAILURE in 1h 10m 59s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/b1e39f117cfc472eaa88c2aa567c7aac : SUCCESS in 1h 15m 31s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/e3bf24d59f1f42a0b9f0728b1d08382f : FAILURE in 55m 09s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/7b0eb3bcd3bb4e5b80db38bb05ed3a76 : FAILURE in 54m 57s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/c1718217a71b4090b8051d93da16679f : FAILURE in 37m 41s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/9f4321481fb74076b528a95a3694d29c : FAILURE in 35m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6ed9aeb1eaa2d0d5c48a01557c229df79daac2a1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-22 02:57:46.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"cd6006106034f6b9113e4e9c5d4f30edb0148e50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 04:12:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7356c9ad2575416485b40fec4956ef9a : SUCCESS in 4m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca6bd48132714eabb17422247deb57cc : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/800b999d926743b7ba155ea686f1c403 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6ab00b3d1c4407d9050d17a1b47ab8b : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd351c848a784d078a3300d9a08fb8c1 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c4e5efa40ee4776905f40d5a25b5c8e : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/264745fee17d462fb0825a45e0967362 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14fba111210d4f48acb730212612bd22 : SUCCESS in 4m 19s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/1486e2eb3a304671b275a1654168ae72 : SUCCESS in 1h 01m 47s\n- patrole-member https://zuul.opendev.org/t/openstack/build/f27ad7af847c43cbbcb82f0b28a7e6c3 : SUCCESS in 1h 03m 12s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/c2af11e2e5764d9bb5a34c46f49ce0c5 : SUCCESS in 1h 14m 22s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/1e43e7adca2e4ca389b891dacc94b293 : SUCCESS in 1h 11m 01s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/8c813bba2c2244e3857c4368719fd98c : SUCCESS in 1h 05m 01s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/e3cbc8cd9ba74b87aa3ed8771c1fd8ef : SUCCESS in 59m 00s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/4b60c675b91d47feb369fc4be53b6aa7 : FAILURE in 1h 08m 57s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/7a3153581299407792f214504a7c36a3 : FAILURE in 34m 45s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/4b11bba382eb49e0bb294121b03625f8 : FAILURE in 33m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7db337f9d06476aa2d8e36ba0af18aa9ed8dea27","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-04-22 23:19:21.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ceb4ad78448705cd56bd01897d7c9f426ea37241","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-23 00:39:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c5f15631874165d59b68f34c53fcf46629f855a8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-23 01:33:41.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nlet\u0027s merge this to make Ussuri release which is due tomorrow.","accounts_in_message":[],"_revision_number":5},{"id":"6fe04e08620990cd85caed8ff6a53f94bdfdaf39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 01:33:53.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"804eabcc0e54f1a70cce032119151a51ce4ea346","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 02:45:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"aeb20a8ad81fa90cc0c2f6ef224d8ca9f477e94c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 02:45:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/678306c152fc4301998dec68689d6f1c : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc2b9ed49d074ab78c20e5bab1a1ce5d : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d689ac03d104ceead9201240a294410 : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4af21705874b46508c3c25031c6f978c : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/131e01ed9a6a49cd99c16ba75c6b9967 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/468de4fe1f154bce8f3b981e6751f9eb : SUCCESS in 4m 24s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/854bfaaed4224a8b98747a309ba39a95 : SUCCESS in 51m 12s\n- patrole-member https://zuul.opendev.org/t/openstack/build/34aad32f9e9c40dcbdaee913e5de9b98 : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":5},{"id":"cbf241b8d064e93245da8c04325b63ba08f4888d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 02:47:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c799b7a825746849a685ef4b01200bf : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47ec6ab894b84caaae301067a7346ffa : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ee53f843fcb19815b94c80b922968fae72e5f789","revisions":{"5225c9b522faa5191f5c47a8491f47e8f0d12784":{"kind":"REWORK","_number":1,"created":"2020-04-16 21:55:52.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/01/720601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/01/720601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/01/720601/1"}}},"commit":{"parents":[{"commit":"2aad8081840452c7dabee870a86c95d22f18323a","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2aad8081840452c7dabee870a86c95d22f18323a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-16 16:52:57.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-16 16:54:13.000000000","tz":-300},"subject":"Update compute tests to adopt new policies","message":"Update compute tests to adopt new policies\n\nNova is moving to new policy defaults in ussuri[1] where\nfew polciies are made more granular to adopt the new defaults.\n\nWith granularity in few policies make change in policy name so\nwe have update the patrole tests to start checking against the\nnew policy names from ussuri onwards.\n\nThis commit updates the security group and server password\npolicy tests to move to new policies from ussuri onwards.\n\nAlso add the already fixed instance action policy in reno\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html\n\nChange-Id: If661299231d548ce40a2e340b1ddb9ebe8d3f964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5225c9b522faa5191f5c47a8491f47e8f0d12784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5225c9b522faa5191f5c47a8491f47e8f0d12784"}]},"branch":"refs/heads/master"},"65b4f0bd08d6efa1807d3508139b8e4416f6b66e":{"kind":"REWORK","_number":2,"created":"2020-04-21 21:41:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/01/720601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/01/720601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/01/720601/2"}}},"commit":{"parents":[{"commit":"2aad8081840452c7dabee870a86c95d22f18323a","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2aad8081840452c7dabee870a86c95d22f18323a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-16 16:52:57.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-21 21:39:39.000000000","tz":-300},"subject":"Update compute tests to adopt new policies","message":"Update compute tests to adopt new policies\n\nNova is moving to new policy defaults in ussuri[1] where\nfew polciies are made more granular to adopt the new defaults.\n\nWith granularity in few policies make change in policy name so\nwe have update the patrole tests to start checking against the\nnew policy names from ussuri onwards.\n\nThis commit updates the security group and server password\npolicy tests to move to new policies from ussuri onwards.\n\nAlso add the already fixed instance action policy in reno\n\nBump the oslo.policy requirement to 2.4.1 for parsing the rule with\nempty string and some deprecated default which is fixed by\n- https://review.opendev.org/#/c/698790\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html\n\nChange-Id: If661299231d548ce40a2e340b1ddb9ebe8d3f964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/65b4f0bd08d6efa1807d3508139b8e4416f6b66e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/65b4f0bd08d6efa1807d3508139b8e4416f6b66e"}]},"branch":"refs/heads/master"},"eb8e3ef15806fac4e7e139168eea3164bc51ab66":{"kind":"REWORK","_number":3,"created":"2020-04-21 23:34:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/01/720601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/01/720601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/01/720601/3"}}},"commit":{"parents":[{"commit":"2aad8081840452c7dabee870a86c95d22f18323a","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2aad8081840452c7dabee870a86c95d22f18323a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-16 16:52:57.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-21 23:33:54.000000000","tz":-300},"subject":"Update compute tests to adopt new policies","message":"Update compute tests to adopt new policies\n\nNova is moving to new policy defaults in ussuri[1] where\nfew polciies are made more granular to adopt the new defaults.\n\nWith granularity in few policies make change in policy name so\nwe have update the patrole tests to start checking against the\nnew policy names from ussuri onwards.\n\nThis commit updates the security group and server password\npolicy tests to move to new policies from ussuri onwards.\n\nAlso add the already fixed instance action policy in reno\n\nAlso fix the gate to parse the combining of deprecated rule\ncheck_str with oslo policy parser instead of string processing.\n\nStory: #2007585\nTask: #39516\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html\n\nChange-Id: If661299231d548ce40a2e340b1ddb9ebe8d3f964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/eb8e3ef15806fac4e7e139168eea3164bc51ab66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/eb8e3ef15806fac4e7e139168eea3164bc51ab66"}]},"branch":"refs/heads/master"},"5f840944fb0e006b983e0c168f498f35bf7f56e5":{"kind":"REWORK","_number":4,"created":"2020-04-22 00:43:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/01/720601/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/01/720601/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/01/720601/4"}}},"commit":{"parents":[{"commit":"2aad8081840452c7dabee870a86c95d22f18323a","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2aad8081840452c7dabee870a86c95d22f18323a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-16 16:52:57.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-22 00:43:08.000000000","tz":-300},"subject":"Gate fix and update compute tests to adopt new policies","message":"Gate fix and update compute tests to adopt new policies\n\nNova is moving to new policy defaults in ussuri[1] where\nfew polciies are made more granular to adopt the new defaults.\n\nWith granularity in few policies make change in policy name so\nwe have update the patrole tests to start checking against the\nnew policy names from ussuri onwards.\n\nThis commit updates the security group and server password\npolicy tests to move to new policies from ussuri onwards.\n\nAlso add the already fixed instance action policy in reno\n\nAlso fix the gate to parse the combining of deprecated rule\ncheck_str with oslo policy parser instead of string processing.\n\nStory: #2007585\nTask: #39516\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html\n\nChange-Id: If661299231d548ce40a2e340b1ddb9ebe8d3f964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5f840944fb0e006b983e0c168f498f35bf7f56e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5f840944fb0e006b983e0c168f498f35bf7f56e5"}]},"branch":"refs/heads/master"},"ee53f843fcb19815b94c80b922968fae72e5f789":{"kind":"REWORK","_number":5,"created":"2020-04-22 02:57:46.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/01/720601/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/01/720601/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/01/720601/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/01/720601/5"}}},"commit":{"parents":[{"commit":"2aad8081840452c7dabee870a86c95d22f18323a","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2aad8081840452c7dabee870a86c95d22f18323a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-16 16:52:57.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-22 02:57:42.000000000","tz":0},"subject":"Gate fix and update compute tests to adopt new policies","message":"Gate fix and update compute tests to adopt new policies\n\nNova is moving to new policy defaults in ussuri[1] where\nfew polciies are made more granular to adopt the new defaults.\n\nWith granularity in few policies make change in policy name so\nwe have update the patrole tests to start checking against the\nnew policy names from ussuri onwards.\n\nThis commit updates the security group and server password\npolicy tests to move to new policies from ussuri onwards.\n\nAlso add the already fixed instance action policy in reno\n\nAlso fix the gate to parse the combining of deprecated rule\ncheck_str with oslo policy parser instead of string processing.\n\nStory: #2007585\nTask: #39516\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html\n\nChange-Id: If661299231d548ce40a2e340b1ddb9ebe8d3f964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ee53f843fcb19815b94c80b922968fae72e5f789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ee53f843fcb19815b94c80b922968fae72e5f789"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
