)]}'
{"id":"openstack%2Fpatrole~720971","triplet_id":"openstack%2Fpatrole~master~I017716842c61b814bbe16cc2b8788f160f4ad9cd","project":"openstack/patrole","branch":"master","topic":"unittest.mock","hashtags":[],"change_id":"I017716842c61b814bbe16cc2b8788f160f4ad9cd","subject":"Use unittest.mock instead of third party mock","status":"MERGED","created":"2020-04-18 16:57:21.000000000","updated":"2020-05-08 23:54:47.000000000","submitted":"2020-05-08 23:53:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720971-1588982005543-b16e9fe7","meta_rev_id":"c4f4f13d6495844b9e668776d52db66a2b633477","_number":720971,"virtual_id_number":720971,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 23:53:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":1,"date":"2020-05-08 21:42:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-08 22:41:30.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-08 22:41:30.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-19 18:08:55.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-05-08 21:42:49.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-05-08 22:41:30.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-05-08 23:53:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60909fd254dc8c933a2fa7affd79d970a96b90b5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-18 16:57:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af68a0153bab714db1b4ed746c6d7c2c661aeb23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 17:17:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2c0c819329843fbb91b12e34d19c1f4 : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a94b29e9fd8c4094954d6e17b51c387b : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b655304d2c2d46adbd559f1259a0ab84 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/850fd6705079438395b320c511b82ab4 : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9246f64b143d4a00ba37c522a53939fa : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffd2675457da4c2f87a42de9cf2016f8 : SUCCESS in 6m 33s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/e6d6cf40177e426bad5cbbbd38e7c614 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f908c6adc594754a21629a924a33cbc : SUCCESS in 9m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71222a59e5534df896263b2b73045ffc : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d8c88623d8ae2b1902b218fcf7c2d1d3ce5c0d38","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-04-19 18:08:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1050cdf2125834e23835a783a744b5b6738eccc","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-04 08:24:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f91c9b28547764b0299469b721baa2e9f97c2779","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-08 20:46:26.000000000","message":"Patch Set 1: Code-Review-1\n\nplease update lower-constraints.txt also","accounts_in_message":[],"_revision_number":1},{"id":"ef5be14765287aa295ed6e9c651f759e922b11ac","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-08 21:13:59.000000000","message":"Patch Set 1:\n\n\u003e please update lower-constraints.txt also\n\nlower-constraints should actually still include packages that are not direct requirements of the project.","accounts_in_message":[],"_revision_number":1},{"id":"d50c96c5a2f3eff71d86f49a2c5e53b54e644979","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-08 21:15:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c3534ec073ae91179189e43dbf086daf1bbb26d","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-08 21:42:49.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0aaa8b610b3b4b6399de984d10c847e2c9d4926c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 22:38:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f92551ca20c4944a539a707086348fa : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b07a3bde1e3400b90efa500c55a51d6 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a970056fb16b4587a9e254d1431118f9 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b4d62738b2f4215b3b0110dc239c6e4 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7c1177d94134123add107921ebb79f4 : SUCCESS in 6m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/76167f26d44c40cf88e26bf20f7f05dc : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0063dd82024e491f9f130f14bad79ca3 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b33e76f50fd944c6a3933686c753257f : SUCCESS in 4m 20s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/8546b4b22df94483857d24f11b5084a3 : SUCCESS in 53m 41s\n- patrole-member https://zuul.opendev.org/t/openstack/build/d62b8aee6e6b497b8c0308892135be66 : SUCCESS in 1h 21m 17s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/3af80b4457074a3793dec3573a990590 : SUCCESS in 1h 07m 24s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/2a3aad2c9d854439afb494339ccb9880 : SUCCESS in 1h 04m 09s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/7a8f000b07fd449089b2bb76872a99f2 : SUCCESS in 1h 20m 05s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/778696062d904cb3a91cea9e675a01e6 : SUCCESS in 1h 10m 49s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/6fa1ba7ec261408ba69cff16b54ff0f1 : SUCCESS in 1h 02m 25s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/c4c24787114a45fc94f994cfffcb0eab : FAILURE in 1h 14m 02s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/720375ef19e24363a97655606c0579de : FAILURE in 40m 46s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/80520ad97a8244f690838969cb3c3d4c : FAILURE in 41m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f252ec9b756a3110f39ae18ea38d41c20af9d836","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-08 22:41:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"76085c228ca2ffa0a13ff8fdb35f2e43bdbb6815","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 22:41:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6bb700770e295581a1232d79bde538f2a0f5eb67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 23:53:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76c405f12ba747de86158ed94c8dac81 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3a38841741843c2bd62357cd100750f : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2913e8c2e4cd44e58323904772d0d584 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22c0485c0b6b456395ac317dd63f47fe : SUCCESS in 3m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8344864af1994ebaaffe4533bf8b71f7 : SUCCESS in 2m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b3cb897a3524bc9a90419e1285da2da : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5748797a26394bd590d65faa02fa84ca : SUCCESS in 3m 08s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/6d36a55afd9e4af1853193a608e11e7c : SUCCESS in 49m 41s\n- patrole-member https://zuul.opendev.org/t/openstack/build/2490dbbd697243e9a09ce00029b33ebe : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":2},{"id":"8351f8bea4b8f7355ed4097a1ae97ab161251a5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 23:53:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f349a1e11241da26317b3854c31b33c90303c021","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 23:54:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dce8496109e4fdcb9f8e5106af9be2a : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d743d8622d014a52aa7e9b3129edfdce : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c5ea6f2cc7004df849aa005fda07d633573089f3","revisions":{"62d042075bb9930fc498fcf57e0658570434292e":{"kind":"REWORK","_number":1,"created":"2020-04-18 16:57:21.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/71/720971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/71/720971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/720971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/720971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/720971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/71/720971/1"}}},"commit":{"parents":[{"commit":"2aad8081840452c7dabee870a86c95d22f18323a","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2aad8081840452c7dabee870a86c95d22f18323a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:57:10.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:57:10.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I017716842c61b814bbe16cc2b8788f160f4ad9cd\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/62d042075bb9930fc498fcf57e0658570434292e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/62d042075bb9930fc498fcf57e0658570434292e"}]},"branch":"refs/heads/master"},"c5ea6f2cc7004df849aa005fda07d633573089f3":{"kind":"REWORK","_number":2,"created":"2020-05-08 21:15:32.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/71/720971/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/71/720971/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/720971/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/720971/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/71/720971/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/71/720971/2"}}},"commit":{"parents":[{"commit":"2aad8081840452c7dabee870a86c95d22f18323a","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2aad8081840452c7dabee870a86c95d22f18323a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:57:10.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 21:15:18.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I017716842c61b814bbe16cc2b8788f160f4ad9cd\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/c5ea6f2cc7004df849aa005fda07d633573089f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/c5ea6f2cc7004df849aa005fda07d633573089f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
