)]}'
{"id":"openstack%2Fpatrole~725023","triplet_id":"openstack%2Fpatrole~master~I4ad6f84c71733b0bcc76d97493aa403d8c38de74","project":"openstack/patrole","branch":"master","topic":"725023","hashtags":[],"change_id":"I4ad6f84c71733b0bcc76d97493aa403d8c38de74","subject":"Add Patrole gate job for stable/ussuri","status":"MERGED","created":"2020-05-02 15:54:37.000000000","updated":"2020-07-06 14:43:09.000000000","submitted":"2020-05-08 18:10:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725023-1588961459079-452ece20","meta_rev_id":"1f2a5fc034697f833865fa7d6294ebda9322918b","_number":725023,"virtual_id_number":725023,"owner":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 18:10:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-05 00:35:58.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-05 00:35:58.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 14:30:39.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-05-05 00:35:58.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-05-08 18:10:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01545415cac898041c7bbd313f561eb9f9686f86","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-02 15:54:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9302ccd48849e54c93911fc85693ad52bf81dac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 17:16:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fc402cda9cf41b89eb9dbb5b9d31284 : SUCCESS in 3m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d2645d614ea488b90d0467ec93d8ffe : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7386c3ad7d574b5597095259d9fa2019 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91223170577d4dec9804370b42613471 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbb78889d2d940cd87ec8f8886645f74 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9c1e1ed86b8469c8c63950308fe83f2 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8020bb0c9879456388e5c514609dae71 : SUCCESS in 5m 07s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/5e80a6e8e1f0440ea3084f7d020ea2e3 : SUCCESS in 58m 34s\n- patrole-member https://zuul.opendev.org/t/openstack/build/2268f40ad802436daa9928d7698aa622 : SUCCESS in 1h 04m 02s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/f3e379adbdf347ce8357f1bbc5ba2020 : SUCCESS in 1h 04m 31s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/788d1ca0bb844f44acd9e7682e3f45dc : SUCCESS in 1h 04m 35s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/b3852f2276c749b28242f1f5ab254b45 : FAILURE in 1h 11m 47s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/205019e48e7449debbfd4dc2d4ac4186 : SUCCESS in 1h 04m 48s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/d3b111d330464af0a9c04d93dd0a60dd : SUCCESS in 1h 17m 28s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/18f66c8968f04eaca0a364fe04d38801 : SUCCESS in 1h 00m 54s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/09e817eceb334616b310ff4082d83327 : FAILURE in 58m 18s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/4be04c4a39ed46c196473d792435fa1a : FAILURE in 35m 21s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/bcdcc91ed1e04e9a956ee6fd5ac462e2 : FAILURE in 35m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0281fd8076a8120ff014a828f4bfa3c3c58229bd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 01:28:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d98e7872c5fa8c5e07638a8e2fdcc43db685f67a","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-04 12:10:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"877be1b82670feda72dc0f4900d46114f5c7ed50","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-04 12:24:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c619df20e977059abdf99b9866de9324cb28cfed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 13:25:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9c825c0ab9e4516902795ff52824cb7 : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c0e47ef688a49d59b6c334daa0a74c6 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0e9bdd51f084e4491d79528ea0305ea : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4112c13221a49ffb5ffaba9aa97a742 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3dff2ef8356c4f7bb1b135589e9362f7 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bbd623d28764478bf3a8b8fbdf5de65 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b95388fec33450f82b7bac0c2ed4e3e : SUCCESS in 11m 41s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/0ae290798ca34c3b897caf0c66c1967b : SUCCESS in 1h 03m 01s\n- patrole-member https://zuul.opendev.org/t/openstack/build/726d78ca5153430c92f1d93f61ce7867 : SUCCESS in 1h 08m 01s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/5e8a3112bfbb4b92bdaef2b1d1f0e791 : SUCCESS in 1h 05m 24s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/4950cc07fbe84550b6926b0a90dbb108 : SUCCESS in 1h 07m 55s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/0e1bf8a7b712484187ca42c260daf191 : SUCCESS in 1h 04m 13s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/ec15675ed4414a1a9683e512a9f57a95 : SUCCESS in 1h 02m 44s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/04046579800b47f9ac81e2f41b6f287e : FAILURE in 1h 02m 26s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/6c86fa97f3b348bdaafad082b5a4daf1 : FAILURE in 33m 24s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/a4c075309cd647babe4abb94f03a9e10 : FAILURE in 37m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a16289067711ca634048fa474e2cda68e3ead866","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 14:00:41.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c83c840efde291d591893c26fef46a96fe3b3b0","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-05-04 14:30:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5799ebac9e3fcdd0dc9d5d4282ced423205e9bba","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-04 17:40:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9f060112854f5de5bdafea4e7b277d4720219c7b","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-04 17:43:16.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c8e041f111881a7ebef03f51d4605fa1743e2d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 19:10:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a06639a74d1e46e2a52944017f91a278 : SUCCESS in 4m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/892c472481da4b9f942e5f6a01e6911e : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614f2bd665974161985984c448cf92c8 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a280ccd2f74a4668998f6d09ed84bb8f : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b55c159503d45e0a2987481712c36e4 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/460321ebb94b41a9bfb976fd88ab34fd : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0343cdc4382040ba8ab30a882d4ca75e : SUCCESS in 5m 50s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/fc25ac8eae9f42c9934b39323157f8c9 : SUCCESS in 57m 06s\n- patrole-member https://zuul.opendev.org/t/openstack/build/7deb93fd97bf43e2883f8b40c49092e5 : SUCCESS in 1h 12m 35s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/c18ef28cb9b84eb38c5adfcaaec2fee8 : SUCCESS in 1h 23m 14s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/5213016557e24085a255c2686babd143 : SUCCESS in 1h 07m 55s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/b7b59d712d334aeabaf883b359a3f74e : FAILURE in 37m 39s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/5db3b7d69e6d4433b2b240a9c86d5fd2 : SUCCESS in 1h 05m 00s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/6490a40b2fe049729115b8fe8508f733 : SUCCESS in 1h 16m 10s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/b0308ad1492041b5bfb91e9f7b1107d0 : FAILURE in 1h 15m 00s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/59c7a7118d3e4e55866316c80a38c08e : FAILURE in 35m 50s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/31bf7dfd2f1a4840be6ccfbec5d79f52 : FAILURE in 41m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ad474fe2bfd7c7860bd265888045cd1c113be5dc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 23:08:57.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"2bf8001e60dbce16ae5f39865a06460d30e30e19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 00:27:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b57c0d388fdc4ddca4d015623f97c798 : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba6fddfd8fe747d1b4ec1560cf50db33 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84fc07bb93c8430c96241cdded9caacd : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2e1b698a407409985a84bd2fa7698e9 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a091eca93b344659afe32355515d4c95 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3ffc4d066b3482ea1090eda1625d8e1 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b2e6c0a89d84021b0f2437b9de472fc : SUCCESS in 5m 30s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/db4bea6974184beebe62d539a195d302 : SUCCESS in 54m 45s\n- patrole-member https://zuul.opendev.org/t/openstack/build/3be28637f95c42b38442ae5ec385d84c : SUCCESS in 1h 00m 42s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/aead3319db314633a67f55cf4cd06cf8 : SUCCESS in 1h 11m 16s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/6c91b84c56774425a06ab16599fb8ab7 : SUCCESS in 1h 07m 01s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/d54406c1eac54343a1e31ad331541f50 : SUCCESS in 1h 06m 46s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/4d84b7b588c54c0589112948cf767eb7 : SUCCESS in 1h 12m 57s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/cc2374c4e7134f5b96ced2094e41b79e : SUCCESS in 1h 06m 16s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/a595c1e9ba814022894f75243d73a2d8 : FAILURE in 1h 00m 23s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/4defebc9351244ab9fce0908dbd513a1 : FAILURE in 32m 56s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/447cfc6910ba4c8eab9b26e57432ba35 : FAILURE in 37m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b7258a642a32aea49338fa740b4247b168882c5e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-05 00:35:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"45d217f38028474fd06b3346a21453520ed0d482","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 16:34:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2b72ab8d9844a59a797cf56ff9db8aee7a7910fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 18:10:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94c5e4df071545a9b1b600f32fd1632e : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e25ea311e6f460a9c3cdd5525e7c453 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89f9cf6bb52c4a158751ce95ec932c68 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/031955f0ebdd4148b5c0294711d3291a : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a043aebee64817bcfd86bbc19552da : SUCCESS in 6m 46s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/50c32963ddcb4553b143e5a6cef697cc : SUCCESS in 1h 05m 05s\n- patrole-member https://zuul.opendev.org/t/openstack/build/8556a53e4f654c89b3b9611d0be14be5 : SUCCESS in 1h 08m 01s","accounts_in_message":[],"_revision_number":4},{"id":"6da6ee555447d23a168f25737caec13b2832c585","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 18:10:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"60e73287fb7c3bfd87842c2095a1c29446b67c36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 18:16:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9352873783a147c2be48eabdc9ef3146 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":4},{"id":"e5a182b26150b9cc81aa8cb3c93e07c366b93637","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2020-07-06 14:43:09.000000000","message":"Patch Set 4:\n\nrecheck / reverify","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a19f3dbe5cfde0e2a4a79b13b80db50f1b71b600","revisions":{"6de4721bb7f1f11f4a3f0d226045088b3822dde6":{"kind":"REWORK","_number":1,"created":"2020-05-02 15:54:37.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/23/725023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/23/725023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/725023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/725023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/725023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/23/725023/1"}}},"commit":{"parents":[{"commit":"1686d23f3881ed3b2d793b71c44e5dde2153a85c","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/1686d23f3881ed3b2d793b71c44e5dde2153a85c"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-02 15:39:33.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-02 15:54:33.000000000","tz":330},"subject":"Add Patrole gate job for stable/ussuri and stable/train","message":"Add Patrole gate job for stable/ussuri and stable/train\n\nNow we have stable/ussuri and stable/train branch ready\nfor devstack and so does for all service projects.\n\nThis commit adds the Patrole testing for stable/ussuri\nand stable/train by adding new jobs running on\nstable/ussuri and stable/train version of openstack.\n\nChange-Id: I4ad6f84c71733b0bcc76d97493aa403d8c38de74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6de4721bb7f1f11f4a3f0d226045088b3822dde6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6de4721bb7f1f11f4a3f0d226045088b3822dde6"}]},"branch":"refs/heads/master"},"195f63dfd626b64ecd68a044d4f1123b11d9af32":{"kind":"REWORK","_number":2,"created":"2020-05-04 12:10:31.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/23/725023/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/23/725023/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/725023/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/725023/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/725023/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/23/725023/2"}}},"commit":{"parents":[{"commit":"18d392a0b5f31a63377c61a750ae6ec583e6e16e","subject":"Add Patrole gate job for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/18d392a0b5f31a63377c61a750ae6ec583e6e16e"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-02 15:39:33.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-04 12:08:45.000000000","tz":330},"subject":"Add Patrole gate job for stable/ussuri","message":"Add Patrole gate job for stable/ussuri\n\nNow we have stable/ussuri branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Patrole testing for stable/ussuri\nby adding new jobs running on stable/ussuri version of\nopenstack.\n\nChange-Id: I4ad6f84c71733b0bcc76d97493aa403d8c38de74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/195f63dfd626b64ecd68a044d4f1123b11d9af32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/195f63dfd626b64ecd68a044d4f1123b11d9af32"}]},"branch":"refs/heads/master"},"01ca4604f8791cf4111521059b0f1894682ef2c0":{"kind":"REWORK","_number":3,"created":"2020-05-04 17:40:40.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/23/725023/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/23/725023/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/725023/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/725023/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/725023/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/23/725023/3"}}},"commit":{"parents":[{"commit":"18d392a0b5f31a63377c61a750ae6ec583e6e16e","subject":"Add Patrole gate job for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/18d392a0b5f31a63377c61a750ae6ec583e6e16e"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-02 15:39:33.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-04 17:40:09.000000000","tz":330},"subject":"Add Patrole gate job for stable/ussuri","message":"Add Patrole gate job for stable/ussuri\n\nNow we have stable/ussuri branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Patrole testing for stable/ussuri\nby adding new jobs running on stable/ussuri version of\nopenstack.\n\nChange-Id: I4ad6f84c71733b0bcc76d97493aa403d8c38de74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/01ca4604f8791cf4111521059b0f1894682ef2c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/01ca4604f8791cf4111521059b0f1894682ef2c0"}]},"branch":"refs/heads/master"},"a19f3dbe5cfde0e2a4a79b13b80db50f1b71b600":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-05-04 23:08:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/23/725023/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/23/725023/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/725023/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/725023/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/725023/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/23/725023/4"}}},"commit":{"parents":[{"commit":"6d97d0f8cb55fc89588e5d340ff4d7a5af00655a","subject":"Add Patrole gate job for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6d97d0f8cb55fc89588e5d340ff4d7a5af00655a"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-05-02 15:39:33.000000000","tz":330},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-04 23:08:57.000000000","tz":0},"subject":"Add Patrole gate job for stable/ussuri","message":"Add Patrole gate job for stable/ussuri\n\nNow we have stable/ussuri branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Patrole testing for stable/ussuri\nby adding new jobs running on stable/ussuri version of\nopenstack.\n\nChange-Id: I4ad6f84c71733b0bcc76d97493aa403d8c38de74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/a19f3dbe5cfde0e2a4a79b13b80db50f1b71b600"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/a19f3dbe5cfde0e2a4a79b13b80db50f1b71b600"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
