)]}'
{"id":"openstack%2Fpatrole~727615","triplet_id":"openstack%2Fpatrole~master~I8c2ad4e15fb42fa5b1aacb3fcde69ecde91c62ce","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"I8c2ad4e15fb42fa5b1aacb3fcde69ecde91c62ce","subject":"Fix hacking min version to 3.0.1","status":"MERGED","created":"2020-05-13 02:52:10.000000000","updated":"2020-05-13 03:54:30.000000000","submitted":"2020-05-13 03:52:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727615-1589341941521-d71d2c75","meta_rev_id":"a951bf6544503637b80e03a1922365a3b8083e90","_number":727615,"virtual_id_number":727615,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 03:52:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":2,"date":"2020-05-13 03:39:45.000000000","_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":1,"date":"2020-05-13 03:39:45.000000000","_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 03:39:45.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-05-13 03:52:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da7d854e0aae899cc7bb99709febeb6ba3dc5779","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-13 02:52:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf953820bad0483e8259159c72a1dce8225574fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 03:06:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10a921fb7ca244259758829171c89179 : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edc5e27ee2264fc686da690a44d3433f : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60e8308364a749789ad950a25477d608 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe07adbdbe914ddcafba43a90c489f74 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc44b32563bf4be9a4b8db90f483048b : SUCCESS in 4m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac4b03097a784d50bfc23600084781e1 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf4ffdc694994b2bada29921cb3ccb30 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b74d254182574deaad4e69f421ebd5de : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a8df9df2130adb7e0e338dd96fd59bf4431ac113","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-05-13 03:39:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"6d3f74f8504265764a541c357a614a343578c6d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 03:39:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"250e68e6e4f80f764a4b6a735ff82ff8b6e92191","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 03:52:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e0a9fec31e64b81866f52b2fe1724ed : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5f486fd36349e48b4b0b29ce5d320b : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bcdcdf193df455ea7f58f21613ca8b6 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdec79805c244ade8fea8b15bddcf9d4 : SUCCESS in 4m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/36015229a3ec4737b27ab7d4de94c288 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5e36417928e4f549f03b242272da7ba : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d70c8594b6b443e89c8722a506771be : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3cd821f1b16f8eb640800738f97ab9448b5525f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 03:52:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a951bf6544503637b80e03a1922365a3b8083e90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 03:54:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a1d438bd2334d6bae5f087ae59996cf : SUCCESS in 1m 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06b3a01781e8418a8cb43045943cbf22 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a0c57c0667fa35f1b86a8e462e538d22c82c307","revisions":{"5a0c57c0667fa35f1b86a8e462e538d22c82c307":{"kind":"REWORK","_number":1,"created":"2020-05-13 02:52:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/15/727615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/15/727615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/15/727615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/15/727615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/15/727615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/15/727615/1"}}},"commit":{"parents":[{"commit":"b429f7916fc1f03e8e244c3299873a7d6f1adc8b","subject":"Merge \"Followup for grouping imports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/b429f7916fc1f03e8e244c3299873a7d6f1adc8b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-13 02:51:09.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-13 02:52:09.000000000","tz":-300},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nflake8 new release 3.8.0 added new checks and gate pep8\njob start failing. hacking 3.0.1 fix the pinning of flake8 to\navoid bringing in a new version with new checks.\n\nThough it is fixed in latest hacking but 2.0 and 3.0 has cap for\nflake8 as \u003c4.0.0 which mean flake8 new version 3.9.0 can also\nbreak the pep8 job if new check are added.\n\nTo avoid similar gate break in future, we need to bump the hacking min\nversion.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014828.html\n\nChange-Id: I8c2ad4e15fb42fa5b1aacb3fcde69ecde91c62ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5a0c57c0667fa35f1b86a8e462e538d22c82c307"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5a0c57c0667fa35f1b86a8e462e538d22c82c307"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
