)]}'
{"id":"openstack%2Fpatrole~732907","triplet_id":"openstack%2Fpatrole~master~Iccbc3087d30712f30617349268b66bb7573f7bd4","project":"openstack/patrole","branch":"master","topic":"drop_future_imports","hashtags":[],"change_id":"Iccbc3087d30712f30617349268b66bb7573f7bd4","subject":"Stop to use the __future__ module.","status":"MERGED","created":"2020-06-02 18:43:12.000000000","updated":"2020-06-19 19:43:52.000000000","submitted":"2020-06-19 19:42:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732907-1592595745682-8182d7df","meta_rev_id":"7deefaa5c6d1b007bd18dd634f12c4238f6f35bb","_number":732907,"virtual_id_number":732907,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 19:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2020-06-10 20:03:14.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":1,"date":"2020-06-19 19:30:03.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 19:30:03.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-06-19 19:42:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cac0817d6533a7d47c57a5d777a73477bd001616","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-02 18:43:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"075d06b3cf686290ccef184ce8d61df1541f3f4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 18:52:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb9396c9c63841678388d655696b0590 : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33ba20c46d664f58b1a2ef7f0b08aba1 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcbce10213704654b14b0934c957b53c : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c5a25748dd647ada7d0aa4134706d43 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26bc23f99aba4bbda3acba9bc7eb49df : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc404c4001fd43249773938c2fcf2539 : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e96a08fcd7f61c2fd653fd5987390b373a49a83b","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-06-10 20:03:14.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"1002051782925d22d3b97e1ac59ed8b324b4f728","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-06-19 19:30:03.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"09178877630ed64166975ef5987338494885faa2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 19:30:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eba655ce62623bcc8e180f1fe0f60309fa997ea1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 19:42:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4d4ad0907864b0ea973523f6d7eb780 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c96aa4d8cd064704ad7263d7c280cac2 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8d9f9d20b37486da2351d0a61b45e87 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d464dd7b784045b68acef6b5110f2275 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be80eeb86df14d0697ff56a097ff7b31 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8c08758f240d346ea1be2a63f8a3246c2a94496c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 19:42:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e560c33bf1312e82b78e3e71f2b5d470c3463b4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 19:43:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eea06aeecc343e680b6c19bf06ee03a : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a69fa081ca18eec487350a8adee6e96e1903a7d","revisions":{"7a69fa081ca18eec487350a8adee6e96e1903a7d":{"kind":"REWORK","_number":1,"created":"2020-06-02 18:43:12.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/07/732907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/07/732907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/07/732907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/07/732907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/07/732907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/07/732907/1"}}},"commit":{"parents":[{"commit":"5d65c587f2aeb97704e8976858c1edb00ac753d7","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5d65c587f2aeb97704e8976858c1edb00ac753d7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 18:43:06.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 18:43:06.000000000","tz":120},"subject":"Stop to use the __future__ module.","message":"Stop to use the __future__ module.\n\nThe __future__ module [1] was used in this context to ensure compatibility\nbetween python 2 and python 3.\n\nWe previously dropped the support of python 2.7 [2] and now we only support\npython 3 so we don\u0027t need to continue to use this module and the imports\nlisted below.\n\nImports commonly used and their related PEPs:\n- `division` is related to PEP 238 [3]\n- `print_function` is related to PEP 3105 [4]\n- `unicode_literals` is related to PEP 3112 [5]\n- `with_statement` is related to PEP 343 [6]\n- `absolute_import` is related to PEP 328 [7]\n\n[1] https://docs.python.org/3/library/__future__.html\n[2] https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n[3] https://www.python.org/dev/peps/pep-0238\n[4] https://www.python.org/dev/peps/pep-3105\n[5] https://www.python.org/dev/peps/pep-3112\n[6] https://www.python.org/dev/peps/pep-0343\n[7] https://www.python.org/dev/peps/pep-0328\n\nChange-Id: Iccbc3087d30712f30617349268b66bb7573f7bd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
