)]}'
{"id":"openstack%2Fpatrole~743064","triplet_id":"openstack%2Fpatrole~master~I3683cca390b44146c217ce8600f63a9894057058","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"I3683cca390b44146c217ce8600f63a9894057058","subject":"Fix gate for multiple issues","status":"MERGED","created":"2020-07-26 17:48:24.000000000","updated":"2020-08-14 18:26:17.000000000","submitted":"2020-08-14 18:24:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743064-1597429481652-ab5395dc","meta_rev_id":"eaeb3a18ca3ac9643d67ec3b1e0decaf8ffca9a9","_number":743064,"virtual_id_number":743064,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-14 18:24:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":2,"date":"2020-08-14 17:10:44.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":1,"date":"2020-08-14 17:10:44.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 05:15:19.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-08-07 13:22:12.000000000","updated_by":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"reviewer":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"state":"REVIEWER"},{"updated":"2020-08-14 18:24:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b833256f14ada6e9f156e7180010d136af5b78a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-26 17:48:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6048a49825ed93f869956ed41db62a9d1dd35833","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-26 19:02:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8f9486cef594626b03702958c30ef8c : FAILURE in 3m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39def9b9a74d440f82182e9620dda6a6 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96fdddc158534ee4b2e1f2f7d1009005 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30a466020eca4d6c9c7656a62b64367d : FAILURE in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3a33c501c054db9b1eecedb7d4b2c48 : FAILURE in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e2859a3deb4c809f6c8782bc802ffa : SUCCESS in 5m 14s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/7adab59a0ed64150bf60d69bc0450049 : SUCCESS in 42m 45s\n- patrole-member https://zuul.opendev.org/t/openstack/build/a7e4412b93144dbbbb42465f4da1d7a6 : SUCCESS in 1h 10m 14s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/71482008b9f34ec594d2d8ad89f8617f : SUCCESS in 1h 12m 04s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/26b88cbacb7a48f1bb517c854b9bd63d : SUCCESS in 1h 07m 17s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/38451e7a7e1c4abba29aefb352ee3aae : SUCCESS in 1h 10m 43s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/5782b9fdb6b044a1a176078e4b05de1e : SUCCESS in 48m 10s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/c5e6f0d63e374de488405f06b92f1474 : SUCCESS in 57m 34s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/92458932852448f09ca420ab44b9178b : FAILURE in 57m 46s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/58a9a49793834777a8200ec847876c50 : FAILURE in 21m 43s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/2296fc4b340d41009bede98db1e574a6 : FAILURE in 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17c7bed029346ad45f2b341e3208f55d43af6505","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 22:50:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b69f6ef9e7215a6907e339b735b224f453e45d1d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 23:23:12.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"62a46a28232f1bc3e79d8e11a3017052a5ab7c11","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-02 23:43:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a8a00299c5c3c21c823f800c1de3ff31f5520363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 00:07:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5216699e9667430d9978202cd71e9ed5 : SUCCESS in 3m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/345476e72146451faaf8d50cc6a54a65 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/683583e4bb494793bff052b50281cd82 : SUCCESS in 2m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4ebafe98f7d40f6ac3826ebdb18ef00 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e05841c9437c4dd1a290d81039cf02d0 : SUCCESS in 3m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8d4314122ee2403881b4badc41efe1df : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/909664b7905e46b9a60854542337f7dc : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07a9670be8e043ed8d79fa0c220a3bd2 : SUCCESS in 2m 37s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/aba04e8cb5074d5fbf003b600446b7ab : SUCCESS in 38m 58s\n- patrole-member https://zuul.opendev.org/t/openstack/build/1d5a9c68905a40ae96c5102af86257c9 : SUCCESS in 1h 05m 06s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/b2b619cb9715403c96ce0cf7dadc4953 : SUCCESS in 1h 07m 02s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/901bd612f2ef4700bee4c1d550966e8e : SUCCESS in 57m 45s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/73cb19d17b9c40028a222c739d8074cc : FAILURE in 1h 04m 13s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/a47b18bd3f8146b6bd88d645745cab04 : FAILURE in 1h 10m 18s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/68f9d9962f00428ca449c4b99feaece9 : SUCCESS in 56m 56s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/540cee9a348b4355842d10ec36a2b35f : FAILURE in 58m 58s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/e288f27def524fdf979b526ca5d348bf : FAILURE in 22m 04s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/89c3389fb4234b6c9cb5dcecd171e058 : FAILURE in 29m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"87d29d91879e269a5b142492a868f9eba18210c2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-03 00:11:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9a9b029dfc8786a64ec030e3b1168da9bd26104e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 01:36:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/900711fc887a4d158b2d3fb3250ab1dc : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0574cd10e144686bb915371bba51e43 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/540b8077d6b84eb9ac1772c898d00617 : SUCCESS in 2m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/717e260eb6fc4610bfce3f294e4bd833 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c4e5a04c1024de09af686b04cf4bf70 : SUCCESS in 3m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/458a363abb1a49328217f6d307f788c4 : SUCCESS in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7d31171503241e5801c3aabab7001fc : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e9506bf0cb5401b9a269ff3f9f1cbd3 : SUCCESS in 3m 22s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/6b2a9671519d4086a9009c8942def8df : SUCCESS in 48m 32s\n- patrole-member https://zuul.opendev.org/t/openstack/build/c41314b305874013b9e77578b79d631b : SUCCESS in 57m 09s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/d538631220de4a68992548c4e387f147 : SUCCESS in 1h 02m 42s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/9beb6a3d889b428aafec6af2273aac96 : SUCCESS in 59m 22s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/637b4f01fc5847af98124023453a4c1b : FAILURE in 1h 05m 43s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/793a4c2436c0440a805bde51c4187f3c : FAILURE in 56m 47s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/878ce5cb9d9146498d6ae3d0431e8d52 : SUCCESS in 1h 13m 43s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/48a8c3be71d545a099a44b6a6701b60e : FAILURE in 45m 35s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/d2ea60962f014e6ab1cda0d4b92b9ef7 : FAILURE in 27m 31s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/0fb0555ccc9b472d98c02c2fb8873351 : FAILURE in 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"720c5d5a2261af7de94f2c8c28ddf69b1e51548a","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-08-03 05:15:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cdf75cae8fc31b2e8a3e490a943cb4cd2867411a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-03 18:25:51.000000000","message":"Patch Set 2:\n\ntrain ansd stein jobs are failing consistently with below error. Not sure why, it is failing well before request of upload image file to glance.\n\nTypeError: object of type \u0027int\u0027 has no len()\n\n- https://github.com/openstack/patrole/blob/7a69fa081ca18eec487350a8adee6e96e1903a7d/patrole_tempest_plugin/tests/api/image/test_images_rbac.py#L41","accounts_in_message":[],"_revision_number":2},{"id":"75c59fa16c0aa90485ad36165095432cadf93096","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-03 18:37:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a7ced00f15abc649fbea2d98fd3ddf2929bc2083","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-03 18:37:25.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"a19c2c8e01a6b153feecb568a872bc8ee923a55c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 20:12:08.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03e688fd905e434ea5253bd77943a6ae : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7b2c59e96f64bdebfef3f1adab8e33c : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f193629b3f5d47e89f0126856b70a278 : FAILURE in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0c9c24a64114315af85abe9fd5f306c : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99d66d33c3cc4c879e9d49b17d423221 : SUCCESS in 4m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f22aca6b902043a4a592963d78aed41b : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198f481a3c754598b0b1ecb0fbc74052 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/605fc672cb1a44ae972541a8f12fca49 : SUCCESS in 6m 36s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/b88417ab15314199900c9aa3798424be : FAILURE in 1h 01m 43s\n- patrole-member https://zuul.opendev.org/t/openstack/build/2488a5b7ba824a89bc10dd6384b9d37f : FAILURE in 1h 04m 38s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/fad1d7c2b8b34f119330bc942e9d8d4d : FAILURE in 1h 27m 14s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/3d02253ae8774f578367cf7aaf6dd982 : FAILURE in 56m 44s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/dfafcac8584c48d9b6605be727a9603e : FAILURE in 1h 10m 11s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/03d0a4d7703b44e7a17c88cef41e25b9 : FAILURE in 1h 02m 44s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/d29b669461c64d19b6b9103d9be89013 : SUCCESS in 1h 04m 17s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/a9d9807916de40889ed074b94ec1a889 : FAILURE in 1h 07m 47s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/3e69776cfeaa481c996a12b6b7158846 : FAILURE in 36m 18s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/248e70860fcf4a4fa82cf924022295c7 : FAILURE in 38m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ab9318b24f81413cdb15577f3553f5eed92604a0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-03 22:39:02.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"806630dff68374304a5ade09bd4055a46bc9ccbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 23:50:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12b86ba205fb480caaa1796302f90876 : SUCCESS in 4m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/872d1110fbe84886bec1ae805843c194 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74e1cf1171aa4de7a043f1f25f2bc510 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f24cc46b36de4bdc8eb7d324b744aec9 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a7d8fa876834ee79b2b0311e43b3383 : SUCCESS in 4m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8ec77c5472b9469da1f5d5b15e50290f : SUCCESS in 1m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf566333c0fd4e9fa4076d7007231f22 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6f5295a609040469f263fdfed93c788 : SUCCESS in 3m 15s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/86984d59cf6347f6b9285c46ca056e91 : SUCCESS in 48m 33s\n- patrole-member https://zuul.opendev.org/t/openstack/build/a8f24a0c6c294fdba2f2c23f2135c409 : SUCCESS in 1h 06m 21s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/c629947de82746ec81cea4704f4474d2 : SUCCESS in 54m 09s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/c32d6094c27a482fbe4f00eb95e36c85 : SUCCESS in 1h 04m 31s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/2f1e7e415b074b62ae54845bd72dbef5 : FAILURE in 1h 03m 08s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/84acfb1ae8ae431e878284fbc58f7a38 : FAILURE in 1h 09m 04s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/58ba11c2b94b48ca81f02a075cba81f2 : SUCCESS in 1h 01m 05s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/6ca855825b38422e975389e347f4977b : FAILURE in 1h 02m 39s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/358568e005fd442abe33270077fb0a35 : FAILURE in 35m 45s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/b03b2592f3434e2b8beebd7a7b6dcc58 : FAILURE in 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ab982ddd122d10dddac89334b3813fc5633a034f","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-06 04:34:11.000000000","message":"Patch Set 4: Code-Review+1\n\ngamin I compare the pip list for the failed run against the successful run and I found that 19.1.0 has a fix for byte array. https://pypi.org/project/pyOpenSSL/19.1.0/\n\nI tried to se the lower constraints but it did not work.  I was wondering if you have any suugetioon of how to not use 19.0.0. and use a  19.1.0 for train and stein.","accounts_in_message":[],"_revision_number":4},{"id":"7c8c072179e7cbacd282385d7e8eded47b1d3f7c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-06 13:47:33.000000000","message":"Patch Set 4:\n\n\u003e gamin I compare the pip list for the failed run against the\n \u003e successful run and I found that 19.1.0 has a fix for byte array.\n \u003e https://pypi.org/project/pyOpenSSL/19.1.0/\n \u003e \n \u003e I tried to se the lower constraints but it did not work.  I was\n \u003e wondering if you have any suugetioon of how to not use 19.0.0. and\n \u003e use a  19.1.0 for train and stein.\n\nThanks for the investigation. We need to bump it on requirement side. let me test with that","accounts_in_message":[],"_revision_number":4},{"id":"7291cffcb6a6aacc100cba2b16d4860c817911f7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-06 13:57:08.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"f214a2596999ab46ddef69b2b663fc771a4bec77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 15:09:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23e7886e239d490c8462ac272e2bd912 : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d84fc722ddde46ccafff5fea1aa0bf74 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbab45fd90b14f6b8f14413fcf05c7c9 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2f0c6560acc42248d463fc2200d798d : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ff2e58cf4e34c3fa8894b12e2c279c5 : SUCCESS in 5m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/351bb5f41321452a9da03df33535e57d : FAILURE in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8befb4f7af6f4eea925abfaca35f1d11 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0377dfd31ce7490b8a5b0a364d547162 : SUCCESS in 3m 22s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/80f3890ac0654637b58f6169987ff36f : SUCCESS in 53m 06s\n- patrole-member https://zuul.opendev.org/t/openstack/build/854bd762d0684d5c8ac83ffd067e43fc : SUCCESS in 1h 07m 02s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/9064ba9daac1497f8ca86763012d6fdb : SUCCESS in 1h 08m 47s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/4b7a215018b3476b917fa0ff6aaaa790 : SUCCESS in 53m 49s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/4402427c71a5417ca8bd4e54ce6cd35c : SUCCESS in 1h 11m 08s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/188abda9d7cd459cb655e717c8976e94 : FAILURE in 1h 06m 51s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/5e763f9b263742fab1d1d3ca052babe7 : SUCCESS in 48m 41s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/c341f86d363448acbde9b2e5695f20e5 : FAILURE in 47m 23s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/996f6fbd4c56463c9e00c2344b3a44b3 : FAILURE in 25m 41s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/3874eeed069d4688a1ec66ea215e6ea3 : FAILURE in 28m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"04f31b427498c8e18c3d110cbc33e4302e940a80","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-06 17:55:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"64449b59321152224aa2f5dd49b448b34b3d1847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 19:09:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa1651149a2c41e3859d65ac00047798 : SUCCESS in 2m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffcb68c8b6924f25aae2a1e5ce9c4c49 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5000d774f08497e9098f4a008d43af0 : SUCCESS in 2m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35ca89b9b7184b70a3192eb3f544be0c : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc4c0904c30240238f3bb63c9c914669 : SUCCESS in 3m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/284a4b09a06142b6a823170dcbba773f : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c3455ddcbe24fa388157a77d2535877 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cd960001a414792b6337b8097671e53 : SUCCESS in 2m 51s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/76de276c6fac4980a10015b530804a48 : SUCCESS in 39m 19s\n- patrole-member https://zuul.opendev.org/t/openstack/build/fd385413f67847508c651daa36f18b54 : SUCCESS in 49m 22s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/3e77c854e39d4a4f87c9dd79e18035bb : SUCCESS in 54m 27s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/5d7fb55b70e64db5a71894a038b5c398 : SUCCESS in 48m 12s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/65878b6e4790458f9f6bb141470cdd84 : SUCCESS in 1h 09m 46s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/5f01870990f244c4af75a6f16e62abed : FAILURE in 1h 08m 31s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/a3bbcf534f144368b5da40207c38300f : SUCCESS in 1h 02m 55s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/d1f7869922e3471cad14a6889212d780 : FAILURE in 1h 07m 40s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/b8fd653d8e0b4319a09b004558cfca63 : FAILURE in 42m 05s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/7a3ec3ce169f4e09a9945b4d2b627887 : FAILURE in 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d13811507293b2a82bbd82e37297c185e6c64b70","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-07 13:22:12.000000000","message":"Patch Set 6: Code-Review+1\n\n\u003e \u003e gamin I compare the pip list for the failed run against the\n \u003e \u003e successful run and I found that 19.1.0 has a fix for byte array.\n \u003e \u003e https://pypi.org/project/pyOpenSSL/19.1.0/\n \u003e \u003e\n \u003e \u003e I tried to se the lower constraints but it did not work.  I was\n \u003e \u003e wondering if you have any suugetioon of how to not use 19.0.0.\n \u003e and\n \u003e \u003e use a  19.1.0 for train and stein.\n \u003e \n \u003e Thanks for the investigation. We need to bump it on requirement\n \u003e side. let me test with that\n\nlooks like the train change worked can you do the same with stein?\nSince patrole and tempest does teh same thing of using master against release branches I was wondering if we would like to containerize tempest and patrole that way it does not use the same pip list that devstack uses? Or can we install in in a virtual-env since it is the client calling the server api and do not have to share code?","accounts_in_message":[],"_revision_number":6},{"id":"8e4393d69e369dcbb7d05d135c85ed7c96db9bbf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-13 16:59:23.000000000","message":"Patch Set 6:\n\nrecheck, req fix merged","accounts_in_message":[],"_revision_number":6},{"id":"cde67d66a0e27e5b81fd2828ea86445264b86f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 18:24:21.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e815fedf2f014efb9aaf7bce3de3e27d : SUCCESS in 3m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d37dc743ce2140649be8d577aee0cbe8 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/639f269aaab345f1822fa605c89cf6f4 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab52217243ae4ba19cfb11188a21f814 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2358d7dde5e46ca88fc7c06ce801a0c : SUCCESS in 4m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/329b24a5c2a8469f968dc8077f7c7001 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a52a61719e1a4c35a52831cbb194b947 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0468d975c01c4ba89557bb6ac4058197 : SUCCESS in 3m 59s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/96004a9c522f42288c0caa07f35835f9 : FAILURE in 50m 33s\n- patrole-member https://zuul.opendev.org/t/openstack/build/0369447c8ca14219b7c827bd7ffc49b5 : FAILURE in 1h 09m 05s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/92f3daad0da84b728adab108e230e4a5 : FAILURE in 54m 43s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/e56d05550139495b9e806c05635a7c08 : SUCCESS in 1h 02m 59s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/fb2d77686a4b450f9f91377a3d4a093e : SUCCESS in 1h 12m 09s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/dac93a2796f047ba9f31024655ac1d4c : SUCCESS in 1h 16m 48s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/0e17069e2c574a4eaa1dc45de516ae7d : SUCCESS in 1h 06m 14s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/183dde571d014a04a113bdedfcefc171 : FAILURE in 1h 15m 54s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/20544017b7614d21afa76f4f4c0bfb41 : FAILURE in 43m 15s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/604681afa2af42ff814c209d7ae868ac : FAILURE in 35m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d6b4907177acd7a56f5c179e4d15ed7d3a87d55e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-13 19:38:06.000000000","message":"Patch Set 6:\n\nok, stable jobs are good now but as anticipated, nova policy changes merged recently need to be fixed here.","accounts_in_message":[],"_revision_number":6},{"id":"d8f2ca43f47c19479d0eb13b983ff977435a0554","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-13 20:08:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a03de17aa3de8777b960877c84eac71486678127","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 21:24:09.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38f7c0dc6cbd4b18ac0f6f5d13fff0d4 : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1891ce4d234e4294820fc7f700f994ec : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c24511dfeef7490482f2b18d723212a2 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/208e8467032b4e1881b98ebefc950ed1 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c38ed37f6b74be1b67a85536bd655c9 : SUCCESS in 6m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/14ca115a64064f9bbf95dfa42db1cd98 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2fc9cd384564f3f9c69a7eb16c97f73 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ced7b000a6414d819276a78814e85edc : SUCCESS in 5m 53s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/7f6784dd27514e0ea61c9a6011798740 : SUCCESS in 1h 00m 51s\n- patrole-member https://zuul.opendev.org/t/openstack/build/7f9670087dae424196a411b8ee014245 : SUCCESS in 1h 10m 18s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/351cd203916947c2bea11bd1deaaa668 : SUCCESS in 1h 13m 33s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/0074f78a96fc49f9baa1f4d4695e7fc8 : FAILURE in 58m 48s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/9d977c845bf44131acbaf4b2ab2be091 : FAILURE in 40m 44s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/1174678b88e846c49f197b2288ef38e0 : FAILURE in 28m 35s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/625a3cdac1b04cbfad92756af6b677cd : SUCCESS in 57m 32s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/33470dc984e74e068f9d9a694cd8e47b : FAILURE in 1h 13m 28s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/565c7aeec8eb4e029e87362c612e6a64 : FAILURE in 41m 59s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/dc63fc41255b414eaff36f8e704f4828 : FAILURE in 32m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e9256e7d2c011b88e4cb366ea8fc8ac2d0fdcb42","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-13 23:53:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5c8c5041e2f9e95477de8e3df7316d22f5aecc35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 01:17:50.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c87fd4e352547a78b3a6ff3df5ae65c : SUCCESS in 2m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f5f29b34c7b4e6193befc4346f8b1cd : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4ba97c22cda4c758c34638ccbcadcaf : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1dc62f5fc50448aaa6c5083cb8078aa : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/134c455088104f7a87099c6979c0e4fb : SUCCESS in 3m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/175175632ab24337877a6667cb0b2b3a : SUCCESS in 2m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a918c365dab4ada89ec90f5dc1dbe29 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9545961574e346018dc94e01565d8e53 : SUCCESS in 3m 07s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/5ef27eeeb3ab4a25880bf42998f74386 : FAILURE in 55m 47s\n- patrole-member https://zuul.opendev.org/t/openstack/build/96f673b0371e4a7d8eff46d175588986 : FAILURE in 1h 13m 36s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/ef7a743f42b940dcb6600902058a7c6e : FAILURE in 1h 10m 36s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/c807fb7ee4e04b9282929c79a826dc89 : SUCCESS in 1h 17m 52s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/394a229d875d46828d8df56bd16aff2d : SUCCESS in 1h 06m 31s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/55f5117ce9164508ae3e5833e7f0f4b7 : SUCCESS in 1h 04m 23s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/7049d871489f42e29a4f50e2a9e9c19e : SUCCESS in 1h 05m 25s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/a522698a9d4b4e43b36dd4d41f52b84c : FAILURE in 58m 16s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/e63d9e5b34aa43f0aa4b0deb2af06aff : FAILURE in 38m 00s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/5ed4d744962b4b4fbc690d41b88e1703 : FAILURE in 47m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c83e61f0db7720ed5899667f28939c4c098ea70c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-14 02:10:53.000000000","message":"Patch Set 8:\n\nmultinic policy changes just merged before this get in.","accounts_in_message":[],"_revision_number":8},{"id":"29a4c54a80979694b7f135bf4785cb9f7fdca288","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-14 02:20:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"862e24383f2292f8d3dd06da018a20291592bd4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 03:37:16.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3da4ebd2d03f4fc2b4da1cf171f4f32b : SUCCESS in 3m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/613c17742cc248ea89d3915c1d64f62e : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6d03ba2f1f7464d86ad4267f27f12a7 : FAILURE in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0d9b85ae7c8433fbcb643d6fe2a8d95 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea55093129f04d7b9b94ac952aa53151 : SUCCESS in 3m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3ddbffcf250e4656aed3c6c7eb86390e : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e19596179bde42d1be8833452e8bf356 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57e787c5905b41e38b7ca6c1b114b0b1 : SUCCESS in 3m 13s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/c966789b65e44190b57e8e41954205b1 : SUCCESS in 52m 19s\n- patrole-member https://zuul.opendev.org/t/openstack/build/e0f47fe9842144d7b90a49d5b95d68f0 : SUCCESS in 1h 07m 43s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/25dd8495dcf044e39b0deb442f478c26 : SUCCESS in 1h 11m 33s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/1524634c6fc74da394eaba4fea566b69 : SUCCESS in 1h 03m 31s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/b5b73aa890e6453690c9bfad69774faa : SUCCESS in 1h 10m 04s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/97052216b0df421eb7fc0e21bd2fd61e : SUCCESS in 59m 23s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/c419444296604764ad7b07e3048db128 : SUCCESS in 57m 40s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/fff0dc803e014543a2bad556d9a8cc28 : FAILURE in 55m 56s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/126d6a92b276491fb2e907a4ca6ff8c5 : FAILURE in 35m 52s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/9c79f2ed88ab493788a777eda64f2da7 : FAILURE in 31m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5cb8a2e4ce6ed1dbca2a473edbc3b0aa4c1e50a3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-14 11:49:40.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"2d775ef947c246724ee57fb9a850122640e5b386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 13:06:55.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4267c74ee0924495bcf607eaaa8d0782 : SUCCESS in 3m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ab353595d924f8091be5ede6c66430b : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9a7d14575df41d5a34690ae2f037868 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8256de9c74214892a83a34aba20bcf4d : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/264762475db24a1cb9de3b320ea9b841 : SUCCESS in 4m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8749598146764b7e8243a302aedaec18 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10aa2d0f66494360a0c2d6c6206499a2 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b34fc702e67944ed939ee8aa54fb5ff5 : SUCCESS in 4m 10s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/730f368abfc340d0a0b819c3fbaa80e3 : SUCCESS in 54m 11s\n- patrole-member https://zuul.opendev.org/t/openstack/build/18ee933734d1419cbb6b9f58e15589d8 : SUCCESS in 1h 09m 24s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/cbba08ad77394d08a5e4f9a2f733ac6e : SUCCESS in 1h 08m 17s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/f7cd7604491d4e3f931049122bc7aa34 : SUCCESS in 1h 09m 18s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/4d67eff4cf6a4ecf81f96183219c4ce9 : SUCCESS in 1h 01m 36s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/bc27f79333aa4f6dbdfc42ccfcf936a7 : SUCCESS in 1h 06m 19s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/88633188f8df47a59d424428537b8f0d : SUCCESS in 55m 39s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/ca1ffeae2d9a489695dc2907905bd8b2 : FAILURE in 54m 54s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/39809a60bb8e47378c60d11ed381de97 : FAILURE in 37m 09s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/f085b1c2bbc940e2b6c16530872b6ee1 : FAILURE in 33m 54s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bd6d891d1764d92fe38b57dd719e07e9ca8ebd18","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-14 14:25:54.000000000","message":"Patch Set 10:\n\nthis is good to merge now.","accounts_in_message":[],"_revision_number":10},{"id":"6f48739141f075b9c49f12d3e8fb14a9fc64dd9a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-14 17:10:44.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nmerging it to unblock the gate as no other core here today.","accounts_in_message":[],"_revision_number":10},{"id":"1766a4fb8646bd9682383a08b6a9dcd1bce6514e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 17:13:28.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"2210a05c465997a409161457262e71ee34540e1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 18:24:41.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94bacd76159541ab83151bb91f3f2243 : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e9ede75d87747cc8fc4776b434942e5 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f990a54fb9e4255bea8c1bd521a1df4 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5019074088c84689a65ef0bf8ac9a38e : SUCCESS in 4m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/36773e33d3774d06b4f72dff511eebd5 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d3f7c205a82496c88b12c0c4b6b0a2d : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21daed4941c844b3a0fbe21a048801fe : SUCCESS in 3m 36s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/64c9580840424100bce6dd3222468127 : SUCCESS in 1h 02m 05s\n- patrole-member https://zuul.opendev.org/t/openstack/build/c17e141804614022ab178695ce612f9f : SUCCESS in 1h 09m 14s","accounts_in_message":[],"_revision_number":10},{"id":"faf704fd8e7e87f53299686b3506eaa2b110cb83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 18:24:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"eaeb3a18ca3ac9643d67ec3b1e0decaf8ffca9a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 18:26:17.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a4a5620b4248c7be1695d92bee5372 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/494493b53654494a88468c6ae9bac5e2 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"cfac16a78c8f05ab2a893cf15ebbcdbf9b7d48fb","revisions":{"f826d035757d31543744f4efd78f285cf7bd9f62":{"kind":"REWORK","_number":1,"created":"2020-07-26 17:48:24.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/743064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/64/743064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/64/743064/1"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-26 17:48:10.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-26 17:48:10.000000000","tz":-300},"subject":"DNM: testing gate","message":"DNM: testing gate\n\nChange-Id: I3683cca390b44146c217ce8600f63a9894057058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f826d035757d31543744f4efd78f285cf7bd9f62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f826d035757d31543744f4efd78f285cf7bd9f62"}]},"branch":"refs/heads/master"},"492340d924e3561f574ce57c1f1a8a9b75ea02f0":{"kind":"REWORK","_number":2,"created":"2020-08-02 22:50:41.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/743064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/64/743064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/64/743064/2"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-26 17:48:10.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 22:50:40.000000000","tz":-300},"subject":"Fix gate to have mock installed for unit tests","message":"Fix gate to have mock installed for unit tests\n\nunit tests jobs use tempest version released in pypi\nwhich has use of mock but in recent changed mock requirement\nis removed from requirements file and it end up failing.\n\n- https://zuul.opendev.org/t/openstack/build/c3a33c501c054db9b1eecedb7d4b2c48\n\nLet\u0027s add mock into the requirement file to be installed for unit tests\njob until we bump the min version of tempest to latest.\n\nChange-Id: I3683cca390b44146c217ce8600f63a9894057058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/492340d924e3561f574ce57c1f1a8a9b75ea02f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/492340d924e3561f574ce57c1f1a8a9b75ea02f0"}]},"branch":"refs/heads/master"},"fd883b2fbf01652de62dfe30cb3d5896e4b83d93":{"kind":"REWORK","_number":3,"created":"2020-08-03 18:37:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/743064/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/64/743064/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/64/743064/3"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-26 17:48:10.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-03 18:37:12.000000000","tz":-300},"subject":"Fix gate to have mock installed for unit tests","message":"Fix gate to have mock installed for unit tests\n\nunit tests jobs use tempest version released in pypi\nwhich has use of mock but in recent changed mock requirement\nis removed from requirements file and it end up failing.\n\n- https://zuul.opendev.org/t/openstack/build/c3a33c501c054db9b1eecedb7d4b2c48\n\nLet\u0027s add mock into the requirement file to be installed for unit tests\njob until we bump the min version of tempest to latest.\n\nChange-Id: I3683cca390b44146c217ce8600f63a9894057058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/fd883b2fbf01652de62dfe30cb3d5896e4b83d93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/fd883b2fbf01652de62dfe30cb3d5896e4b83d93"}]},"branch":"refs/heads/master"},"bdcb969002a29b7c7a1eb5f53f4c8d8c827303cd":{"kind":"REWORK","_number":4,"created":"2020-08-03 22:39:02.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/743064/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/64/743064/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/64/743064/4"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-26 17:48:10.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-03 22:39:00.000000000","tz":0},"subject":"Fix gate to have mock installed for unit tests","message":"Fix gate to have mock installed for unit tests\n\nunit tests jobs use tempest version released in pypi\nwhich has use of mock but in recent changed mock requirement\nis removed from requirements file and it end up failing.\n\n- https://zuul.opendev.org/t/openstack/build/c3a33c501c054db9b1eecedb7d4b2c48\n\nLet\u0027s add mock into the requirement file to be installed for unit tests\njob until we bump the min version of tempest to latest.\n\nChange-Id: I3683cca390b44146c217ce8600f63a9894057058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/bdcb969002a29b7c7a1eb5f53f4c8d8c827303cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/bdcb969002a29b7c7a1eb5f53f4c8d8c827303cd"}]},"branch":"refs/heads/master"},"bb0f2472bd4c4844474be0a38e8d5cfbeafc8484":{"kind":"REWORK","_number":5,"created":"2020-08-06 13:57:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/743064/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/64/743064/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/64/743064/5"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-26 17:48:10.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-06 13:56:52.000000000","tz":0},"subject":"Fix gate to have mock installed for unit tests","message":"Fix gate to have mock installed for unit tests\n\nunit tests jobs use tempest version released in pypi\nwhich has use of mock but in recent changed mock requirement\nis removed from requirements file and it end up failing.\n\n- https://zuul.opendev.org/t/openstack/build/c3a33c501c054db9b1eecedb7d4b2c48\n\nLet\u0027s add mock into the requirement file to be installed for unit tests\njob until we bump the min version of tempest to latest.\n\nDepends-On: https://review.opendev.org/#/c/745158/\n\nChange-Id: I3683cca390b44146c217ce8600f63a9894057058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/bb0f2472bd4c4844474be0a38e8d5cfbeafc8484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/bb0f2472bd4c4844474be0a38e8d5cfbeafc8484"}]},"branch":"refs/heads/master"},"3ee8017a1804dd60dacba7a4d572fc011ec481b6":{"kind":"REWORK","_number":6,"created":"2020-08-06 17:55:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/743064/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/64/743064/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/64/743064/6"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-26 17:48:10.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-06 17:55:20.000000000","tz":-300},"subject":"Fix gate to have mock installed for unit tests","message":"Fix gate to have mock installed for unit tests\n\nunit tests jobs use tempest version released in pypi\nwhich has use of mock but in recent changed mock requirement\nis removed from requirements file and it end up failing.\n\n- https://zuul.opendev.org/t/openstack/build/c3a33c501c054db9b1eecedb7d4b2c48\n\nLet\u0027s add mock into the requirement file to be installed for unit tests\njob until we bump the min version of tempest to latest.\n\nDepends-On: https://review.opendev.org/#/c/745158/\n\nChange-Id: I3683cca390b44146c217ce8600f63a9894057058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/3ee8017a1804dd60dacba7a4d572fc011ec481b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/3ee8017a1804dd60dacba7a4d572fc011ec481b6"}]},"branch":"refs/heads/master"},"0d4b6886f12617e7a908d13593b428d4fad2ac7d":{"kind":"REWORK","_number":7,"created":"2020-08-13 20:08:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/743064/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/64/743064/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/64/743064/7"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-26 17:48:10.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-13 20:07:09.000000000","tz":-300},"subject":"Fix gate for multiple issues","message":"Fix gate for multiple issues\n\n1. To have mock installed for unit tests\n\nunit tests jobs use tempest version released in pypi\nwhich has use of mock but in recent changed mock requirement\nis removed from requirements file and it end up failing.\n\n- https://zuul.opendev.org/t/openstack/build/c3a33c501c054db9b1eecedb7d4b2c48\n\nLet\u0027s add mock into the requirement file to be installed for unit tests\njob until we bump the min version of tempest to latest.\n\n2. Nova policy granular work\nhttps://review.opendev.org/#/q/topic:bp/policy-defaults-refresh-deprecated-apis+(status:open+OR+status:merged)\n\nDepends-On: https://review.opendev.org/#/c/745158/\n\nChange-Id: I3683cca390b44146c217ce8600f63a9894057058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/0d4b6886f12617e7a908d13593b428d4fad2ac7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/0d4b6886f12617e7a908d13593b428d4fad2ac7d"}]},"branch":"refs/heads/master"},"0dfc63d12d88853db771de9788c539f670e6fc31":{"kind":"REWORK","_number":8,"created":"2020-08-13 23:53:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/743064/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/64/743064/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/64/743064/8"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-26 17:48:10.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-13 23:53:04.000000000","tz":-300},"subject":"Fix gate for multiple issues","message":"Fix gate for multiple issues\n\n1. To have mock installed for unit tests\n\nunit tests jobs use tempest version released in pypi\nwhich has use of mock but in recent changed mock requirement\nis removed from requirements file and it end up failing.\n\n- https://zuul.opendev.org/t/openstack/build/c3a33c501c054db9b1eecedb7d4b2c48\n\nLet\u0027s add mock into the requirement file to be installed for unit tests\njob until we bump the min version of tempest to latest.\n\n2. Nova policy granular work\nhttps://review.opendev.org/#/q/topic:bp/policy-defaults-refresh-deprecated-apis+(status:open+OR+status:merged)\nAdding new flag to handle the policy changed in Victoria.\n\nDepends-On: https://review.opendev.org/#/c/745158/\n\nChange-Id: I3683cca390b44146c217ce8600f63a9894057058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/0dfc63d12d88853db771de9788c539f670e6fc31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/0dfc63d12d88853db771de9788c539f670e6fc31"}]},"branch":"refs/heads/master"},"b5308d0fa91b4a235bec240f7c56428b6a08a46f":{"kind":"REWORK","_number":9,"created":"2020-08-14 02:20:55.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/743064/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/64/743064/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/64/743064/9"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-26 17:48:10.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-14 02:20:48.000000000","tz":-300},"subject":"Fix gate for multiple issues","message":"Fix gate for multiple issues\n\n1. To have mock installed for unit tests\n\nunit tests jobs use tempest version released in pypi\nwhich has use of mock but in recent changed mock requirement\nis removed from requirements file and it end up failing.\n\n- https://zuul.opendev.org/t/openstack/build/c3a33c501c054db9b1eecedb7d4b2c48\n\nLet\u0027s add mock into the requirement file to be installed for unit tests\njob until we bump the min version of tempest to latest.\n\n2. Nova policy granular work\nhttps://review.opendev.org/#/q/topic:bp/policy-defaults-refresh-deprecated-apis+(status:open+OR+status:merged)\nAdding new flag to handle the policy changed in Victoria.\n\nDepends-On: https://review.opendev.org/#/c/745158/\n\nChange-Id: I3683cca390b44146c217ce8600f63a9894057058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/b5308d0fa91b4a235bec240f7c56428b6a08a46f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/b5308d0fa91b4a235bec240f7c56428b6a08a46f"}]},"branch":"refs/heads/master"},"cfac16a78c8f05ab2a893cf15ebbcdbf9b7d48fb":{"kind":"REWORK","_number":10,"created":"2020-08-14 11:49:40.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/743064/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/64/743064/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/64/743064/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/64/743064/10"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-26 17:48:10.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-14 11:49:39.000000000","tz":0},"subject":"Fix gate for multiple issues","message":"Fix gate for multiple issues\n\n1. To have mock installed for unit tests\n\nunit tests jobs use tempest version released in pypi\nwhich has use of mock but in recent changed mock requirement\nis removed from requirements file and it end up failing.\n\n- https://zuul.opendev.org/t/openstack/build/c3a33c501c054db9b1eecedb7d4b2c48\n\nLet\u0027s add mock into the requirement file to be installed for unit tests\njob until we bump the min version of tempest to latest.\n\n2. Nova policy granular work\nhttps://review.opendev.org/#/q/topic:bp/policy-defaults-refresh-deprecated-apis+(status:open+OR+status:merged)\nAdding new flag to handle the policy changed in Victoria.\n\nDepends-On: https://review.opendev.org/#/c/745158/\n\nChange-Id: I3683cca390b44146c217ce8600f63a9894057058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/cfac16a78c8f05ab2a893cf15ebbcdbf9b7d48fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/cfac16a78c8f05ab2a893cf15ebbcdbf9b7d48fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
