)]}'
{"id":"openstack%2Fpatrole~744378","triplet_id":"openstack%2Fpatrole~master~I58359093054d68373d4585f93131756b21096bc7","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"I58359093054d68373d4585f93131756b21096bc7","subject":"Shorten Identity project name due to 64 char limit.","status":"ABANDONED","created":"2020-08-02 16:18:20.000000000","updated":"2020-08-07 19:49:00.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0ca67c2f09ecf71f82401ec94bf49c1899abd557","_number":744378,"virtual_id_number":744378,"owner":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-07 03:06:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":30745,"name":"Kelly Carter","email":"KC755D@att.com","username":"kc755d"},{"_account_id":26365,"name":"Bryan Strassner","email":"strassner.bryan@gmail.com","username":"bryan-strassner"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30745,"name":"Kelly Carter","email":"KC755D@att.com","username":"kc755d"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26365,"name":"Bryan Strassner","email":"strassner.bryan@gmail.com","username":"bryan-strassner"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":30745,"name":"Kelly Carter","email":"KC755D@att.com","username":"kc755d"},{"_account_id":26365,"name":"Bryan Strassner","email":"strassner.bryan@gmail.com","username":"bryan-strassner"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":26365,"name":"Bryan Strassner","email":"strassner.bryan@gmail.com","username":"bryan-strassner"},{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},{"_account_id":30745,"name":"Kelly Carter","email":"KC755D@att.com","username":"kc755d"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-02 23:34:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-08-03 14:34:16.000000000","updated_by":{"_account_id":30745,"name":"Kelly Carter","email":"KC755D@att.com","username":"kc755d"},"reviewer":{"_account_id":30745,"name":"Kelly Carter","email":"KC755D@att.com","username":"kc755d"},"state":"REVIEWER"},{"updated":"2020-08-04 00:07:07.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2020-08-04 00:07:07.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2020-08-04 00:07:07.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-08-04 00:07:07.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2020-08-04 00:07:07.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2020-08-04 00:07:07.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2020-08-04 00:07:07.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-08-04 00:07:07.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-08-06 13:28:39.000000000","updated_by":{"_account_id":26365,"name":"Bryan Strassner","email":"strassner.bryan@gmail.com","username":"bryan-strassner"},"reviewer":{"_account_id":26365,"name":"Bryan Strassner","email":"strassner.bryan@gmail.com","username":"bryan-strassner"},"state":"REVIEWER"},{"updated":"2020-08-06 14:09:16.000000000","updated_by":{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},"reviewer":{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},"state":"REVIEWER"},{"updated":"2020-08-07 03:04:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3631c5d47dfed25a9ec44e9d492a3b44b3798864","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-02 16:18:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c9688fcf83a233da1efe5ebc3adf6deed95abdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 17:34:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1f819dc86cb478aba850e342be5adb7 : FAILURE in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/158ab0853d494d729ac18470e062097a : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b1ed17891d94398bbe74964f0d65c08 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa5c586b520348dc88f8ba20fc464b97 : FAILURE in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a7864611ac84aa2ad2917b7d25ce9d5 : FAILURE in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adc0c0864e6a4e93af7216ea15d75f2a : SUCCESS in 4m 48s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/22849aa6a80f44fa866c415fa7c2bfcb : SUCCESS in 47m 10s\n- patrole-member https://zuul.opendev.org/t/openstack/build/9d7a2e7e071a4fd29c062bddae3b0a6a : SUCCESS in 1h 05m 02s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/31a652c5d7664c4ab137fe49d78619c1 : SUCCESS in 1h 05m 40s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/70691e7c6b9c4f2d8d1e7d1f712bd797 : SUCCESS in 59m 55s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/85983d733b6246eaa4cc45a43d6d2b57 : FAILURE in 1h 02m 16s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/8695bce4f638424dbb6186b4607f8dc2 : FAILURE in 1h 07m 19s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/d65a9b85a5074343a501c4780ce7f6e2 : SUCCESS in 58m 34s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/1e0f0674fb4943119125d1069c4ec740 : FAILURE in 55m 40s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/f6742945bc704f85813eceab49dc0e2b : FAILURE in 22m 45s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/cbbfe3f515344bdfb28302e36291de6d : FAILURE in 37m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc7eb7510f34be2a1f9c38dfdb7f74d82e127a10","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-02 19:24:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e7cc77a3ab22aa976be13f4f2d5f145d4750b48c","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-02 19:28:53.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"95f650684162cd41a981e642dee43109ce92302b","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-02 19:39:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d523d77e927b9e0e43579627b0cd39f2aed474ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 20:59:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f761db4c61046c7822eab77472d575b : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78b6df7193ab4af3989463d914436a6a : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f44fd9a0c614f30bd44d410593eac36 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01786549bb014f5a8ecb17a0f4e14653 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0aca2627c0c54b4fa6a0c74175000143 : SUCCESS in 3m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/af7da196626f44a4a906a27453d5b5d7 : FAILURE in 1m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a25b97a8cad744b2a280af1cbef1f1b2 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6b253a8c83e43bd999500035341f06b : SUCCESS in 2m 59s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/f6e3bf4cb36743e4a211493b911dd65c : SUCCESS in 52m 12s\n- patrole-member https://zuul.opendev.org/t/openstack/build/300ad634bd044f2186a412b73b244900 : SUCCESS in 1h 12m 44s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/3c51ec315f8942feb525fc4b0e1ae165 : SUCCESS in 1h 04m 48s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/125c1ec4c44b45079644f5bdb50c45b1 : SUCCESS in 1h 13m 08s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/b71364f043b4483d89b94d1526cc59d2 : FAILURE in 1h 04m 25s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/d6be0a55fba94f03ae966ec72b492bb1 : FAILURE in 1h 12m 57s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/f8f9fdc5fc0f44d69265a126c625b298 : SUCCESS in 1h 10m 50s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/048b7d5204174e85afd4d46637d571c0 : FAILURE in 58m 24s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/c987f76b73454674a120f20d2aa88644 : FAILURE in 1h 02m 36s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/b3d8e73d0da84dadb69da16c233033f0 : FAILURE in 23m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"84a5a317819f63d251fc97d634686a3fcc73f04c","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-02 23:27:57.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"d791dd73071219b0f9eb5d341fe5ad6a3dc99531","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 23:34:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"67163dccd53b3bb5fdb2e51c0aa73789d78e058b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 01:00:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89a1305c4aef46658a35561f0ddba57f : SUCCESS in 2m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d24ef386e7964fe3ad77382b73c41652 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f4de342c4954e5f94a1bcbd54a0783c : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d64c307a9162473ab5ed0f780f06fd87 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d28074a45e94d8c8b53143b1b81eb33 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa32164842c4c659e3cc40ac06ebc96 : SUCCESS in 8m 25s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/d2cf76baf6dd42fda5f45307ada1c025 : SUCCESS in 35m 47s\n- patrole-member https://zuul.opendev.org/t/openstack/build/bf5e357540a8486f9adefb0110ba99cc : SUCCESS in 1h 01m 17s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/cd30c226c024440bb8cda746345786ca : SUCCESS in 1h 04m 24s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/87f0c00478a9400aa7c4d85608729810 : SUCCESS in 1h 05m 02s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/3806514a8c104158a7af13a05d5bc9f8 : FAILURE in 1h 11m 31s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/07f5015769154b2eb6b0c942837f0ebe : FAILURE in 1h 04m 51s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/153d22cf8a4e46019211b25586cd08c6 : SUCCESS in 1h 07m 53s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/0e6f3d47512a4cd6a903a5d294be33cb : FAILURE in 57m 48s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/8b82353e09a146fbb227c9da4a59146b : FAILURE in 37m 08s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/2c74123013374bb0809e90a21eab8c2a : FAILURE in 37m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"77d5a90c5bd6aecad33e5425d45b214a1b160cff","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-03 11:32:05.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"2d9a423fd690745601c1a9c80bfecdeb448e1847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 12:51:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34c07e58b06645a2ae08063f3423e31b : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/052168742b7b4f399abf2c4e4758da75 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da430d35f57244be89b3877d2316384d : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6a0df869acc44ecb20e423cb616e097 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91e291d255e1457d9c9bdf895e0552f5 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be46aa135985486dbce92043e010bc73 : SUCCESS in 6m 53s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/bec20a6a9ec44db4b00aebf777d89ef1 : SUCCESS in 56m 16s\n- patrole-member https://zuul.opendev.org/t/openstack/build/fd90fa937929474382c8f811686bc34f : SUCCESS in 1h 05m 44s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/ab7ccc55effa464ea725a3c855227180 : SUCCESS in 1h 05m 54s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/94e2a4375e9f47c4ba112b4d2a7060b4 : SUCCESS in 47m 39s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/8bbf27db4ff54f74910aff384f5c4348 : FAILURE in 1h 09m 46s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/cb91b3d594c04e8d9a25daa4d0694848 : FAILURE in 1h 06m 39s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/3119b94e35884d0996308e53382972ab : SUCCESS in 1h 15m 39s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/8cc358722917448b840bc446f122616c : FAILURE in 56m 13s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/346decb2dc504a1dbe2775597653dbfd : FAILURE in 29m 17s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/c4ec2e90e5374cb39150e51eb2b019d1 : FAILURE in 35m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"06a190208908c95d37fe87f67036b6ac844fc275","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-03 14:42:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b437fd2efc6a088f0880596890ec416cf2e09b87","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-03 15:00:36.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"93d743643de88ee292df36b1c79ca48971d3dd63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 16:26:37.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3cf195018784d6992672b9cb008db00 : SUCCESS in 3m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26c597e53637413faa349bb128e77108 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efcb164e777845c6b5520f636d8548af : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf805d83da3046c9b8c00a0983a12f00 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39e3ed3211124211831b29fc490cd664 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/484063d159bf40d595fc757e39fa8764 : SUCCESS in 6m 10s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/337615b4fbbe4e9bab5955345aa4c958 : SUCCESS in 53m 00s\n- patrole-member https://zuul.opendev.org/t/openstack/build/c053b7063af24d2497077f087301b3e2 : SUCCESS in 1h 09m 21s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/a62e559f6e2b4aa086543a8262743c76 : SUCCESS in 1h 05m 31s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/95e3d5fa80ec49739a927186220e4e94 : SUCCESS in 1h 16m 01s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/6483e0e5b1ba46b28844436074938de9 : FAILURE in 1h 18m 33s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/4434f216abcb47e8aba95e5d03443cbc : FAILURE in 1h 06m 43s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/91ca6d06ce394ae6a1342ca35a964c3c : SUCCESS in 47m 31s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/c7bc380879c9477588eb4cb717859525 : FAILURE in 58m 15s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/e6b5ac92947d48fba04d65bc0edad6af : FAILURE in 28m 51s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/0ad8e3903a2a4330ae6b26deb2cc2329 : FAILURE in 35m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6fa829e883513bddef931b508f421a2bffc7fea8","author":{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},"date":"2020-08-03 22:23:44.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"b60db70d9243f2f3cee22cea73d9d10427dc83ff","author":{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},"date":"2020-08-03 22:23:59.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"fca96e5035c2af49c8c9eb26ac11817fb7534c51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 23:33:01.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa97d3ac63ce45a49bca83a7fb47d919 : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47ac30ad65004f61998ef6764f8c357f : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7975090a98024aa192f6eea6c43b2a7d : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70c0d7da75fe4d8b875434657f383fa2 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60a7d84df587423d8ff12173d45a4443 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/271589f6e8d9424cb592fe0239880419 : SUCCESS in 6m 20s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/bf5b54c2b8734d6f994b1e083b35ee0f : SUCCESS in 51m 32s\n- patrole-member https://zuul.opendev.org/t/openstack/build/e9804587f6a84cc89d40593ace69b98d : SUCCESS in 1h 06m 48s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/1ec290bb7ee64e6090ee0c05893626e6 : SUCCESS in 1h 03m 03s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/75a7cd4ac33a402d86a827ce7e591ab5 : SUCCESS in 1h 03m 29s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/f32772b9010047bfb2d39f5d5e39d2a6 : FAILURE in 1h 02m 48s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/dbcd51dc626f46b1a0a7b93891e47426 : FAILURE in 1h 00m 14s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/2cc98cd17a214f65bb5b266848c03e73 : SUCCESS in 53m 23s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/e203fdbdb4a24664b140eb5b00d309a1 : FAILURE in 55m 07s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/66cbde41ee3f4b659a9e621aa67bcaae : FAILURE in 30m 28s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/073c8b1a26454323b3eb36be4a0aafe6 : FAILURE in 31m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4965a994c4f8aab026de4bfbdd7b18d3296f1a43","author":{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},"date":"2020-08-04 20:15:27.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"087168e44c6539f8c90a3bbedffb8545d3507b69","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-04 20:29:13.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ad4584288661b0b5bb08abe613e2f303c6aa0ab1","author":{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},"date":"2020-08-04 20:31:41.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"a4437d365eaab4f575803c63ad52bca59430485e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 21:54:31.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0796fe5840b04738a72ab2c7c8f9888c : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b7b3b8d2f7146c39ffd6972e5d64802 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eab6cf8a976466c93cb8f454734c579 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6aab99bd27c439b8245aae53a960b62 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/901716bb317a4a0081f253dcf921ce4d : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d116c6c91a16402dae555d7202b3c872 : SUCCESS in 6m 01s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/59616096c19b44ab98fd005be825748e : SUCCESS in 1h 04m 17s\n- patrole-member https://zuul.opendev.org/t/openstack/build/13ec90e0c8324dff986099ac8ec18585 : SUCCESS in 1h 14m 54s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/1d22c3f32a3b4bceadc04891a8696f1c : SUCCESS in 1h 14m 57s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/8adc2bb0a822422ca0c8b1be4059627f : SUCCESS in 1h 10m 26s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/df46c0f4ebdc41a7a886b98805d70247 : FAILURE in 1h 18m 43s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/49b35a1dbda643238240e411fc8ad2e5 : FAILURE in 1h 16m 13s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/abf8dbd40ef84a8a94253fab5e18c657 : SUCCESS in 53m 56s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/596b7a72395749408a12cd279ecea91e : FAILURE in 58m 29s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/95d2ec04c7d64c79b144dd9ef0f0eba8 : FAILURE in 34m 59s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/d7c4a37bb43b4e6ab5eb8a2b5690722e : FAILURE in 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f6ac8941f52e7046dc4b903be3398a87e778b26f","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-04 22:42:51.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"8bb4d37adf613e9300c823794dbf63201c680db6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 00:14:15.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb440aed1d4940c69d54570e23867a7c : SUCCESS in 3m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87368078314d4ce5874e440c682e56a3 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b05cf38adca14c72a025c0f066c3ffde : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99b454b8cdae45399477d3586ec3b7ac : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a075c96aa3a4b5cb313cc9427434221 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58eb72c2e79340bdb96a80a9f97fcff2 : SUCCESS in 6m 04s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/c35b22a819f64fbb8c5e3ad160e06f6b : SUCCESS in 53m 48s\n- patrole-member https://zuul.opendev.org/t/openstack/build/80caad0d73c54ff4ac84cdf14b215c4c : SUCCESS in 1h 05m 04s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/ca9b0772ad554c2cb86d65d8c12f417b : SUCCESS in 1h 02m 16s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/e04d09ad218547cdbcccf2a777c68598 : SUCCESS in 1h 01m 32s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/f3532f33f9c74d7b9dd8954724e5c339 : FAILURE in 1h 27m 51s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/dab3c31281bd4dcfb4e654e9066f6ab8 : FAILURE in 1h 18m 25s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/067052cbe11b4596b299f1fb2c4335f2 : SUCCESS in 59m 37s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/53ab472ffa844a4499a7ea8cc60e6f0d : FAILURE in 1h 15m 16s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/d96ce340110c4c9d86927b984e077bf9 : FAILURE in 46m 04s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/08b81fec7d6d40b2a916705af4d9bbb3 : FAILURE in 35m 11s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"039030f887a5c35993570429388ccb2444519951","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-06 00:43:38.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"546e8d3029448386f2f0bd1131c3b0062bd83238","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-06 00:44:20.000000000","message":"Patch Set 11: Code-Review-1 Workflow-1\n\njust try to see if I can get the gate working","accounts_in_message":[],"_revision_number":11},{"id":"c4083579185cb375d94b64e99de9a5e559d670d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 01:58:22.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d6ee05104f9482ca85594ffc3ae550c : SUCCESS in 2m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9136cdc12ae41008cf699a0f9cff3cf : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7053a8aac4040d9b65dbf30b8de8754 : SUCCESS in 2m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e429e82e06a445f6a7fef0c3ecb8d859 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c7c66d15c934f80a0a5829e4422534c : SUCCESS in 4m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2411304d53bb4b398c29d906a052def1 : FAILURE in 1m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea8399335374a6d9291b9dfbfb67737 : SUCCESS in 5m 13s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/8311747e742b426c856d83a8c522769b : SUCCESS in 52m 14s\n- patrole-member https://zuul.opendev.org/t/openstack/build/a7e1d525301c4414b6c4b2067934e9fd : SUCCESS in 1h 05m 43s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/7941ce5545b54c85ac8a8bd521340602 : SUCCESS in 53m 43s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/3bf2388ae2b041008648e02d8e2d95fe : SUCCESS in 1h 06m 38s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/8342b4fd78944620a02f4c8953dc381b : FAILURE in 1h 03m 16s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/32a5ab0ac88e4eaea1425713bea679ed : FAILURE in 1h 08m 32s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/bb2ba42757ee43cdbc5336ede9563b11 : SUCCESS in 58m 43s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/2df64716a6824a7584aa378eb330dde6 : FAILURE in 1h 01m 07s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/e1abc683fc85451098ef4d48ff930974 : FAILURE in 33m 48s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/cd2d495232904b74b1a573d8a76edc48 : FAILURE in 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7623132694c2ac4890dcd2c2318dbaefa22aca57","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-06 01:58:42.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"39fa0b33ade8ef0cc393328a3b3f3cbcce927bbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 03:28:19.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe4852e0fc20459e8b351f5a3eded49d : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ada31440bf94694a4b05b2071c32de1 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79a3b38cbb1b442fbe1c76c91d8f2efb : SUCCESS in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8daad743fa08400b8f268125d79bb1ba : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c69fe29874b4f1cbe309d1dec116ef3 : SUCCESS in 3m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/97fb34017f8c40aba2dbe4af100ac9d2 : SUCCESS in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d1a1ebce21d47a2bf431fb6611cb905 : SUCCESS in 4m 37s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/f659dac4c75d4337b7a4213d77f4f02a : SUCCESS in 56m 23s\n- patrole-member https://zuul.opendev.org/t/openstack/build/f3ef5163e29e41f293de1a08264e65d0 : SUCCESS in 1h 22m 46s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/0b975989c722495fa3e83afaba6ef7dc : SUCCESS in 1h 09m 33s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/e6b41557e1fe44e0b00167de279afdba : SUCCESS in 1h 10m 00s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/d68d6ea26b57401ca2d087df97206831 : FAILURE in 1h 03m 35s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/e3a74f17e556454f95d2c62f6a5dbdf6 : FAILURE in 1h 07m 01s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/a52763058df24478952160680240c096 : SUCCESS in 1h 12m 18s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/1376467ac3f548109ea445a11028b030 : FAILURE in 1h 04m 04s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/80c984055d0e4b4cac2224b29fecf64f : FAILURE in 32m 50s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/ddf90780c5a84adca8c67df11a44a7b7 : FAILURE in 34m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"bc6c78587740b990a91ccde36e0a8c9aee53b3de","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-06 04:29:57.000000000","message":"Patch Set 11: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":11},{"id":"124cee705d51a16f16e3f48ffcdeb03a082de21f","author":{"_account_id":26365,"name":"Bryan Strassner","email":"strassner.bryan@gmail.com","username":"bryan-strassner"},"date":"2020-08-06 13:28:39.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"ad4d264f1a0f40bb2b069911ed028e243e3ea6b4","author":{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},"date":"2020-08-06 14:09:16.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"9b91a0d4c53a98b42e9045477bfeb8d081bd4462","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-06 22:03:52.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"55493c238e2a770c8b40cf458d1ac12b9cb653b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 23:20:58.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6565c0feda9d49069026b3284b310932 : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fa66f5eff5a4258ac2c37305b846e9f : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c911ba43a88c4c8ead537656b02612d2 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb15b10812f3474b916f0206e1d37930 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f061771fb30346f09dee55501decaeac : SUCCESS in 3m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8705d341e1de4e6382fc40a3b5d0d0e8 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0eefcd65e06444499fe291f655b5dd1 : SUCCESS in 4m 58s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/6cd4da7aa0eb40ffab683cfb656790fd : SUCCESS in 59m 12s\n- patrole-member https://zuul.opendev.org/t/openstack/build/15c6c774e2404228a0b52e622f7c2a13 : SUCCESS in 1h 06m 37s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/2127e93607b94e7d901e2f3767776380 : SUCCESS in 1h 10m 14s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/9bab9b3d1d6b4d22890e025c8da6a069 : SUCCESS in 50m 47s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/0a7fcb3d8b964bf5b4c72d40f3d75e81 : SUCCESS in 1h 09m 37s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/020a5000abc544a9b7b87d0b12964989 : FAILURE in 1h 00m 05s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/e2f7acaf8a334ce283ea7c1c24e9fe15 : FAILURE in 47m 33s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/d4bb99a8b3b34026b82caffe2540a7e9 : FAILURE in 1h 04m 12s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/78422fa1b9ec40ed8e68a99830c24d5f : FAILURE in 24m 40s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/1a928f4ffd0b4b23ab23580664153ad4 : FAILURE in 40m 20s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"74693f93c04ce1a2ed5035311049104cff2477b6","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-06 23:27:27.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e5cbf32b318a321fb8f3e4ecbe0bcc1be9b272b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 23:27:55.000000000","message":"Patch Set 13: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"d9497e5c8ec70338d55d1f7e2207520a2a34d682","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-07 02:33:58.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"c734d4047dc34bec0c9f765ae0e15a5a6314e664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 02:34:13.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"fb24df2b35efd7146fc89c4b4ff6ddb4c2ecc5c9","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-07 03:00:52.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"62280e6b2949f4711f8ffd39ada5e6fd0c1998dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 03:01:08.000000000","message":"Patch Set 15: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"42f3a51f89d8a6b3a95c394442cbabc10602a74d","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-07 03:04:20.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"81f0d1c46cdfbe08cfd7d820ff43b03446e90f86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 03:04:32.000000000","message":"Patch Set 16: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"0d3aa7fced2cc54468c88b2afbb86542f41a2f8f","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-07 03:06:50.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"cec9adf51b989b8db56ec986cdb967637708506a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 03:07:04.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":17},{"id":"0ca67c2f09ecf71f82401ec94bf49c1899abd557","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-07 19:49:00.000000000","message":"Abandoned\n\nsubmitting a cleaner PS","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"6b020107e4390f0e2a929599958a0bcaf09dc528","revisions":{"5fd6ab3d919a50fbcd3dc3f410ea07e5d5479940":{"kind":"REWORK","_number":1,"created":"2020-08-02 16:18:20.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/1"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:18:17.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5fd6ab3d919a50fbcd3dc3f410ea07e5d5479940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5fd6ab3d919a50fbcd3dc3f410ea07e5d5479940"}]},"branch":"refs/heads/master"},"0c0263e62496391a6f9981ffb70826ed8a7397a9":{"kind":"REWORK","_number":2,"created":"2020-08-02 19:24:23.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/2"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 19:19:30.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py I  will submit a change to tempest to use\nunittest mock that is in 3.3 on and delete the mock in tempest\nand when the next tempest in pypi I will delete the this mock for\nunittest mock.\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/0c0263e62496391a6f9981ffb70826ed8a7397a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/0c0263e62496391a6f9981ffb70826ed8a7397a9"}]},"branch":"refs/heads/master"},"e6bec71add5a5930b577488e5da18cb274d67021":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-02 19:28:53.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/3"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 19:25:22.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\nimport mock library and does not used unittest mock class.\nThe current tempest master supports unittest mock but it is not in\npypi at thie time.\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/e6bec71add5a5930b577488e5da18cb274d67021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/e6bec71add5a5930b577488e5da18cb274d67021"}]},"branch":"refs/heads/master"},"305e3fdd17ba3142c70b7a6988160e38b5b49e5e":{"kind":"REWORK","_number":4,"created":"2020-08-02 19:39:15.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/4"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 19:38:49.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\nimport mock library and does not used unittest mock class.\nThe current tempest master supports unittest mock but it is not in\npypi at thie time.\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/305e3fdd17ba3142c70b7a6988160e38b5b49e5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/305e3fdd17ba3142c70b7a6988160e38b5b49e5e"}]},"branch":"refs/heads/master"},"7ee551fd5d8ee919228f7f7c4a00df4fa99c212e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-08-02 23:27:57.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/5"}}},"commit":{"parents":[{"commit":"492340d924e3561f574ce57c1f1a8a9b75ea02f0","subject":"Fix gate to have mock installed for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/492340d924e3561f574ce57c1f1a8a9b75ea02f0"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 23:27:57.000000000","tz":0},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\nimport mock library and does not used unittest mock class.\nThe current tempest master supports unittest mock but it is not in\npypi at thie time.\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7ee551fd5d8ee919228f7f7c4a00df4fa99c212e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7ee551fd5d8ee919228f7f7c4a00df4fa99c212e"}]},"branch":"refs/heads/master"},"2ec77604f232c228a623de19bb39020878b3bde4":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-08-03 11:32:05.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/6"}}},"commit":{"parents":[{"commit":"492340d924e3561f574ce57c1f1a8a9b75ea02f0","subject":"Fix gate to have mock installed for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/492340d924e3561f574ce57c1f1a8a9b75ea02f0"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-03 11:30:56.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\ndo not pip install mock it is only in tempest test-requirements.txt\nSo I rebase off gmann change to correct mock.\n\nHaving issue with _image_store_file on all calls so trying to remove\nsix wrapper\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2ec77604f232c228a623de19bb39020878b3bde4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2ec77604f232c228a623de19bb39020878b3bde4"}]},"branch":"refs/heads/master"},"aaf822aada0b64753916088d5d319c046174042b":{"kind":"REWORK","_number":7,"created":"2020-08-03 14:42:43.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/7"}}},"commit":{"parents":[{"commit":"492340d924e3561f574ce57c1f1a8a9b75ea02f0","subject":"Fix gate to have mock installed for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/492340d924e3561f574ce57c1f1a8a9b75ea02f0"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-03 14:42:22.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nAlso removing -test_domain and -test_update_domain sufix from\ndomain name since they have a 64 char limit also.  Thisshould\nbe the last ones from our testing.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\ndo not pip install mock it is only in tempest test-requirements.txt\nSo I rebase off gmann change to correct mock.\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/aaf822aada0b64753916088d5d319c046174042b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/aaf822aada0b64753916088d5d319c046174042b"}]},"branch":"refs/heads/master"},"5f99ffe2973f0ff6f23f3f26c23453de47252efe":{"kind":"REWORK","_number":8,"created":"2020-08-03 15:00:36.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/8"}}},"commit":{"parents":[{"commit":"492340d924e3561f574ce57c1f1a8a9b75ea02f0","subject":"Fix gate to have mock installed for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/492340d924e3561f574ce57c1f1a8a9b75ea02f0"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-03 14:59:19.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nAlso removing -test_domain and -test_update_domain sufix from\ndomain name since they have a 64 char limit also.  Thisshould\nbe the last ones from our testing.\n\nremove -IdentityCosumer from test.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\ndo not pip install mock it is only in tempest test-requirements.txt\nSo I rebase off gmann change to correct mock.\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5f99ffe2973f0ff6f23f3f26c23453de47252efe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5f99ffe2973f0ff6f23f3f26c23453de47252efe"}]},"branch":"refs/heads/master"},"c086d3b70bc5e7485ce7b01547d8a31efe1647e8":{"kind":"REWORK","_number":9,"created":"2020-08-04 20:29:13.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/9"}}},"commit":{"parents":[{"commit":"492340d924e3561f574ce57c1f1a8a9b75ea02f0","subject":"Fix gate to have mock installed for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/492340d924e3561f574ce57c1f1a8a9b75ea02f0"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-04 20:28:23.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nAlso removing -test_domain and -test_update_domain sufix from\ndomain name since they have a 64 char limit also.  Thisshould\nbe the last ones from our testing.\n\nremove -IdentityCosumer from test.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\ndo not pip install mock it is only in tempest test-requirements.txt\nSo I rebase off gmann change to correct mock.\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/c086d3b70bc5e7485ce7b01547d8a31efe1647e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/c086d3b70bc5e7485ce7b01547d8a31efe1647e8"}]},"branch":"refs/heads/master"},"8c038bbc030f3cd380e78adeada4ef37101bd002":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2020-08-04 22:42:51.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/10"}}},"commit":{"parents":[{"commit":"492340d924e3561f574ce57c1f1a8a9b75ea02f0","subject":"Fix gate to have mock installed for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/492340d924e3561f574ce57c1f1a8a9b75ea02f0"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-04 22:38:53.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nAlso removing -test_domain and -test_update_domain sufix from\ndomain name since they have a 64 char limit also.\n\nremove -IdentityCosumer from test test_oauth_consumers_rbac.py.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\ndo not pip install mock it is only in tempest test-requirements.txt\nSo I rebase off gmann change to correct mock.\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\nDepends-On: https://review.opendev.org/#/c/743064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/8c038bbc030f3cd380e78adeada4ef37101bd002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/8c038bbc030f3cd380e78adeada4ef37101bd002"}]},"branch":"refs/heads/master"},"5f13eb2b14d7ec8986c068a5d9861fbed7f9c92e":{"kind":"REWORK","_number":11,"created":"2020-08-06 00:43:38.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/11"}}},"commit":{"parents":[{"commit":"492340d924e3561f574ce57c1f1a8a9b75ea02f0","subject":"Fix gate to have mock installed for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/492340d924e3561f574ce57c1f1a8a9b75ea02f0"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-06 00:42:54.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nDNM Trying to get the train gates running\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nAlso removing -test_domain and -test_update_domain sufix from\ndomain name since they have a 64 char limit also.\n\nremove -IdentityCosumer from test test_oauth_consumers_rbac.py.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\ndo not pip install mock it is only in tempest test-requirements.txt\nSo I rebase off gmann change to correct mock.\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\nDepends-On: https://review.opendev.org/#/c/743064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5f13eb2b14d7ec8986c068a5d9861fbed7f9c92e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5f13eb2b14d7ec8986c068a5d9861fbed7f9c92e"}]},"branch":"refs/heads/master"},"cf9ceca59748c36cf2ede5aa405db4928409b31d":{"kind":"REWORK","_number":12,"created":"2020-08-06 01:58:42.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/12"}}},"commit":{"parents":[{"commit":"492340d924e3561f574ce57c1f1a8a9b75ea02f0","subject":"Fix gate to have mock installed for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/492340d924e3561f574ce57c1f1a8a9b75ea02f0"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-06 01:35:55.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nAlso removing -test_domain and -test_update_domain sufix from\ndomain name since they have a 64 char limit also.\n\nremove -IdentityCosumer from test test_oauth_consumers_rbac.py.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\ndo not pip install mock it is only in tempest test-requirements.txt\nSo I rebase off gmann change to correct mock.\n\nhad to set low requirements for pyOpenSSL to 19.1.0 to fix bytearry\nissue in 19.0.0 https://pypi.org/project/pyOpenSSL/19.1.0/\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\nDepends-On: https://review.opendev.org/#/c/743064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/cf9ceca59748c36cf2ede5aa405db4928409b31d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/cf9ceca59748c36cf2ede5aa405db4928409b31d"}]},"branch":"refs/heads/master"},"4233513b38ba3ccb32a4d52a7deb2203b67aa0e2":{"kind":"REWORK","_number":13,"created":"2020-08-06 23:27:27.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/13"}}},"commit":{"parents":[{"commit":"492340d924e3561f574ce57c1f1a8a9b75ea02f0","subject":"Fix gate to have mock installed for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/492340d924e3561f574ce57c1f1a8a9b75ea02f0"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-06 23:26:25.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nAlso removing -test_domain and -test_update_domain sufix from\ndomain name since they have a 64 char limit also.\n\nremove -IdentityCosumer from test test_oauth_consumers_rbac.py.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\ndo not pip install mock it is only in tempest test-requirements.txt\nSo I rebase off gmann change to correct mock.\n\nTrying to get around the train and stein gate issue\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\nDepends-On: https://review.opendev.org/#/c/743064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4233513b38ba3ccb32a4d52a7deb2203b67aa0e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4233513b38ba3ccb32a4d52a7deb2203b67aa0e2"}]},"branch":"refs/heads/master"},"ab0a0afb34b9d933bdfc3b1a6f00b01812740b43":{"kind":"NO_CHANGE","_number":14,"created":"2020-08-07 02:33:58.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/14"}}},"commit":{"parents":[{"commit":"492340d924e3561f574ce57c1f1a8a9b75ea02f0","subject":"Fix gate to have mock installed for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/492340d924e3561f574ce57c1f1a8a9b75ea02f0"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-07 02:33:42.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nAlso removing -test_domain and -test_update_domain sufix from\ndomain name since they have a 64 char limit also.\n\nremove -IdentityCosumer from test test_oauth_consumers_rbac.py.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\ndo not pip install mock it is only in tempest test-requirements.txt\nSo I rebase off gmann change to correct mock.\n\nTrying to get around the train and stein gate issue\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\nDepends-On: https://review.opendev.org/#/c/743064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ab0a0afb34b9d933bdfc3b1a6f00b01812740b43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ab0a0afb34b9d933bdfc3b1a6f00b01812740b43"}]},"branch":"refs/heads/master"},"aeff226cd7faa2cc8f8ff25d8b64fe8527d4f3b0":{"kind":"REWORK","_number":15,"created":"2020-08-07 03:00:52.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/15"}}},"commit":{"parents":[{"commit":"492340d924e3561f574ce57c1f1a8a9b75ea02f0","subject":"Fix gate to have mock installed for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/492340d924e3561f574ce57c1f1a8a9b75ea02f0"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-07 03:00:21.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nAlso removing -test_domain and -test_update_domain sufix from\ndomain name since they have a 64 char limit also.\n\nremove -IdentityCosumer from test test_oauth_consumers_rbac.py.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\ndo not pip install mock it is only in tempest test-requirements.txt\nSo I rebase off gmann change to correct mock.\n\nTrying to get around the train and stein gate issue\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\nDepends-On: https://review.opendev.org/#/c/743064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/aeff226cd7faa2cc8f8ff25d8b64fe8527d4f3b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/aeff226cd7faa2cc8f8ff25d8b64fe8527d4f3b0"}]},"branch":"refs/heads/master"},"ec63d0cca8ffdc19fc8ffc00557c130bfb73edd9":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2020-08-07 03:04:20.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/16"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-07 03:04:20.000000000","tz":0},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nAlso removing -test_domain and -test_update_domain sufix from\ndomain name since they have a 64 char limit also.\n\nremove -IdentityCosumer from test test_oauth_consumers_rbac.py.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\ndo not pip install mock it is only in tempest test-requirements.txt\nSo I rebase off gmann change to correct mock.\n\nTrying to get around the train and stein gate issue\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\nDepends-On: https://review.opendev.org/#/c/743064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ec63d0cca8ffdc19fc8ffc00557c130bfb73edd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ec63d0cca8ffdc19fc8ffc00557c130bfb73edd9"}]},"branch":"refs/heads/master"},"6b020107e4390f0e2a929599958a0bcaf09dc528":{"kind":"NO_CHANGE","_number":17,"created":"2020-08-07 03:06:50.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/744378/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/78/744378/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/78/744378/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/78/744378/17"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-02 16:06:37.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-07 03:06:38.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nAlso removing -test_domain and -test_update_domain sufix from\ndomain name since they have a 64 char limit also.\n\nremove -IdentityCosumer from test test_oauth_consumers_rbac.py.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\ndo not pip install mock it is only in tempest test-requirements.txt\nSo I rebase off gmann change to correct mock.\n\nTrying to get around the train and stein gate issue\n\nChange-Id: I58359093054d68373d4585f93131756b21096bc7\nDepends-On: https://review.opendev.org/#/c/743064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6b020107e4390f0e2a929599958a0bcaf09dc528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6b020107e4390f0e2a929599958a0bcaf09dc528"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
