)]}'
{"id":"openstack%2Fpatrole~745405","triplet_id":"openstack%2Fpatrole~master~Ic4014938c4a2dae64892954d8638f9d8d519a234","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"Ic4014938c4a2dae64892954d8638f9d8d519a234","subject":"Shorten Identity project name due to 64 char limit.","status":"MERGED","created":"2020-08-07 19:49:08.000000000","updated":"2020-08-17 04:01:00.000000000","submitted":"2020-08-17 03:59:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745405-1597636786309-c9d102ff","meta_rev_id":"d79e7c21c45730e76585de18568ceb22c005567e","_number":745405,"virtual_id_number":745405,"owner":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-17 03:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"value":0,"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"date":"2020-08-15 22:59:26.000000000","_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":2,"date":"2020-08-16 01:06:07.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"value":0,"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":2,"date":"2020-08-17 03:00:27.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"value":0,"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":1,"date":"2020-08-17 03:00:27.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 20:53:50.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2020-08-14 20:53:50.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2020-08-14 20:53:50.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-08-14 20:53:50.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2020-08-14 20:53:50.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2020-08-14 20:53:50.000000000","updated_by":{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},"reviewer":{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},"state":"REVIEWER"},{"updated":"2020-08-14 20:53:50.000000000","updated_by":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"reviewer":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"state":"REVIEWER"},{"updated":"2020-08-14 20:53:50.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2020-08-14 20:53:50.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-08-14 20:53:50.000000000","updated_by":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"reviewer":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"state":"REVIEWER"},{"updated":"2020-08-14 20:55:24.000000000","updated_by":{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},"reviewer":{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},"state":"REVIEWER"},{"updated":"2020-08-14 21:32:53.000000000","updated_by":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"reviewer":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"state":"REVIEWER"},{"updated":"2020-08-16 01:06:07.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-08-17 03:00:27.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-08-17 03:59:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f744e18c2f1c1eeb0b80b2b9fd8db2d0df818de","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-07 19:49:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1325dccf1bc50694588c5ed70cb82ab803999caf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 21:28:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b467e25a7764433683d972c0c7e2be4d : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f8baeb7a41847a5ad6319f43314c8c9 : RETRY_LIMIT in 0s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16ea9f18a21f4e3eb8981c71a773cb7d : RETRY_LIMIT in 0s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6b27a1a1b7c456aa7378bba849cc460 : RETRY_LIMIT in 0s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97b181000f9644bebd1053faa29abc3f : RETRY_LIMIT in 0s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1611a66d73ca473e9109c213529cbb1c : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42513b66e5024e8ab7b91d1b90b30cb6 : RETRY_LIMIT in 0s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4773e7073da04ebea8e458ebd58e2a14 : RETRY_LIMIT in 0s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/e6cf2d71ed504a2f9a858753cc332b6e : FAILURE in 1h 20m 49s\n- patrole-member https://zuul.opendev.org/t/openstack/build/ffa0432d9cdc4fe08b29f8f362b4f856 : FAILURE in 1h 36m 54s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/5fffd751dc87423087988de8d2f10191 : RETRY_LIMIT in 0s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/138860a232774a13a28e261571fdc932 : SUCCESS in 1h 29m 38s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/a156c0e0766e4ac195a7fea176cacdab : FAILURE in 54m 31s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/f178ff9d1d994b6eb99dfb4f1b3e24ed : RETRY_LIMIT in 0s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/cf8e7ba9b14e4a61977c69feaf8ccc36 : SUCCESS in 57m 41s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/1fa96de463034f7fa281f244ac9e4038 : FAILURE in 1h 28m 14s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/82ca24ad2af54ce29c4445239efccbc0 : RETRY_LIMIT in 0s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/e0edf1191f624157888cebf505622deb : FAILURE in 39m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72e17fd7ef7246a12e8568ae2ff3bb63e2c9f887","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-08 12:55:16.000000000","message":"Patch Set 1:\n\nRecheck","accounts_in_message":[],"_revision_number":1},{"id":"74bbc0d5c53e9aeaecc7e9a0db61be8d21375c31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 14:14:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48f8e582a8e643b8a3dd9d0ef33db462 : SUCCESS in 2m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1582963053b248898d9a70299e660649 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93717712d99448fb99bf380ac894375b : SUCCESS in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e47b509b5ea4536aa843203f523d404 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2eb63648fe5f4fc4b6725bd74e3deed2 : SUCCESS in 3m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9cf01cc5219e418da2ddc174e80701b7 : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3847dda588b4095a2ee6a5051557cce : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98d33dd22cbc4634aa1cd0d8679b21ce : SUCCESS in 3m 01s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/c7b3b5a5430c4e518f1c6917aa133100 : FAILURE in 55m 05s\n- patrole-member https://zuul.opendev.org/t/openstack/build/e661057ec1d849d7850af36d51c91070 : FAILURE in 1h 08m 57s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/a5ed05c5a6cd488cbc1fcf81c31c1cf9 : FAILURE in 1h 07m 13s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/02f0d8ebcbcc4baab8a70b1c3989caa4 : SUCCESS in 1h 13m 05s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/3debd24e036c4e9c82f1df8a8d865c62 : FAILURE in 1h 08m 54s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/16c58acf7b7b4be8b54230c2f43225e7 : FAILURE in 1h 08m 53s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/775317180dd049c39bfd065420ecd408 : SUCCESS in 1h 03m 39s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/5032617a60104e278c5d0a9199c44f85 : FAILURE in 1h 04m 04s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/5834469f2c8748468d07cb90d893d99f : FAILURE in 37m 10s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/fdc2fb0783e04f299479e3ddf943ee04 : FAILURE in 59m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd63ae0b7dd1896eeaeb98251c4c81085e93ba23","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-10 23:37:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f7b13f17ba45c66b9118eb14d10d4650b18ee2e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 00:47:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f754254740814b598d686b3483e80d28 : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66355d834d4d4c63af77715e0bc90904 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92156f81d5244bec998d2d54f3fb6bda : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d3503196f2741faa1412d850012b2d7 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60407c87a52a45ef98a52aacf89ae24d : SUCCESS in 3m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6c715d98ec7848c4b12edf8c4e5d45b1 : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5f579c85ad407bb5e396a75abeec1f : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/471c5673d63b4ccf883d1d96b220a3de : SUCCESS in 3m 31s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/cce9e04d6c0845b9a4ed8d237482c296 : FAILURE in 54m 08s\n- patrole-member https://zuul.opendev.org/t/openstack/build/1979fc88aca94d68859be61d7cdfe9d9 : FAILURE in 1h 04m 33s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/9ddb07c1c4a74bbc9b3ca2aa8a132734 : FAILURE in 1h 02m 49s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/8cc7a698fa254adca8328aa8496e95fe : SUCCESS in 1h 08m 30s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/33d3604397e24daba67ad3e6204e5a56 : SUCCESS in 1h 04m 37s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/2a8af69da19b47bb8ebdfefa3ea2ed95 : FAILURE in 1h 01m 01s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/cb0d7d926ced449db423730b8b3bb5a3 : SUCCESS in 59m 35s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/87bd4f002ea5424984e753acf4210b0e : FAILURE in 59m 56s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/6fc05bde1c4b45f7b252a94e089eb39d : FAILURE in 34m 20s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/24c935fc8e724462bef646d20720eaae : FAILURE in 33m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90abf2a3c75cd4f5ff55b92b97ef288ce1ac9fb4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-14 18:43:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a20796b69ffe1f549641bb29e496dd68490d60ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 20:06:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d181adf3477a473e961547276d5529ff : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19b1d9a24c0c47ccbfd41b27217a2dd2 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ac960a7d26b4b2584da64d1317e024c : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98d929c82d2a4dc9a509206bd7178e6f : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90c3f921ad304ac4a43403e0b494c7e4 : SUCCESS in 3m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eb773287daf448538e44f0aed73bcad8 : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43580e566ba64690a18a7ac3d28eb147 : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21fef1c6e1d1474987907cb7aede36f1 : SUCCESS in 3m 02s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/674508acd5ec4d5ba90a48171db8b0b9 : SUCCESS in 1h 00m 15s\n- patrole-member https://zuul.opendev.org/t/openstack/build/a2c9783e0277469d90a4434b2d6f6169 : SUCCESS in 1h 21m 15s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/c461370d242b44798e35b89b18c0f28e : SUCCESS in 1h 14m 21s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/f31ce9bf4a4f4d67bb615a4006185035 : SUCCESS in 1h 21m 38s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/718111e5d864452c98998b87957385b1 : SUCCESS in 1h 11m 38s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/f5cc83424d0d4022b24664cbad30c5f2 : SUCCESS in 51m 29s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/c0a6e1e1c8f6424fb44f97c7ca17aeb8 : SUCCESS in 44m 55s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/992683a91e9a42dfba33a20ea8199f8e : FAILURE in 1h 14m 38s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/044d173e41e74c1c88572f0e56904aa3 : FAILURE in 48m 55s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/16cd1b4ccbbe4794afb80a21e9e3e200 : FAILURE in 24m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffe851eac1be6f4db088864556f28e3e027f719d","author":{"_account_id":30690,"name":"manasareddybethi","email":"mb711d@att.com","username":"mb711d"},"date":"2020-08-14 20:55:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d540385c4d805f128526ffbc8d0a66e6d23ca00b","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2020-08-14 21:32:53.000000000","message":"Patch Set 1:\n\nIf there is a 64 size limit, let\u0027s explicitly use `[:64]`?","accounts_in_message":[],"_revision_number":1},{"id":"819226b861a8c90ecf7a09afb0c086510c8aa7cc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-14 23:38:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9eb0472cbdcab9a5ecd32b31b20f853d0e56341","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-08-15 18:29:48.000000000","message":"Patch Set 1: Code-Review+1\n\nAgree with gmann, mock changes were already fixed by him. Else LGTM as this is an innocuous change.","accounts_in_message":[],"_revision_number":1},{"id":"69c1c57216f706b037d47ba21ba104140b408175","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-15 22:58:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"35755d7c3ad6e982848716c9e2b6ab1da8acbf16","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-15 22:59:12.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\ndone","accounts_in_message":[],"_revision_number":2},{"id":"a5f22e866cd7e659ad2c3bce1ebd3607dc2ea83b","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-15 22:59:26.000000000","message":"Patch Set 2:\n\n\u003e Agree with gmann, mock changes were already fixed by him. Else LGTM\n \u003e as this is an innocuous change.\n\ndone","accounts_in_message":[],"_revision_number":2},{"id":"4cdbb12a4ac301431e187d00edf13297bc1d8147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 00:18:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8c35eace092422182d4683136fe439d : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9171ed17182047449becd22dd7b0d0ad : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b46f0e856eda4cbca38402188e9f1f37 : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cc9653a12e946f1ae25efa0fce6014c : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b894768be1ad4066a079145f67d97d5d : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7f3a0bad138497199bacdf135fb3471 : SUCCESS in 5m 20s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/213b04d6cb4a4b34970e3c0dda5b035c : SUCCESS in 56m 43s\n- patrole-member https://zuul.opendev.org/t/openstack/build/31a5f427d7df4afd8952cf97d811c8a2 : SUCCESS in 1h 12m 52s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/952d2ffcfaf1496694f8eeeec5731cc4 : SUCCESS in 1h 11m 14s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/1bcfca928d694f73a9a1542e5a998544 : SUCCESS in 1h 06m 30s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/eed7701b0b68429084cb1d45f8e5c561 : SUCCESS in 1h 07m 04s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/5df2724f2bec40dd93ef5944614d7bac : SUCCESS in 1h 14m 19s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/d363e7641b9a40bbaf4b7a800b57e5ba : SUCCESS in 1h 00m 50s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/35f8fdad3c46417597dca949e0f2074d : FAILURE in 1h 03m 30s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/75009392ca844f969bbae36ed59b0458 : FAILURE in 36m 37s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/2e75d7a68be84f1eac08903ea21cca5a : FAILURE in 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"53a16da0246dc8fd6b2d6e1f5cb1e76dd2481d39","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-16 01:06:07.000000000","message":"Patch Set 2: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"9778328ad6be56144b32e42922dadb3ce4dfb078","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-08-17 03:00:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3d49d2ba81350549ac43d50f4f7ad0e4db431d41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 03:00:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"421c82c99821ffc7171438f0f0e78c0376ecab1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 03:59:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1aaf0738dda5456eadbe098580f71f25 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29fd48cc5d094b5390c8c7781e76d488 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c57dba6b5a0d48a6867062cd44b0613e : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36191a38440c4c01be5e70f349a0deb7 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7944a8f76a4d208cfa453b97c216e1 : SUCCESS in 5m 54s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/e419779eebab4329832ce65fc59cb6f8 : SUCCESS in 56m 11s\n- patrole-member https://zuul.opendev.org/t/openstack/build/52195bf2ab384bb0995cc9ee446b0111 : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":2},{"id":"029032f4b0dfad4e783e4d9541cfe97e89942eb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 03:59:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d79e7c21c45730e76585de18568ceb22c005567e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 04:01:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d6812cea65340009929e076e234a28f : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5b51b6f33359fc8c4f50f43eae7a1dc3f66044e","revisions":{"d8a14ac503981f4abae5a1f0f47237733e3533d8":{"kind":"REWORK","_number":1,"created":"2020-08-07 19:49:08.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/05/745405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/05/745405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/05/745405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/05/745405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/05/745405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/05/745405/1"}}},"commit":{"parents":[{"commit":"7a69fa081ca18eec487350a8adee6e96e1903a7d","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/7a69fa081ca18eec487350a8adee6e96e1903a7d"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-07 19:48:08.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-07 19:48:08.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nAlso removing -test_domain and -test_update_domain sufix from\ndomain name since they have a 64 char limit also.\n\nremove -IdentityCosumer from test test_oauth_consumers_rbac.py.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nAlso had to add mock to test-requirements.txt since patrole extends\ntempest tests base.py and the greatest version of tempest 24.0.0\ndo not pip install mock it is only in tempest test-requirements.txt\n\nChange-Id: Ic4014938c4a2dae64892954d8638f9d8d519a234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/d8a14ac503981f4abae5a1f0f47237733e3533d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/d8a14ac503981f4abae5a1f0f47237733e3533d8"}]},"branch":"refs/heads/master"},"a5b51b6f33359fc8c4f50f43eae7a1dc3f66044e":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2020-08-15 22:58:00.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/05/745405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/05/745405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/05/745405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/05/745405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/05/745405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/05/745405/2"}}},"commit":{"parents":[{"commit":"cfac16a78c8f05ab2a893cf15ebbcdbf9b7d48fb","subject":"Fix gate for multiple issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/cfac16a78c8f05ab2a893cf15ebbcdbf9b7d48fb"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-07 19:48:08.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-15 22:57:30.000000000","tz":-300},"subject":"Shorten Identity project name due to 64 char limit.","message":"Shorten Identity project name due to 64 char limit.\n\nRemoved -testproject from identity project created using\nsetup_test_project since rand_name it already has tempest-\nin front of the name. Adding -testproject to a project\nprovides no additional value on a name with restricted size limit.\n\nAlso removing -test_domain and -test_update_domain sufix from\ndomain name since they have a 64 char limit also.\n\nremove -IdentityCosumer from test test_oauth_consumers_rbac.py.\n\nOur downstream tooling we replace tempest- with shorter word\nand unique id to find project, global and orphaned project resources\nin test and production site. This will allow us to find resource\nleaks and submit upstream corrections.\n\nChange-Id: Ic4014938c4a2dae64892954d8638f9d8d519a234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/a5b51b6f33359fc8c4f50f43eae7a1dc3f66044e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/a5b51b6f33359fc8c4f50f43eae7a1dc3f66044e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
