)]}'
{"id":"openstack%2Fpatrole~746480","triplet_id":"openstack%2Fpatrole~master~Id8a599c8754dfc10ffa5fa18c232a9afff180a8e","project":"openstack/patrole","branch":"master","topic":"bug/2008018","hashtags":[],"change_id":"Id8a599c8754dfc10ffa5fa18c232a9afff180a8e","subject":"Fix Extension gates by supporting Skip Execption in test case.","status":"MERGED","created":"2020-08-17 11:07:37.000000000","updated":"2020-09-08 20:08:50.000000000","submitted":"2020-09-08 20:06:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746480-1599595614141-210ac784","meta_rev_id":"6bb138cce06f1dadf4d508f7c5967b4a4e725a3a","_number":746480,"virtual_id_number":746480,"owner":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-08 20:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"value":0,"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":2,"date":"2020-09-08 15:59:36.000000000","_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":2,"date":"2020-09-08 19:01:20.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"value":0,"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":1,"date":"2020-09-08 19:01:20.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"value":0,"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-17 22:07:26.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-08-21 12:52:32.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2020-08-21 12:52:32.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2020-08-21 12:52:32.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-08-21 12:52:32.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2020-08-21 12:52:32.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2020-08-21 12:52:32.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2020-08-28 11:03:56.000000000","updated_by":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"reviewer":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"state":"REVIEWER"},{"updated":"2020-08-28 11:03:56.000000000","updated_by":{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},"reviewer":{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},"state":"REVIEWER"},{"updated":"2020-08-28 11:03:56.000000000","updated_by":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"reviewer":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"state":"REVIEWER"},{"updated":"2020-09-01 08:52:23.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-09-08 15:59:36.000000000","updated_by":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"reviewer":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"state":"REVIEWER"},{"updated":"2020-09-08 19:01:20.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-09-08 20:06:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e0bff2ed7cf542d15160e3a910347566aa48bc4","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-17 11:07:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e890b7e6f8f5c120c92d07b8994dc046551ca30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 12:21:00.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45c56fe10bb04c6598f0fe5df3ffcaa4 : SUCCESS in 3m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fcd3d3eecae4634985b1e12c0306234 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1adb92fb0d684060b75d1e55025b3057 : FAILURE in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8a22ed7c50b40b4a926dadacc2bdb8c : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4923c50981844f7b219c420d51112ee : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5864213af8e4472097b4bc6ff1609f55 : SUCCESS in 4m 42s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/b10cc2477b19483b90f3bb5bb45c89b3 : SUCCESS in 52m 24s\n- patrole-member https://zuul.opendev.org/t/openstack/build/09a98b0b9ddd4bdfbaff9982bdf814f0 : SUCCESS in 1h 11m 45s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/c61e72e456d2483d950c881ab9c41ccb : SUCCESS in 55m 42s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/c905b4b631cf49dc843b38acad340e03 : SUCCESS in 51m 38s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/c922c317f1e340dfaffbf0317648a68a : SUCCESS in 55m 44s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/446d13f55d6c4ff9813ebd1dfd63346e : SUCCESS in 56m 06s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/3e080ac654a94e2a9d6239640ac2c3d9 : SUCCESS in 1h 02m 52s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/9aa7df30fc5c46749d60aff3b1d9c37e : FAILURE in 59m 43s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/ebfa6ef8f24f49f287a5fd59d2be788d : FAILURE in 38m 12s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/43c3ff525571410e84cc6dd8d6cc2fc9 : FAILURE in 38m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45cb31e7adca4a05fff9d9bf16e0950e714e889d","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-17 16:21:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5643b846f6fc3fcbea0a3e5ca4a9b4a895d84848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 18:13:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5e787d8eb964e02ab0ee868e3461be7 : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b109ca49e33d46c0a4cb25b94757b3e1 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e85fef9f39794975b093f102f9a59f60 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1632f2d37f3142c3ad96fae8ea4188b1 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c6634ac95aa4e3fac09cd9517bf2fb3 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69a02e7caee4d3baffe163ae986e1f9 : SUCCESS in 8m 01s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/e35aa8fed09747b388e38b9df33c4bee : SUCCESS in 46m 31s\n- patrole-member https://zuul.opendev.org/t/openstack/build/ab4fa3f35b9c47e68fdd56b6d6209882 : SUCCESS in 56m 13s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/61b3828b51f443a39c98442a10f992af : SUCCESS in 1h 01m 18s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/72356a2fec0847a394795b18a2407ac6 : SUCCESS in 52m 43s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/b055ee6365784691a0dbc97c015018e0 : SUCCESS in 1h 01m 57s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/f3a07cbc7cfe4af6b405947a9a04757f : SUCCESS in 54m 14s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/010df0fc9cf84057872cedbb5c1d64c9 : SUCCESS in 1h 36m 53s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/35b695e0b5eb413186cc13627358fdf6 : FAILURE in 49m 16s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/7021c8c26db0492e81d67dad13f963c4 : FAILURE in 25m 28s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/4aa56b0c214a4f50943e65566bb8bc94 : FAILURE in 33m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d00520772e28d681b63f6e5b89e6baad3f5d5cbe","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-17 19:54:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f3e7a544bb3857c42c64b3b21d5a488f42eac8e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 21:16:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62c5a12483ff4cb7893c74d0e708c34d : SUCCESS in 2m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8b11f846df54bb9a3c5717095441da8 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/458e8a7f0ce2452f961feaf87494d385 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d19ac6f95b84985a9c16ed57735fc43 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8a3765dfba043f9ab28c7bd66585275 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e48c9068637491492dcfe112c113ad4 : SUCCESS in 8m 07s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/c070e934ec1c4b4eaf8440b602e4af54 : SUCCESS in 1h 07m 59s\n- patrole-member https://zuul.opendev.org/t/openstack/build/ce5c39dd85214616b7ba956fd8fc7b12 : SUCCESS in 55m 10s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/443d4534d19548759bab0982fbda13ec : SUCCESS in 56m 16s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/842485bd1be04bc78e5af456dcf83c6b : SUCCESS in 55m 33s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/b7d0c72c5ce14e70849547b43e2cd2dd : SUCCESS in 57m 22s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/0c40a13ece85404b98ec8ef3e715f3a0 : SUCCESS in 56m 49s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/780f1b04e39f4848a5f303e8535c2161 : SUCCESS in 54m 40s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/ffaa88197df54a61b383b918fc3987fa : FAILURE in 1h 16m 31s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/26d226da84334024a3d7a190c69e1ce4 : SUCCESS in 32m 54s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/cf6d229fc6364019935e5615bdd22d39 : SUCCESS in 39m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"24a16d519a65578c47c6b8af129be70dc7a15f6b","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-08-17 22:07:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nThis makes sense to me but due to some existing Tempest issues with the exception class type the solution requires a bit more handling","accounts_in_message":[],"_revision_number":3},{"id":"bac11f4ac24195ff1e1a753a9022b11a42e0909c","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-18 00:09:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"38a80908e6f09921db67cb5a8f84ff74972e6f82","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-18 02:36:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8fa2b6447faaee1fe8cd37d680d3a9a9709fcb22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 03:47:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da71e16353fd41ea9764e59fa981e7e4 : SUCCESS in 3m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b13d8de2ac3d4aa191cafeb9c1a39ff4 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bde5dbd46934948ab57d84fe5e1d2e9 : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5d4dc7f90014237b459f700c0f0d952 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11bff16e25444b6386bf6e35883859ef : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20c8e4412b124480b6755be8460849a0 : SUCCESS in 6m 13s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/8699f63e97fc4092ae2e924f0d8b3548 : SUCCESS in 42m 21s\n- patrole-member https://zuul.opendev.org/t/openstack/build/1bf7dc1f33464a8fb95c54ad3eb8dd78 : SUCCESS in 53m 23s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/901b295dc29d452cb20a99cc5060c150 : SUCCESS in 53m 11s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/22b14abeae3c472dba91675d43e124a2 : SUCCESS in 56m 19s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/1f9f874518ac40caabbbdf49e7e54114 : SUCCESS in 51m 29s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/884e3636a14f480e8f5c6a75117aabce : SUCCESS in 51m 53s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/aba0f0d41c094302b705ce19416a73cc : SUCCESS in 1h 02m 23s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/abd990e280ee42ca8bc805093b377dcc : FAILURE in 1h 04m 52s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/5b0219bf0a064402bbd04be78c16246f : SUCCESS in 33m 28s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/8164faeaf8074ab399c09f332e0ba348 : SUCCESS in 40m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5aa0f475ff5b2db64358e9130e5a53049d822ca6","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-21 11:26:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bd6000cf14b2917c8109954b7679ae2b0e4a07fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 12:54:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47bb365484994d5fa280d8a76fcdbad6 : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ada0bc74b024c5ba0d57ddd6f64fe93 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84db5fea80694de7b95ca13ac704ff65 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e1ca885386643ada87715a7624fd7f1 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3bbc183b4aa45a885e8b3dddcf41fe4 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85eb17b726304ea4930febe93457210e : SUCCESS in 6m 20s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/2a9a2511cb95433da61b13c4bc3fe9de : SUCCESS in 52m 15s\n- patrole-member https://zuul.opendev.org/t/openstack/build/a8ebcdfea38b405a8a85926e83afeca2 : SUCCESS in 1h 11m 37s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/9c519c7022d2470c8d43adcec652a67b : SUCCESS in 1h 06m 27s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/b7874f06b9da4acb8cdba6c281409ee5 : SUCCESS in 1h 03m 26s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/c3c68689bc3f4b519cbf0800e1d8ed51 : SUCCESS in 1h 10m 43s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/1355b3b0fee04336be2cb8150d6f3450 : SUCCESS in 1h 11m 18s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/5ef1dcfc8e744e99896f9b9984d9e32d : SUCCESS in 1h 03m 29s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/6a9710cd62014b7d9e420c839bf68b12 : FAILURE in 1h 00m 25s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/fe562e8f3b1f4198afefcee4cc9387ef : SUCCESS in 38m 11s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/6d190226ae5c42c4a28607fbb506ee58 : SUCCESS in 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b91aa51e3e83b0c6028c766d928442849d68987e","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-28 16:36:29.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"3645153e5e967bff570999c7e37035d1a96a2bfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 17:46:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f7e23a2b1a34bc0ad456e57d6b8e027 : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8177089d02d41008cd9e8fa2020adae : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff41cb0346741469fe583453af41146 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4b79fc8dc494a188ea60f894ce8cebb : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c1a2790afe242c1b242214519e0b445 : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c3c0637ab94a569c320d84fd47690d : SUCCESS in 7m 00s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/cb4b827255fe42b7b45de0bea8e4fe6b : SUCCESS in 48m 09s\n- patrole-member https://zuul.opendev.org/t/openstack/build/9cddba9eb5e9445283dd22125a0c381a : SUCCESS in 1h 03m 17s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/8c54d59e79904cdf9fdeccd16e97a994 : SUCCESS in 59m 17s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/87a5703436f740699d3573a3ad661cb3 : SUCCESS in 54m 52s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/d538e87a6adf418f9f66b1aaac16e232 : SUCCESS in 56m 43s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/7b9c1b1ad2604182990fac2bf424a585 : SUCCESS in 54m 56s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/a7e1cec0955e4e60bea5e539af2e6b0a : SUCCESS in 57m 55s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/fbf2b16fa2904042a18b510ca9ea9ce4 : FAILURE in 59m 29s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/96b56484ed5a475fb4476fa78f7e7940 : SUCCESS in 34m 34s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/d55147f18d4741f39d009feb96115d0a : SUCCESS in 37m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1a3a43ef5dc8a064fe135e3800731fe6b7a1ce47","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-29 11:28:49.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"75f9858cfbdb0507f5b161fd4644398151913304","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2020-08-31 18:37:01.000000000","message":"Patch Set 6: Code-Review+2\n\n(6 comments)\n\nJust minor nits, nothing to hold back from going forward. But just marked them in case a new PS is requested in subsequent review.","accounts_in_message":[],"_revision_number":6},{"id":"5405b09b151a8849041100dc117bab23c4eab7d0","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-31 23:48:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9707bcb5a11c683ba055849b87119860de729f79","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-31 23:48:18.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"861aa1917fa109872bb332a00b61acd4b51ab12c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 00:49:50.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3236c8249f10462796f95941a7bb73cf : SUCCESS in 3m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09a3c0fda4154ef884915191d5ff137d : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92e4f67daa5446319bd6a35d9cd186c5 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80d3481ebca147d385a4d3b5ea87c62e : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5039bc823aff45c6841530e4cefcf22b : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13feff72ad4d4c82b4fcd9d60098b5d7 : SUCCESS in 6m 27s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/a75dae20cb0649f69d147d293d9d665c : SUCCESS in 46m 41s\n- patrole-member https://zuul.opendev.org/t/openstack/build/8864c8cb9d7c4c65a62ffc152ff7a9c8 : SUCCESS in 53m 26s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/014926e744784214a10b05d82c96c795 : SUCCESS in 49m 16s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/01ed029db3c8429d9e5d259ec99e5257 : FAILURE in 28m 49s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/72d1c83babee456c955fa04513a8fd47 : FAILURE in 31m 01s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/e25060b2ae2b4bf19d05edb68c2321f4 : FAILURE in 17m 29s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/5ad48f9b92e14e60b90e3ac8f74a0789 : SUCCESS in 55m 18s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/57012eaa38194269bfbb161bdd6a3c3c : FAILURE in 57m 09s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/e4681780f6284f09ae56d95952d4a8bc : SUCCESS in 31m 34s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/419d6eae3b5e4a7b8e3fef670ef47777 : SUCCESS in 31m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5a7323171069a95202264833ff76c86d11c8db18","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-01 08:52:23.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f0273d569451779f8d3eca2bd10d4749090a3828","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-09-02 10:59:18.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"b6382fac5b309545b29e370ca0a9e36858fba03b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 12:46:57.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05fb1a337e444bdea9a78c4db96b2811 : SUCCESS in 3m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae3dcca8d2d34de58d5e67ab4a74de5d : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a13b9b91ebf4376bab627279c0ccadb : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/996ac9f5d1e64b3195c3ff14c4ef8d9c : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dff0d71d73404791a40437bffaf2f1b7 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a77bc19f70b047f58a2ed1b8affcca99 : SUCCESS in 5m 45s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/4774cd8dd49b4d1c9d524affa738c391 : SUCCESS in 46m 24s\n- patrole-member https://zuul.opendev.org/t/openstack/build/f8fd7cdf8d6a4cf0955e27f60be6d774 : SUCCESS in 53m 04s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/639428559b33478a89c2bb85f1c890bf : SUCCESS in 52m 29s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/7c54c34e1bed4093852da4e5533e16b9 : FAILURE in 31m 33s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/8bd33f0ce208431f9f0b2852f916a5dc : FAILURE in 34m 01s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/0f8cc9267b414ae1acdd731bc716755b : FAILURE in 31m 00s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/01022a498ce84565bd192ef7669d82ae : SUCCESS in 56m 32s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/90f18c14d32841fa911a7b820044dd81 : FAILURE in 1h 02m 14s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/1f5e16e958454429b6329111d8e53faa : SUCCESS in 34m 13s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/17baa8ce9acd4feb9a73126aebefcbae : SUCCESS in 37m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ff14c236d10e1657773491155043d9a0731d976b","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-09-08 01:26:55.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"b8c700ee7c0a48e2991dd154a1cedf189194ac32","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-09-08 01:32:04.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e77c688f6b6fe8302971efc4e0eb739056ab17f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 03:30:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48773fc3fd4b4989828e7de3261a722e : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a78782c479894093a2bd338d491217d5 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b73541252f346b790de40908188f785 : SUCCESS in 2m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cf44861d2314cea8c39ee906bdd1ad5 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b987927cf4f402a9dc4388404a3a2c7 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c86a0fb6e9a42948b4391d500b3f26b : SUCCESS in 6m 14s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/108f0c5a4729408a84e4ad4fef543807 : SUCCESS in 48m 05s\n- patrole-member https://zuul.opendev.org/t/openstack/build/5975dc61d87c448fafbb4e5255352340 : SUCCESS in 1h 56m 14s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/50298d5975394a46a161b73544eb6052 : SUCCESS in 46m 31s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/8810676aea68421b829b9607837b3246 : SUCCESS in 43m 26s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/358bba22f6b24560afd79a43a0619e3c : SUCCESS in 1h 02m 41s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/924f3e6ade3441dea68b0eec8fbe0423 : SUCCESS in 50m 18s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/ef6cce40d11941669a4624b9de546903 : FAILURE in 50m 46s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/d8459fe97cbc4943bbb27530e9b7e2a0 : FAILURE in 58m 01s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/3cee8ae2ba954d728fd845568d943f16 : SUCCESS in 32m 04s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/04f3e2a6c1f24aca9b66f9cc535519d3 : SUCCESS in 34m 23s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1b23e62da3a811e7c08be1673e9437e1efafc1af","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2020-09-08 15:59:36.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"6e61fd8534bc313beb3b39b75970d2b87f9b72b3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-08 19:01:20.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nlgtm, let\u0027s fix multinode also later.","accounts_in_message":[],"_revision_number":8},{"id":"a40013173c0a3f293bfe86cbdeb3b0e58cd1ee66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 19:02:18.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"10b1f1ab6cc0f59953d7a38a81cc087df1b9bcd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 20:06:53.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e437ddde8964593829e20bec4fc7065 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d04bd111ff64729ad935b3e599a063f : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/030d652b311e4643833c167ed742edf9 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3293612ed3b64ca4b29f425c7e51e341 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65b64034c24c47f99d2823ba9e6012a8 : SUCCESS in 6m 00s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/b3546737121c4e65b639e3f26b4efeaf : SUCCESS in 51m 51s\n- patrole-member https://zuul.opendev.org/t/openstack/build/a6bf98f3ddb64efcbe0a7332e8d14fc1 : SUCCESS in 56m 27s","accounts_in_message":[],"_revision_number":8},{"id":"5c9e16be1ce42b17b7082f8622c1f9a9fcd9f47b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 20:06:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"80db3e84d01cbf61362b9886737567f2339fc7c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 20:08:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf082fb6a0d4efa8551c91641e2a5b5 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"89d9ff89fed8f09e5987646b04ccaaa4d34af833","revisions":{"029046a86291a9fb451c428f88d4bdc96cd983b6":{"kind":"REWORK","_number":1,"created":"2020-08-17 11:07:37.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/80/746480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/80/746480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/80/746480/1"}}},"commit":{"parents":[{"commit":"318baaa6abe0dfdb2ecebdaa47ea0160c11c2690","subject":"Set admin, member and readeer gates to be the same","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/318baaa6abe0dfdb2ecebdaa47ea0160c11c2690"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-17 10:59:35.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-17 10:59:35.000000000","tz":-300},"subject":"Fix Extension gates by supporting SKipExecption in test case.","message":"Fix Extension gates by supporting SKipExecption in test case.\n\nThe test test_delete_auto_allocated_topology is the only\ntest in patrole that does a Skip test exception in the test_ method\nso I had to add support to the rbac_rule_validation decorator\nso the framework would allow the skip execption to be thrown\nthrough the decorator and make it throught to the unittest\nframework as a skip.\n\nhttps://storyboard.openstack.org/#!/story/2008018\n\nChange-Id: Id8a599c8754dfc10ffa5fa18c232a9afff180a8e\nCloses-Bug:2008018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/029046a86291a9fb451c428f88d4bdc96cd983b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/029046a86291a9fb451c428f88d4bdc96cd983b6"}]},"branch":"refs/heads/master"},"5f5fff48cddc16c18aa1f986c802f89fe66b212a":{"kind":"REWORK","_number":2,"created":"2020-08-17 16:21:21.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/80/746480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/80/746480/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/80/746480/2"}}},"commit":{"parents":[{"commit":"318baaa6abe0dfdb2ecebdaa47ea0160c11c2690","subject":"Set admin, member and readeer gates to be the same","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/318baaa6abe0dfdb2ecebdaa47ea0160c11c2690"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-17 10:59:35.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-17 16:21:06.000000000","tz":-300},"subject":"Fix Extension gates by supporting SKipExecption in test case.","message":"Fix Extension gates by supporting SKipExecption in test case.\n\nThe test test_delete_auto_allocated_topology is the only\ntest in patrole that does a Skip test exception in the test_ method\nso I had to add support to the rbac_rule_validation decorator\nso the framework would allow the skip execption to be thrown\nthrough the decorator and make it throught to the unittest\nframework as a skip.\n\nhttps://storyboard.openstack.org/#!/story/2008018\n\nChange-Id: Id8a599c8754dfc10ffa5fa18c232a9afff180a8e\nCloses-Bug:2008018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5f5fff48cddc16c18aa1f986c802f89fe66b212a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5f5fff48cddc16c18aa1f986c802f89fe66b212a"}]},"branch":"refs/heads/master"},"10140b269cb7965ff67af93a9941990cdccdedf8":{"kind":"REWORK","_number":3,"created":"2020-08-17 19:54:09.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/80/746480/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/80/746480/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/80/746480/3"}}},"commit":{"parents":[{"commit":"318baaa6abe0dfdb2ecebdaa47ea0160c11c2690","subject":"Set admin, member and readeer gates to be the same","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/318baaa6abe0dfdb2ecebdaa47ea0160c11c2690"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-17 10:59:35.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-17 18:31:22.000000000","tz":-300},"subject":"Fix Extension gates by supporting SKipExecption in test case.","message":"Fix Extension gates by supporting SKipExecption in test case.\n\nThe test test_delete_auto_allocated_topology is the only\ntest in patrole that does a Skip test exception in the test_ method\nso I had to add support to the rbac_rule_validation decorator\nso the framework would allow the skip execption to be thrown\nthrough the decorator and make it throught to the unittest\nframework as a skip.\n\nhttps://storyboard.openstack.org/#!/story/2008018\n\nChange-Id: Id8a599c8754dfc10ffa5fa18c232a9afff180a8e\nCloses-Bug:2008018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/10140b269cb7965ff67af93a9941990cdccdedf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/10140b269cb7965ff67af93a9941990cdccdedf8"}]},"branch":"refs/heads/master"},"cca0b31bd5be3fda7ac6d124734afe7b4655ac95":{"kind":"REWORK","_number":4,"created":"2020-08-18 02:36:25.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/80/746480/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/80/746480/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/80/746480/4"}}},"commit":{"parents":[{"commit":"318baaa6abe0dfdb2ecebdaa47ea0160c11c2690","subject":"Set admin, member and readeer gates to be the same","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/318baaa6abe0dfdb2ecebdaa47ea0160c11c2690"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-17 10:59:35.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-17 23:50:06.000000000","tz":-300},"subject":"Fix Extension gates by supporting SKip Execption in test case.","message":"Fix Extension gates by supporting SKip Execption in test case.\n\nThe test test_delete_auto_allocated_topology is the only\ntest in patrole that does a Skip test exception in the test_ method\nso I had to add support to the rbac_rule_validation decorator\nso the framework would allow the skip execption to be thrown\nthrough the decorator and make it throught to the unittest\nframework as a skip. Let me know if anyone has a better\nway to do this.  I am just trying to get all of the gates running.\n\nFYI I am doown to one testing in multinode that is only\nsupported by system scoped token in Nova so I think that\nwe will have to wait until tempest supports scope token and\nthen when patrole support them we should be able to get all\ngate and non voting gates to work.\n\nhttps://storyboard.openstack.org/#!/story/2008018\n\nChange-Id: Id8a599c8754dfc10ffa5fa18c232a9afff180a8e\nCloses-Bug:2008018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/cca0b31bd5be3fda7ac6d124734afe7b4655ac95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/cca0b31bd5be3fda7ac6d124734afe7b4655ac95"}]},"branch":"refs/heads/master"},"75f038e8d2feb7bdc39875c58d93718aefc4eee8":{"kind":"REWORK","_number":5,"created":"2020-08-21 11:26:45.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/80/746480/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/80/746480/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/80/746480/5"}}},"commit":{"parents":[{"commit":"318baaa6abe0dfdb2ecebdaa47ea0160c11c2690","subject":"Set admin, member and readeer gates to be the same","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/318baaa6abe0dfdb2ecebdaa47ea0160c11c2690"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-17 10:59:35.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-21 11:26:27.000000000","tz":-300},"subject":"Fix Extension gates by supporting SKip Execption in test case.","message":"Fix Extension gates by supporting SKip Execption in test case.\n\nThe test test_delete_auto_allocated_topology is the only\ntest in patrole that does a Skip test exception in the test_ method\nso I had to add support to the rbac_rule_validation decorator\nso the framework would allow the skip execption to be thrown\nthrough the decorator and make it throught to the unittest\nframework as a skip. Let me know if anyone has a better\nway to do this.  I am just trying to get all of the gates running.\n\nFYI I am doown to one testing in multinode that is only\nsupported by system scoped token in Nova so I think that\nwe will have to wait until tempest supports scope token and\nthen when patrole support them we should be able to get all\ngate and non voting gates to work.\n\nhttps://storyboard.openstack.org/#!/story/2008018\n\nChange-Id: Id8a599c8754dfc10ffa5fa18c232a9afff180a8e\nCloses-Bug:2008018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/75f038e8d2feb7bdc39875c58d93718aefc4eee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/75f038e8d2feb7bdc39875c58d93718aefc4eee8"}]},"branch":"refs/heads/master"},"9425f9d13a3010989f658eb61467e0af1a5ba8a7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-08-28 16:36:29.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/80/746480/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/80/746480/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/80/746480/6"}}},"commit":{"parents":[{"commit":"3ab156a54579aa4e97ed9d715148d652cc4bdb17","subject":"Merge \"Set admin, member and reader gates to be the same\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/3ab156a54579aa4e97ed9d715148d652cc4bdb17"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-17 10:59:35.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-28 16:36:29.000000000","tz":0},"subject":"Fix Extension gates by supporting SKip Execption in test case.","message":"Fix Extension gates by supporting SKip Execption in test case.\n\nThe test test_delete_auto_allocated_topology is the only\ntest in patrole that does a Skip test exception in the test_ method\nso I had to add support to the rbac_rule_validation decorator\nso the framework would allow the skip execption to be thrown\nthrough the decorator and make it throught to the unittest\nframework as a skip. Let me know if anyone has a better\nway to do this.  I am just trying to get all of the gates running.\n\nFYI I am doown to one testing in multinode that is only\nsupported by system scoped token in Nova so I think that\nwe will have to wait until tempest supports scope token and\nthen when patrole support them we should be able to get all\ngate and non voting gates to work.\n\nhttps://storyboard.openstack.org/#!/story/2008018\n\nChange-Id: Id8a599c8754dfc10ffa5fa18c232a9afff180a8e\nCloses-Bug:2008018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/9425f9d13a3010989f658eb61467e0af1a5ba8a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/9425f9d13a3010989f658eb61467e0af1a5ba8a7"}]},"branch":"refs/heads/master"},"a8be9165159baf1d9b483ed516c9251a25f4c22e":{"kind":"REWORK","_number":7,"created":"2020-08-31 23:48:08.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/80/746480/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/80/746480/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/80/746480/7"}}},"commit":{"parents":[{"commit":"3ab156a54579aa4e97ed9d715148d652cc4bdb17","subject":"Merge \"Set admin, member and reader gates to be the same\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/3ab156a54579aa4e97ed9d715148d652cc4bdb17"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-17 10:59:35.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-31 23:46:45.000000000","tz":-300},"subject":"Fix Extension gates by supporting Skip Execption in test case.","message":"Fix Extension gates by supporting Skip Execption in test case.\n\nThe test test_delete_auto_allocated_topology is the only\ntest in patrole that does a Skip test exception in the test_ method\nso I had to add support to the rbac_rule_validation decorator\nso the framework would allow the skip exception to be thrown\nthrough the decorator and make it through to the unittest\nframework as a skip. Let me know if anyone has a better\nway to do this.  I am just trying to get all of the gates running.\n\nFYI I am down to one testing in multinode that is only\nsupported by system scoped token in Nova so I think that\nwe will have to wait until tempest supports scope token and\nthen when patrole support them we should be able to get all\nvoting gate and non voting gates to work.\n\nhttps://storyboard.openstack.org/#!/story/2008018\n\nChange-Id: Id8a599c8754dfc10ffa5fa18c232a9afff180a8e\nCloses-Bug:2008018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/a8be9165159baf1d9b483ed516c9251a25f4c22e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/a8be9165159baf1d9b483ed516c9251a25f4c22e"}]},"branch":"refs/heads/master"},"89d9ff89fed8f09e5987646b04ccaaa4d34af833":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":8,"created":"2020-09-08 01:32:04.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/80/746480/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/80/746480/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/80/746480/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/80/746480/8"}}},"commit":{"parents":[{"commit":"8746b77bfed8aa79ded324cc74ddb96e9cda97f7","subject":"fix patrole-multinode-member non-voting gate by skip_because a bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/8746b77bfed8aa79ded324cc74ddb96e9cda97f7"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-17 10:59:35.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-09-08 01:30:39.000000000","tz":-300},"subject":"Fix Extension gates by supporting Skip Execption in test case.","message":"Fix Extension gates by supporting Skip Execption in test case.\n\nThe test test_delete_auto_allocated_topology is the only\ntest in patrole that does a Skip test exception in the test_ method\nso I had to add support to the rbac_rule_validation decorator\nso the framework would allow the skip exception to be thrown\nthrough the decorator and make it through to the unittest\nframework as a skip. Let me know if anyone has a better\nway to do this.  I am just trying to get all of the gates running.\n\nFYI I am down to one testing in multinode that is only\nsupported by system scoped token in Nova so I think that\nwe will have to wait until tempest supports scope token and\nthen when patrole support them we should be able to get all\nvoting gate and non voting gates to work.\n\nhttps://storyboard.openstack.org/#!/story/2008018\n\nChange-Id: Id8a599c8754dfc10ffa5fa18c232a9afff180a8e\nStory: 2008018\nTask: 40670\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/89d9ff89fed8f09e5987646b04ccaaa4d34af833"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/89d9ff89fed8f09e5987646b04ccaaa4d34af833"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
