)]}'
{"id":"openstack%2Fpatrole~747528","triplet_id":"openstack%2Fpatrole~master~I6963fdff199fca342620ab15948ce2c1d29c2c7a","project":"openstack/patrole","branch":"master","topic":"bug/2008051","hashtags":[],"change_id":"I6963fdff199fca342620ab15948ce2c1d29c2c7a","subject":"fix patrole-multinode-member non-voting gate by skip_because a bug","status":"MERGED","created":"2020-08-22 14:43:30.000000000","updated":"2020-09-07 14:16:56.000000000","submitted":"2020-09-07 14:15:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747528-1599488115672-69a35627","meta_rev_id":"5e751dfa1b5318fefddc6eafeee3e883bfe26d9f","_number":747528,"virtual_id_number":747528,"owner":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-07 14:15:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"value":0,"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":2,"date":"2020-09-02 10:58:29.000000000","_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-02 10:58:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"value":0,"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":2,"date":"2020-09-02 10:58:29.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":1,"date":"2020-09-02 10:58:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"value":0,"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":1,"date":"2020-09-07 13:12:45.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-27 14:29:00.000000000","updated_by":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"reviewer":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"state":"REVIEWER"},{"updated":"2020-08-28 11:05:08.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2020-08-28 11:05:08.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2020-08-28 11:05:08.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-08-28 11:05:08.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2020-08-28 11:05:08.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2020-08-28 11:05:08.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-08-28 11:05:08.000000000","updated_by":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"reviewer":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"state":"REVIEWER"},{"updated":"2020-08-28 11:05:08.000000000","updated_by":{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},"reviewer":{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},"state":"REVIEWER"},{"updated":"2020-08-28 11:05:08.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2020-08-28 11:05:08.000000000","updated_by":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"reviewer":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"state":"REVIEWER"},{"updated":"2020-08-28 11:05:08.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-08-29 12:59:35.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-09-07 13:12:45.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-09-07 14:15:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b115df9f2c8d0785a7403ef9e5d768d3dd075d34","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-22 14:43:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15ab6da04e2fcb20be4ccf1409e68624e9ba1a17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 15:59:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/473385f2a4db412ab82c973d03081a18 : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73998e51b52343bc9ed2e3cf40d5f9c9 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f89b38b7cd640fbb5274f93b96ae11d : SUCCESS in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b99de1008b1c43a49fb62d7a622a7191 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/026cf74ac1f74c4993eb2fc22081c8c9 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7df5c67081b47ec8caec18280a71cae : SUCCESS in 5m 19s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/6ab104c33fbb4074b0c3395b2ac683fd : FAILURE in 50m 57s\n- patrole-member https://zuul.opendev.org/t/openstack/build/5a9ff64c63e14d9e9ca58fd883287d64 : FAILURE in 1h 10m 29s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/4abf25c1b6774159be1d3956ea777399 : FAILURE in 1h 08m 37s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/faaf308364ca478b993fdf9d30253cf8 : SUCCESS in 1h 05m 42s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/f2af364d7de34f47b6f14f42d19d1304 : SUCCESS in 1h 02m 07s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/ffcbfb5fc9b649c4bd8172b7b735c536 : SUCCESS in 1h 03m 50s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/f4966e3656154c30a901ba4da0ed01ad : SUCCESS in 58m 09s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/ef3e314fd9d04300a74c38d9764ba30b : SUCCESS in 57m 13s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/b4dd7b62d3064c8ba0d057201c336a66 : FAILURE in 33m 48s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/07ed68cbd201466590deed789c2a0ca1 : FAILURE in 34m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ead379cef6e49a9283a8804529c2c1748fd589a","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-23 12:24:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6cb9fcf9c52a64ffd38c420c34d77acc6dda197b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 13:43:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/400d03fcad234e339eec2f7b851f03e1 : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15d693f88ec642288349fec71e1c6aae : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e3219d9382488c834c224cc384a0a6 : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e07d0673912a4ca08c98385470ec29d4 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e184e07f5c6b45049d5a7ef176dfa03d : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6163fa1f71ec4d1e9e314f2cc20785bc : SUCCESS in 7m 07s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/c08c67e8a38241ac94127e7f24396c95 : FAILURE in 1h 06m 36s\n- patrole-member https://zuul.opendev.org/t/openstack/build/50a46f1d65594ebd98b04ad1eaebd260 : FAILURE in 1h 13m 00s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/dd07dea900014158a3c7985a2e51e269 : FAILURE in 1h 07m 18s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/c4b184efb89b41f3bd48f3c10268d438 : SUCCESS in 1h 13m 55s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/884d83896b284e7d9ce17058ddd83f3b : SUCCESS in 55m 34s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/5ba02acf38a343e38c76a5a8db31e19d : SUCCESS in 1h 18m 38s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/fabc45616d1046b2aa24a3e7a3347994 : SUCCESS in 1h 05m 13s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/c7dedfea210742dc873424c9803f4a73 : SUCCESS in 1h 04m 28s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/7cf54370c3a1486995aaacd2ea878dd7 : FAILURE in 35m 56s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/f360fb280a404237b549765d71a343df : FAILURE in 45m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e4ee095f0141ba6b968ee4a9560ef2c82a4f641","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-23 21:38:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e85ed85cbe6e51fcb2b1eae32d07fc7f6db6ef9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 22:42:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a10026cc05834c96b2dc92e61770cc3c : SUCCESS in 2m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d9eedaca0974dc388b215eeb64f02fe : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b12afb668c6b457b8e132e88968b9abd : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32b0cdf0454e45e6b0a80bafcc00c59f : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e7541d94c2f4a3d8c3a74d40452172e : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47593c04d3ef4eb38eae95dff746b86e : SUCCESS in 4m 57s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/009089f0f933452bb58c1507f86f98a0 : FAILURE in 42m 21s\n- patrole-member https://zuul.opendev.org/t/openstack/build/4ccedd9bbf0d45e293a2e43bdebc87d7 : FAILURE in 1h 00m 45s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/e0d02e75a3ab48a7afa304edf08cac9e : FAILURE in 52m 35s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/13354568477046d6a9490686c3319da1 : SUCCESS in 1h 00m 23s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/b085a60999b44eeea9c181248f0fc594 : SUCCESS in 50m 59s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/b55257a875ae4de7a939689aab13cdbf : SUCCESS in 1h 02m 15s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/08b402ee78a14d5f9d836ff397c0cf86 : SUCCESS in 1h 02m 40s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/6c2376b1ef194794815e0b383703e17a : SUCCESS in 58m 27s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/c2619c14f1f343aaaab112669fa8f26b : FAILURE in 22m 51s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/947ee152e08948babd1099e6c9fb7e73 : FAILURE in 32m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"937ae8fc25d570565f4e22909c00e0161a5c6559","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-24 02:39:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b6d76df08aa4142085460005bd2bcc7f215ee61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 03:56:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a8f8769f9674b648b53829020683767 : SUCCESS in 2m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcfc5adeb6774a9d97017994c36b6aae : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bda869052a749a985ffe8a961900845 : SUCCESS in 2m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d64f26b513f54d85840c4d7cd0da0c24 : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e46556cfea3422a96f9747f0d7c04e3 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/038e280f861c41f098c24d812cafd7cd : SUCCESS in 4m 38s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/99940b40d3c147e6ace471497292fd28 : FAILURE in 55m 34s\n- patrole-member https://zuul.opendev.org/t/openstack/build/c38fa5b681d7474297be7f4690fade01 : FAILURE in 1h 09m 53s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/c1114f010df74911947e59c4dcb36fc2 : FAILURE in 1h 01m 24s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/d1a434643d664244a05371b6fd947275 : SUCCESS in 1h 00m 10s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/45d24e2e5dde4526ae460d467dbc2685 : SUCCESS in 1h 10m 57s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/2fce0b0eda3d472491bac47dfc488281 : SUCCESS in 1h 05m 44s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/1ac26468c58f483c9f9a1d2bc435e58d : SUCCESS in 55m 41s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/327ad2e8498044cca596135e515f264f : SUCCESS in 57m 11s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/07c22141246648b888e0829c7c24f974 : FAILURE in 31m 56s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/b9311d3bd4de40c99770c0908a931e32 : FAILURE in 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97bf9760dd80741183bb96dcf2ca1bda5c3c618e","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-24 18:34:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5bd5ebbd667e4d616a29022f167604a9bd8e5591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 19:55:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a48a80ba7311443e8659f6d20c55d395 : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ee2b4bbd0654c5c81c4b06c71c11a04 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7da25106351747219a65724e9cb91e7e : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66425a8179b746cd8340557e482af9a1 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06abbc56be0242a980c070de9d599eb1 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ea8f6efb6e4e95bc3ed6f77eb0f859 : SUCCESS in 6m 20s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/a4896687a1ab4221a786adc0cad84f40 : SUCCESS in 51m 27s\n- patrole-member https://zuul.opendev.org/t/openstack/build/e725cf9f6a2a49699f7915451f3707c5 : SUCCESS in 1h 06m 07s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/0e7332b3832149308a44477a901e1f08 : SUCCESS in 1h 04m 27s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/b14457f9ddbe4d23b1014f29d681534d : SUCCESS in 1h 06m 08s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/f671dbd6f2c34ad29c4129994a3d9b72 : FAILURE in 1h 14m 43s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/10f2642c51f645b298ab77cef16946f2 : SUCCESS in 1h 09m 58s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/ef5d189861344fec8f2102eee8e2ec4e : SUCCESS in 50m 15s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/5bdf1618399e491993a8f91911921401 : SUCCESS in 1h 11m 28s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/016cc4ebbcf74b6783e84154ba7370a0 : FAILURE in 35m 20s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/e5fa59c0877541669a894cb5153cbecf : FAILURE in 53m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b4d5d3ed56dca7a39291610b94a1a884d3c9a5f5","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-24 21:29:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3db378daceccd1964d40a942db8c6b3db52471fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 22:53:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da3f55aff6064fc7ab80cf001e38a90a : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37f949a77f5948b18385c7205e888aa2 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aaf747cbcab4e7d9dc6731c326e8b96 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdc877cbe51c49b7bcd00e10bd6af226 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e92eab39660a4f06a0061c90d66c39a1 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c69caa59f741588d09f80d529ab20c : SUCCESS in 7m 03s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/974f58904a0245b49059c2a9a7f5d55a : SUCCESS in 1h 07m 14s\n- patrole-member https://zuul.opendev.org/t/openstack/build/e79188d550bd483bb0faca158266c1bb : SUCCESS in 1h 07m 06s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/de4229fe8da8456fb1b7ddb30b1d455b : SUCCESS in 1h 06m 20s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/5b5aefb3f71741f9adce01e82e449363 : SUCCESS in 1h 03m 49s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/712d7298b0e24c10bf47c1cf6c39eced : SUCCESS in 1h 07m 56s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/accd652a96044759897356c3b6742da6 : SUCCESS in 1h 14m 25s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/78331032d9aa4ca389f719a8de832efd : SUCCESS in 1h 07m 01s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/259a92c50235405fa142c1d7b9863eb7 : SUCCESS in 1h 10m 00s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/79e34ddc8f02454bb9ad07198e1c0e2a : FAILURE in 37m 14s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/34c49fcdf60548819d0977ccded091a8 : FAILURE in 40m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bec1d40cd441410cd249f36c71897c1d49be6325","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-26 12:47:18.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b923e25fc47d5e3e7eb09d8fb78f4b2b3c13dff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 15:22:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6667dd0dbc34d4eae6d0aba4a5da35a : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b382a3ec726479e89a9c834b192b448 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ec7478cbbd345c3a0498acbc64be310 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3d78fb0c2b34e43b4a6b62937298709 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/068525c545f44140bd9985c73528b180 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37edf2516407455196f7935156435c25 : SUCCESS in 8m 29s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/166d8a566de3432e806065d2720b4636 : SUCCESS in 2h 16m 30s\n- patrole-member https://zuul.opendev.org/t/openstack/build/6ae182cf2fa34855a4c9235cd1e3d410 : SUCCESS in 1h 07m 27s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/3e1e7e7228b8457c8f806e707dc7268e : SUCCESS in 1h 06m 45s (non-voting)\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/b32ead37f39246908c017b0df3e3722a : SUCCESS in 1h 05m 57s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/04bd012e7cdb4515ae8097c8311b0fc7 : SUCCESS in 1h 08m 00s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/39762c00baf3410296b18deec55e7a24 : SUCCESS in 1h 05m 09s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/c88a293ed8bf408ebf1ede05d2203821 : FAILURE in 56m 56s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/cf8a007750da472baa731387d3629939 : SUCCESS in 1h 07m 52s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/f636f7a87e4442dda0010f333539b692 : FAILURE in 32m 06s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/b4573b770ec044f7b0ef0bce76323d0e : FAILURE in 41m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5a0764a0736778c33dab3585abb7903b961e3f69","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2020-08-27 14:21:57.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a49fd238ed73d885a6881e28549d133f2fc0a4ea","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2020-08-27 14:29:00.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"67e3b439dbced1d5e54fd08f9b787a0b38f99b39","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-29 11:23:56.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"961bf127dbce2fa4082acb3c7f4fd0c256acdba3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-08-29 12:59:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2b106286179f7f61912eae662d699208615042a0","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-31 07:30:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"018b2d0ed437a08de459775eda8146a604f10145","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2020-08-31 12:48:03.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"afe167039551c48b31dd8dc65259c4be1a42d922","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 12:48:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c8361253dada641c4021ad05141a264cbca0309f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:04:28.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8b1c958331d4a9b874aa14fde786ac6 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/974c3762e04f4b10a1373fb0074874fa : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fadaf4d14b14deb91eaf1446d298c92 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0836811599c0447abdcb686bc02467b5 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03016c09780d431b9913eee0157f8306 : SUCCESS in 4m 29s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/11861a7d49264c60b6a7bd838fd3226d : FAILURE in 14m 50s\n- patrole-member https://zuul.opendev.org/t/openstack/build/c73c9393833c4f51baae2a2eb225590a : FAILURE in 14m 07s","accounts_in_message":[],"_revision_number":3},{"id":"282c85a74a12c0fa161c894b37a26dd117eeaa2d","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-31 18:03:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3250d9cf9164b3fa7ea0a00c3851406c455212aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 19:10:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/622225e2e8494eecae0ab12cd79c3a6d : SUCCESS in 3m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32d15c04bde240bd9dfdd4e986fb5849 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ea7a0872f14261af37e094fd0fb186 : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0d24ac1d35143528bcc5b5cc20eee51 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b61d37c234dc44cfa1f1bbbef955bea2 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85b96ce7a6914efabc54d1f40a942e9a : SUCCESS in 4m 57s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/7cd5e3224deb44cd809fbaac29e6eca1 : SUCCESS in 35m 12s\n- patrole-member https://zuul.opendev.org/t/openstack/build/dc6d522017bc4d0884248d83d775a04f : SUCCESS in 36m 04s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/af940a2c99ac481da8272007a0567483 : SUCCESS in 1h 04m 42s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/b3836b78b2914a5b8a49814e4be01754 : FAILURE in 11m 41s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/ae52a7a566b84928ba8499fcf59227b3 : FAILURE in 24m 21s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/06d10061991f4a1281eba35a508914d0 : FAILURE in 13m 21s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/ef7d61a9b07540e59757c6374f670d6b : SUCCESS in 1h 04m 27s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/6925779c82234ab2bdc9b545e9dd6a8f : SUCCESS in 44m 16s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/1eeba12e85d74b8d930e6c99861d4134 : FAILURE in 37m 36s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/3c3b6f1000034fd9a2dc04be290c8478 : FAILURE in 24m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"34b453de4c8ca40c3c9d8f7d60ae5dca96ded76f","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-31 23:42:42.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"09e236a38e314e8473087a04081a94428efafd80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 00:41:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/276ac11a1421494687bcc5e6f22a4032 : SUCCESS in 3m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f9aa62e09334107ad8f757d59d3b42f : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6998892e3d02478190e83ac5b34fd6b1 : SUCCESS in 2m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b5fcaec897d48bb8b48aef44be9a8c1 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7362883610be42788909698b1d8fbb0e : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66caaf0733204232869bdf9c80a24c48 : SUCCESS in 6m 27s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/b6b4d6f2baaf412db934334867976915 : SUCCESS in 44m 24s\n- patrole-member https://zuul.opendev.org/t/openstack/build/5c375100feac466bafa9f0bdffef433f : SUCCESS in 58m 02s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/62585008ab984e90adc96e1aded2e2f6 : SUCCESS in 52m 44s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/1775b60f06784c1d9192f1fa5f7ef1ca : FAILURE in 31m 19s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/13956b28d72544c5a1c66c7d2d6349ee : FAILURE in 31m 11s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/e237e7c74cee47fe850811ff00a9e959 : FAILURE in 13m 08s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/2d7afa118db44139810c832abb5e62dc : SUCCESS in 57m 37s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/95ddc49340f84defbe372482c8db815b : SUCCESS in 46m 07s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/b2334477aaf44620a5f9903584115f60 : FAILURE in 31m 00s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/4606f23f50ce448cafa25ee647c0c13a : FAILURE in 22m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c9216ce5f6c6717efdcabeb65441ca21f4daf66f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-01 08:25:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"657559c7f4370da09e188b0f0b549c8657e142bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 09:39:22.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df380922d35d4af18198b82c7bf9ee74 : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bc5dd9903ca43b79b4c1259f67d4140 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b50acbc7673499281553ec66150d848 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a99b73c209d46228ffb1151ded58b5a : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b06420ed358c4be986c10010bc878698 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2256cf332fd4c0aa744bf072935ff69 : SUCCESS in 6m 29s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/a55010cfedb4429cb6131284b5b8aaaf : SUCCESS in 53m 40s\n- patrole-member https://zuul.opendev.org/t/openstack/build/bbbfac8258da4218851cedc741fd3c6c : SUCCESS in 55m 25s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/3aaabb9ca4694ac0b8b277a47dee8e65 : SUCCESS in 1h 00m 23s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/ecbc319adff148e1a7fc7d7092ca9681 : FAILURE in 34m 13s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/ca60e87db3424898a17c4e17b04a62ca : FAILURE in 34m 29s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/458458ba680046598cd7c0341906dafb : FAILURE in 21m 37s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/6426a0e89d5b4a478200300a81f461a3 : SUCCESS in 1h 01m 44s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/a84ab43bbc4a4819829164f1c74ae73c : SUCCESS in 58m 34s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/5b14dec9cfd84ca1ade21434a77c08a8 : FAILURE in 23m 51s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/bffbc5530157484f94a73ff903797590 : FAILURE in 38m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"db3e188696cdb1951046e8cc9b840c7484ecfcc5","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-09-01 11:12:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7de5980a7b826f7c2a0a300096b76e3d24d1a37b","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-09-01 11:23:22.000000000","message":"Patch Set 3:\n\nhow do i pin setup tools https://github.com/mozilla/DeepSpeech/issues/3295 in patrole","accounts_in_message":[],"_revision_number":3},{"id":"114fc09a6bb213e05ac64dafd4030e204e676e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 12:23:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aeeef7f04f474677a0c2885a7a7fa333 : SUCCESS in 4m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d476d31a9fa4aa78ebbd06324ec1363 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8687ce125b504057b2535d52014ce076 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1a4026627ec4ee2b3d3bb25f796085f : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0db26ae6e8a244e1b5f68a3061b649f8 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13b6738083c540fca9b686add9b2a5b0 : SUCCESS in 4m 54s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/068cc4376a884e478a28cde26007b4ce : FAILURE in 46m 18s\n- patrole-member https://zuul.opendev.org/t/openstack/build/d72c86267f8b4204b701125b0ecbe05b : SUCCESS in 39m 50s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/17f27d2eda5a46c7be3f5923262bcd99 : SUCCESS in 50m 55s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/9eb0b528b4d443a880b06b05caed6d4d : FAILURE in 21m 59s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/752dd7d8ccb347ebacd276ac53e73e0b : FAILURE in 35m 43s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/d1294adf6205422f8ba66356e60ee275 : FAILURE in 15m 13s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/b0dd32f5f42e436c81215477d8429445 : SUCCESS in 1h 01m 51s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/b50de06c36aa48f4b92c5f9e6ab1af3b : SUCCESS in 1h 02m 07s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/ccaefaae551544dab7561407990f1ee3 : FAILURE in 34m 00s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/443119f31ae6433ea19f82146f308440 : FAILURE in 36m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8abb7bc72e1a2a1b4f2c59873b11eb2dd48e151d","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-09-01 13:18:13.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ea4390e94c9705e7e6a18516dada7c1b75931fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 14:30:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5139d31ca4cb4d8fb11cab940ad57e78 : SUCCESS in 3m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c668c8458951430fa5dd584a96a877f8 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8291a047ab5446bba1b80dc8a4f48c3 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cd2cadbeee4487f87bd2458632e03d8 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41aa8ad9941d4698a7dbd145571cc814 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0215d9565a0d4388893713b7ddaeb401 : SUCCESS in 6m 58s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/8e398995ed4141c29a89501093c35338 : SUCCESS in 57m 05s\n- patrole-member https://zuul.opendev.org/t/openstack/build/be36bafd507d463cb12641ae4c652f27 : SUCCESS in 55m 17s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/1a2ef11e060d43e7adc5f998a17f3d26 : SUCCESS in 56m 04s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/e176dbae260c4d9e80700143c61ce164 : FAILURE in 37m 43s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/3f75f879bace4d3b8cee702e2df6d004 : FAILURE in 29m 30s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/9b472e60d8554784a81dc8789856e5b7 : FAILURE in 19m 32s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/fd2d5115888d43c4b134ee36173b9346 : SUCCESS in 58m 32s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/47c13315614941e8a9333154f53aee3e : SUCCESS in 1h 01m 07s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/66f4e27842e34664a43078f2ea905d9b : FAILURE in 34m 54s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/00c38357a9b94a08be2788197140a13d : FAILURE in 43m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"902d9b481658e3f07ba10b46907eb90f9028d11c","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-09-02 03:22:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"aada6da4ccff332a8602031990a05e5026a30780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 04:40:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b21c5df3ef4c437fbf71f5c1864cf347 : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3f3d9dcae174b6faba0f9443b6daaba : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b2d5219d0943c8838abc77d0db1020 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f611e9e858564ec58e85612c1503722d : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b773a6d859ed4a8c916bd9e912ebd196 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d5865cec0584b3094557e0132ec0e4b : SUCCESS in 6m 08s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/ca96e0a9c0f4448bb3548baa0e1dfd9c : SUCCESS in 43m 40s\n- patrole-member https://zuul.opendev.org/t/openstack/build/c5a18f9ef7b64b69b997c959b5314e93 : SUCCESS in 1h 16m 26s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/509f1534375a4760996d32d78ab16166 : SUCCESS in 51m 30s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/9027b49772194e86b372d81927734a1f : FAILURE in 51m 00s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/96ea85959560446695151578fa019b0a : FAILURE in 30m 12s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/12f4afd123fb4e94b7cbdaa0c0b57e6b : FAILURE in 51m 03s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/c616844396664c78a7e191b96908f3c1 : SUCCESS in 56m 59s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/bedbf2a3476b4a1aaeb142b3f3e45540 : SUCCESS in 53m 35s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/434143405c054c839012aa620570a994 : FAILURE in 31m 20s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/295a719eca704ddabfea626d4f788e77 : FAILURE in 33m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4ef8b9727f942528b7284bb19481085752ed6027","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-09-02 10:58:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"81515726e527258a3b225a0efd5b53fa7568807f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 12:21:51.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c1a1861b5b34cbaa27525fcd383fca8 : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84b4119abe4843118bd14411ad19d3a6 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e46ba400e4074e5cae29f0540cb0aa12 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f815c7d47ced409794e02236212737f5 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4aa540cb8dd49788db1331b5e932282 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c5775ebb924e288f650c620e4341a4 : SUCCESS in 5m 49s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/845c0b4114d3459bbb7d9d6296514057 : SUCCESS in 44m 42s\n- patrole-member https://zuul.opendev.org/t/openstack/build/ae6dfc9df9d24e0f97e64b5b007b157e : SUCCESS in 58m 33s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/4f451833ee2d499d84228d2d81687ff4 : SUCCESS in 55m 26s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/277327fd3fdd4ad7b5bd945c9282074e : FAILURE in 35m 16s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/81ec9b33a0984c5aabda166f2f9e5430 : FAILURE in 38m 41s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/631290af870d46c4994fc1f62f4d5c31 : FAILURE in 21m 19s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/ee4986584b8b4543b7e96f6c880fd198 : SUCCESS in 1h 15m 06s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/4feea86a80fb4ff092bf9947a7d4db86 : SUCCESS in 57m 43s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/996cad9f63024bf3b4df8afa1d900a5c : FAILURE in 33m 52s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/50b9cc922163462aa51ccd58959fd097 : FAILURE in 31m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"33bed1caf2da536652af33f99e3b1d46ea9bd07d","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-06 09:27:30.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"c5d219513ff36cd8089be6a130beed7fcd7faafd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 10:29:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af9186e65b994852bdcd3939d56c6d1a : SUCCESS in 2m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64a21f0b27ab4e15abb3a1ac13383d10 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/599a2069688a40a5946e2fe381197164 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfac688aca8f4f799030fd03e65009b0 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/156d68261fb94d67ba46422552c99de1 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05490751535c44c0a4b7dc9083779926 : SUCCESS in 6m 01s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/9ed676deb4d04c87a8f298f24c757203 : SUCCESS in 50m 44s\n- patrole-member https://zuul.opendev.org/t/openstack/build/3d895cebb62644a5b3239d8eeac4ba41 : SUCCESS in 49m 43s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/a1da1f37ed194ef8888f97722ab078ed : SUCCESS in 53m 06s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/2535e202a1234e879ce1382157569fdc : SUCCESS in 52m 46s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/c531fc496dac49549e1624494a90a9c8 : SUCCESS in 1h 00m 15s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/d84244bb52ca47239bbca88bfc025771 : SUCCESS in 53m 56s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/cc6caaf3a52d4b10bfbc111afcd7dced : FAILURE in 57m 34s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/418469d775de4d9b941b6b095f422461 : FAILURE in 56m 57s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/c8b3fc1c67064fe98f8d8a3cda45718f : FAILURE in 26m 57s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/7e428a8e0ce64e139bc2b3fa9fe06ea4 : FAILURE in 32m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"90d6a6f87ded2ff97f155f189b1757d89e1614f1","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-07 13:12:45.000000000","message":"Patch Set 6: Workflow+1\n\n+w got removed when the patch was rebased","accounts_in_message":[],"_revision_number":6},{"id":"beefa6dcbc0c9cc5a5b0915716e81824592d4c82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 13:12:57.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9575349b8f6dfb124df24838c587fd8e45d02e3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 14:15:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"be83cd506be377e71d2c19095705572d41d759b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 14:15:15.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8430e204a5284e55af90586ad959fca4 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41ceebe7d5a744b297cbfec392372ec4 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57010ae057b043c1bfbd55dea5ff6129 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5ad260aaf474249a76c71caf6e58e44 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d942d7abd8ff4cbca52c1c0bfd6ae1c8 : SUCCESS in 6m 11s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/7ece75657ff3436f9b1d865519143f39 : SUCCESS in 52m 00s\n- patrole-member https://zuul.opendev.org/t/openstack/build/11af4d5a2987445096d917bfda59dd1d : SUCCESS in 59m 50s","accounts_in_message":[],"_revision_number":6},{"id":"2fee0c5c77626a5c006e900b0eb84baec8e49521","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 14:16:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a9ec673bdc4c7899ce6f4a5bd1f627 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8746b77bfed8aa79ded324cc74ddb96e9cda97f7","revisions":{"ad4857a1208c9c847946908fea053a1a9cc9ba5a":{"kind":"REWORK","_number":1,"created":"2020-08-22 14:43:30.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/28/747528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/28/747528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/28/747528/1"}}},"commit":{"parents":[{"commit":"91fd70b98aea02d59887c491b5c0940f63c15363","subject":"Merge \"[goal] Migrate patrole jobs to focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/91fd70b98aea02d59887c491b5c0940f63c15363"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-22 14:37:36.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-22 14:37:36.000000000","tz":-300},"subject":"fix patrole-multinode-member non-voting gate by skip_because a bug","message":"fix patrole-multinode-member non-voting gate by skip_because a bug\n\nthe bug is due to a nova api that is allowed with a system scoped\ntoken and patrole does not support them at this time.\n\nhttps://storyboard.openstack.org/#!/story/2008051\n\nChange-Id: I6963fdff199fca342620ab15948ce2c1d29c2c7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ad4857a1208c9c847946908fea053a1a9cc9ba5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ad4857a1208c9c847946908fea053a1a9cc9ba5a"}]},"branch":"refs/heads/master"},"4d2ca699c6ee47c481a64ad81990bb80a43e98ca":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-24 18:34:33.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/28/747528/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/28/747528/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/28/747528/2"}}},"commit":{"parents":[{"commit":"2c80e38e7cb3bc4f40ce7d3a5f1d0399ea903670","subject":"Get the patrole gate working due to Nova category changes in Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2c80e38e7cb3bc4f40ce7d3a5f1d0399ea903670"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-22 14:37:36.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-24 18:34:33.000000000","tz":0},"subject":"fix patrole-multinode-member non-voting gate by skip_because a bug","message":"fix patrole-multinode-member non-voting gate by skip_because a bug\n\nthe bug is due to a nova api that is allowed with a system scoped\ntoken and patrole does not support them at this time.\n\nhttps://storyboard.openstack.org/#!/story/2008051\n\nChange-Id: I6963fdff199fca342620ab15948ce2c1d29c2c7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4d2ca699c6ee47c481a64ad81990bb80a43e98ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4d2ca699c6ee47c481a64ad81990bb80a43e98ca"}]},"branch":"refs/heads/master"},"3428e96a71caeab947ce22396c506669657b2ce4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-26 12:47:18.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/28/747528/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/28/747528/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/28/747528/3"}}},"commit":{"parents":[{"commit":"2c80e38e7cb3bc4f40ce7d3a5f1d0399ea903670","subject":"Get the patrole gate working due to Nova category changes in Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2c80e38e7cb3bc4f40ce7d3a5f1d0399ea903670"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-22 14:37:36.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-26 12:46:51.000000000","tz":-300},"subject":"fix patrole-multinode-member non-voting gate by skip_because a bug","message":"fix patrole-multinode-member non-voting gate by skip_because a bug\n\nthe bug is due to a nova api that is allowed with a system scoped\ntoken and patrole does not support them at this time.\n\nhttps://storyboard.openstack.org/#!/story/2008051\nCloses-Bug:2008051\n\nChange-Id: I6963fdff199fca342620ab15948ce2c1d29c2c7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/3428e96a71caeab947ce22396c506669657b2ce4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/3428e96a71caeab947ce22396c506669657b2ce4"}]},"branch":"refs/heads/master"},"4d26304ed66d54b2db2e0d3bfe5c8d5fc8fa58d0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-09-01 13:18:13.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/28/747528/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/28/747528/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/28/747528/4"}}},"commit":{"parents":[{"commit":"3ab156a54579aa4e97ed9d715148d652cc4bdb17","subject":"Merge \"Set admin, member and reader gates to be the same\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/3ab156a54579aa4e97ed9d715148d652cc4bdb17"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-22 14:37:36.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-09-01 13:18:13.000000000","tz":0},"subject":"fix patrole-multinode-member non-voting gate by skip_because a bug","message":"fix patrole-multinode-member non-voting gate by skip_because a bug\n\nthe bug is due to a nova api that is allowed with a system scoped\ntoken and patrole does not support them at this time.\n\nhttps://storyboard.openstack.org/#!/story/2008051\nCloses-Bug:2008051\n\nChange-Id: I6963fdff199fca342620ab15948ce2c1d29c2c7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4d26304ed66d54b2db2e0d3bfe5c8d5fc8fa58d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4d26304ed66d54b2db2e0d3bfe5c8d5fc8fa58d0"}]},"branch":"refs/heads/master"},"d0ba3e96b6189e6dfab7a919b85470cf8fc641fc":{"kind":"REWORK","_number":5,"created":"2020-09-02 03:22:04.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/28/747528/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/28/747528/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/28/747528/5"}}},"commit":{"parents":[{"commit":"3ab156a54579aa4e97ed9d715148d652cc4bdb17","subject":"Merge \"Set admin, member and reader gates to be the same\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/3ab156a54579aa4e97ed9d715148d652cc4bdb17"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-22 14:37:36.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-09-02 03:22:02.000000000","tz":-300},"subject":"fix patrole-multinode-member non-voting gate by skip_because a bug","message":"fix patrole-multinode-member non-voting gate by skip_because a bug\n\nthe bug is due to a nova api that is allowed with a system scoped\ntoken and patrole does not support them at this time.\n\nhttps://storyboard.openstack.org/#!/story/2008051\nCloses-Bug:2008051\n\nChange-Id: I6963fdff199fca342620ab15948ce2c1d29c2c7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/d0ba3e96b6189e6dfab7a919b85470cf8fc641fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/d0ba3e96b6189e6dfab7a919b85470cf8fc641fc"}]},"branch":"refs/heads/master"},"8746b77bfed8aa79ded324cc74ddb96e9cda97f7":{"kind":"REWORK","_number":6,"created":"2020-09-02 10:58:29.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/28/747528/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/28/747528/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/747528/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/28/747528/6"}}},"commit":{"parents":[{"commit":"3ab156a54579aa4e97ed9d715148d652cc4bdb17","subject":"Merge \"Set admin, member and reader gates to be the same\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/3ab156a54579aa4e97ed9d715148d652cc4bdb17"}]}],"author":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-08-22 14:37:36.000000000","tz":-300},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2020-09-02 10:51:59.000000000","tz":-300},"subject":"fix patrole-multinode-member non-voting gate by skip_because a bug","message":"fix patrole-multinode-member non-voting gate by skip_because a bug\n\nthe bug is due to a nova api that is allowed with a system scoped\ntoken and patrole does not support them at this time.\n\nhttps://storyboard.openstack.org/#!/story/2008051\nCloses-Bug:2008051\n\nChange-Id: I6963fdff199fca342620ab15948ce2c1d29c2c7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/8746b77bfed8aa79ded324cc74ddb96e9cda97f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/8746b77bfed8aa79ded324cc74ddb96e9cda97f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
