)]}'
{"id":"openstack%2Fpatrole~769309","triplet_id":"openstack%2Fpatrole~master~I948c6ee0e2b1b021eae1393caa2cc91fdb477e06","project":"openstack/patrole","branch":"master","topic":"remove-policy","hashtags":[],"change_id":"I948c6ee0e2b1b021eae1393caa2cc91fdb477e06","subject":"Add skip for Nova policy \"os_compute_api:os-admin-actions:reset_network\"","status":"ABANDONED","created":"2021-01-05 10:22:11.000000000","updated":"2021-01-06 20:09:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"231b157588a39ed9c79758f914cb1348af6a5e19","_number":769309,"virtual_id_number":769309,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-05 12:09:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-05 12:09:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d38d6f2290b68f88459d8abc5cfd3e8033e8923e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-01-05 10:22:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"486de8b00dae2762dfc1e54331086e6f9910b47e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 12:09:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e924939c81c4b0b90fcd8f0a8418635 : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d8c727ad7c44e68a078e91914b27b6e : FAILURE in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f5354dfbc142e483744a6b7009f4a2 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1a571fea0d442e09fdf2b942d506c80 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fe678fb49254e2c9d5079993b1182de : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d5756807ccd4637a5cc8ddaafa0cdc9 : SUCCESS in 8m 03s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/36c06613b7d54fa299f6deecb1fd1f8e : SUCCESS in 51m 57s\n- patrole-member https://zuul.opendev.org/t/openstack/build/57408627b3ba47b584f8281ed1920982 : SUCCESS in 48m 29s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/213ab6ded79e430ea026d258c42eb446 : SUCCESS in 58m 07s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/79582701bcf64b7bb6a4d0f0e8dade14 : SUCCESS in 58m 40s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/a34256a016d84af286b27b20ebd4fe1f : SUCCESS in 54m 52s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/9ddd8d886391480d9d8194c8e7b02fcd : SUCCESS in 1h 00m 41s\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/182d10704f5e4fee811466e17507376c : FAILURE in 21m 01s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/e44316e283734a96a09c8ac07374ac35 : SUCCESS in 1h 04m 42s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/767180cf2d8c4b6ba7f3fe4646d673e3 : SUCCESS in 1h 42m 28s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/0b616ebb65b540b28c371893dd139f8f : SUCCESS in 1h 04m 09s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/484d9ee0d9ae41aab66fd0e36e26247e : SUCCESS in 50m 00s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/5c15d37fdbe6449991b759b158915028 : SUCCESS in 44m 27s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/849735c1100247ce890042d813eac41f : SUCCESS in 38m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"231b157588a39ed9c79758f914cb1348af6a5e19","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-01-06 20:09:56.000000000","message":"Abandoned\n\nIncluded in https://review.opendev.org/c/openstack/patrole/+/768119","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc0939f70da2b0e83b9450fe79873ac183f3ff61","revisions":{"fc0939f70da2b0e83b9450fe79873ac183f3ff61":{"kind":"REWORK","_number":1,"created":"2021-01-05 10:22:11.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/09/769309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/09/769309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/09/769309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/09/769309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/09/769309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/09/769309/1"}}},"commit":{"parents":[{"commit":"5c466a2301ecf6471a7f36633ffa994de9a4abf8","subject":"Add releasenote to tag the end of support for Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5c466a2301ecf6471a7f36633ffa994de9a4abf8"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-01-05 10:19:57.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-01-05 10:19:57.000000000","tz":780},"subject":"Add skip for Nova policy \"os_compute_api:os-admin-actions:reset_network\"","message":"Add skip for Nova policy \"os_compute_api:os-admin-actions:reset_network\"\n\nThe policy has been removed in\nhttps://review.opendev.org/c/openstack/nova/+/749315\n\nChange-Id: I948c6ee0e2b1b021eae1393caa2cc91fdb477e06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/fc0939f70da2b0e83b9450fe79873ac183f3ff61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/fc0939f70da2b0e83b9450fe79873ac183f3ff61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
