)]}'
{"id":"openstack%2Fpatrole~770472","triplet_id":"openstack%2Fpatrole~master~I0e200584c36c758ec9a9a08d1301bee7de522c29","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"I0e200584c36c758ec9a9a08d1301bee7de522c29","subject":"DNM: Testing without assigning admin role to test creds","status":"ABANDONED","created":"2021-01-12 20:32:01.000000000","updated":"2023-03-11 02:13:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8f05ee145f4c2c0a9f34bb2d533031f8b935040e","_number":770472,"virtual_id_number":770472,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-12 21:35:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 20:39:02.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2021-01-12 21:35:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"90221300dd0ff4bd3dbc7cc2f15e7be7dca71011","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-12 20:32:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2a926f749997f1eea6c2a1a1bed6ec96da074df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 21:35:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ea39a2ec370481393a9abeb4e8348bb : FAILURE in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/123c1427eb5e421b87f4f428fb3c1ee2 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36fa8ab0f75144469b18789aab517e33 : FAILURE in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bb396034a4946a089ff5fa1ad0e1091 : FAILURE in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06fc18e94b6e45a4bd8b74ca53244590 : SUCCESS in 7m 43s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/8b650b21afcc4a1580868c88780f6cba : FAILURE in 40m 01s\n- patrole-member https://zuul.opendev.org/t/openstack/build/ff838c0a1e0f444faa3257b218d019b8 : FAILURE in 31m 10s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/c07e5a5353264e9e9f6d64930f2107d0 : FAILURE in 39m 48s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/6da1c74362cb4f5494fa6a61628839f1 : FAILURE in 50m 58s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/67b41dad5655419cb5a445f96a90837d : FAILURE in 32m 02s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/eca824fe104344b3aa22d2416c98537c : FAILURE in 27m 59s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/81f5f8c40f9a4d07b4428c89b9b976cd : FAILURE in 38m 46s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/98d5998ac6f34918b7e1b826894c2b9c : FAILURE in 54m 33s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/8b96a28f56074d778b05daba171bd9bc : FAILURE in 46m 59s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/3b7947e2089d4a60a1b5d4008530f36d : FAILURE in 44m 41s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/a6e10bc58bb94741a59aa159b539d4c8 : FAILURE in 43m 44s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/57f8d8e85d444ceda0ded1f49fe446c9 : FAILURE in 36m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f05ee145f4c2c0a9f34bb2d533031f8b935040e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-11 02:13:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"421e1d5289368218270af812ed381ab885db6de0","revisions":{"421e1d5289368218270af812ed381ab885db6de0":{"kind":"REWORK","_number":1,"created":"2021-01-12 20:32:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/72/770472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/72/770472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/72/770472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/72/770472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/72/770472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/72/770472/1"}}},"commit":{"parents":[{"commit":"700c1db070bd6a5ac331dc5a6b6fc59bb2c43105","subject":"Remove stable/stein testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/700c1db070bd6a5ac331dc5a6b6fc59bb2c43105"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-01-12 20:31:26.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-01-12 20:31:26.000000000","tz":-360},"subject":"DNM: Testing without assigning admin role to test creds","message":"DNM: Testing without assigning admin role to test creds\n\nChange-Id: I0e200584c36c758ec9a9a08d1301bee7de522c29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/421e1d5289368218270af812ed381ab885db6de0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/421e1d5289368218270af812ed381ab885db6de0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
