)]}'
{"id":"openstack%2Fpatrole~770523","triplet_id":"openstack%2Fpatrole~master~I965d9e280ef0a85dfe6b2e4000eb47e1d06fed77","project":"openstack/patrole","branch":"master","topic":"restore-orig-roles","hashtags":[],"change_id":"I965d9e280ef0a85dfe6b2e4000eb47e1d06fed77","subject":"Restore test user\u0027s original roles during clean up","status":"MERGED","created":"2021-01-13 02:04:43.000000000","updated":"2021-01-26 00:29:32.000000000","submitted":"2021-01-26 00:27:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770523-restore-orig-roles","meta_rev_id":"d3cfab2d3381de670725675877729c5adb6cefdb","_number":770523,"virtual_id_number":770523,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 00:27:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2021-01-25 23:16:01.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-18 18:31:53.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-25 23:17:01.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-13 03:30:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-14 20:43:41.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-01-21 21:29:10.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2021-01-25 23:16:01.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"8f44970375d6ce9fd38198ef50664192395d7105","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-01-13 02:04:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19e1b35e16924755a5fbdfaef67fd8470d9b6b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 03:30:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/227b4278d8a2437d90241a09a21da72b : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee71797d3d5b4075a328653274fff53d : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/638d169ac7aa42f798aee4079350a927 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c044e50772f84e13b6f58e5c150d5e78 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf0d64f4bbc14f6c910745931c335315 : SUCCESS in 13m 34s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/550bd016a5234d8a8487c8ae6ac8f1e9 : FAILURE in 1h 07m 48s\n- patrole-member https://zuul.opendev.org/t/openstack/build/c41dabbccbeb4ca1add541f6b66a2794 : FAILURE in 1h 13m 58s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/88d94a35ab684909a31d07b60bdfb94d : FAILURE in 1h 16m 37s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/ecdcaa8ac4484da7bcc0d29aa2fcdea9 : FAILURE in 1h 09m 38s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/92e36742ccbe494480e0deee6b8a3caa : SUCCESS in 50m 01s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/a4ae0c0fad6244918840d122ce371251 : SUCCESS in 1h 01m 46s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/041374fb8b944984a98aabf5afb13a14 : SUCCESS in 57m 21s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/8a0c0b98ccad4c2a9492326fbee4fd54 : SUCCESS in 1h 01m 38s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/bf8cec5d6f22445d85af9704b436c288 : SUCCESS in 1h 07m 15s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/eac07fa2451a46ad9f3bab1c9ed91df7 : SUCCESS in 50m 32s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/21b48d51fd4a49118021594fcb7ba74f : SUCCESS in 50m 37s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/8982e61b6dcd4e4091e2544e427f6eaa : SUCCESS in 51m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0460e984dcdf2918cbce959aefb32b7b8690ac6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-01-13 09:29:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"15f435d0d64e537a3da2d9e31f29201f4aca68e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 10:47:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02b5edd348b740b388e7aaf9576e0e42 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fb44462f4384a6b8e96c20c051d4279 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/955a692ae91d413fadcc0efe1e6ce76f : RETRY_LIMIT in 2m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e14207935119498e8956724afca4b9cf : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a8f5bdfe55e41e49e67b4eb4581c93d : SUCCESS in 16m 05s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/1eb3ed02ef9847bab8a2980c3d94b03d : FAILURE in 4m 28s\n- patrole-member https://zuul.opendev.org/t/openstack/build/4704ce6a3ee64708b1f89b23508fe67e : FAILURE in 4m 14s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/488ead3cd5bd4ad59f9129882b33632d : FAILURE in 4m 10s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/afbd4e3c039845da862665050f5ea777 : SUCCESS in 42m 16s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/e038a86a079f4dbb935f208fcf88961d : SUCCESS in 1h 08m 41s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/394b1e19c2bb4923a46ac1cb806d03db : SUCCESS in 51m 11s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/e202243eca5f417682120c69f13163fc : SUCCESS in 57m 07s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/2f250bf2982f4248a46e6fdf6297c952 : SUCCESS in 1h 09m 15s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/6d95df59d7004b549f38b1589235b402 : SUCCESS in 1h 03m 31s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/4602d5d47435462c9aa0e62175c58e29 : FAILURE in 4m 44s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/0ed9e4896b6e486ba3d161bdba4cfd1b : FAILURE in 6m 09s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/fcff77be1f034970a686a969adf4b11b : SUCCESS in 38m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5bffc4a8b7f1ba3edabf6e1ceb60f3f6d5fde784","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-01-13 20:21:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"03a25b3b53003af8cbaaa5eb8904fa83d6d525ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 21:48:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e633c8b16caf4bcc9d1f749518d94860 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8bbfa9e0fe649ab9eb08b4bf16b5357 : SUCCESS in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40f1e440cc974a96bd50f5f7ef66c471 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9df8a6a78bb34e988899c0498753c5ab : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b114097a4067452fa65599964f6401f8 : SUCCESS in 9m 31s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/10c0c5e0b84d426b8c84f0fe95a23699 : SUCCESS in 56m 25s\n- patrole-member https://zuul.opendev.org/t/openstack/build/4b7f53657f0344d88bbd0e69b9aeebe0 : SUCCESS in 56m 50s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/45bfbbb18c7e4c0fbfd96c9e00ee67e5 : SUCCESS in 1h 06m 49s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/953f6fda0e604e1cbd0b3edfe42873bb : SUCCESS in 1h 01m 24s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/062381b813fc43dca0c4dfb094db55f4 : SUCCESS in 1h 04m 58s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/c13287d39c35432e8f1b6963bba62faa : SUCCESS in 55m 25s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/b79ca666f93845beadc5d890372f2a03 : SUCCESS in 1h 19m 50s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/b3845c2ded234cbdb9b39dc18deb8bcc : SUCCESS in 1h 14m 03s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/a48634f90fbd44af98009cdcacac9592 : SUCCESS in 1h 13m 31s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/570283043a4f4cf08d1ddb96a262bff7 : SUCCESS in 36m 56s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/15402bfd602340cb94bd23cb4f0d20ad : SUCCESS in 46m 01s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/99053d8bd6e44822b396132f998181b0 : SUCCESS in 45m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d77f02da51294f264914508e8fff434d29cd5cf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-14 20:52:23.000000000","message":"Patch Set 2:\n\napproach looks good, can you add some unit test for that by checking the orig role exist or not.\n\nalso did not get why to delete the FIp and VIF tests","accounts_in_message":[],"_revision_number":2},{"id":"97720d665315cc6f9801be55ee1867de95f0441c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-01-15 21:43:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3116b7914651a9624b06ededd40c463ac33700bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 23:06:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7fba0260afa45f584507a924c416fac : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a2ba7e224a46498d8eec19cec31a95 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6821101e050c4d91999d966790229e00 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85283ee8902245a49b798a20fdb0485e : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c50e21c9246d4edeb5df291056347f83 : SUCCESS in 6m 25s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/88478919aa70458baf4f494ab0f706ab : SUCCESS in 48m 27s\n- patrole-member https://zuul.opendev.org/t/openstack/build/45dda71965074bdbbfec85314564da95 : SUCCESS in 52m 01s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/8644f8d148f84f9ab731d2a4d2049ff1 : SUCCESS in 56m 35s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/5054fcf97c7e402b9f32dc4762c9643a : SUCCESS in 1h 02m 12s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/eb3c20bd043a440f889d31a172d7e51b : FAILURE in 49m 17s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/d9e4d624da0f43d0a74aeb8779604acd : SUCCESS in 58m 36s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/eb8d4660ea58413595bfadd9f03f2898 : SUCCESS in 1h 04m 38s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/8f459e197a92465c9c663f1ed596d1b2 : SUCCESS in 1h 01m 16s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/9409edd30f114fb3bdb0c3e7528bda44 : SUCCESS in 1h 04m 36s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/916ed85aa0d545daae4f6395046f8acb : SUCCESS in 29m 00s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/b3bab4047f374f45b628452e4793b10a : SUCCESS in 32m 49s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/9f35da0407b244b099e672616a51650d : SUCCESS in 33m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"43684771e7c6c1955067762dc5ff8df2d804c5b4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-01-15 23:23:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0b6586a85db5ba874688a001e163c47b94fbfeee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 00:39:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31f60f36507a458eb960557207be6562 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02b232511c5f4e588276e9edd331eb0e : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0a974fba1054703830dcfa1d9e90044 : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f98a9065a6c84c55a857ade521d2d898 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80f382d40bf04c6884a55753fb3cce18 : SUCCESS in 7m 08s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/8af796861eda418aab6ecdf619c05e3d : SUCCESS in 57m 09s\n- patrole-member https://zuul.opendev.org/t/openstack/build/c33f1299aaa44a67b914d7561a7f0e8c : SUCCESS in 51m 21s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/1208c205d63c4f23b395e11941e7681b : SUCCESS in 57m 10s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/254f4ae4598e4ef387f88a918c51910f : SUCCESS in 57m 11s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/69a689c3cd1242d08ea19800546eda61 : SUCCESS in 49m 52s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/f6484bc9360c4781800bd2a407a6a633 : SUCCESS in 50m 37s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/9fe5542262a344b48f5a52ac0b387c75 : SUCCESS in 1h 06m 50s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/dc8655d1c1c24a35a4c9f0d383d58823 : SUCCESS in 1h 08m 08s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/c1743c268f084e1ca089cdef12eb3d81 : SUCCESS in 1h 04m 12s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/edbc1bd79d9a41f4829e2ba4a544a8ae : SUCCESS in 38m 27s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/91ef8418c1be4696847cca2b042322cb : SUCCESS in 35m 32s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/b302b22cf63f4168ae449febc491b6a0 : SUCCESS in 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5061e0fc6f0e9363ceae9e797c5e2a60cf6a3ad6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-18 18:31:53.000000000","message":"Patch Set 3: Code-Review+2\n\nthanks, lgtm","accounts_in_message":[],"_revision_number":3},{"id":"637995d6a81ce3ec75f550c28ce260a7425f9880","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-01-25 23:16:01.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"ee57f83712d744c735672070be4eae7b1ee06ca8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-25 23:17:01.000000000","message":"Patch Set 3: Workflow+1\n\nlet\u0027s merge it","accounts_in_message":[],"_revision_number":3},{"id":"9c64854e752d4e3bcff659f44dee089bb5d94e53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 23:17:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ced44f72c4a560dd6cd8edf5b2a893d8ab91784e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 00:27:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4711b8657b73496ea3c1964ee1d49cdb : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb6f849690c040e8a70d65be5186166e : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bbf19a76d554d248c211a12d9c8817b : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8db19ba4613047e39b814a65c3f926a1 : SUCCESS in 7m 30s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/c9ca57ac5bb94d77a720ba646740ec19 : SUCCESS in 59m 03s\n- patrole-member https://zuul.opendev.org/t/openstack/build/08f048a45b66447ebd410a6ee38005ed : SUCCESS in 53m 43s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/bc1c691b9c324ccd85604d16db82797c : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":3},{"id":"0dc54f5a89d6fd38c6b3f596108bfacda57c7cee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 00:27:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d3cfab2d3381de670725675877729c5adb6cefdb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 00:29:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a052fefa4a4b49c3a49b5b87a60cf936 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b73e1088e01329684c0724081840af3870870f52","revisions":{"1b012a1556831ddb30fcbbd64f945338dfc937c7":{"kind":"REWORK","_number":1,"created":"2021-01-13 02:04:43.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/23/770523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/23/770523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/770523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/770523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/770523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/23/770523/1"}}},"commit":{"parents":[{"commit":"700c1db070bd6a5ac331dc5a6b6fc59bb2c43105","subject":"Remove stable/stein testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/700c1db070bd6a5ac331dc5a6b6fc59bb2c43105"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-01-13 02:04:39.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-01-13 02:04:39.000000000","tz":780},"subject":"[WIP] Restore test user\u0027s original roles during clean up","message":"[WIP] Restore test user\u0027s original roles during clean up\n\nChange-Id: I965d9e280ef0a85dfe6b2e4000eb47e1d06fed77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/1b012a1556831ddb30fcbbd64f945338dfc937c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/1b012a1556831ddb30fcbbd64f945338dfc937c7"}]},"branch":"refs/heads/master"},"96b7ddee5d8591c9cf3bba5e5d55e632b403156c":{"kind":"REWORK","_number":2,"created":"2021-01-13 09:29:15.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/23/770523/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/23/770523/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/770523/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/770523/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/770523/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/23/770523/2"}}},"commit":{"parents":[{"commit":"700c1db070bd6a5ac331dc5a6b6fc59bb2c43105","subject":"Remove stable/stein testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/700c1db070bd6a5ac331dc5a6b6fc59bb2c43105"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-01-13 02:04:39.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-01-13 09:29:10.000000000","tz":780},"subject":"[WIP] Restore test user\u0027s original roles during clean up","message":"[WIP] Restore test user\u0027s original roles during clean up\n\nChange-Id: I965d9e280ef0a85dfe6b2e4000eb47e1d06fed77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/96b7ddee5d8591c9cf3bba5e5d55e632b403156c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/96b7ddee5d8591c9cf3bba5e5d55e632b403156c"}]},"branch":"refs/heads/master"},"b73e1088e01329684c0724081840af3870870f52":{"kind":"REWORK","_number":3,"created":"2021-01-15 21:43:18.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/23/770523/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/23/770523/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/770523/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/770523/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/23/770523/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/23/770523/3"}}},"commit":{"parents":[{"commit":"f5699576d85bf29e9bc9f9160307db8d00ad5086","subject":"Remove the tests for unsupported Nova APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f5699576d85bf29e9bc9f9160307db8d00ad5086"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-01-13 02:04:39.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-01-15 21:42:30.000000000","tz":780},"subject":"Restore test user\u0027s original roles during clean up","message":"Restore test user\u0027s original roles during clean up\n\nDepends-On: https://review.opendev.org/c/openstack/patrole/+/770867\nChange-Id: I965d9e280ef0a85dfe6b2e4000eb47e1d06fed77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/b73e1088e01329684c0724081840af3870870f52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/b73e1088e01329684c0724081840af3870870f52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
