)]}'
{"id":"openstack%2Fpatrole~770867","triplet_id":"openstack%2Fpatrole~master~Ib582c984796cbcd93fdb3a505d0b3fcdf6a2fbf2","project":"openstack/patrole","branch":"master","topic":"remove-unsupported-api-tests","hashtags":[],"change_id":"Ib582c984796cbcd93fdb3a505d0b3fcdf6a2fbf2","subject":"Remove the tests for unsupported Nova APIs","status":"MERGED","created":"2021-01-14 21:22:27.000000000","updated":"2021-01-25 17:43:41.000000000","submitted":"2021-01-25 17:41:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770867-remove-unsupported-api-tests","meta_rev_id":"b3c8e5a9631f284f7159c83c570f896d1807bda2","_number":770867,"virtual_id_number":770867,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-25 17:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2021-01-25 16:34:03.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2021-01-14 22:05:30.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2021-01-25 16:34:03.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-14 22:05:30.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-01-14 23:42:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-21 21:29:04.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2021-01-25 16:34:03.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"51052946448388aa99a57bc46275b875a8d9e104","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-01-14 21:22:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b269f301a03a332b1c6888c0694556c927a220eb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-14 22:05:30.000000000","message":"Patch Set 1: Code-Review+2\n\nnova-network is all gone from all supported branches which are train, ussuri and victoria. we are good to remove these.","accounts_in_message":[],"_revision_number":1},{"id":"a3e80d27a08840fb378f15334ee3fdac26948e39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 23:42:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/080a15f49bb6453cb922422f4437a79b : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7705c45dfc147d9abd0a80b74a5f50a : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1849508517340109697b449b8c63d46 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16bd28f1a5d9412c829e965e5e7c2f30 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb7d30e6018c4bc19709472a6578a50b : SUCCESS in 5m 08s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/a729788bb9974a378687f98fb20a2565 : SUCCESS in 47m 48s\n- patrole-member https://zuul.opendev.org/t/openstack/build/fe04303c8295452889f54c2801d9e58a : SUCCESS in 49m 37s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/1f645f8781004d44b5a81712714d5128 : SUCCESS in 49m 33s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/5ad63c3c3bca4a5ca11bb2dec7e107cb : SUCCESS in 48m 05s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/dd8a1abb53e04039af0b2eef63d9aa2c : SUCCESS in 50m 38s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/8aab6663b71745ba97540d1799466eaa : SUCCESS in 53m 28s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/8e60435f0a2a42f7af73a58ee4ccd04c : SUCCESS in 1h 04m 07s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/2158ac5e03b24a6b993c771471d62eb3 : SUCCESS in 2h 07m 21s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/c96e4783b1444ba68cdade373f029e51 : SUCCESS in 2h 05m 18s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/77cbb84b157b431eae65a17f0f7abf4a : SUCCESS in 1h 22m 06s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/806be610dcd84fdba8759d611236c00f : SUCCESS in 32m 43s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/6cac970e6ec24367835753a8ced7d8df : SUCCESS in 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07c2afdb8522c2a6a8ee00f26040f08f47705d28","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-01-25 16:34:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nlgtm, the maximum microversion set is too old in the deleted test classes","accounts_in_message":[],"_revision_number":1},{"id":"f7cda716a35cc953b556ad5c63d4088b82f3000f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 16:36:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25d7712093ae90e21e86c242bff19d9d87febcb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 17:41:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/315ba85eb2f4463b8e9501ab13b7530b : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/685b4ab28d92435ea3798678cef90299 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bde96fb489ff49d0a586fdccca52d532 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102c06aad7d64a2fb4ad5b4f5fc16939 : SUCCESS in 5m 22s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/b356b8dc5346476f8136a69392084e4c : SUCCESS in 1h 02m 30s\n- patrole-member https://zuul.opendev.org/t/openstack/build/fae902163b94414889f7634ca1fae902 : SUCCESS in 56m 00s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/5172951eaab54c93810b62e220fd0e9c : SUCCESS in 55m 12s","accounts_in_message":[],"_revision_number":1},{"id":"058cf923666bef9a9e2f8976f1759f9aa8dbe09c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 17:41:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b3c8e5a9631f284f7159c83c570f896d1807bda2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 17:43:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0eb7d2a5b1545958cb65fe0dbda6ab0 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5699576d85bf29e9bc9f9160307db8d00ad5086","revisions":{"f5699576d85bf29e9bc9f9160307db8d00ad5086":{"kind":"REWORK","_number":1,"created":"2021-01-14 21:22:27.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/67/770867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/67/770867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/67/770867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/67/770867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/67/770867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/67/770867/1"}}},"commit":{"parents":[{"commit":"700c1db070bd6a5ac331dc5a6b6fc59bb2c43105","subject":"Remove stable/stein testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/700c1db070bd6a5ac331dc5a6b6fc59bb2c43105"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-01-14 21:22:21.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-01-14 21:22:21.000000000","tz":780},"subject":"Remove the tests for unsupported Nova APIs","message":"Remove the tests for unsupported Nova APIs\n\nChange-Id: Ib582c984796cbcd93fdb3a505d0b3fcdf6a2fbf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f5699576d85bf29e9bc9f9160307db8d00ad5086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f5699576d85bf29e9bc9f9160307db8d00ad5086"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
