)]}'
{"id":"openstack%2Fpatrole~777628","triplet_id":"openstack%2Fpatrole~master~Iad89625e7514d1046e0340db96a73960a17023cf","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"Iad89625e7514d1046e0340db96a73960a17023cf","subject":"Changed minversion in tox to 3.18.0","status":"MERGED","created":"2021-02-25 16:59:43.000000000","updated":"2021-02-26 17:04:15.000000000","submitted":"2021-02-26 17:02:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777628","meta_rev_id":"8927fd69d74ff6959c2e55f0a8b2313278d169df","_number":777628,"virtual_id_number":777628,"owner":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-26 17:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2021-02-26 15:50:44.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2021-02-26 15:50:44.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-25 18:42:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-26 15:50:44.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"59449641b9afe259600bcb3c2bcec1813e7abfc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-02-25 16:59:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58625d767944a1770be08ef4336dd2085e5acec3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 18:42:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef4aceeb59f64f27ab3dd67d39f05c06 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6413287797654942b5cf2af9734c6f36 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4a6cfffcb384c19824e6bb791e59dcc : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e16871942f441488d4300445b521c2e : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e4aa45baba54ba48d68e7fe53b2acee : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81f0b6bbdf5a4b76b080a14f404bd376 : SUCCESS in 4m 14s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/e761d455a733482dbbe5fa502916d8fa : SUCCESS in 50m 09s\n- patrole-member https://zuul.opendev.org/t/openstack/build/72d5463b36a2437b90517343db9274ab : SUCCESS in 53m 12s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/425da9762058408693eeb84912adcd6f : SUCCESS in 1h 07m 32s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/6d421963778c475489a061a5a78c0c49 : SUCCESS in 1h 02m 41s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/54f6549391054a6b9319d09c0ca1df42 : SUCCESS in 1h 01m 27s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/bc0c318da70f4275ad027f8f36010b7b : SUCCESS in 55m 43s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/ade4526a88014483a94ba7e9f5001a4c : FAILURE in 1h 30m 26s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/0c11510249ea42a2bc9d46e782a0571f : SUCCESS in 1h 10m 53s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/65578f93213f46fc887295a2f0a1ce94 : SUCCESS in 1h 17m 30s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/8cbf22aebc1744fe9006dc76d979520d : SUCCESS in 32m 47s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/2fe38c4cd6e64855a02e9ebe1c307f23 : FAILURE in 33m 22s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/852f1f44038e4af48d577be1cb6e597a : FAILURE in 33m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf8d7500a86a8d68fa86f8a0d2c1c02d3a38de24","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-26 15:50:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bdc683d7f7fab80a3af54b2418c1dda05154faa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 15:51:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64cc03074f313a206b3d02781eab49e26d680875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 17:02:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef9c3c594f74315aec168f7b1d33caa : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e76b25168044c1e9294f83e7e43c189 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e187f73ff7e1474d817e4be0b541601a : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b24d731a8d14cb7ab0189d398941f24 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c43f45d916b648f8bba8b2f1da8b67ba : SUCCESS in 4m 58s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/27e3b28e4cae4fcfb02d1388896e0f9c : SUCCESS in 52m 22s\n- patrole-member https://zuul.opendev.org/t/openstack/build/26296d360a2d49d0b3024743c669ead0 : SUCCESS in 1h 09m 35s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/dcd907788e79480ab2afa27b53268f95 : SUCCESS in 1h 08m 28s","accounts_in_message":[],"_revision_number":1},{"id":"184345a9663ba86e8b14cfa1f2500159c3cc73c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 17:02:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8927fd69d74ff6959c2e55f0a8b2313278d169df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 17:04:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3655629827f14678beda75983a5acf11 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b2c0bb3cdbe4efc9ab097a112d642fa : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ec7c465620dd44d392ba12ecb58bfadd97f3260","revisions":{"9ec7c465620dd44d392ba12ecb58bfadd97f3260":{"kind":"REWORK","_number":1,"created":"2021-02-25 16:59:43.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/28/777628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/28/777628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/777628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/777628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/28/777628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/28/777628/1"}}},"commit":{"parents":[{"commit":"0bfba962e8d3331b440d6d5f75b793903b33339c","subject":"Merge \"Reuse tempest to create admin client manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/0bfba962e8d3331b440d6d5f75b793903b33339c"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-02-25 16:58:02.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-02-25 16:59:38.000000000","tz":330},"subject":"Changed minversion in tox to 3.18.0","message":"Changed minversion in tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 python in order to\nreplace tox\u0027s whitelist_externals by allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nChange-Id: Iad89625e7514d1046e0340db96a73960a17023cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/9ec7c465620dd44d392ba12ecb58bfadd97f3260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/9ec7c465620dd44d392ba12ecb58bfadd97f3260"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
