)]}'
{"id":"openstack%2Fpatrole~787169","triplet_id":"openstack%2Fpatrole~master~Ie18dd47beef0d15e168c3284c4cf0d3564467eff","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"Ie18dd47beef0d15e168c3284c4cf0d3564467eff","subject":"Use py3 as the default runtime for tox","status":"MERGED","created":"2021-04-20 08:54:11.000000000","updated":"2021-09-29 03:21:06.000000000","submitted":"2021-09-29 03:19:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787169","meta_rev_id":"fdfbf683bc78d943348d9e3674a9c11000da66a4","_number":787169,"virtual_id_number":787169,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-29 03:19:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2021-09-29 02:12:04.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2021-09-29 02:12:04.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 10:17:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-29 12:23:48.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"CC"},{"updated":"2021-09-29 02:12:04.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"2c54c3c3eeaf9b7e4e43ef556b4ae2bfa651a75c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-04-20 08:54:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cc44c2710654560ca040cfedb0c8c9da7177e15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 10:17:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7b12b7570db4fa4b314e37810246834 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65b7ce916e704329b1552b672674cccb : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49e0f441b7bc4df988d18fc8e7f55892 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83eeba920ffd4c8f95499a4e544228ca : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26a95680d7448c38726d90835f88bcb : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f2cba37c7ad4c8ebee3b912445ebde3 : SUCCESS in 5m 42s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/4cc418de0e2a4fcc8f6769176513d026 : SUCCESS in 57m 24s\n- patrole-member https://zuul.opendev.org/t/openstack/build/b94774b746f84420af618a86670e16a5 : SUCCESS in 1h 06m 46s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/f811582938584637a61c9dac62000ec1 : SUCCESS in 54m 42s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/85e6a726dd8747b6aa3e3835a23c4666 : SUCCESS in 1h 20m 04s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/13e37b39ee9f447487a3b845c24a5219 : FAILURE in 49m 44s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/1b7ad3629a0947bfb24e682ad468cade : SUCCESS in 52m 09s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/ebacb0be3c9f48eb87e0bba4bdbff27a : SUCCESS in 57m 43s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/dcc2329fc22746cebf87f9e7d10dd01f : SUCCESS in 1h 02m 58s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/5174bb9f5557483da8499cddbf9fc170 : SUCCESS in 1h 02m 55s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/7b33cb06497545158f6ba9e13bdf26f1 : SUCCESS in 29m 01s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/b1aad28568b54326ad46aca3de19a5ee : FAILURE in 29m 05s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/a134f49adc8c4d90bc9e0c3b3741a8c5 : FAILURE in 30m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51d33be44b5489aefc66a9d0e315ca63a6d23ee6","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-04-29 12:23:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d1a1893a327df763729ca18b9d6f955cb2eb51a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 13:42:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54836f5a881e412ba0a5aafa394966bb : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/345df3ae30b4457f9b901ebbea0dc5b1 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a665715e180437197ae90beecae7b56 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4559b9d6238144c8b4e370918a8d9025 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce032dacd984f55ab77ade31485a97d : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b63b97b80bce4578999cfcdfd72b1271 : SUCCESS in 4m 24s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/b18033382b2844248eeea07c6d2e9f3f : FAILURE in 54m 37s\n- patrole-member https://zuul.opendev.org/t/openstack/build/f07665c498c345bea9e938320a47b3e4 : SUCCESS in 54m 43s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/6d083c0d1e8e45c1bd8bc863ca929970 : FAILURE in 1h 03m 12s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/c23c3be65399444491976beb5df87730 : SUCCESS in 1h 07m 08s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/05c8ff64972b4477b08880a049d6bc32 : SUCCESS in 1h 00m 18s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/25cf6d952e1a48268846cfa27b9b7c92 : SUCCESS in 1h 10m 06s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/e2a2d9aa8bb143ccb0a50920e8ebb347 : SUCCESS in 1h 04m 29s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/41e33c491eee446cb75671f94b5de748 : SUCCESS in 59m 38s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/f561a39e81b84826a255d688fb425ce5 : SUCCESS in 1h 00m 19s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/9c38d488b62f4d909b814930a104d379 : SUCCESS in 30m 17s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/9fca5860ec274d3fa3627a02f6a7781b : FAILURE in 39m 41s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/0be6eba66bf84fbf82af4c25e72a6b16 : FAILURE in 31m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78eb310a612e474ec1712f680a195b3ed866a444","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-27 12:31:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"25084c000970df96cb0944a33cf215854a664e82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 14:47:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/307dd3147d994cf08f8498c9026465ee : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78637b74ef914aaa8da921203fa69092 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ef84c61f5ab4e53b5dae51455d693f3 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d881a1a16f8540ed9dcfec5956fc4305 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26271623149c4ee1bdc5eeda78f2cd69 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bbd0e10ef194f8f95eb915e8d0e7251 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ce890c932b548d781925d5ea366b594 : SUCCESS in 4m 27s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/690e44c2441f40ef9d6a47e6f3012120 : SUCCESS in 39m 54s\n- patrole-member https://zuul.opendev.org/t/openstack/build/d271f7144dd84299aa3cddadfcc7ee20 : SUCCESS in 59m 31s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/4f7e72c62848444a9f35aad5a1afe1b8 : SUCCESS in 49m 26s\n- patrole-member-wallaby https://zuul.opendev.org/t/openstack/build/e5ec153418d549e7a866a20b79632c05 : SUCCESS in 54m 49s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/3bee3295533b4c0cba8b22eb373e2505 : SUCCESS in 59m 40s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/e58ad62c4f7c4de0b605ebe4cb8b45e1 : SUCCESS in 51m 59s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/73ff0c72c470438cb5e6db7df3acb2cc : TIMED_OUT in 2h 14m 11s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/30d27ca3dd2a4b1ebe6cae3193237f91 : SUCCESS in 1h 04m 00s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/27b39bd62f6e4eef9d159745331c2b69 : SUCCESS in 58m 40s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/7a839676c64d4ec4bf5a8e983bc97f96 : FAILURE in 30m 07s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/aac23bde00d54cbfa4d303bfe5ed85b8 : FAILURE in 29m 10s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/643e6cbc19764e5caa7979f0aebb349c : FAILURE in 31m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9c05680ea591984290932505f7392d5dfcda93d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-29 02:12:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"395d4f3d522c547abba5ed3dae04db32e3e9a496","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 02:13:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7edbb4a85725894d253b7fa483b90c33d7cd8e55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 03:19:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e91dc4f52e94f15812083a7769a36ee : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e7573f067eb4d35a361f6925e625c1a : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c012350e70984b6c876418c03caa2322 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f789564427224782b0b9b25517da3ae8 : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cf5dd62b38b473e9c9d0480ff796322 : SUCCESS in 5m 57s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/6c82b0042db9429598a8a2cb4187ed75 : SUCCESS in 46m 10s\n- patrole-member https://zuul.opendev.org/t/openstack/build/8ef3df272a2f4f509b9418f06b73d732 : SUCCESS in 59m 55s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/64fe2063d74545a2a2c2d40570291715 : SUCCESS in 51m 07s","accounts_in_message":[],"_revision_number":1},{"id":"414b307218123f68be929c333ab626f9ea11c0a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 03:19:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fdfbf683bc78d943348d9e3674a9c11000da66a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 03:21:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf5d9b208d842808e3f59b51494b6ab : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b20388a40cf440248bb775d63ccb791c : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3a7ce27a26c6e7c7eeb9e35b6c83256a469db10","revisions":{"e3a7ce27a26c6e7c7eeb9e35b6c83256a469db10":{"kind":"REWORK","_number":1,"created":"2021-04-20 08:54:11.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/69/787169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/69/787169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/69/787169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/69/787169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/69/787169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/69/787169/1"}}},"commit":{"parents":[{"commit":"f4bae14ce972cf83081eddc1a3fb7f082d01a42e","subject":"Fixed issue with backup delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f4bae14ce972cf83081eddc1a3fb7f082d01a42e"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 08:53:19.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 08:54:05.000000000","tz":480},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: Ie18dd47beef0d15e168c3284c4cf0d3564467eff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/e3a7ce27a26c6e7c7eeb9e35b6c83256a469db10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/e3a7ce27a26c6e7c7eeb9e35b6c83256a469db10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
