)]}'
{"id":"openstack%2Fpatrole~807131","triplet_id":"openstack%2Fpatrole~master~I886fdf8488cc25e8d9a93af7560f655e9db92a21","project":"openstack/patrole","branch":"master","topic":"loggable_Resource_test","attention_set":{},"removed_from_attention_set":{"30742":{"account":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"last_update":"2023-04-27 19:19:14.000000000","reason":"Change was abandoned"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2021-11-11 19:58:22.000000000","reason":"Ghanshyam replied on the change"}},"hashtags":[],"change_id":"I886fdf8488cc25e8d9a93af7560f655e9db92a21","subject":"Add tests for loggable resource","status":"ABANDONED","created":"2021-09-02 12:43:01.000000000","updated":"2023-04-27 19:19:14.000000000","total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"25a062c906e0a7978361808c216cf0ba311bf633","_number":807131,"virtual_id_number":807131,"owner":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-11 11:16:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-1,"date":"2021-11-11 19:58:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-02 14:06:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-29 02:50:54.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"a605460b735b3cd2bfd1661b70a3b3afb78e2474","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-02 12:43:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1497b846e9f92ee6f32dc504b30126080231046e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 14:06:15.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bce07f309baa446e84048f505a9ade8d : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8add97170f8144569a0b0222ebf411b2 : FAILURE in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37f685285905466ea2e723db29f93d45 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7a16ad0a0544fbe8c8116c936006f85 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66a70c4b489548adb690dee30fae6d9b : SUCCESS in 6m 03s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/09cdc60e52384d18abaf981443f6d318 : FAILURE in 40m 03s\n- patrole-member https://zuul.opendev.org/t/openstack/build/478b5d588c8e4e47854cac9f442f758e : FAILURE in 50m 16s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/ee11fe0dc251411a9f6d7c5dd07bf35e : FAILURE in 54m 46s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/5e6e1275879f43729b98514fdd751551 : SUCCESS in 1h 17m 02s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/4bcea011a3444818a76fd7aece60b476 : FAILURE in 50m 10s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/57bc7164113f462cb2419f7f3d6e2495 : SUCCESS in 1h 02m 37s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/4173f4e9766c46b1b35ca158f82b0452 : SUCCESS in 57m 15s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/2774b3d4c72f481b948c9f81b436e316 : SUCCESS in 1h 20m 40s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/b5380c74af1a44bfb6f52f9119126919 : SUCCESS in 1h 01m 49s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/78e8b761dc29467d8210357e9508e616 : FAILURE in 29m 35s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/301df1b16ed9489d8a5e1dfda83efff3 : FAILURE in 29m 40s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/b9f61ea182ce4c6e8068ae99d80bb543 : FAILURE in 30m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec8cd04d8441e5e7749aba753c3dd095f88e91e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-02 14:22:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ca972a50aafef4d012c007fd20cba28d1a1ed161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 15:44:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76faaf0a57df46eb9dfd1e307aea1063 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5fef0e3b3094d78849ccb06932a8df6 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02bbe1eedc214edea739943a5e1b3613 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6a0fef9bb4547239024fe01f73ab019 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4caa2feb6fb4ae4aa86c79e1f33b5d7 : SUCCESS in 7m 18s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/8dd00a936a5141d2a5e6218cdcd3aba0 : FAILURE in 57m 04s\n- patrole-member https://zuul.opendev.org/t/openstack/build/73e2f2c06ca34ce986bcebf9fbcdb554 : FAILURE in 56m 11s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/db3e760164a944d3846382288f08a374 : FAILURE in 55m 05s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/5ca532d7a9a0438eadac613e5be7cd5b : SUCCESS in 54m 15s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/1b620ab6fb224d41a91d2eef77da7fba : SUCCESS in 58m 07s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/5b0106fc7bb1402d9e53827b1a409c34 : SUCCESS in 1h 02m 27s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/22efbc29a66c4cb7b1934eefa3e1a713 : SUCCESS in 59m 22s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/0889a4720cbb4eb49fd26f02e3df5428 : SUCCESS in 1h 17m 25s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/27a79fb6dcf645ca89337f8c1895e2de : SUCCESS in 1h 12m 39s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/b0786a2ad716415a8e2b65d38f8edef7 : FAILURE in 33m 42s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/6bc477e864a44e4894c825c5cd597108 : FAILURE in 34m 41s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/e4d814aaa8f643abb2363f1879494c1f : FAILURE in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"437d734ffd3bc475e5e1f5131bec9db0f05a0526","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-13 12:34:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"724baddb43d9c076aec0707c482baf9865fd1c01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 13:53:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d5daedc49944e9bae266e80e4e390da : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62446c61509a41f1b3e30fd265c456b9 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c76a2641e3b422a97dd5a490aba79cb : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/511ad33c82b44a10986378f1e87c34b5 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7ef97dc9ee1421e964895b1b48125fc : SUCCESS in 5m 43s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/d1313dd25d264432ada3fa9f9192f2d8 : SUCCESS in 41m 17s\n- patrole-member https://zuul.opendev.org/t/openstack/build/d5fc14ecca0349ada27edda2b3cf8e0c : SUCCESS in 45m 43s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/9e061a2f2897469f8fb26ee1525beed8 : SUCCESS in 47m 28s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/652bc0c4774a4a29b67f6fa12251a8e4 : SUCCESS in 53m 32s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/5a88b9ef01174bc4bcb637f5286967b1 : FAILURE in 47m 58s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/68091319b124477893747d0557825c78 : SUCCESS in 50m 24s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/54611e061835486e9949446ada547e69 : SUCCESS in 52m 57s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/59e6aa7873dc46d9b96fd33149e80842 : SUCCESS in 1h 03m 08s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/c11d29eefef242428ef7e7a96eda35ee : SUCCESS in 1h 03m 02s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/d2fa5cedbf084790907d4d1dba999798 : FAILURE in 26m 39s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/a2aa5509246249c6a1bb28c5c3cf4fc9 : FAILURE in 27m 02s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/981970fbddf8477eaab929f4f11cae55 : FAILURE in 28m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"987f534bd15e3d02bc91b3998d335cc9aeb584f6","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-23 05:10:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"225b33fbd9471e2ff0f933dcf5540a574c5036b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 06:24:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47d202f36ce64e90878dd0d93814f87c : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2463b4d279b24fc2ac85aef513c72a04 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f10dbeff27e045e3b4e8f5ab91934c6e : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ec5f6f6c920440b99cd46e2bdd050f3 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad710ea7ef534a88aaf47d5f8a174386 : SUCCESS in 8m 18s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/4e8069cddd4644c3ac85f851da9ae6b5 : SUCCESS in 48m 35s\n- patrole-member https://zuul.opendev.org/t/openstack/build/b1b9b6341d8d4d81b9a49828de6d1473 : SUCCESS in 47m 39s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/1fa86e0a6192484d90834d9532413f26 : FAILURE in 55m 39s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/80f648e60105440c92691b457b4281de : FAILURE in 1h 00m 21s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/f031e3242f464f528139f626b2e36957 : SUCCESS in 52m 38s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/bdf92b1b912b483b89f59f288ced9c38 : SUCCESS in 59m 04s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/ed319c56096147f9838afb3a0296e55d : SUCCESS in 56m 48s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/f55aef9adff44e60bc6c3082d953ce60 : SUCCESS in 1h 08m 02s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/b83d07c41fc2438b923832a059535dc7 : SUCCESS in 1h 03m 21s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/f276668620d140e1ba345dab37ad2852 : FAILURE in 34m 21s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/f07a050fe78c45d795484ef18fe816a5 : FAILURE in 41m 31s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/5ec5d37748fc4e8fbc4044cfa9fd7598 : FAILURE in 35m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"389b39e68a619152fe2f45e65556f539e698deac","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-23 13:40:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e06c9302317578a5eb5ccff96517edc008b6ccf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 14:51:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07de40c9f72e44f48efb5ae8810afd17 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ceb9d89c79c46b98e86550f2d823d7c : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7b312b23f4e49baabe7595d0808d20b : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/188e4690a13244dc9788435691ec04ee : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f0e510c21d4add8c0c773dce759e66 : SUCCESS in 7m 46s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/52adc0019fca4da192b95abd590017c9 : SUCCESS in 46m 25s\n- patrole-member https://zuul.opendev.org/t/openstack/build/ec6e60861daf4ec599eecb59fadeb094 : SUCCESS in 55m 52s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/c0e0a35e62cd43beac0228a0ba3c2030 : SUCCESS in 50m 58s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/fd737e17916c416fa8acf9c27b2d575b : SUCCESS in 1h 01m 01s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/2232dddcf867432fa80fe0adb8e01226 : SUCCESS in 58m 37s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/f65149e3631d4541939cd6667c19aff7 : FAILURE in 1h 01m 06s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/cd97417e541c417b8f012713d097b895 : SUCCESS in 1h 03m 34s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/0d8ddcb7f32a4b77a53863b12b908e37 : SUCCESS in 1h 06m 40s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/4fdaf9ae39b14964a02656975223d514 : SUCCESS in 1h 03m 40s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/b79bc407b9f74aa78d5d42beaffe001b : FAILURE in 30m 18s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/d95bf9a2ff2c4c25920f18f71bf18e3e : FAILURE in 30m 21s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/9a00e6a0ba434766a6c70ea722755eab : FAILURE in 38m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cc5edeae3cc6482b464028b66bec598df6635d5b","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-23 15:15:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4363d227c4ac971452984adaccca8fa1c2865a01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 16:34:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcacbb2492e14407881ce95276405450 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df3e4aada8e4484d9914bbb66ac73fc0 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e70a4e27ee75459b8a9132e0f634133f : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad9f41dc7524420caf0252c04773b578 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c48d0ca951224394aeb5a4e191cf471f : SUCCESS in 6m 21s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/8bdf821ce4e949e6aa2122f5d0f3cc79 : SUCCESS in 49m 56s\n- patrole-member https://zuul.opendev.org/t/openstack/build/1b3d2dacdad24de0ad0acb31311d7c94 : SUCCESS in 52m 09s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/47758879cee347058296b4b94252c732 : FAILURE in 56m 30s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/059710cc9d3d4ba5b0d53381ef7872c7 : SUCCESS in 56m 27s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/edffbf5e10554fc69d2d519e21750f9a : FAILURE in 1h 00m 28s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/5357df019c394aa5b68335f728a72411 : FAILURE in 55m 30s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/2fe7c4942095486c994445cfe53f47f2 : SUCCESS in 1h 11m 45s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/b60260228a51481fb2362c075cbec04b : SUCCESS in 1h 04m 02s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/11e5ce38c4584b45a56dfd2b939dd630 : SUCCESS in 1h 03m 46s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/84a74718fdfc4fb297d171aedd827202 : FAILURE in 31m 01s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/0ffe4ef7233b4e7b83e662010cfb3bde : FAILURE in 30m 14s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/e396d16d24084b03a43850241b46ecce : FAILURE in 31m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9e7dafc9c1574bf7bc0ced20a34ca26029e676df","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-24 05:08:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f4c139ae187a10224104645aa03907aab9565002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 06:36:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/970e18a44e4b46c48051e643ca32edf3 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e022ef52d22747bb8d0809e842b4cb6a : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8fa3f64c50f4df79be0d3ff2b9c7561 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78e73683a18947d58de0b78d4b081c81 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41ef957d8c41491b93ad0a3f272782d4 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5eaa0b97ae742d59ac2ecf0508f8748 : SUCCESS in 6m 33s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/137bfb20f7a0487fafd06182c6035ffb : SUCCESS in 58m 51s\n- patrole-member https://zuul.opendev.org/t/openstack/build/29d2bcdd41e04821b9a0e8daf0eaade0 : FAILURE in 47m 18s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/c745fdf7af764f8583615c9ff5133018 : SUCCESS in 49m 15s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/9c2e5372d564473481a62113864b54b0 : FAILURE in 48m 44s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/04c3e1778bb14e3ba16191aa549ffc27 : SUCCESS in 1h 03m 15s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/28cc06fecbff4a7fa388ffc79d1cb800 : SUCCESS in 1h 00m 18s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/5b6ff752df594981a4750c7acc2bb021 : SUCCESS in 1h 02m 57s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/3f4491a186cd4393828dc1a2a86949fb : SUCCESS in 1h 25m 50s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/dd8edc0802004862ac8c1fe34b748f68 : FAILURE in 30m 28s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/6daad5ea983f4f64b38d8f2f595fb15c : FAILURE in 29m 52s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/bb9ac3512c314af99d55a8e11e135b1b : FAILURE in 31m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"40a626babd4237b7afc16bda9b73d58e37c8335d","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-24 12:49:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"64bdbdedfb2baaffa21a4009bae6ec793409f6c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 14:03:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/939319f02ac848669ea421dc96cb340c : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d152b4b45a8e4128a29139ac3e00d0e2 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e4b4c3eeb4c4dc98f08af9ad86658c8 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c2b3b8781bd4f9ba0bf432f36f80e7f : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/083897001f5e46b2aa58df682990a9be : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/630e680101b44976be54beffded43756 : SUCCESS in 6m 30s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/0b4524e8de574e7799fd0af7f06c3ffb : SUCCESS in 43m 33s\n- patrole-member https://zuul.opendev.org/t/openstack/build/096943d772c440ffbd0fb7a641b6cfbd : SUCCESS in 55m 04s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/8df8c8b614df42288669f1a6abda9d3a : FAILURE in 54m 08s\n- patrole-member-wallaby https://zuul.opendev.org/t/openstack/build/04d38a18a7994b588c7080ea3c674a6c : SUCCESS in 59m 06s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/4d848a90034f4829ba23f29c3cfcc2d1 : FAILURE in 1h 01m 56s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/74d477eea579478da933d30062277818 : FAILURE in 53m 39s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/41b2e79a09f245a8a18484549bc8861d : SUCCESS in 1h 06m 23s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/93510f3dea4a43a0b5240a06d1659f2c : SUCCESS in 1h 02m 55s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/c3b4f30ffb004aa6a40b282860751acc : SUCCESS in 1h 02m 03s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/8229940eb43b4b929046bf02d5619eec : FAILURE in 34m 41s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/bb6768be1b8f41608187d4b279ca23e7 : FAILURE in 34m 06s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/394e6cf018f840b18e6febfbe47b7add : FAILURE in 35m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e0282c51a3f1e6bd5fd900f20a43342f512c5a6d","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-27 04:11:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"26311b286a1b4a3f7f8ef33d6eba2880829309ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 05:24:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dadbd3d4769d4a9f8e35fa4fb18dfd67 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d55baa7c1e6c4d4d9817caf4da30933b : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cffe5699a404dd7a16a0a48523d6042 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca12e639d0f34477b330dd26c466acc9 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d07a9697c8c94573be4a2890c0892d64 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/348698b286b7430da7e7a10042e36ff8 : SUCCESS in 5m 36s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/8704cb5a4be4419da0f6c9cb324a7196 : SUCCESS in 43m 46s\n- patrole-member https://zuul.opendev.org/t/openstack/build/415daebbbee04b998e72bc9f28506883 : SUCCESS in 54m 04s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/3de9696938534d81b61c583ecbbefd57 : FAILURE in 51m 35s\n- patrole-member-wallaby https://zuul.opendev.org/t/openstack/build/5bb46cfa15c747c796de65d2381b5e95 : SUCCESS in 58m 17s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/cb45534ff36e4330afdeacb37fb9d8ba : SUCCESS in 49m 41s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/85c454011c6a423992b339895674b2ce : SUCCESS in 59m 52s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/51e48282fe594da38decb3a1a875898e : SUCCESS in 59m 41s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/fd944febb43b47db9aac9926362926df : SUCCESS in 1h 05m 49s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/6338b8e950fe466cba7e9f3475964162 : SUCCESS in 1h 03m 21s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/6f735fe05d464744bace48c175908be7 : FAILURE in 32m 53s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/bfdc4d473d204909809430ca9ae51aeb : FAILURE in 38m 39s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/30a982cd8fbd4d61b9996cd3d3e532b7 : FAILURE in 30m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5dec2476558a840ed4d64a64ddc42bf69b76e110","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-27 07:54:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0db0fd8aa840ebfeb83cea0102a0505fdf57e09e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 09:04:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/410e0b3c17f6431d8fbaa87d6b7bffde : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa274c2a0e5745c3a544e864a6c289ae : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f182fb1457a4176a93984db0c5e006f : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6908b29b8814137b05d22221550dd36 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ce64ae9559d496d932c544d10bc4da0 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d63c869d3ddf44ae84129fb853a8aba4 : SUCCESS in 7m 12s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/07116e05b6ad4f23bcd87d61cd591997 : SUCCESS in 45m 21s\n- patrole-member https://zuul.opendev.org/t/openstack/build/fabacd5802944ae4a7ffd36851c75866 : SUCCESS in 54m 32s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/d9800a1d50b147778484224c2bf5d33b : FAILURE in 50m 33s\n- patrole-member-wallaby https://zuul.opendev.org/t/openstack/build/c2c59bea7ef54106ba71419f7378a4f1 : SUCCESS in 59m 05s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/19573682b20740b48f7bd7b5edc0d5fd : SUCCESS in 54m 38s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/352d197f382b49369c73789f8c8e2187 : SUCCESS in 58m 03s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/1fa4e0f4b78a479698adc0621e2ea259 : SUCCESS in 1h 07m 24s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/6d23baadc858473d904d341622b9e25f : SUCCESS in 1h 07m 00s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/8c2c392dca494b149854f8e4249de35b : SUCCESS in 1h 02m 15s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/cf2960e76600440e83d83fa75d796308 : FAILURE in 38m 42s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/41759c2d0c444b75977f407fd1bd8983 : FAILURE in 31m 49s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/762e69ee76aa48f4a02a8a76d06a71fe : FAILURE in 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"45fcb53632ea73ce6a58ee6de4ef9b44260738cd","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-27 10:39:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4d003d3acf014312bdf00c03cb18125eb80ad8c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:55:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7065c9e9a2414a37a2c53fda683b48b4 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54de2e9fc19f44c0979266523ac31e6a : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/916dd9c1b2d34604b531d6a1bd22b6bc : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8242d47b9f8e4a269c2a565048816609 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e03bc9b7d3874b78afde320c25c21b14 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dd3cde85dfe4959a983b193fbb95fde : SUCCESS in 7m 50s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/a7bcc532483946c1b4e3433b230c137e : SUCCESS in 53m 13s\n- patrole-member https://zuul.opendev.org/t/openstack/build/144a786b8b6745a59cf63d3cdfbe7621 : SUCCESS in 59m 03s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/1fafe43ad3ea4ebea7ceabed9b29d80d : FAILURE in 58m 15s\n- patrole-member-wallaby https://zuul.opendev.org/t/openstack/build/1b5b4d9c7bc44a7891a3296b5e586ce9 : SUCCESS in 55m 26s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/82c0ad3b768647eab0be7d81043f950d : SUCCESS in 59m 02s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/e528376dccf74043a5e42da24fa2a940 : FAILURE in 53m 56s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/970679e46cf14bbf8edb51ded74a459d : SUCCESS in 1h 04m 35s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/44365d0e716a424084967b8d3a3f8327 : SUCCESS in 1h 12m 25s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/fe4884893fbf4da88542a2c8508b7b55 : SUCCESS in 1h 14m 02s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/8948cf9beac4412e85799b042df002f0 : FAILURE in 36m 39s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/ac0be05a708b4a049f157a584979ff1a : FAILURE in 31m 56s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/cbf514af71af4493b959828f25697ecb : FAILURE in 31m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f956f26db3946c0a54af73098ae8dc12acd3b514","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-27 12:04:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6a78e5cbd6307ed69b299b6fdcbca875cf37c8fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 13:18:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78008d7814ad4258900081388a601cdc : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a5f15585b9a4231954a4f6b7e5b23c3 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b66a3d05bdd44b77adb3c12d862d4760 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13f5f579e9a343208e17513d06133549 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67e0fbb2466e483b99b9ff89cfd7f298 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a464ff9416d842e78017aad2f2f95cb2 : SUCCESS in 6m 09s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/5f9022357c2a4d37b336ddca0f57b297 : SUCCESS in 47m 35s\n- patrole-member https://zuul.opendev.org/t/openstack/build/afdfe9a8957d48079e43fd354d037cb2 : SUCCESS in 55m 26s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/06eae1c377414c238a16e2317d660c61 : SUCCESS in 49m 49s\n- patrole-member-wallaby https://zuul.opendev.org/t/openstack/build/73ae5e59076e46409770e8fe82a0b665 : SUCCESS in 1h 00m 29s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/3e9fc9e4e6ce4eeebb09c1f9bc777bcd : SUCCESS in 53m 35s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/df57ddf7139a4b228182b1877d5aa7bd : SUCCESS in 58m 20s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/6fba4a096ed04b9484a640bb5c64d01d : SUCCESS in 1h 03m 31s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/722216906a6948e59a3116a409d60e20 : SUCCESS in 1h 07m 52s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/cbd20b29c9b44aeb9bafb8e1cb6d763e : SUCCESS in 1h 02m 33s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/37168cca0a56485a8b4c011afd6ff706 : FAILURE in 35m 50s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/4c7dc08f1c9544cd994a13e16d3fb2de : FAILURE in 32m 40s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/50e7f387a54c4b139fec329eb030542c : FAILURE in 36m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aedc9ae86564e407cb14852ca65d30160bc4d2a7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-29 02:50:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nplease check if test is running or not, we need to show the test result before we merge it","accounts_in_message":[],"_revision_number":2},{"id":"69c344ffbc44ece96e788ff122fe0acb2ed2690a","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-10-04 14:03:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"12217aa332a94e04742031b9a575d945d79ed286","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-10-05 14:15:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"476c4c552340d6be44f94cef8cef7803c5febf14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-10-07 10:40:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"47b0ed080677f7dfb6e8c40d8eac21a05ca96270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 11:59:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4950145ef587428397cbef6d1371d97f : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb72782aa7274ee9ae563497da69eb92 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e03b1fc99504be2ba2e202ac83ddf29 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b27f10ec59f4434abb9574d5db8a116 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56aafc84109d4a1887f0c16efa46420b : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7337c644650245499ce446212ebee93f : SUCCESS in 6m 47s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/23e842af7a6e428d98295bc94c1d7965 : SUCCESS in 44m 54s\n- patrole-member https://zuul.opendev.org/t/openstack/build/bb6a5a6a292a4bbc8cea17ef3b5ee84b : FAILURE in 53m 53s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/47cbc1a0a20a4826b7c39b9f41f1d12b : SUCCESS in 53m 12s\n- patrole-member-wallaby https://zuul.opendev.org/t/openstack/build/626654b7ef7c440fb9593a71aa7a8d2a : SUCCESS in 58m 41s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/3c832b0d1fee4a64a9bd102c83612f39 : SUCCESS in 56m 47s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/3c913a9427c44f5886cb4193d59096a0 : FAILURE in 1h 01m 56s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/fd019bdedcfb40b18363423c96619e2d : SUCCESS in 1h 08m 10s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/c433a5ee91a64df395b35d98bda7973c : SUCCESS in 1h 10m 34s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/ba204bf8a223402daf2befc58ea49342 : SUCCESS in 1h 00m 10s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/ac204944815a472284015a06dfdafac2 : FAILURE in 29m 46s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/1cf6fb0b7a8f4f67bb363ca3099b4441 : FAILURE in 36m 18s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/1efb257688b94c37b892bc10eab658c5 : FAILURE in 40m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c701f7e4e274438bfd174eb6000b8681ab4f704d","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-10-11 07:17:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"12d25360166b2588a39383f1e585c6dc16c1778f","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-10-11 07:52:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ce614db29ce2e9dd322553bbd472f4e996a2609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 09:16:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5797d2dfa0bf44ae89cfc3ca50b2352a : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/883ae216f852410496d33aa49a1c06c3 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/859f24d22cb742899519076cd472dabd : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/951748e2ebe5491fbb1f5ec3c024dcd4 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d63fa576df4469da41f8965fb3551f0 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a8580f15a914809aeb46590d5131e53 : SUCCESS in 6m 30s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/310751c37270468aa1e332a8f1e9a318 : SUCCESS in 42m 59s\n- patrole-member https://zuul.opendev.org/t/openstack/build/6d0357359f6f4d5c88ea9f48d7efa75e : FAILURE in 53m 31s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/807b660697f44d30ac04b97663725dcf : SUCCESS in 59m 08s\n- patrole-member-wallaby https://zuul.opendev.org/t/openstack/build/c42af3252cf848a69c2bbbc1dc6c5907 : SUCCESS in 55m 19s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/e2252d4e995c4442ac38a0d10bf4800a : SUCCESS in 55m 41s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/ddf3e419df9143d9b342c22617698f07 : SUCCESS in 1h 16m 35s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/0e003c9648f34bd49324b11c502fdf2c : SUCCESS in 1h 03m 10s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/edd77858dd3845e98f911ccf8a106104 : SUCCESS in 1h 05m 14s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/3ffb781d7b084ef3930f17ee9f3f4672 : SUCCESS in 1h 04m 08s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/0dd013f62ec0494e9b594c68c446774f : FAILURE in 29m 49s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/7d88c9ffc1c94ca6a14dbdfd18105c42 : FAILURE in 33m 52s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/e7d60328c16f445d8b229f67932d26d6 : FAILURE in 33m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e19e87f1626d4c1c52eb53579f5b5035784e1fc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-10-11 10:08:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1cacb1a21707505bd002e3a80baf40c2a596bc6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 11:16:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f85b88801ed64d229598c7280f83ae8d : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88fdfbb67f934547ad7e5b4cdffdc056 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a15193df8f6d4a3b80144b8aa4b99f26 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/318d88ef52cb4c58a07d55cca734cb19 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb57349ec7e34018950a962cbf60bcba : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/885bec4970f24e7d8d69229ae1f63a49 : SUCCESS in 5m 32s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/6c0eb0e4ca1042dcb6f2e3b6bb971a5f : FAILURE in 38m 29s\n- patrole-member https://zuul.opendev.org/t/openstack/build/17a2c024153c4cfba25ff4a73b9af8ea : FAILURE in 41m 07s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/72a04f92ae5546d4b1bfc0b34a83efc6 : FAILURE in 50m 32s\n- patrole-member-wallaby https://zuul.opendev.org/t/openstack/build/dfa77476d00d455383e690d756b2cd09 : SUCCESS in 46m 33s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/8633aaee49da4f33a365b399e30ccfba : SUCCESS in 58m 37s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/8af67cde08fd491fbcd70eb5cb4fd69a : SUCCESS in 58m 57s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/072f9ffd7dc843c998fab101ee96a61f : SUCCESS in 1h 02m 24s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/d78d515581f64f4e87a708dd990b65f9 : SUCCESS in 50m 02s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/dd3a6c47a0b441ad990ad16d39a118d3 : SUCCESS in 1h 04m 33s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/43c0848802da4dd9b1bf07a9a80ef317 : FAILURE in 22m 45s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/c0876b397e8644a09970a43501e70322 : FAILURE in 31m 42s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/d0603d9374ee4643b257db6cff8ed948 : FAILURE in 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3b10325a828755183aea9c26146bfebae6072cf1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-11 19:58:22.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"25a062c906e0a7978361808c216cf0ba311bf633","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-04-27 19:19:14.000000000","message":"Abandoned\n\nThis project is retired, for more information, refer to the https://github.com/openstack-archive/patrole","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"38639c3d1ceaa8737c21d6e1c1d856da84ed0753","revisions":{"f53cd86fdfaf38701e6b4fcba02d2064a48d2e58":{"kind":"REWORK","_number":1,"created":"2021-09-02 12:43:01.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/807131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/31/807131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/31/807131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/31/807131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/31/807131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/31/807131/1"}}},"commit":{"parents":[{"commit":"82fd18ba85e0034bae443bfb55f9b0589a383c93","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/82fd18ba85e0034bae443bfb55f9b0589a383c93"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-09-02 12:09:05.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-09-02 12:09:05.000000000","tz":330},"subject":"Add tests for loggable resource","message":"Add tests for loggable resource\n\nThis patch adds tests for neutron loggable resource api[0].\nPart of \"Increase Neutron RBAC Coverage\" initiative [1]\n\n[0] https://docs.openstack.org/api-ref/network/v2/#loggable-resource\n[1] https://storyboard.openstack.org/#!/story/2002641\n\nStory: 2002641\nTask: 29352\n\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\n\nChange-Id: I886fdf8488cc25e8d9a93af7560f655e9db92a21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f53cd86fdfaf38701e6b4fcba02d2064a48d2e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f53cd86fdfaf38701e6b4fcba02d2064a48d2e58"}]},"branch":"refs/heads/master"},"521274d55aecbe9befc168ea9e9748ee27568af7":{"kind":"REWORK","_number":2,"created":"2021-09-02 14:22:27.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/807131/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/31/807131/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/31/807131/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/31/807131/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/31/807131/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/31/807131/2"}}},"commit":{"parents":[{"commit":"82fd18ba85e0034bae443bfb55f9b0589a383c93","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/82fd18ba85e0034bae443bfb55f9b0589a383c93"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-09-02 12:09:05.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-09-02 14:22:10.000000000","tz":330},"subject":"Add tests for loggable resource","message":"Add tests for loggable resource\n\nThis patch adds tests for neutron loggable resource api[0].\nPart of \"Increase Neutron RBAC Coverage\" initiative [1]\n\n[0] https://docs.openstack.org/api-ref/network/v2/#loggable-resource\n[1] https://storyboard.openstack.org/#!/story/2002641\n\nStory: 2002641\nTask: 29352\n\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\n\nChange-Id: I886fdf8488cc25e8d9a93af7560f655e9db92a21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/521274d55aecbe9befc168ea9e9748ee27568af7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/521274d55aecbe9befc168ea9e9748ee27568af7"}]},"branch":"refs/heads/master"},"aeed7fb98eb3e52fb02d0934efb023e0f5be4905":{"kind":"REWORK","_number":3,"created":"2021-10-07 10:40:58.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/807131/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/31/807131/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/31/807131/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/31/807131/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/31/807131/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/31/807131/3"}}},"commit":{"parents":[{"commit":"82fd18ba85e0034bae443bfb55f9b0589a383c93","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/82fd18ba85e0034bae443bfb55f9b0589a383c93"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-09-02 12:09:05.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-10-07 10:30:49.000000000","tz":330},"subject":"Add tests for loggable resource","message":"Add tests for loggable resource\n\nThis patch adds tests for neutron loggable resource api[0].\nPart of \"Increase Neutron RBAC Coverage\" initiative [1]\n\n[0] https://docs.openstack.org/api-ref/network/v2/#loggable-resource\n[1] https://storyboard.openstack.org/#!/story/2002641\n\nStory: 2002641\nTask: 29352\n\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\n\nChange-Id: I886fdf8488cc25e8d9a93af7560f655e9db92a21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/aeed7fb98eb3e52fb02d0934efb023e0f5be4905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/aeed7fb98eb3e52fb02d0934efb023e0f5be4905"}]},"branch":"refs/heads/master"},"38639c3d1ceaa8737c21d6e1c1d856da84ed0753":{"kind":"REWORK","_number":4,"created":"2021-10-11 10:08:52.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/807131/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/31/807131/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/31/807131/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/31/807131/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/31/807131/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/31/807131/4"}}},"commit":{"parents":[{"commit":"82fd18ba85e0034bae443bfb55f9b0589a383c93","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/82fd18ba85e0034bae443bfb55f9b0589a383c93"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-09-02 12:09:05.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-10-11 10:07:56.000000000","tz":330},"subject":"Add tests for loggable resource","message":"Add tests for loggable resource\n\nThis patch adds tests for neutron loggable resource api[0].\nPart of \"Increase Neutron RBAC Coverage\" initiative [1]\n\n[0] https://docs.openstack.org/api-ref/network/v2/#loggable-resource\n[1] https://storyboard.openstack.org/#!/story/2002641\n\nStory: 2002641\nTask: 29352\n\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\n\nChange-Id: I886fdf8488cc25e8d9a93af7560f655e9db92a21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/38639c3d1ceaa8737c21d6e1c1d856da84ed0753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/38639c3d1ceaa8737c21d6e1c1d856da84ed0753"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
