)]}'
{"id":"openstack%2Fpatrole~810274","triplet_id":"openstack%2Fpatrole~master~I4be60e3e92704f8e55d3acdb0e025078ae5b21f1","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"I4be60e3e92704f8e55d3acdb0e025078ae5b21f1","subject":"[Fix gate]: Cinder policy change handling in tests","status":"MERGED","created":"2021-09-21 15:12:12.000000000","updated":"2021-09-22 17:17:57.000000000","submitted":"2021-09-22 16:02:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810274","meta_rev_id":"a9f6fb9afabb3b5e59f9122ea4f778985cb45172","_number":810274,"virtual_id_number":810274,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-22 16:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2021-09-22 14:10:34.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2021-09-22 14:10:34.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"CC":[{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-21 16:22:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-22 09:40:36.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"CC"},{"updated":"2021-09-22 14:10:34.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"cabb2aeabd2178cdccda303ed4f4dc05371794e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-21 15:12:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d30b29394e2391efb684d13f43a198d5e545ebbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 16:22:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5da0ecf20c684a219150f83419ad20d5 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3475a612a31a4d59b368a5ca45fd841d : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04a0114813a641bd853da6ff5d38fc14 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35d102b283b645bd902a0c37f8f14898 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/458d1f7a20ea48bdb5cb9d0a8268d005 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8fc022b55bd458297ad52ebd0198677 : SUCCESS in 4m 37s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/39be46e20e7a4a3a9b7fb4a87a794228 : FAILURE in 49m 15s\n- patrole-member https://zuul.opendev.org/t/openstack/build/c91a320610034781a6e9febaf9de94e6 : FAILURE in 54m 43s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/a85335e0cc094c3891a8e80c6ca98912 : FAILURE in 56m 30s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/32c5625550114de589b1fb9e32f9d885 : SUCCESS in 56m 33s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/5bcda5354fd24639a84a9d3ea8a58576 : SUCCESS in 52m 31s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/77209775274b4a29980ccc84643a7ddb : FAILURE in 54m 43s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/bb2d512229074f548d771b9dd121f771 : SUCCESS in 1h 04m 03s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/89f2fa0fa8a14ba49edfb2788f875fea : SUCCESS in 1h 03m 51s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/cb30d666b00f4a0e810e113188fa52e2 : FAILURE in 47m 40s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/6a2c78d3285d4c35aca27eac6b33fc7b : FAILURE in 39m 40s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/1f3c796fde924d5d8ff51ae76ba528c4 : FAILURE in 35m 08s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/ef431bfdb5d34152a0ae41294cbd709b : FAILURE in 36m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5375adb14f84aec8e3a13d1d0a87a02bebfec5ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-21 16:52:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a10be89bda5d2eaca74fa31bb99122ed1db25688","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 18:14:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e05c484e1264bd99a3e3620fb51b4c4 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c237ef4d95ee4262b9778efe372864a0 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f4bbb66fdbb42cdbd05899153336af9 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/179a48d74f544be6911b7a215fa27c4d : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/099e9999ccb44618bf8aa0d77f8a2d66 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c52780a4a53a43aa8bfa7b8081b53767 : SUCCESS in 4m 16s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/f2e3ff6aca674f4dbc8c1e1fe5749e5a : SUCCESS in 44m 55s\n- patrole-member https://zuul.opendev.org/t/openstack/build/79a51c92bd0443c8a3fec31acb8c33d4 : FAILURE in 54m 43s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/727b8ddd90294e6aa368f915371682d0 : FAILURE in 52m 40s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/4f6d41366e8a47c0a02fc4b500cfc75e : FAILURE in 53m 09s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/81d21d86e2a34c1586e39809f00fb4cb : SUCCESS in 51m 19s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/76b9882b25714262b1edfe3cbc4d8af4 : SUCCESS in 54m 30s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/7022626d529c4196a11a22326da02a3e : SUCCESS in 1h 09m 39s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/126acac0027b4e1bb7943ed3a3ea3515 : SUCCESS in 1h 14m 58s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/03e2d86c73f148c7837d19279a51a985 : SUCCESS in 1h 00m 01s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/b63272c65c9b4d6fa200694b5c8b6785 : FAILURE in 34m 25s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/75007ba0ad43452e8c5c44b673d2af60 : FAILURE in 33m 59s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/46b89e3e4dd4492486e650506e5d690a : FAILURE in 36m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f85332ecc3dd7d02f550edb5e0bbda0b29458c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-21 21:28:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a5089635c6d2d985aee710880d5e22161bdd67c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 22:24:47.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/268fa2c958554f4d887f3268ebe24a8b : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecbf7900116748d0a3bad88878d85cfe : FAILURE in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b2ceec2187146a6b319d575c23b3a02 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c04f7dda765040f8b651708ef112ade2 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ded9a051b1349aea48bb1e763bf04b3 : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a469797e410d488080cd62627ef93a73 : SUCCESS in 4m 31s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/2f496d4ab7074b57a611cfbbd783b77a : FAILURE in 29m 00s\n- patrole-member https://zuul.opendev.org/t/openstack/build/82064f5346604808bebee2df0cd7b9e8 : FAILURE in 26m 40s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/deae946f1ec64e38bfce5f7491b73c71 : FAILURE in 23m 50s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/781be6e2636c4232ab79d39f78741ba5 : FAILURE in 25m 42s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/51475360cb364097b99623710a9162b0 : FAILURE in 31m 20s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/529c4af95c5741a1a1186f01f09a3afe : FAILURE in 28m 47s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/6817f2bd072b417281c28c1fdafdd85c : FAILURE in 41m 53s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/941ba81ffd024d81b4a649886731a46f : FAILURE in 53m 06s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/ea0fd31ba88448c6a5afe1178158b888 : FAILURE in 39m 49s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/ce29627e62e8440f97175998fb7a6675 : FAILURE in 25m 32s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/829aa04176c742c7ac81dbe80d706980 : FAILURE in 24m 30s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/df8c4f4dfe7a449195d8951674caed20 : FAILURE in 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5e24880c7c07c5c9edebb718deb94a99e3468a96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-21 22:38:40.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"2f8b1a1a0b227ee2f5a49baaf1bb7841f38b2fb1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-22 00:48:08.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0dfafb90ffb9c460be01971202ad9d9c3cce3dc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 01:50:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7e51b3915df45acbcb17e751fd82dde : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94ed878925d94f7182f0f2faa2b391c3 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76a33b2b20404fb8af284153356c87c2 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91a67b3881754badb8b713a9a90bdcdc : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d2934a8c3d401da01b5c4535f05e89 : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f54b813e4654ba19915ae2929cadeff : SUCCESS in 4m 07s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/d2da9c1ccef24d38a3cff4a8db945c4e : SUCCESS in 42m 44s\n- patrole-member https://zuul.opendev.org/t/openstack/build/cb317e0e98bc4da79b5cd8711935e3da : SUCCESS in 46m 54s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/1e5967a01cd24f95971f7db71699ea0e : FAILURE in 54m 59s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/a0e6d0f150f8489b94fa86af79b6b275 : SUCCESS in 56m 44s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/57c988d88e7f45fd9ecb685700c5dc29 : SUCCESS in 52m 02s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/cbfde692fe314e56b7df435ca643fdcb : SUCCESS in 51m 34s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/112aef75a65a4e388614c2d3cb6ffa0c : SUCCESS in 59m 30s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/204ab251a727471a94e380991aaa36bd : SUCCESS in 1h 00m 27s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/299eef23a4a3416db49f0d1278fbdaa9 : SUCCESS in 59m 23s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/a266e4b0a79c4249a320f0ccbff20dc4 : FAILURE in 34m 13s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/722926e67dcc4108bd574d789acab66c : FAILURE in 32m 14s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/1f8e17277d214f72b67ba7d4bb0ecdfc : FAILURE in 31m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4c15ebcda10d93e0347bf79699cb75cc53f4fcf7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-22 02:22:40.000000000","message":"Patch Set 4: Review-Priority+2\n\nrecheck volume timeout","accounts_in_message":[],"_revision_number":4},{"id":"a67f5447276b63592aa4674757eeba08c227529d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 03:34:48.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/930e5b8c70184a6990ac7efb350417d8 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8525c37b68e41b99eb57d7083cc0b1c : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be0ec72fc66d4b019dfa040129ad49dd : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2fb9c176c4c4eff988ef6db0a3545f5 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be5f488a922340f4b49bcc736f8b4ee5 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/206df03680d344f7ae61d8566e5fbdab : SUCCESS in 3m 45s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/17073295bdda4361902c2eebedec7713 : SUCCESS in 47m 42s\n- patrole-member https://zuul.opendev.org/t/openstack/build/15b1107b115e45da878c2fc569721951 : SUCCESS in 54m 16s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/dffa56bc44504ee1a4c2eb342edd37cc : SUCCESS in 57m 22s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/064526631b8d4961b2cfcb9f4bfb1f5e : SUCCESS in 1h 02m 39s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/dafdd8918a824f5d91ec53d88a9a8a94 : SUCCESS in 59m 03s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/753db42790de4891a6f3ddd841accb7c : FAILURE in 59m 41s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/2059230b4b68493590d0a19054bf4d28 : SUCCESS in 1h 03m 19s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/a21de36a472b43a6ad851ffe2b318c65 : SUCCESS in 1h 03m 38s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/72761bf1ae4946b5a7517ff44f9d4e8a : SUCCESS in 1h 05m 13s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/4ea75c4f25da471fa770e8aa94cf6390 : FAILURE in 33m 58s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/3e758e0f106e42829a451ba833697705 : FAILURE in 38m 13s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/3528a7d8fad947b6843b1f8aab5344e7 : FAILURE in 35m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c0b206d462e5d6c305706f8d7d18151f0f8a4f8e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-22 03:35:57.000000000","message":"Patch Set 4:\n\nrecheck network issue","accounts_in_message":[],"_revision_number":4},{"id":"c9864b6cec5bd3e3b23f21c2d29b30f1321669c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 04:42:03.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fea236e24ba4991a8f30c320a5f12b9 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80e5949ef17048fbb989f80e5d1d51df : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/738bb2ee0f4340d29843c258384bd4b6 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49703a83c8db4795ab2bdf34701e2172 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5ff44c2bcc3448a8b069e7d50a253ea : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38583809b60f4f47a2e7bd9ee678e419 : SUCCESS in 4m 30s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/bc2ec1557c274e70b0b303d2691d6935 : SUCCESS in 38m 35s\n- patrole-member https://zuul.opendev.org/t/openstack/build/2cdf07f61d2e458cb08a0a167cc4c872 : SUCCESS in 42m 08s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/ce229225f2b44dbcb33c485bd19f11a6 : SUCCESS in 43m 56s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/eff9b287f37e4db3a8f3515e9cd67656 : SUCCESS in 1h 00m 02s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/994d0cd185e14eb58b63c8d7b33daa40 : SUCCESS in 42m 56s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/baec480783f74388b1d7ddf9b2dfe0cf : FAILURE in 58m 20s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/9141b26d0f324818a692f7f3d44d6dd6 : SUCCESS in 50m 45s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/4de168becc574f0aa89826398ed908f0 : SUCCESS in 1h 00m 54s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/059d280ae3834afda12e3ba365bd9cbe : SUCCESS in 1h 00m 03s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/ad6ea4b2e91d4488bb522923a225f0c3 : FAILURE in 24m 25s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/69a5722bc22744f39a9340c56be77a92 : FAILURE in 24m 37s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/1e9173dc8acf412b964021e03df4253f : FAILURE in 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"db39ad275259db179d27d70f3da25d02da18bb54","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-22 09:40:36.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"dcdf7d8010c486a0b2dd1a7802a99a7edca622cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 10:57:48.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6644b66266e4d2e9266baa1cb7b109b : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f90025aea9d4adfb9cb06816133724c : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d98fabfcb9a54337a728844ebe5d869e : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13a875d7bd9443438487d01f025e6d5d : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5cdf88c933640e484a0920a8d32d7b6 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afe9c93f5ee84c509baeec829849bf40 : SUCCESS in 4m 37s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/a18926f09eb04209825225f00841e45e : SUCCESS in 45m 42s\n- patrole-member https://zuul.opendev.org/t/openstack/build/df4bbd59805e4e838e048d6efee4c439 : SUCCESS in 56m 40s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/a9ad0bf5e7644d3a8fe6e9af46e33032 : SUCCESS in 50m 14s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/6e520b9cec6a445a82e05ffa057e107f : SUCCESS in 54m 03s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/a9a3a5c5260f460f8b32167db5fbd50b : SUCCESS in 53m 07s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/91d6ba1049fc4305899ad2dd7dc9b032 : FAILURE in 1h 12m 08s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/c27424e6b2d1454db0d770580cc6c969 : SUCCESS in 1h 01m 34s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/c6c54c2095634449b927e7759aeda1a4 : SUCCESS in 1h 03m 59s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/c98bd16529074f2fa95183c627e95e5f : SUCCESS in 1h 06m 41s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/673487e7868b44f8b95ef23c9e9160a7 : FAILURE in 28m 57s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/f035dee6c6784e398fc0047120373ad3 : FAILURE in 30m 37s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/a29d1a6a3c8c412697da7ffb0ce010ce : FAILURE in 30m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4771378f0beea2fb8031275b85e52af8af21ae7a","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-22 11:58:19.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"51e7670d68c97c6a97f26ee8736dfdccc6341eff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 13:23:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8a7925bbc1e482eaa0cccb279b02883 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f4e37b44c4c408faa43f02280b844ce : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc33738db8f2410490305da94c57dd58 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4b0862271e048529084af87745cc298 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6f203686c044656a1ce589d1bd2881f : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b82ad866ff554c91b69c19c3dc65fcda : SUCCESS in 4m 50s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/5f1e0269d16a4663872b8512c70f79f5 : SUCCESS in 43m 08s\n- patrole-member https://zuul.opendev.org/t/openstack/build/6ba5890b5d79402eafe9aa5eeb2cd15c : SUCCESS in 58m 25s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/57475598e03046139b22bc9cab786cb9 : SUCCESS in 1h 00m 29s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/2bde916bde31446c9071e6ca4aa16593 : SUCCESS in 1h 05m 49s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/27895f05c2e340d69080e76da31a97cd : SUCCESS in 55m 46s\n- patrole-member-train https://zuul.opendev.org/t/openstack/build/ba5de5d28d9e436fab7eab128f94bcec : SUCCESS in 1h 04m 18s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/c57713709954418f91b88b84a06f8786 : SUCCESS in 1h 09m 33s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/df64c299a0124a90ba242a8c0d00509a : SUCCESS in 1h 03m 10s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/d413b6416ecb4dfba5baf0009f3932b8 : SUCCESS in 1h 07m 03s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/fd4e3874c39d4ef9b942607513bab019 : FAILURE in 42m 26s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/fb6d5883cdda43ae8857663625453415 : FAILURE in 33m 02s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/ff48f292171b4297be7446410e8ad7bf : FAILURE in 37m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6a747ea8a068f9555517f9aef44b3a46efb28cbb","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-09-22 14:10:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLGTM, let\u0027s merge while gates are happy","accounts_in_message":[],"_revision_number":4},{"id":"6718cf7ae8334fc4d751cb63c82d291aaf53e49b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 15:06:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b93062e2105eccc800dddad4f8f0840be34ba1fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 16:02:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6f5cd894b5c48da8e0ae143ab069ebe : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a312e07ecf042dbbf3609f1ad4a91b1 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c3be8eeca8e4dee8e0e07e41d77c6f6 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb3b4983a5204dd294226588626a3231 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54de17955d6c479fb1f327e3a96e9bb0 : SUCCESS in 3m 44s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/dc5cd92dd7ea4196b61c68a445c98ae8 : SUCCESS in 40m 03s\n- patrole-member https://zuul.opendev.org/t/openstack/build/b7f677e2e3464a37a705b582a7e012be : SUCCESS in 45m 43s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/69a02e34c1b743fa8d91ea26e2bd0a1d : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":4},{"id":"e66c810e0f2ba19e228ebef64c55a78604b91350","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 16:02:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a9f6fb9afabb3b5e59f9122ea4f778985cb45172","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 17:17:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c366858e5afc4ef69edf1b8e386370f8 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80d0e222c4d842d1ae54e1318414d33d : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"588c33d6d7312c3bea0a9ceaab80ad000b1b928e","revisions":{"1a06ee4bc521ff6a09d43dea8b6cb8a4d661e885":{"kind":"REWORK","_number":1,"created":"2021-09-21 15:12:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/810274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/74/810274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/74/810274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/74/810274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/74/810274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/74/810274/1"}}},"commit":{"parents":[{"commit":"f304d7aaac2eea3f81c0edebf32a7c4f1b32f753","subject":"Add release notes page for version 0.13.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f304d7aaac2eea3f81c0edebf32a7c4f1b32f753"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-09-21 15:09:44.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-09-21 15:09:44.000000000","tz":-300},"subject":"[Fix gate]: Cinder policy change handling in tests","message":"[Fix gate]: Cinder policy change handling in tests\n\nCinder policies are made more granular and now we need\nto adjust the patrole tests to handle those changed policies.\n\nThis commit introduces a new flag so that we test the old\npolicies in stable branches and new one in Xena onwards.\n\nChange-Id: I4be60e3e92704f8e55d3acdb0e025078ae5b21f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/1a06ee4bc521ff6a09d43dea8b6cb8a4d661e885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/1a06ee4bc521ff6a09d43dea8b6cb8a4d661e885"}]},"branch":"refs/heads/master"},"d533d1858dad5a669b878111b1e9df2b4a7b3176":{"kind":"REWORK","_number":2,"created":"2021-09-21 16:52:26.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/810274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/74/810274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/74/810274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/74/810274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/74/810274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/74/810274/2"}}},"commit":{"parents":[{"commit":"f304d7aaac2eea3f81c0edebf32a7c4f1b32f753","subject":"Add release notes page for version 0.13.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f304d7aaac2eea3f81c0edebf32a7c4f1b32f753"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-09-21 15:09:44.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-09-21 16:52:18.000000000","tz":-300},"subject":"[Fix gate]: Cinder policy change handling in tests","message":"[Fix gate]: Cinder policy change handling in tests\n\nCinder policies are made more granular and now we need\nto adjust the patrole tests to handle those changed policies.\n\nThis commit introduces a new flag so that we test the old\npolicies in stable branches and new one in Xena onwards.\n\nChange-Id: I4be60e3e92704f8e55d3acdb0e025078ae5b21f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/d533d1858dad5a669b878111b1e9df2b4a7b3176"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/d533d1858dad5a669b878111b1e9df2b4a7b3176"}]},"branch":"refs/heads/master"},"fd7d0eb3f8a05b38cd5d1136fe2ddca72846fb88":{"kind":"REWORK","_number":3,"created":"2021-09-21 21:28:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/810274/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/74/810274/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/74/810274/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/74/810274/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/74/810274/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/74/810274/3"}}},"commit":{"parents":[{"commit":"f304d7aaac2eea3f81c0edebf32a7c4f1b32f753","subject":"Add release notes page for version 0.13.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f304d7aaac2eea3f81c0edebf32a7c4f1b32f753"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-09-21 15:09:44.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-09-21 21:28:04.000000000","tz":-300},"subject":"[Fix gate]: Cinder policy change handling in tests","message":"[Fix gate]: Cinder policy change handling in tests\n\nCinder policies are made more granular and now we need\nto adjust the patrole tests to handle those changed policies.\n\nThis commit introduces a new flag so that we test the old\npolicies in stable branches and new one in Xena onwards.\n\nChange-Id: I4be60e3e92704f8e55d3acdb0e025078ae5b21f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/fd7d0eb3f8a05b38cd5d1136fe2ddca72846fb88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/fd7d0eb3f8a05b38cd5d1136fe2ddca72846fb88"}]},"branch":"refs/heads/master"},"588c33d6d7312c3bea0a9ceaab80ad000b1b928e":{"kind":"REWORK","_number":4,"created":"2021-09-21 22:38:40.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/810274/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/74/810274/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/74/810274/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/74/810274/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/74/810274/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/74/810274/4"}}},"commit":{"parents":[{"commit":"f304d7aaac2eea3f81c0edebf32a7c4f1b32f753","subject":"Add release notes page for version 0.13.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f304d7aaac2eea3f81c0edebf32a7c4f1b32f753"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-09-21 15:09:44.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-09-21 22:38:37.000000000","tz":0},"subject":"[Fix gate]: Cinder policy change handling in tests","message":"[Fix gate]: Cinder policy change handling in tests\n\nCinder policies are made more granular and now we need\nto adjust the patrole tests to handle those changed policies.\n\nThis commit introduces a new flag so that we test the old\npolicies in stable branches and new one in Xena onwards.\n\nChange-Id: I4be60e3e92704f8e55d3acdb0e025078ae5b21f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/588c33d6d7312c3bea0a9ceaab80ad000b1b928e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/588c33d6d7312c3bea0a9ceaab80ad000b1b928e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
