)]}'
{"id":"openstack%2Fpatrole~874817","triplet_id":"openstack%2Fpatrole~master~Ib310212696e0b08df80eb75f3a1b2d7d37c715fa","project":"openstack/patrole","branch":"master","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-02-24 19:39:46.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"Ib310212696e0b08df80eb75f3a1b2d7d37c715fa","subject":"Fix tox4 error","status":"ABANDONED","created":"2023-02-22 23:58:35.000000000","updated":"2023-04-27 19:19:24.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2fa6c7bfce6ca59581bf873280baf08b744a353b","_number":874817,"virtual_id_number":874817,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-23 01:10:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-23 01:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7e16dbf68f29bcd1d67897d206fb982c261d5a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-22 23:58:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"420531878374c0d27cf0df3a1c89c45dfc4cc528","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 01:10:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fd8b3ca9a604e9b86191f8b86411a7c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dadfecefffc14a6aa446d9650f00b885 : FAILURE in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4da0345b0d475b91e02ba096671695 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c80845540e8e45d2a6bd294a9cffc09b : FAILURE in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/897d2a886ef1481ea038370e94bb8e5b : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b90b3132c5b4f12b689223af3316c86 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bbcd2ad23e64c79b321d05afa931291 : SUCCESS in 2m 44s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/24e6cb15be4745cea2c62f806b23073e : FAILURE in 34m 24s\n- patrole-member https://zuul.opendev.org/t/openstack/build/72bb8be53c08463ba6ce94613856fe24 : FAILURE in 44m 07s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/73060e2b78b64774b8c702eb880aee86 : FAILURE in 39m 21s\n- patrole-member-wallaby https://zuul.opendev.org/t/openstack/build/64d916cbbada4d998d0a4f265786ed3f : FAILURE in 24m 56s\n- patrole-member-victoria https://zuul.opendev.org/t/openstack/build/1bab7930c7134258836903094edb03ef : FAILURE in 46m 06s\n- patrole-member-ussuri https://zuul.opendev.org/t/openstack/build/673c2b1697714b6cb8b1f9bad48aa318 : FAILURE in 1h 01m 56s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/b34ff7d9745b40e4bd712cbe6577fe02 : SUCCESS in 53m 42s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/844d898d705f448d95d54982c0972433 : FAILURE in 56m 53s (non-voting)\n- patrole-multinode-reader https://zuul.opendev.org/t/openstack/build/d6ae239ba19141eb84967ef80a9376a8 : SUCCESS in 54m 06s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/f904777ffb954487aa369f9a1449346e : FAILURE in 25m 52s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/5a5f417291284610b9498f71a9da3fd1 : FAILURE in 29m 13s (non-voting)\n- patrole-extension-reader https://zuul.opendev.org/t/openstack/build/5e55996648354979be758ea6eecc801e : FAILURE in 26m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3083b4f1156043941e1621cc1239d70467c758b6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-24 19:39:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2fa6c7bfce6ca59581bf873280baf08b744a353b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-04-27 19:19:24.000000000","message":"Abandoned\n\nThis project is retired, for more information, refer to the https://github.com/openstack-archive/patrole","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"137bc19a46a01b30796c7da9a1035d187f9a285f","revisions":{"137bc19a46a01b30796c7da9a1035d187f9a285f":{"kind":"REWORK","_number":1,"created":"2023-02-22 23:58:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/874817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/17/874817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/17/874817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/17/874817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/17/874817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/17/874817/1"}}},"commit":{"parents":[{"commit":"aa7bc0dd9a7e79f0ea346838617fe66df0ed0de4","subject":"Merge \"Add cleanup for test_show_auto_allocated_topology test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/aa7bc0dd9a7e79f0ea346838617fe66df0ed0de4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-02-22 23:56:42.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-02-22 23:58:34.000000000","tz":-480},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox \u003e4 due to \u0027skipsdist\u0027 option.\nFixing it to work with tox4.\n\nChange-Id: Ib310212696e0b08df80eb75f3a1b2d7d37c715fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/137bc19a46a01b30796c7da9a1035d187f9a285f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/137bc19a46a01b30796c7da9a1035d187f9a285f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
