)]}'
{"id":"openstack%2Fpaunch~642910","triplet_id":"openstack%2Fpaunch~master~Ia3fe06dd221d0894a5de2c18beaa48ba29440003","project":"openstack/paunch","branch":"master","topic":"Ia3fe06dd221d0894a5de2c18beaa48ba29440003","hashtags":[],"change_id":"Ia3fe06dd221d0894a5de2c18beaa48ba29440003","subject":"Handle defined containers that are stopped.","status":"MERGED","created":"2019-03-12 22:22:15.000000000","updated":"2019-08-27 23:20:53.000000000","submitted":"2019-08-08 12:18:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"642910-1565266723885-62203689","meta_rev_id":"15e0a1d192877171b95ff79442a54a2fde7e815b","_number":642910,"virtual_id_number":642910,"owner":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-31 12:17:57.000000000","_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":2,"date":"2019-08-08 12:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-30 19:11:27.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":2,"date":"2019-07-31 19:46:00.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-08-07 22:28:17.000000000","_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2019-08-07 22:28:17.000000000","_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 19:11:27.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-31 19:46:00.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-08-07 22:28:17.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2019-08-08 12:18:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9809cb5127a8a68ead6a40acd37dcf3ba49acb69","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-03-12 22:22:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"432ab728478bfa172871882099952dacc88443ad","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-03-12 23:46:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a16e42abd33058ed7378f8db210068551847928","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 00:03:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/1/check/tripleo-ci-centos-7-undercloud-containers/d4ec78e/ : FAILURE in 21m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/1/check/tripleo-ci-centos-7-standalone/ff9f646/ : FAILURE in 25m 02s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/1/check/tripleo-ci-fedora-28-standalone/85fcc3f/ : FAILURE in 30m 27s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/10/642910/1/check/openstack-tox-py35/8924a25/ : FAILURE in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/1/check/openstack-tox-py36/2598fb8/ : FAILURE in 5m 11s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/1/check/openstack-tox-py37/b0d101c/ : FAILURE in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/10/642910/1/check/openstack-tox-docs/261163b/html/ : SUCCESS in 6m 12s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/1/check/openstack-tox-pep8/dbe2822/ : FAILURE in 7m 10s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/1/check/openstack-tox-py27/dbb913c/ : FAILURE in 6m 42s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container tripleo-ci-centos-7-scenario008-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone tripleo-ci-centos-7-scenario001-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone tripleo-ci-centos-7-scenario002-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone tripleo-ci-centos-7-scenario003-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone tripleo-ci-centos-7-scenario004-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario012-standalone tripleo-ci-centos-7-scenario012-standalone : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10ac48b3ccba48edf5ceb45141d079c9f4049006","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-13 00:15:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/642910/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ada099c/ : FAILURE in 19m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/642910/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/7af6c0d/ : FAILURE in 19m 57s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/10/642910/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/45446ea/ : FAILURE in 18m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d0f672980294f092a1e2553f16592f4168e6579c","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-03-21 17:00:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"143d0f80a8a580bd40c2c6996d6c348626102caf","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-03-21 17:02:31.000000000","message":"Patch Set 2: Code-Review-1\n\nThanks to Steve\u0027s help, I have fixed the existing idempotency test. I still need to add in a new integration test for this functionality.","accounts_in_message":[],"_revision_number":2},{"id":"d7f190a720500da9dab36d542d5ec39c9d4ceb75","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-21 17:54:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 tripleo-ci-centos-7-multinode-1ctlr-featureset010 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"1abcf9e6d6446a14b9583b66b30b60a1e6bf5fa7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 20:31:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/2/check/tripleo-ci-centos-7-undercloud-containers/13b102a/ : SUCCESS in 1h 27m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/2/check/tripleo-ci-centos-7-standalone/2cabd8f/ : SUCCESS in 1h 33m 32s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/2/check/tripleo-ci-fedora-28-standalone/89ecc3a/ : SUCCESS in 1h 31m 55s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/10/642910/2/check/openstack-tox-py35/e7e8314/ : SUCCESS in 6m 40s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/2/check/openstack-tox-py36/8f394cb/ : SUCCESS in 6m 13s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/2/check/openstack-tox-py37/5646dc9/ : SUCCESS in 5m 51s\n- openstack-tox-docs http://logs.openstack.org/10/642910/2/check/openstack-tox-docs/dec3d23/html/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/2/check/openstack-tox-pep8/d82a1de/ : FAILURE in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/2/check/openstack-tox-py27/1fddde4/ : SUCCESS in 5m 30s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container tripleo-ci-centos-7-scenario008-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone tripleo-ci-centos-7-scenario001-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone tripleo-ci-centos-7-scenario002-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone tripleo-ci-centos-7-scenario003-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone tripleo-ci-centos-7-scenario004-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario012-standalone tripleo-ci-centos-7-scenario012-standalone : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca4bbdd5ee0cc13bb5ba1dbed41b82076c212093","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-04-10 23:58:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c51a8ba47fb18da0cb2b7d1f8c3e573bf2c1935e","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-04-11 00:05:01.000000000","message":"Patch Set 3: Code-Review-1\n\nI added the related test for my function. It is only half way done. I\u0027m having difficulty figuring out the next method to Mock. Any extra eyes on this would be greatly appreciated.","accounts_in_message":[],"_revision_number":3},{"id":"673f8254b2b79be4dfe793eeca2697e7f95f4692","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 00:42:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/3/check/tripleo-ci-centos-7-undercloud-containers/b9f5cdf/ : FAILURE in 25m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/3/check/tripleo-ci-centos-7-standalone/4cd2971/ : FAILURE in 21m 48s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/3/check/tripleo-ci-fedora-28-standalone/d2010cb/ : FAILURE in 20m 41s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/10/642910/3/check/openstack-tox-py35/62bfe2a/ : FAILURE in 6m 33s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/3/check/openstack-tox-py36/43a67a3/ : FAILURE in 4m 45s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/3/check/openstack-tox-py37/3a76362/ : FAILURE in 7m 06s\n- openstack-tox-docs http://logs.openstack.org/10/642910/3/check/openstack-tox-docs/3dfbcab/html/ : SUCCESS in 5m 57s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/3/check/openstack-tox-pep8/c0cad47/ : FAILURE in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/3/check/openstack-tox-py27/bc18c56/ : FAILURE in 6m 10s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container tripleo-ci-centos-7-scenario008-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone tripleo-ci-centos-7-scenario001-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone tripleo-ci-centos-7-scenario002-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone tripleo-ci-centos-7-scenario003-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone tripleo-ci-centos-7-scenario004-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario012-standalone tripleo-ci-centos-7-scenario012-standalone : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e93d20b11ddcaba1cddb71c32ea568ef550fe43","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-11 00:45:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/642910/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2444c03/ : FAILURE in 23m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/642910/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/fee3a81/ : FAILURE in 25m 38s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/10/642910/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/6002129/ : FAILURE in 17m 14s","accounts_in_message":[],"_revision_number":3},{"id":"802c25a7b7392d7d1c9382949e7806c9062b9f5e","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-04-29 20:08:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fbe71787a43b09547cb435762cc29ee87f705fd1","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-04-29 20:10:01.000000000","message":"Patch Set 4:\n\nThe test is still broken but is almost to completion. We can see from running ` python -m testtools.run paunch.tests.test_builder_podman.TestPodmanBuilder.test_recreate_stopped_container` that it correctly detects that we need to delete a stopped container. However, I am having difficulties Mocking the \"remove_container\" method.","accounts_in_message":[],"_revision_number":4},{"id":"09b018168592764a39b848da01fff65638dc23a9","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-04-29 20:10:24.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"98478615139cbbe5874015d79c8214eeb1dd1c4a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-29 20:29:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/642910/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/fdc9688/ : FAILURE in 19m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/642910/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/825ee2f/ : FAILURE in 18m 09s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/10/642910/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/aef6301/ : FAILURE in 14m 03s","accounts_in_message":[],"_revision_number":4},{"id":"792dfc758ea659168778b638a02d022013ad2c1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 20:57:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/4/check/tripleo-ci-centos-7-undercloud-containers/dc01843/ : FAILURE in 22m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/4/check/tripleo-ci-centos-7-standalone/5a04bf8/ : FAILURE in 21m 30s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/4/check/tripleo-ci-fedora-28-standalone/58e28e2/ : FAILURE in 26m 14s\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/10/642910/4/check/openstack-tox-py35/190304f/ : FAILURE in 6m 21s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/4/check/openstack-tox-py36/1ebeefb/ : FAILURE in 5m 57s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/4/check/openstack-tox-py37/b53a958/ : FAILURE in 6m 34s\n- openstack-tox-docs http://logs.openstack.org/10/642910/4/check/openstack-tox-docs/c500909/html/ : SUCCESS in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/4/check/openstack-tox-pep8/a07d5ab/ : FAILURE in 5m 44s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/4/check/openstack-tox-py27/bcb4394/ : FAILURE in 5m 56s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container tripleo-ci-centos-7-scenario008-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone tripleo-ci-centos-7-scenario001-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone tripleo-ci-centos-7-scenario002-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone tripleo-ci-centos-7-scenario003-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone tripleo-ci-centos-7-scenario004-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario012-standalone tripleo-ci-centos-7-scenario012-standalone : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8708b9f7cad31b0793f68864ae240038bad4fe1d","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-06-27 20:30:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"37ad244b7d0eed77392f061e3e35ed62091c7dc8","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-06-27 20:33:30.000000000","message":"Patch Set 5:\n\nLocal tox tests are passing locally now thanks to help from Alex.\n\nI still need to rebase, possibly refactor some mocks and assertations, and remove the work-in-progress title.","accounts_in_message":[],"_revision_number":5},{"id":"49e79bf0fa39e28e2ae6b0eb57126b967a381979","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-06-27 22:27:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ea09c388c16d74ad4da40d005c3d16623bca453e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-28 01:27:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/642910/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b97a576/ : SUCCESS in 2h 59m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/642910/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/89c2d3a/ : SUCCESS in 2h 58m 49s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/10/642910/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/71f1d49/ : SUCCESS in 1h 17m 42s","accounts_in_message":[],"_revision_number":6},{"id":"5f8e12ed3b7181dd9aa98d50fa17a315845f76d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 04:01:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/6/check/tripleo-ci-centos-7-undercloud-containers/4f7a852/ : SUCCESS in 1h 53m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/6/check/tripleo-ci-centos-7-standalone/898eeeb/ : SUCCESS in 2h 27m 34s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/6/check/tripleo-ci-fedora-28-standalone/397b7e8/ : FAILURE in 25m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/10/642910/6/check/tripleo-ci-centos-7-standalone-upgrade-stein/f67b5ce/ : SUCCESS in 2h 28m 37s\n- openstack-tox-py35 http://logs.openstack.org/10/642910/6/check/openstack-tox-py35/22b966a/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/6/check/openstack-tox-py36/b776518/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/6/check/openstack-tox-py37/13b21c4/ : SUCCESS in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/10/642910/6/check/openstack-tox-docs/6ec882c/html/ : SUCCESS in 2m 45s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/6/check/openstack-tox-pep8/71f05aa/ : FAILURE in 2m 35s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/6/check/openstack-tox-py27/717af0b/ : SUCCESS in 3m 36s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone tripleo-ci-centos-7-scenario001-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone tripleo-ci-centos-7-scenario002-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone tripleo-ci-centos-7-scenario003-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone tripleo-ci-centos-7-scenario004-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario012-standalone tripleo-ci-centos-7-scenario012-standalone : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d369b87b951b5c0064f3ca4865e0a8c3584653f3","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-06-28 13:02:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"db24daf054943cd76bcce994634aee4e3f3e2ac8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 16:03:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-containers-multinode/19c2244/ : SUCCESS in 2h 24m 18s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-undercloud-containers/d37d0b0/ : SUCCESS in 1h 33m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3cafef3/ : SUCCESS in 2h 28m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/77ce214/ : SUCCESS in 2h 25m 54s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-standalone/0b7f59c/ : SUCCESS in 2h 08m 57s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-fedora-28-standalone/445bd4d/ : SUCCESS in 1h 35m 53s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-standalone-upgrade/7cb0324/ : SUCCESS in 2h 00m 56s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/156e5e2/ : SUCCESS in 2h 23m 02s\n- openstack-tox-py35 http://logs.openstack.org/10/642910/7/check/openstack-tox-py35/d9a61f9/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/7/check/openstack-tox-py36/4372947/ : SUCCESS in 3m 24s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/7/check/openstack-tox-py37/7a29b0d/ : SUCCESS in 3m 27s\n- openstack-tox-docs http://logs.openstack.org/10/642910/7/check/openstack-tox-docs/2d1222a/html/ : SUCCESS in 3m 02s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/7/check/openstack-tox-pep8/f12ea00/ : SUCCESS in 2m 28s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/7/check/openstack-tox-py27/dfb05f8/ : SUCCESS in 3m 03s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/5beddfd/ : SUCCESS in 2h 19m 05s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/6eb2558/ : FAILURE in 2h 41m 04s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario001-standalone/760971e/ : SUCCESS in 2h 10m 34s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario002-standalone/c2584c7/ : SUCCESS in 2h 07m 19s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario003-standalone/95f941a/ : SUCCESS in 1h 25m 50s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario004-standalone/98fbbef/ : SUCCESS in 2h 06m 01s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario012-standalone/cf5adf8/ : FAILURE in 25m 25s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5f9de42/ : FAILURE in 1h 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"71fd15d3aeed627763701873d25284928be46647","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-28 16:39:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/642910/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/83be2ac/ : FAILURE in 2h 46m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze02.softwarefactory-project.io:7979/7829231753804a928dae20748daec193 : POST_FAILURE in 3h 35m 53s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 finger://ze07.softwarefactory-project.io:7979/35f60ca753a64af7b30a48b5f97e11a4 : POST_FAILURE in 2h 23m 36s","accounts_in_message":[],"_revision_number":7},{"id":"2569588488ac1123a9c719d1a97a354dd2d65694","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-06-28 18:00:22.000000000","message":"Patch Set 7:\n\nThe CI errors look to be false-positives. Examples:\nhttp://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/6eb2558/logs/undercloud/home/zuul/overcloud_deploy.log.txt.gz \u003d Network configuration issue.\nhttp://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5f9de42/logs/undercloud/home/zuul/undercloud_upgrade.log.txt.gz \u003d I keep seeing this error in CI. A file cannot be found for an unrelated Ansible task.","accounts_in_message":[],"_revision_number":7},{"id":"61253473d00b5a86b1854ca7efea822556b58763","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-01 14:15:07.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"b35763ce9a6fbb9eb1a08c03262da565577b46d7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 17:50:22.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze06.softwarefactory-project.io:7979/63f9f6a844764a2a9fb077639efd2d00 : POST_FAILURE in 3h 34m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze05.softwarefactory-project.io:7979/ed63a33fb4304c9ab7abd624f399a6d9 : POST_FAILURE in 3h 30m 01s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/10/642910/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/d539d18/ : SUCCESS in 2h 14m 49s","accounts_in_message":[],"_revision_number":7},{"id":"b6addd773722d64a32f1a3ff48b742a1fe718514","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 20:04:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-containers-multinode/a6a5f1d/ : SUCCESS in 2h 05m 56s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-undercloud-containers/f720fc3/ : SUCCESS in 1h 50m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/cbec007/ : SUCCESS in 2h 31m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/67b18cc/ : SUCCESS in 2h 30m 54s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-standalone/0063fa1/ : SUCCESS in 2h 03m 49s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-fedora-28-standalone/85237ec/ : SUCCESS in 1h 48m 37s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-standalone-upgrade/39f9aaf/ : SUCCESS in 2h 28m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/670bcae/ : SUCCESS in 1h 47m 15s\n- openstack-tox-py35 http://logs.openstack.org/10/642910/7/check/openstack-tox-py35/b38bd03/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/7/check/openstack-tox-py36/754955c/ : SUCCESS in 3m 38s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/7/check/openstack-tox-py37/bcb71fe/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/10/642910/7/check/openstack-tox-docs/8e8fd68/html/ : SUCCESS in 2m 50s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/7/check/openstack-tox-pep8/8b4b6fa/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/7/check/openstack-tox-py27/a0efd22/ : SUCCESS in 3m 30s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/907a46b/ : SUCCESS in 2h 28m 58s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/8a3fd71/ : SUCCESS in 2h 13m 51s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario001-standalone/ff7274b/ : SUCCESS in 1h 48m 33s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario002-standalone/4c9b85b/ : SUCCESS in 1h 41m 07s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario003-standalone/676b5aa/ : SUCCESS in 1h 22m 22s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario004-standalone/7a90efa/ : SUCCESS in 2h 10m 27s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario012-standalone/a155477/ : FAILURE in 31m 32s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/4f19113/ : FAILURE in 1h 47m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"45304a4832f186bddab85c0598f6f28f0767f50f","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-03 14:51:05.000000000","message":"Patch Set 7:\n\nCI failures are false-positives. See http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-scenario012-standalone/a155477/logs/undercloud/home/zuul/standalone_deploy.log.txt.gz#_2019-07-01_17_38_55 and http://logs.openstack.org/10/642910/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/4f19113/logs/undercloud/home/zuul/undercloud_upgrade.log.txt.gz#_2019-07-01_18_29_03","accounts_in_message":[],"_revision_number":7},{"id":"6ecd7ff561dee59be63181f59546846f7ab09493","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-07-03 21:01:15.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)\n\nLooks good, just needs some cleanup of the debug statements","accounts_in_message":[],"_revision_number":7},{"id":"06179ea46848ab5ac5fb68322bce3342f61b7cbf","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-05 15:33:29.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"060707bf05478c63cdbb50d3b8003d6166347627","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-05 16:16:06.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ed344f81dbf461f3e63194b48a1f913809d72c08","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-05 19:04:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/642910/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6c88059/ : SUCCESS in 2h 32m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/642910/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/acb7e61/ : SUCCESS in 2h 45m 54s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/10/642910/8/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/39c29fd/ : SUCCESS in 1h 18m 16s","accounts_in_message":[],"_revision_number":8},{"id":"ec46a45a769e8101999737992d42ece61c9954a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 19:34:58.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-containers-multinode/8d89ad0/ : SUCCESS in 2h 19m 14s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-undercloud-containers/b31e42c/ : SUCCESS in 2h 08m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e2a7a53/ : SUCCESS in 2h 07m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/df01c43/ : SUCCESS in 2h 45m 32s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-standalone/bc7b531/ : SUCCESS in 2h 05m 14s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/8/check/tripleo-ci-fedora-28-standalone/6f06f42/ : SUCCESS in 1h 56m 28s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-standalone-upgrade/abdff7b/ : SUCCESS in 2h 11m 31s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-standalone-upgrade-stein/255bf7b/ : FAILURE in 1h 21m 52s\n- openstack-tox-py35 http://logs.openstack.org/10/642910/8/check/openstack-tox-py35/483ad90/ : SUCCESS in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/8/check/openstack-tox-py36/4ca0932/ : SUCCESS in 3m 30s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/8/check/openstack-tox-py37/741bb68/ : SUCCESS in 3m 59s\n- openstack-tox-docs http://logs.openstack.org/10/642910/8/check/openstack-tox-docs/d41e4bd/html/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/8/check/openstack-tox-pep8/ad4576a/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/8/check/openstack-tox-py27/57b59c5/ : SUCCESS in 3m 38s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/be84f91/ : SUCCESS in 2h 35m 35s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/2bb4405/ : SUCCESS in 2h 37m 56s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-scenario001-standalone/acc4970/ : SUCCESS in 1h 40m 35s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-scenario002-standalone/580d3ac/ : SUCCESS in 1h 46m 23s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-scenario003-standalone/514d870/ : SUCCESS in 1h 25m 24s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-scenario004-standalone/a960225/ : SUCCESS in 1h 52m 14s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-scenario012-standalone/1b8b69a/ : SUCCESS in 2h 00m 54s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/e419cbf/ : FAILURE in 2h 06m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"78fd943ccef09f7e6cf0d2aeaf97229c5e9daa14","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-05 21:13:27.000000000","message":"Patch Set 7:\n\nhttp://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-standalone-upgrade-stein/255bf7b/logs/undercloud/home/zuul/standalone_upgrade.log.txt.gz#_2019-07-05_17_49_47 \u003d This is not relevant. It\u0027s an unrelated registry error.\nhttp://logs.openstack.org/10/642910/8/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/e419cbf/job-output.txt.gz#_2019-07-05_18_45_00_405642 \u003d The Swift containers are reported as failed. I\u0027m leaning towards this not being an issue with this patch since all other (and a few similar) CI tests are passing.","accounts_in_message":[],"_revision_number":7},{"id":"fc6e1991f126494877d7ea45b37f68f8562bf104","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-10 12:14:57.000000000","message":"Patch Set 8:\n\nI\u0027ve been seeing the Swift errors in other patch sets on Gerrit so none of the CI issues appear to be related.","accounts_in_message":[],"_revision_number":8},{"id":"0b2c4d4fe17175f317687129ea2a6abd236326b0","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-07-12 15:09:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4cfad3cff152fc4a4ab3d99e1e6f00d473ebbc55","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-07-15 21:47:22.000000000","message":"Patch Set 8: Code-Review-1\n\nwe likely need to create the container if inspect_container is None.","accounts_in_message":[],"_revision_number":8},{"id":"05c39eeb2720f3d96e79a262288694600fc71dd9","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-16 19:59:47.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a88cde824501b79e3b5a6734a2978c364ab1815f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-16 20:18:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/10/642910/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/9f0343f/ : FAILURE in 17m 16s","accounts_in_message":[],"_revision_number":9},{"id":"285ab9a2ddcdef7376aa0479e597eb5580a3aea4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 23:17:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/9/check/tripleo-ci-centos-7-undercloud-containers/b0a5c49/ : FAILURE in 25m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/9/check/tripleo-ci-centos-7-standalone/f5bd2ad/ : FAILURE in 22m 21s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/9/check/tripleo-ci-fedora-28-standalone/90bbf09/ : FAILURE in 26m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/10/642910/9/check/openstack-tox-py35/16b20a5/ : FAILURE in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/9/check/openstack-tox-py36/4043073/ : FAILURE in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/9/check/openstack-tox-py37/7b3cec1/ : FAILURE in 4m 22s\n- openstack-tox-docs http://logs.openstack.org/10/642910/9/check/openstack-tox-docs/1206541/html/ : SUCCESS in 2m 47s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/9/check/openstack-tox-pep8/de2a683/ : FAILURE in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/9/check/openstack-tox-py27/6d651e6/ : FAILURE in 4m 05s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone tripleo-ci-centos-7-scenario001-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone tripleo-ci-centos-7-scenario002-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone tripleo-ci-centos-7-scenario003-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone tripleo-ci-centos-7-scenario004-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario012-standalone tripleo-ci-centos-7-scenario012-standalone : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"391a438800d563177d8234e7e3049612b5f1f361","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-17 15:58:35.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4f7e4fe885e3a171068cb7c3efe0f75a96b663e6","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-17 16:02:09.000000000","message":"Patch Set 10:\n\n(2 comments)\n\nAll of the previous concerns should be addressed with the code now being simplified. I also tweaked the idempotency test to work as I expect it to.","accounts_in_message":[],"_revision_number":10},{"id":"04fa5b8f0104df13dd9addda2f08a410d155510f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:16:25.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/10/check/tripleo-ci-centos-7-undercloud-containers/e11cb62/ : SUCCESS in 1h 35m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/10/check/tripleo-ci-centos-7-standalone/5b1e63b/ : FAILURE in 1h 10m 20s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/10/check/tripleo-ci-fedora-28-standalone/e3d6efd/ : SUCCESS in 1h 50m 42s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/10/642910/10/check/openstack-tox-py35/26aec08/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/10/check/openstack-tox-py36/c35b9e4/ : SUCCESS in 3m 09s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/10/check/openstack-tox-py37/100567d/ : SUCCESS in 3m 37s\n- openstack-tox-docs http://logs.openstack.org/10/642910/10/check/openstack-tox-docs/b313aad/html/ : SUCCESS in 2m 41s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/10/check/openstack-tox-pep8/0624477/ : FAILURE in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/10/check/openstack-tox-py27/e1f2425/ : SUCCESS in 3m 31s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone tripleo-ci-centos-7-scenario001-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone tripleo-ci-centos-7-scenario002-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone tripleo-ci-centos-7-scenario003-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone tripleo-ci-centos-7-scenario004-standalone : SKIPPED\n- tripleo-ci-centos-7-scenario012-standalone tripleo-ci-centos-7-scenario012-standalone : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2f7e579c6dbadb1b4478ff90dcfe8091c55689b2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 18:30:31.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/10/642910/10/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/023ebf0/ : SUCCESS in 2h 29m 57s","accounts_in_message":[],"_revision_number":10},{"id":"619950fe98ecb9abeb76230750c34b286f06e7fc","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-17 18:43:45.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"4d92d7464a28a392d5c7efba39522c69c4cdd0b5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 21:03:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/10/642910/11/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/0ee78bc/ : SUCCESS in 2h 18m 42s","accounts_in_message":[],"_revision_number":11},{"id":"b14c3525cc7951388f7056cbc74578c03887bead","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 21:29:24.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-containers-multinode/1afbadc/ : FAILURE in 2h 04m 45s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-undercloud-containers/d42b717/ : SUCCESS in 1h 30m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2c59c8c/ : SUCCESS in 2h 31m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f84c440/ : FAILURE in 2h 28m 09s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-standalone/cf3ecc0/ : FAILURE in 1h 08m 21s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-fedora-28-standalone/13719f5/ : SUCCESS in 1h 52m 11s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-standalone-upgrade/6cab914/ : FAILURE in 2h 03m 00s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/10/642910/11/check/openstack-tox-py35/744449e/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/11/check/openstack-tox-py36/c617635/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/11/check/openstack-tox-py37/15e8bfe/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/10/642910/11/check/openstack-tox-docs/8f9f332/html/ : SUCCESS in 3m 05s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/11/check/openstack-tox-pep8/d282cf1/ : SUCCESS in 3m 48s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/11/check/openstack-tox-py27/3cc7dc9/ : SUCCESS in 3m 57s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/91a1bf1/ : SUCCESS in 2h 23m 48s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/fdd7855/ : FAILURE in 1h 54m 53s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario001-standalone/2f19adf/ : FAILURE in 1h 30m 48s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario002-standalone/c827fa7/ : FAILURE in 1h 34m 59s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario003-standalone/b3c8cb0/ : FAILURE in 1h 15m 57s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario004-standalone/01f56d0/ : FAILURE in 1h 30m 12s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario012-standalone/edcdf57/ : SUCCESS in 1h 50m 03s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/4ef63c7/ : FAILURE in 2h 02m 37s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4bb8698c0028becc7ca2db6a2f067763d818e343","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-18 15:28:27.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"0e609d6c8061566eaeba5a0989d5d8637ab29b29","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 17:33:50.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/10/642910/11/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/f9792dd/ : FAILURE in 1h 46m 08s","accounts_in_message":[],"_revision_number":11},{"id":"850d3380a41ee66bcd2b022773f58a8ba808a938","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 20:20:58.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-containers-multinode/75c08c3/ : SUCCESS in 2h 17m 33s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-undercloud-containers/9d80646/ : SUCCESS in 1h 40m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a00b494/ : SUCCESS in 2h 28m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/89079ba/ : FAILURE in 2h 34m 49s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-standalone/0d24e49/ : SUCCESS in 1h 59m 19s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-fedora-28-standalone/53d0b9a/ : FAILURE in 27m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-standalone-upgrade/287df33/ : FAILURE in 2h 22m 39s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/10/642910/11/check/openstack-tox-py35/89ffdd3/ : SUCCESS in 3m 51s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/11/check/openstack-tox-py36/4009e53/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/11/check/openstack-tox-py37/092460a/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/10/642910/11/check/openstack-tox-docs/f43d5b7/html/ : SUCCESS in 3m 00s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/11/check/openstack-tox-pep8/154195f/ : SUCCESS in 2m 26s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/11/check/openstack-tox-py27/31967ba/ : SUCCESS in 3m 47s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/865281f/ : SUCCESS in 2h 31m 15s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/5a72bdd/ : FAILURE in 1h 04m 32s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario001-standalone/e79d4c3/ : SUCCESS in 2h 00m 17s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario002-standalone/a75801c/ : FAILURE in 1h 01m 44s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario003-standalone/a5cd6e7/ : FAILURE in 50m 58s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario004-standalone/e9b36ac/ : FAILURE in 1h 03m 55s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario012-standalone/5b7f9f7/ : SUCCESS in 1h 53m 54s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/e1952fa/ : FAILURE in 2h 27m 47s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"32ec2a5226a2124dfd1bf07ee0cd1b3609d125c3","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-22 14:14:27.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"0faa6915d98ba5554851637a8cf9624640eb1905","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 15:48:09.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/10/642910/11/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/a0c2963/ : FAILURE in 1h 31m 40s","accounts_in_message":[],"_revision_number":11},{"id":"91f708c8610a67f61431945985eede2e91e6a619","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 18:39:56.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-containers-multinode/e88880e/ : SUCCESS in 2h 23m 43s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-undercloud-containers/af1f137/ : SUCCESS in 1h 38m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3fe3052/ : SUCCESS in 2h 34m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c1fb74e/ : SUCCESS in 2h 36m 55s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-standalone/8870cbc/ : SUCCESS in 1h 53m 06s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-fedora-28-standalone/ef2d4ac/ : SUCCESS in 1h 41m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-standalone-upgrade/67ddc3a/ : FAILURE in 1h 53m 05s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/10/642910/11/check/openstack-tox-py35/c6cf3da/ : SUCCESS in 3m 40s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/11/check/openstack-tox-py36/d53b1b0/ : SUCCESS in 4m 58s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/11/check/openstack-tox-py37/0539b8d/ : SUCCESS in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/10/642910/11/check/openstack-tox-docs/2a2b462/html/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/11/check/openstack-tox-pep8/5ca8195/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/11/check/openstack-tox-py27/b3deb55/ : SUCCESS in 3m 56s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/a724c0c/ : SUCCESS in 2h 23m 29s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/e12da34/ : FAILURE in 1h 31m 25s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario001-standalone/b79a231/ : FAILURE in 1h 10m 52s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario002-standalone/e392b57/ : FAILURE in 1h 02m 03s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario003-standalone/6fceb51/ : SUCCESS in 1h 17m 51s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario004-standalone/68d0a67/ : SUCCESS in 1h 50m 42s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-scenario012-standalone/332ed69/ : SUCCESS in 1h 42m 33s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/10/642910/11/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/4f6575e/ : FAILURE in 1h 52m 17s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b81ac629881d3bedc8f57dd0c53023bc0e5a8e1f","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-30 16:41:01.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"05d592364c3c744b3583d21a1440ec9d15b1f3f9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-30 19:11:27.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/10/642910/12/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/dbcab23/ : SUCCESS in 2h 28m 14s","accounts_in_message":[],"_revision_number":12},{"id":"108d12a3ce956ab8629f64fe391761d9702b7b9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 00:32:51.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-containers-multinode/03df08f/ : SUCCESS in 2h 39m 09s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-undercloud-containers/902518d/ : SUCCESS in 1h 32m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1cdc7c7/ : SUCCESS in 2h 14m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/2c15f48/ : SUCCESS in 2h 23m 29s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-standalone/603f582/ : SUCCESS in 1h 36m 26s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/10/642910/12/check/tripleo-ci-fedora-28-standalone/93b3daa/ : FAILURE in 2h 00m 12s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-standalone-upgrade/fe6af49/ : SUCCESS in 2h 04m 46s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/10/642910/12/check/openstack-tox-py35/b537bdd/ : SUCCESS in 4m 09s\n- openstack-tox-py36 http://logs.openstack.org/10/642910/12/check/openstack-tox-py36/3c58c1a/ : SUCCESS in 4m 30s\n- openstack-tox-py37 http://logs.openstack.org/10/642910/12/check/openstack-tox-py37/4004360/ : SUCCESS in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/10/642910/12/check/openstack-tox-docs/8c52eb4/html/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/10/642910/12/check/openstack-tox-pep8/3205150/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/10/642910/12/check/openstack-tox-py27/dcf104c/ : SUCCESS in 4m 20s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/ba4702f/ : SUCCESS in 2h 38m 01s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/fcc4a9c/ : FAILURE in 1h 20m 54s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-scenario001-standalone/66cd730/ : SUCCESS in 1h 42m 12s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-scenario002-standalone/3faf43d/ : SUCCESS in 1h 36m 36s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-scenario003-standalone/d1189e4/ : SUCCESS in 1h 16m 09s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-scenario004-standalone/a4587b0/ : SUCCESS in 1h 34m 39s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-scenario012-standalone/50afc38/ : SUCCESS in 1h 43m 10s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ab1937e/ : FAILURE in 2h 01m 47s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"64834fef18fcfce6c0ff0408a43539315cd6c53d","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-31 12:17:57.000000000","message":"Patch Set 12:\n\nThe 3 non-voting errors are false-positives (not related to this patch set): (1) http://logs.openstack.org/10/642910/12/check/tripleo-ci-fedora-28-standalone/93b3daa/logs/undercloud/home/zuul/tempest.log.txt.gz#_2019-07-30_23_27_41 (2) http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/fcc4a9c/logs/undercloud/home/zuul/overcloud_deploy.log.txt.gz#_2019-07-30_23_11_27 (3) http://logs.openstack.org/10/642910/12/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ab1937e/job-output.txt.gz#_2019-07-31_00_00_47_904750","accounts_in_message":[],"_revision_number":12},{"id":"8c629f09bf8b4a884aea2f2c3393b683e50e8373","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-07-31 19:46:00.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"e4129e65a0020071d8d70bf664c0a63f43f760ed","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-07 22:28:17.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\nNice work, thanks","accounts_in_message":[],"_revision_number":12},{"id":"b87ce020011c6ddf573dcc404e18a560862d0981","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 07:01:37.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"99d88212a05968f8f63b66737f1ad80e53e9473b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 12:18:43.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/10/642910/12/gate/tripleo-ci-centos-7-containers-multinode/e442fc4/ : SUCCESS in 2h 50m 36s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/10/642910/12/gate/tripleo-ci-centos-7-standalone/44116ba/ : SUCCESS in 2h 00m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/10/642910/12/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1b359fc/ : SUCCESS in 2h 37m 17s\n- openstack-tox-py35 https://logs.opendev.org/10/642910/12/gate/openstack-tox-py35/da311dc/ : SUCCESS in 4m 06s\n- openstack-tox-py36 https://logs.opendev.org/10/642910/12/gate/openstack-tox-py36/de049b8/ : SUCCESS in 4m 31s\n- openstack-tox-py37 https://logs.opendev.org/10/642910/12/gate/openstack-tox-py37/59ecc96/ : SUCCESS in 6m 21s\n- openstack-tox-docs https://logs.opendev.org/10/642910/12/gate/openstack-tox-docs/0a824b8/html/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://logs.opendev.org/10/642910/12/gate/openstack-tox-pep8/78b6777/ : SUCCESS in 5m 53s\n- openstack-tox-py27 https://logs.opendev.org/10/642910/12/gate/openstack-tox-py27/e600bc4/ : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":12},{"id":"1500cba6a692c58c4cec17463fecdae3241b5c57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 12:18:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"57902423e60f592abfb1bd684a47beffd9e9e18f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-27 14:12:02.000000000","message":"Created a revert of this change as I8a27b92498d832698ba4d55bed1b2adcf2c1e3eb","accounts_in_message":[],"_revision_number":12},{"id":"f002e207f8d6f255cef988077399fa1218315333","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-27 14:28:32.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"fec22810c10d3fa8f5f82824923038e6718b6ae0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-27 23:20:53.000000000","message":"Created a revert of this change as I22d3eb35dced8178fe6beb0c76c855746efdf94d","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"85fb2ed4d3f7e42840842f2bdec617e33d6ba4ef","revisions":{"fc67de1bf614c88a94a0993ffc70b3344a592a51":{"kind":"REWORK","_number":1,"created":"2019-03-12 22:22:15.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/10/642910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/10/642910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/10/642910/1"}}},"commit":{"parents":[{"commit":"8c478a46003e8fae67ae97f8e61dd51d99ef56fe","subject":"Merge \"Keep the old paunch behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/8c478a46003e8fae67ae97f8e61dd51d99ef56fe"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-03-12 22:21:47.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-03-12 22:22:15.000000000","tz":-240},"subject":"Handle defined containers that are stopped.","message":"Handle defined containers that are stopped.\n\nDelete the stopped container and recreate it to ensure\nthere are no local modifications to the container.\n\nChange-Id: Ia3fe06dd221d0894a5de2c18beaa48ba29440003\nResolves: rhbz#1674517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/fc67de1bf614c88a94a0993ffc70b3344a592a51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/fc67de1bf614c88a94a0993ffc70b3344a592a51"}]},"branch":"refs/heads/master"},"5e95d88732dcacc1459396659673ababb5454520":{"kind":"REWORK","_number":2,"created":"2019-03-21 17:00:58.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/10/642910/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/10/642910/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/10/642910/2"}}},"commit":{"parents":[{"commit":"8c478a46003e8fae67ae97f8e61dd51d99ef56fe","subject":"Merge \"Keep the old paunch behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/8c478a46003e8fae67ae97f8e61dd51d99ef56fe"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-03-12 22:21:47.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-03-21 17:00:52.000000000","tz":-240},"subject":"WIP: Handle defined containers that are stopped.","message":"WIP: Handle defined containers that are stopped.\n\nDelete the stopped container and recreate it to ensure\nthere are no local modifications to the container.\n\nChange-Id: Ia3fe06dd221d0894a5de2c18beaa48ba29440003\nResolves: rhbz#1674517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5e95d88732dcacc1459396659673ababb5454520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5e95d88732dcacc1459396659673ababb5454520"}]},"branch":"refs/heads/master"},"eb82d4350b8794d9d06e8f2f8833ad17d253b130":{"kind":"REWORK","_number":3,"created":"2019-04-10 23:58:28.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/10/642910/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/10/642910/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/10/642910/3"}}},"commit":{"parents":[{"commit":"8c478a46003e8fae67ae97f8e61dd51d99ef56fe","subject":"Merge \"Keep the old paunch behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/8c478a46003e8fae67ae97f8e61dd51d99ef56fe"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-03-12 22:21:47.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-04-10 23:56:25.000000000","tz":-240},"subject":"WIP: Handle defined containers that are stopped.","message":"WIP: Handle defined containers that are stopped.\n\nDelete the stopped container and recreate it to ensure\nthere are no local modifications to the container.\n\nChange-Id: Ia3fe06dd221d0894a5de2c18beaa48ba29440003\nResolves: rhbz#1674517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/eb82d4350b8794d9d06e8f2f8833ad17d253b130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/eb82d4350b8794d9d06e8f2f8833ad17d253b130"}]},"branch":"refs/heads/master"},"34ce70229cd5e0ebf92cc66fe15e1495ea885a3c":{"kind":"REWORK","_number":4,"created":"2019-04-29 20:08:22.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/10/642910/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/10/642910/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/10/642910/4"}}},"commit":{"parents":[{"commit":"8c478a46003e8fae67ae97f8e61dd51d99ef56fe","subject":"Merge \"Keep the old paunch behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/8c478a46003e8fae67ae97f8e61dd51d99ef56fe"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-03-12 22:21:47.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-04-29 20:07:58.000000000","tz":-240},"subject":"WIP: Handle defined containers that are stopped.","message":"WIP: Handle defined containers that are stopped.\n\nDelete the stopped container and recreate it to ensure\nthere are no local modifications to the container.\n\nChange-Id: Ia3fe06dd221d0894a5de2c18beaa48ba29440003\nResolves: rhbz#1674517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/34ce70229cd5e0ebf92cc66fe15e1495ea885a3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/34ce70229cd5e0ebf92cc66fe15e1495ea885a3c"}]},"branch":"refs/heads/master"},"7a7ab4e2b29d34a143906f23029840f901821166":{"kind":"REWORK","_number":5,"created":"2019-06-27 20:30:38.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/10/642910/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/10/642910/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/10/642910/5"}}},"commit":{"parents":[{"commit":"8c478a46003e8fae67ae97f8e61dd51d99ef56fe","subject":"Merge \"Keep the old paunch behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/8c478a46003e8fae67ae97f8e61dd51d99ef56fe"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-03-12 22:21:47.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-06-27 20:29:50.000000000","tz":-240},"subject":"WIP: Handle defined containers that are stopped.","message":"WIP: Handle defined containers that are stopped.\n\nDelete the stopped container and recreate it to ensure\nthere are no local modifications to the container.\n\nChange-Id: Ia3fe06dd221d0894a5de2c18beaa48ba29440003\nResolves: rhbz#1674517\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/7a7ab4e2b29d34a143906f23029840f901821166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/7a7ab4e2b29d34a143906f23029840f901821166"}]},"branch":"refs/heads/master"},"21c342fff98f4698276d88f8c509e57ac4a3cbf8":{"kind":"REWORK","_number":6,"created":"2019-06-27 22:27:05.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/10/642910/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/10/642910/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/10/642910/6"}}},"commit":{"parents":[{"commit":"40d951047e33805ecf9990e75240c64d027d9ef0","subject":"Merge \"Quote the health check command argument during a debug with print-cmd.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/40d951047e33805ecf9990e75240c64d027d9ef0"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-03-12 22:21:47.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-06-27 22:26:47.000000000","tz":-240},"subject":"Handle defined containers that are stopped.","message":"Handle defined containers that are stopped.\n\nDelete the stopped container and recreate it to ensure\nthere are no local modifications to the container.\n\nChange-Id: Ia3fe06dd221d0894a5de2c18beaa48ba29440003\nResolves: rhbz#1674517\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/21c342fff98f4698276d88f8c509e57ac4a3cbf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/21c342fff98f4698276d88f8c509e57ac4a3cbf8"}]},"branch":"refs/heads/master"},"778d9b69f3d90862c5e0925e6b0a8646ff30d7c7":{"kind":"REWORK","_number":7,"created":"2019-06-28 13:02:25.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/10/642910/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/10/642910/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/10/642910/7"}}},"commit":{"parents":[{"commit":"40d951047e33805ecf9990e75240c64d027d9ef0","subject":"Merge \"Quote the health check command argument during a debug with print-cmd.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/40d951047e33805ecf9990e75240c64d027d9ef0"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-03-12 22:21:47.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-06-28 13:02:13.000000000","tz":-240},"subject":"Handle defined containers that are stopped.","message":"Handle defined containers that are stopped.\n\nDelete the stopped container and recreate it to ensure\nthere are no local modifications to the container.\n\nChange-Id: Ia3fe06dd221d0894a5de2c18beaa48ba29440003\nResolves: rhbz#1674517\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/778d9b69f3d90862c5e0925e6b0a8646ff30d7c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/778d9b69f3d90862c5e0925e6b0a8646ff30d7c7"}]},"branch":"refs/heads/master"},"2165130bba0b30ee1dd75b731a6f9dfa773f58be":{"kind":"REWORK","_number":8,"created":"2019-07-05 16:16:06.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/10/642910/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/10/642910/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/10/642910/8"}}},"commit":{"parents":[{"commit":"40d951047e33805ecf9990e75240c64d027d9ef0","subject":"Merge \"Quote the health check command argument during a debug with print-cmd.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/40d951047e33805ecf9990e75240c64d027d9ef0"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-07-05 15:31:13.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-07-05 16:15:38.000000000","tz":-240},"subject":"Handle defined containers that are stopped.","message":"Handle defined containers that are stopped.\n\nDelete the stopped container and recreate it to ensure\nthere are no local modifications to the container.\n\nChange-Id: Ia3fe06dd221d0894a5de2c18beaa48ba29440003\nResolves: rhbz#1674517\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2165130bba0b30ee1dd75b731a6f9dfa773f58be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2165130bba0b30ee1dd75b731a6f9dfa773f58be"}]},"branch":"refs/heads/master"},"c2a8f0fa2ebb8f10800ad9c027f25f2e345e83ac":{"kind":"REWORK","_number":9,"created":"2019-07-16 19:59:47.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/10/642910/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/10/642910/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/10/642910/9"}}},"commit":{"parents":[{"commit":"40d951047e33805ecf9990e75240c64d027d9ef0","subject":"Merge \"Quote the health check command argument during a debug with print-cmd.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/40d951047e33805ecf9990e75240c64d027d9ef0"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-07-05 15:31:13.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-07-16 19:59:41.000000000","tz":-240},"subject":"Handle defined containers that are stopped.","message":"Handle defined containers that are stopped.\n\nDelete the stopped container and recreate it to ensure\nthere are no local modifications to the container.\n\nChange-Id: Ia3fe06dd221d0894a5de2c18beaa48ba29440003\nResolves: rhbz#1674517\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/c2a8f0fa2ebb8f10800ad9c027f25f2e345e83ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/c2a8f0fa2ebb8f10800ad9c027f25f2e345e83ac"}]},"branch":"refs/heads/master"},"9bdde9b12acd20e828ca77867df41e5d35588f95":{"kind":"REWORK","_number":10,"created":"2019-07-17 15:58:35.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/10/642910/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/10/642910/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/10/642910/10"}}},"commit":{"parents":[{"commit":"40d951047e33805ecf9990e75240c64d027d9ef0","subject":"Merge \"Quote the health check command argument during a debug with print-cmd.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/40d951047e33805ecf9990e75240c64d027d9ef0"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-07-05 15:31:13.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-07-17 15:58:25.000000000","tz":-240},"subject":"Handle defined containers that are stopped.","message":"Handle defined containers that are stopped.\n\nDelete the stopped container and recreate it to ensure\nthere are no local modifications to the container.\n\nChange-Id: Ia3fe06dd221d0894a5de2c18beaa48ba29440003\nResolves: rhbz#1674517\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/9bdde9b12acd20e828ca77867df41e5d35588f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/9bdde9b12acd20e828ca77867df41e5d35588f95"}]},"branch":"refs/heads/master"},"cfffa2f73f92020b66b0ff319366c78f8b6796b5":{"kind":"REWORK","_number":11,"created":"2019-07-17 18:43:45.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/10/642910/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/10/642910/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/10/642910/11"}}},"commit":{"parents":[{"commit":"40d951047e33805ecf9990e75240c64d027d9ef0","subject":"Merge \"Quote the health check command argument during a debug with print-cmd.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/40d951047e33805ecf9990e75240c64d027d9ef0"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-07-05 15:31:13.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-07-17 18:43:42.000000000","tz":-240},"subject":"Handle defined containers that are stopped.","message":"Handle defined containers that are stopped.\n\nDelete the stopped container and recreate it to ensure\nthere are no local modifications to the container.\n\nChange-Id: Ia3fe06dd221d0894a5de2c18beaa48ba29440003\nResolves: rhbz#1674517\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/cfffa2f73f92020b66b0ff319366c78f8b6796b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/cfffa2f73f92020b66b0ff319366c78f8b6796b5"}]},"branch":"refs/heads/master"},"85fb2ed4d3f7e42840842f2bdec617e33d6ba4ef":{"kind":"REWORK","_number":12,"created":"2019-07-30 16:41:01.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/10/642910/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/10/642910/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/10/642910/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/10/642910/12"}}},"commit":{"parents":[{"commit":"6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce","subject":"Check if volume paths exist before executing Docker.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-07-05 15:31:13.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-07-30 16:40:51.000000000","tz":-240},"subject":"Handle defined containers that are stopped.","message":"Handle defined containers that are stopped.\n\nDelete the stopped container and recreate it to ensure\nthere are no local modifications to the container.\n\nChange-Id: Ia3fe06dd221d0894a5de2c18beaa48ba29440003\nResolves: rhbz#1674517\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/85fb2ed4d3f7e42840842f2bdec617e33d6ba4ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/85fb2ed4d3f7e42840842f2bdec617e33d6ba4ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
