)]}'
{"id":"openstack%2Fpaunch~645550","triplet_id":"openstack%2Fpaunch~master~Icbf2b81477902e3d7ff9e064bf2408c2fc7e510e","project":"openstack/paunch","branch":"master","topic":"bug/1821241","hashtags":[],"change_id":"Icbf2b81477902e3d7ff9e064bf2408c2fc7e510e","subject":"systemd: switch KillMode to \u0027none\u0027","status":"MERGED","created":"2019-03-22 12:18:13.000000000","updated":"2019-03-22 19:39:00.000000000","submitted":"2019-03-22 19:39:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"645550-1553283540694-46cdd86e","meta_rev_id":"c52c6a71f2a882e5c17883a1629cff88f85821f3","_number":645550,"virtual_id_number":645550,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":2,"date":"2019-03-22 19:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-03-22 15:19:08.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":2,"date":"2019-03-22 15:06:05.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-03-22 12:19:21.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"date":"2019-03-22 12:38:03.000000000","_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":1,"date":"2019-03-22 17:31:19.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"date":"2019-03-22 12:38:08.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-22 12:19:21.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-03-22 12:38:03.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2019-03-22 12:38:08.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-03-22 15:19:08.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-22 17:31:19.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-03-22 19:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e2f71cc76d9f2040096dd0052473cbb0d6d2a52","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-03-22 12:18:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63f9734fe56d80d2ad9e03028487201d06f48c01","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-03-22 12:19:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"692638962aa1ba1a8925143e6f7a4aa5962329e7","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-03-22 15:06:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"83685ea74befd5b2c1d2be19d6df352f4a127c40","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-22 15:19:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/645550/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b68932f/ : SUCCESS in 2h 44m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/50/645550/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a11fbcb/ : FAILURE in 2h 58m 11s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/50/645550/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/ce308c4/ : SUCCESS in 1h 24m 55s","accounts_in_message":[],"_revision_number":1},{"id":"66d54a053f614930e9f2477ebf326849790e6203","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 15:31:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-containers-multinode/dce75f5/ : SUCCESS in 2h 07m 23s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-undercloud-containers/ba62bef/ : SUCCESS in 1h 33m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/61d4342/ : SUCCESS in 2h 31m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4e6e1c5/ : SUCCESS in 2h 45m 18s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-standalone/d0e47f7/ : SUCCESS in 1h 49m 02s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/50/645550/1/check/tripleo-ci-fedora-28-standalone/515708a/ : SUCCESS in 1h 40m 54s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-standalone-upgrade/1e86865/ : FAILURE in 1h 44m 52s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/50/645550/1/check/openstack-tox-py35/f068fae/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/50/645550/1/check/openstack-tox-py36/5cfe3d7/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/50/645550/1/check/openstack-tox-py37/0b7f2ca/ : SUCCESS in 5m 11s\n- openstack-tox-docs http://logs.openstack.org/50/645550/1/check/openstack-tox-docs/20ee1f8/html/ : SUCCESS in 7m 19s\n- openstack-tox-pep8 http://logs.openstack.org/50/645550/1/check/openstack-tox-pep8/4336caf/ : SUCCESS in 7m 30s\n- openstack-tox-py27 http://logs.openstack.org/50/645550/1/check/openstack-tox-py27/00a7aac/ : SUCCESS in 8m 18s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/aea4f1c/ : SUCCESS in 2h 12m 04s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/ab39888/ : FAILURE in 1h 22m 44s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/0fd7f85/ : FAILURE in 1h 55m 05s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-scenario001-standalone/66455ab/ : SUCCESS in 1h 30m 50s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-scenario002-standalone/484976f/ : SUCCESS in 1h 32m 31s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-scenario003-standalone/3e10d9f/ : SUCCESS in 1h 12m 01s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-scenario004-standalone/26b39a8/ : SUCCESS in 1h 27m 57s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-scenario012-standalone/17a245b/ : FAILURE in 27m 37s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/50/645550/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/4f15cda/ : SUCCESS in 2h 07m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4569cd995a4f5cb26c3a2f0e57590c157357457e","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-03-22 17:31:19.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"74551b1d64d86f0960b2e06ccfe1563ecb79f90f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 17:31:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80491a9dc39adb1f1a10d74cab506d90ee672e3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 19:39:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/645550/1/gate/tripleo-ci-centos-7-containers-multinode/6e36599/ : SUCCESS in 2h 04m 51s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/50/645550/1/gate/tripleo-ci-centos-7-standalone/62760ed/ : SUCCESS in 1h 25m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/50/645550/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1b0aa94/ : SUCCESS in 2h 00m 38s\n- openstack-tox-py35 http://logs.openstack.org/50/645550/1/gate/openstack-tox-py35/9e75fdc/ : SUCCESS in 7m 25s\n- openstack-tox-py36 http://logs.openstack.org/50/645550/1/gate/openstack-tox-py36/fa73df5/ : SUCCESS in 6m 51s\n- openstack-tox-py37 http://logs.openstack.org/50/645550/1/gate/openstack-tox-py37/86ce844/ : SUCCESS in 7m 13s\n- openstack-tox-docs http://logs.openstack.org/50/645550/1/gate/openstack-tox-docs/fba537a/html/ : SUCCESS in 5m 58s\n- openstack-tox-pep8 http://logs.openstack.org/50/645550/1/gate/openstack-tox-pep8/309b960/ : SUCCESS in 6m 25s\n- openstack-tox-py27 http://logs.openstack.org/50/645550/1/gate/openstack-tox-py27/55713fd/ : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4c428892e8ba633ebe61dc0b5ad2b7f2e4960655","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 19:39:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"110a9496b2c87a4d41d863aa918c2c4535b32e22","revisions":{"110a9496b2c87a4d41d863aa918c2c4535b32e22":{"kind":"REWORK","_number":1,"created":"2019-03-22 12:18:13.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/50/645550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/50/645550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/50/645550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/50/645550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/50/645550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/50/645550/1"}}},"commit":{"parents":[{"commit":"8a61464c92d5aa1450c95543ad57eb9e6b6bdc7b","subject":"Merge \"Delay check CI jobs until the pep8/unit passes..\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/8a61464c92d5aa1450c95543ad57eb9e6b6bdc7b"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-03-22 12:14:48.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-03-22 12:17:27.000000000","tz":-240},"subject":"systemd: switch KillMode to \u0027none\u0027","message":"systemd: switch KillMode to \u0027none\u0027\n\nWhen running the services with KillMode\u003dprocess, there is a race\ncondition between ExecStop and the command specified in ExecStart.\nThe ExecStop seems faster and the container is killed then cleaned up.\nHowever podman started by ExecStart is still running and systemd kills\nit as soon as the ExecStop finished.\n\nSince we rely on Podman to manage the containers \u0026 processes, let\u0027s\nswitch to KillMode\u003dnone.\n\nCredits to Giuseppe Scrivano for explaining the root cause.\n\nChange-Id: Icbf2b81477902e3d7ff9e064bf2408c2fc7e510e\nCloses-Bug: #1821241\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/110a9496b2c87a4d41d863aa918c2c4535b32e22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/110a9496b2c87a4d41d863aa918c2c4535b32e22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
