)]}'
{"id":"openstack%2Fpaunch~665739","triplet_id":"openstack%2Fpaunch~master~I9b7f1df0f466f8e5cd44f0c8e4465618fe4fc8a1","project":"openstack/paunch","branch":"master","topic":"refactor/exists","hashtags":[],"change_id":"I9b7f1df0f466f8e5cd44f0c8e4465618fe4fc8a1","subject":"runner: refactor image_exist","status":"MERGED","created":"2019-06-17 18:04:15.000000000","updated":"2019-07-08 18:04:18.000000000","submitted":"2019-07-08 06:43:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665739-1562568218991-5870c8eb","meta_rev_id":"b9b5a7c7f4f5052f9fb667679ef8a564592d9e2b","_number":665739,"virtual_id_number":665739,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-07-08 06:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-04 22:30:38.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-04 19:38:38.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-08 03:55:25.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":1,"date":"2019-07-04 19:38:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-04 19:38:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-07-08 03:59:52.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 13:42:57.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-26 16:25:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-06-28 14:59:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-04 22:30:38.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-08 03:55:25.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-07-08 06:43:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"526f32ae18678695e97079f3e69ba5c063ef08c0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-17 18:04:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c45d4efd474dc3b981beb5942c81b6c7d4ea13d7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-17 19:23:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/39/665739/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3a72e9d/ : FAILURE in 1h 03m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/39/665739/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3d8d0c2/ : FAILURE in 1h 06m 10s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/39/665739/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/55ba6b0/ : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":1},{"id":"28e63ba26bb773e52c054f68d938d9156ea428dd","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-17 20:02:11.000000000","message":"Patch Set 1: Workflow-1\n\npodman image exists is too slow we might need to refactor again","accounts_in_message":[],"_revision_number":1},{"id":"72508c8f064d601352175f1742114093762c87d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:54:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-containers-multinode/cc2d083/ : SUCCESS in 1h 58m 16s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-undercloud-containers/a1aa8e4/ : SUCCESS in 1h 30m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b08a9f1/ : SUCCESS in 2h 21m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/187b798/ : SUCCESS in 2h 30m 06s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-standalone/c5809fd/ : SUCCESS in 1h 56m 59s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/39/665739/1/check/tripleo-ci-fedora-28-standalone/8074bac/ : SUCCESS in 2h 03m 55s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-standalone-upgrade/63ad8de/ : SUCCESS in 2h 02m 17s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/828d376/ : SUCCESS in 2h 10m 56s\n- openstack-tox-py35 http://logs.openstack.org/39/665739/1/check/openstack-tox-py35/963d9ac/ : SUCCESS in 6m 14s\n- openstack-tox-py36 http://logs.openstack.org/39/665739/1/check/openstack-tox-py36/61946bb/ : SUCCESS in 7m 38s\n- openstack-tox-py37 http://logs.openstack.org/39/665739/1/check/openstack-tox-py37/d0b84fe/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/39/665739/1/check/openstack-tox-docs/74b8c43/html/ : SUCCESS in 4m 38s\n- openstack-tox-pep8 http://logs.openstack.org/39/665739/1/check/openstack-tox-pep8/2b94282/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/39/665739/1/check/openstack-tox-py27/48762cf/ : SUCCESS in 7m 19s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/8f7e7ed/ : SUCCESS in 2h 11m 37s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/ed439bc/ : SUCCESS in 2h 10m 15s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-scenario001-standalone/8da7fec/ : SUCCESS in 1h 48m 10s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-scenario002-standalone/fd53962/ : SUCCESS in 1h 40m 53s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-scenario003-standalone/ae8cccc/ : SUCCESS in 1h 19m 55s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-scenario004-standalone/0e0f0b0/ : SUCCESS in 1h 41m 18s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-scenario012-standalone/45b4f7a/ : FAILURE in 25m 35s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/39/665739/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5ead837/ : FAILURE in 1h 20m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"644cd49901a6247c446fa52ef79beefd8e5fd827","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-19 15:15:14.000000000","message":"Abandoned\n\nnot in the plans for now","accounts_in_message":[],"_revision_number":1},{"id":"f424b3fed519264cf1f14c23a20075562b5e0457","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-21 15:14:01.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"67ba9b91b176826d976aa2d22dff330a8befa924","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-21 15:14:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"12d8bf7eac664939d39c0d4ef5490fc4498fe81b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-21 17:51:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/39/665739/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/fb1aa24/ : SUCCESS in 2h 36m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/39/665739/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/e6cac93/ : SUCCESS in 2h 30m 41s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/39/665739/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/0383abb/ : SUCCESS in 1h 42m 41s","accounts_in_message":[],"_revision_number":2},{"id":"6a42361fdfaa942db920371a34281f29f1d1438d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 20:16:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-containers-multinode/55280e1/ : SUCCESS in 2h 09m 49s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-undercloud-containers/560016d/ : SUCCESS in 1h 43m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/cc00646/ : SUCCESS in 2h 12m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/40d6c78/ : SUCCESS in 2h 23m 28s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-standalone/20fa686/ : SUCCESS in 1h 45m 46s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/39/665739/2/check/tripleo-ci-fedora-28-standalone/98fce32/ : SUCCESS in 1h 39m 24s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-standalone-upgrade/37e0da5/ : SUCCESS in 2h 03m 55s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/9d20d97/ : SUCCESS in 2h 11m 04s\n- openstack-tox-py35 http://logs.openstack.org/39/665739/2/check/openstack-tox-py35/43004aa/ : SUCCESS in 5m 04s\n- openstack-tox-py36 http://logs.openstack.org/39/665739/2/check/openstack-tox-py36/5a11706/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/39/665739/2/check/openstack-tox-py37/6fdb2d9/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/39/665739/2/check/openstack-tox-docs/bddd9f2/html/ : SUCCESS in 5m 05s\n- openstack-tox-pep8 http://logs.openstack.org/39/665739/2/check/openstack-tox-pep8/b9008d3/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/39/665739/2/check/openstack-tox-py27/1d703b3/ : SUCCESS in 5m 01s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/ba67162/ : SUCCESS in 2h 12m 27s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/dd8c713/ : SUCCESS in 2h 06m 33s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-scenario001-standalone/44a173c/ : SUCCESS in 1h 39m 08s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-scenario002-standalone/0c13b0d/ : SUCCESS in 1h 36m 27s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-scenario003-standalone/9bbe9b1/ : SUCCESS in 1h 13m 24s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-scenario004-standalone/8bc29eb/ : SUCCESS in 1h 49m 24s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-scenario012-standalone/458c46a/ : FAILURE in 30m 22s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/39/665739/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/7042060/ : FAILURE in 1h 25m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac5836a6a4062a85b0139e5482c1ddd090439811","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-25 10:07:49.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nHello,\n\nSome comments inline.\nThoughts?","accounts_in_message":[],"_revision_number":2},{"id":"e2c3a02f9615b89193fa792442080101bbeb8742","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-25 12:13:34.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nThanks for the review! I\u0027ll push it right away.","accounts_in_message":[],"_revision_number":2},{"id":"af5cbf36fd1ef1919a003093ffe9756e2d25eb44","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-25 12:15:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"84ced891b2fce7e22b9bd28441438c5d61611ccc","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-25 12:36:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"edf6c79abdde6c500cb9def19e8bb7508b609814","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-25 12:50:07.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a30bd3db96de86ed6024a73afa740c33acd2a4e8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-25 15:17:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/39/665739/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d6e46c3/ : SUCCESS in 2h 56m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/39/665739/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/665f297/ : SUCCESS in 3h 00m 35s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/39/665739/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/c9cdc19/ : SUCCESS in 1h 28m 18s","accounts_in_message":[],"_revision_number":3},{"id":"73d0f8764d0ab820bdae8d30c14d4de42a572781","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-25 17:10:11.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e177c4d7d5c9e0f146e3571eae79cb06ab206fb1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-25 17:10:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4774c3e0aeaa90d8436a05c54b82d133424b0b0e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-25 18:06:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b1183767c8e68665f151e4912f1ed5c1d81dc9e7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-25 18:09:29.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c81ee908033b3f50481bdf73531a055aa447629b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-25 19:49:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/39/665739/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cf60563/ : SUCCESS in 2h 38m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/39/665739/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/7370e92/ : SUCCESS in 2h 36m 35s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/39/665739/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/e4a41ca/ : SUCCESS in 1h 20m 42s","accounts_in_message":[],"_revision_number":4},{"id":"ff927fe28c083bdbb1a0e93f270da952af824a04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 03:40:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-containers-multinode/7f1d34a/ : SUCCESS in 2h 19m 23s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-undercloud-containers/6a533e9/ : SUCCESS in 1h 59m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/fdae2bf/ : SUCCESS in 2h 14m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/e59f9b6/ : FAILURE in 1h 54m 15s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-standalone/03f4e64/ : SUCCESS in 1h 55m 47s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/39/665739/4/check/tripleo-ci-fedora-28-standalone/0479e8e/ : SUCCESS in 1h 47m 53s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-standalone-upgrade/f7ff54f/ : SUCCESS in 2h 19m 37s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-standalone-upgrade-stein/dab10bb/ : SUCCESS in 2h 12m 28s\n- openstack-tox-py35 http://logs.openstack.org/39/665739/4/check/openstack-tox-py35/b634989/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/39/665739/4/check/openstack-tox-py36/bfdaa62/ : SUCCESS in 3m 34s\n- openstack-tox-py37 http://logs.openstack.org/39/665739/4/check/openstack-tox-py37/275ee95/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/39/665739/4/check/openstack-tox-docs/49cbe7c/html/ : SUCCESS in 2m 46s\n- openstack-tox-pep8 http://logs.openstack.org/39/665739/4/check/openstack-tox-pep8/8b60c13/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/39/665739/4/check/openstack-tox-py27/ed6f67b/ : SUCCESS in 3m 24s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/f8395de/ : SUCCESS in 2h 35m 01s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/4262ca8/ : FAILURE in 2h 06m 15s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-scenario001-standalone/4cfaae2/ : SUCCESS in 1h 54m 32s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-scenario002-standalone/7bf0a95/ : SUCCESS in 1h 43m 44s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-scenario003-standalone/12845c4/ : SUCCESS in 1h 15m 04s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-scenario004-standalone/2262fff/ : SUCCESS in 1h 47m 13s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-scenario012-standalone/7452a05/ : FAILURE in 22m 58s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/39/665739/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/cc5666e/ : FAILURE in 1h 15m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8d1382d1ca9fd0562f14b2e006deb07a071ecb38","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-26 13:41:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"133548f899498a44ae7e29d6fa8508813b3e790c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-26 13:41:52.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"cdd40808257dbeaac783db1595a7163b3995338f","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-26 13:42:57.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d03c9563aad771dab2118d175260a21820d8e354","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-26 16:25:49.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"79ae1ac72d808bef51e983ff3b652c442a14c02b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 16:40:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/39/665739/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6231fbc/ : SUCCESS in 2h 42m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/39/665739/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/50be35c/ : SUCCESS in 2h 56m 51s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/39/665739/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/0cffb64/ : SUCCESS in 1h 28m 33s","accounts_in_message":[],"_revision_number":5},{"id":"9f73c99f3e1385bb1f9af6a9362023d8bf11e950","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 22:24:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-containers-multinode/9c29de7/ : SUCCESS in 2h 04m 29s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-undercloud-containers/51c425a/ : SUCCESS in 1h 45m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2ba5111/ : SUCCESS in 2h 24m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5c5d972/ : SUCCESS in 2h 18m 22s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-standalone/8c8c1ac/ : SUCCESS in 1h 50m 15s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/39/665739/5/check/tripleo-ci-fedora-28-standalone/bdb7793/ : SUCCESS in 2h 06m 41s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-standalone-upgrade/a6e7d62/ : SUCCESS in 2h 36m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-standalone-upgrade-stein/a6959b2/ : SUCCESS in 2h 09m 11s\n- openstack-tox-py35 http://logs.openstack.org/39/665739/5/check/openstack-tox-py35/280f7c1/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/39/665739/5/check/openstack-tox-py36/d453f6c/ : SUCCESS in 3m 24s\n- openstack-tox-py37 http://logs.openstack.org/39/665739/5/check/openstack-tox-py37/2ceeae0/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/39/665739/5/check/openstack-tox-docs/d4624f7/html/ : SUCCESS in 2m 39s\n- openstack-tox-pep8 http://logs.openstack.org/39/665739/5/check/openstack-tox-pep8/e69f213/ : SUCCESS in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/39/665739/5/check/openstack-tox-py27/0f47657/ : SUCCESS in 2m 57s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/ebfa9c2/ : SUCCESS in 2h 17m 56s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/4206570/ : SUCCESS in 2h 13m 32s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-scenario001-standalone/bf6698f/ : SUCCESS in 1h 59m 26s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-scenario002-standalone/aa96c98/ : SUCCESS in 1h 56m 11s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-scenario003-standalone/59d1017/ : SUCCESS in 1h 19m 08s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-scenario004-standalone/6a43dd1/ : SUCCESS in 1h 48m 11s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-scenario012-standalone/268187c/ : FAILURE in 24m 36s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/39/665739/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/6aec666/ : FAILURE in 1h 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"712a8498b0afeb6019d0b5641261eddef4883e1f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-28 14:59:21.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fe7e70f120ed2627842d8958357224c4dac1805a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-04 19:38:38.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"80471631a9e4d415d328edb0771dd3439650e990","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-04 22:30:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/39/665739/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3aae762/ : SUCCESS in 2h 31m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/39/665739/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9836db7/ : SUCCESS in 2h 50m 33s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/39/665739/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/a294ff4/ : SUCCESS in 1h 27m 32s","accounts_in_message":[],"_revision_number":6},{"id":"30728bb16a4ac2c053646cc4bba2007f3b96f9da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 22:46:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-containers-multinode/1a8738e/ : SUCCESS in 2h 29m 39s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-undercloud-containers/641989e/ : SUCCESS in 1h 34m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/0514cfe/ : SUCCESS in 2h 26m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/76c5a05/ : SUCCESS in 2h 21m 01s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-standalone/b02742c/ : SUCCESS in 2h 10m 57s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/39/665739/6/check/tripleo-ci-fedora-28-standalone/4f4d267/ : FAILURE in 32m 50s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-standalone-upgrade/f474f8a/ : SUCCESS in 2h 11m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-standalone-upgrade-stein/32647b6/ : SUCCESS in 2h 15m 03s\n- openstack-tox-py35 http://logs.openstack.org/39/665739/6/check/openstack-tox-py35/8cb2b8b/ : SUCCESS in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/39/665739/6/check/openstack-tox-py36/e150a94/ : SUCCESS in 3m 19s\n- openstack-tox-py37 http://logs.openstack.org/39/665739/6/check/openstack-tox-py37/9d6b359/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/39/665739/6/check/openstack-tox-docs/e26d716/html/ : SUCCESS in 2m 27s\n- openstack-tox-pep8 http://logs.openstack.org/39/665739/6/check/openstack-tox-pep8/ae8d4d0/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/39/665739/6/check/openstack-tox-py27/6b5b2ab/ : SUCCESS in 3m 18s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/722c615/ : SUCCESS in 2h 45m 13s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/e54add6/ : SUCCESS in 2h 04m 58s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-scenario001-standalone/acfbe5a/ : SUCCESS in 1h 43m 35s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-scenario002-standalone/8f09a2a/ : SUCCESS in 1h 38m 51s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-scenario003-standalone/6b24969/ : SUCCESS in 1h 16m 14s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-scenario004-standalone/e99cd88/ : SUCCESS in 1h 34m 11s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-scenario012-standalone/7ebf5fc/ : SUCCESS in 1h 33m 41s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/39/665739/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/6ac9373/ : FAILURE in 2h 06m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"55d3c02f59020aba868861fca8f048ba1f62d282","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-08 03:55:25.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"3fd8aee9e50a0771e4f55c8e249236e4d2ced964","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-08 03:59:52.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"a90e60bf3128961c3692134aee15ef3bde45d18c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 04:00:03.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e21840fa7ae32e84ddc11a8c9b7d296d476a8da5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 06:43:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"d83131e74a33a550673a91e94707482ca83126ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 06:43:38.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/665739/6/gate/tripleo-ci-centos-7-containers-multinode/fbead13/ : SUCCESS in 2h 04m 03s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/39/665739/6/gate/tripleo-ci-centos-7-standalone/7634ae0/ : SUCCESS in 1h 48m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/39/665739/6/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/97fc1c7/ : SUCCESS in 2h 36m 42s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/39/665739/6/gate/tripleo-ci-centos-7-standalone-upgrade-stein/1a42b02/ : SUCCESS in 2h 31m 35s\n- openstack-tox-py35 http://logs.openstack.org/39/665739/6/gate/openstack-tox-py35/54b5b04/ : SUCCESS in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/39/665739/6/gate/openstack-tox-py36/43ba68a/ : SUCCESS in 2m 49s\n- openstack-tox-py37 http://logs.openstack.org/39/665739/6/gate/openstack-tox-py37/caa0c5a/ : SUCCESS in 3m 14s\n- openstack-tox-docs http://logs.openstack.org/39/665739/6/gate/openstack-tox-docs/d4780da/html/ : SUCCESS in 2m 17s\n- openstack-tox-pep8 http://logs.openstack.org/39/665739/6/gate/openstack-tox-pep8/8927908/ : SUCCESS in 2m 41s\n- openstack-tox-py27 http://logs.openstack.org/39/665739/6/gate/openstack-tox-py27/6e6fe65/ : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":6},{"id":"2cca4e094d646a78a9a7cd0ac150a553f28c565c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-08 18:04:18.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 45ea759f9a82562f432f64d068edf778a0805816","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"43468616c89e695f65befd4d49b8c1c7bdcd0b2c","revisions":{"29bb794a0f5796113ccc737f3aaef9d4581f5bf0":{"kind":"REWORK","_number":1,"created":"2019-06-17 18:04:15.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/39/665739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/39/665739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/39/665739/1"}}},"commit":{"parents":[{"commit":"87ab8dc78b1a226be04b5a38e577f504709f6bb8","subject":"Add upper constraints for testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/87ab8dc78b1a226be04b5a38e577f504709f6bb8"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:01:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:01:42.000000000","tz":-240},"subject":"runner: refactor image_exist","message":"runner: refactor image_exist\n\niamge_exist only makes sense in podman where the inspect command\nrequires the image to be here.\n\nAlso, rather than returning a return code, return a bool to make the use\nof image_exist simpler in the code.\n\nThis patch is preparing the code to have container_exist and re-use the\nsame logic later.\n\nChange-Id: I9b7f1df0f466f8e5cd44f0c8e4465618fe4fc8a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/29bb794a0f5796113ccc737f3aaef9d4581f5bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/29bb794a0f5796113ccc737f3aaef9d4581f5bf0"}]},"branch":"refs/heads/master"},"ed21d4f6276a2cd892ab70bcefddd9d2e0f774f0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-21 15:14:05.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/39/665739/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/39/665739/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/39/665739/2"}}},"commit":{"parents":[{"commit":"01eb542554db45ccfc7b004196d1271fccaa5797","subject":"Merge \"Revert \"docker/compose: quote health-cmd\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/01eb542554db45ccfc7b004196d1271fccaa5797"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:01:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-21 15:14:05.000000000","tz":0},"subject":"runner: refactor image_exist","message":"runner: refactor image_exist\n\niamge_exist only makes sense in podman where the inspect command\nrequires the image to be here.\n\nAlso, rather than returning a return code, return a bool to make the use\nof image_exist simpler in the code.\n\nThis patch is preparing the code to have container_exist and re-use the\nsame logic later.\n\nChange-Id: I9b7f1df0f466f8e5cd44f0c8e4465618fe4fc8a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/ed21d4f6276a2cd892ab70bcefddd9d2e0f774f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/ed21d4f6276a2cd892ab70bcefddd9d2e0f774f0"}]},"branch":"refs/heads/master"},"bd32de1f209e13fd275ae8b96b6489c6a1102f66":{"kind":"REWORK","_number":3,"created":"2019-06-25 12:15:15.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/39/665739/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/39/665739/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/39/665739/3"}}},"commit":{"parents":[{"commit":"d1eef2930ea2ac3610c9c32d3ab172ddbecf665d","subject":"Use k8s-file log driver with podman","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/d1eef2930ea2ac3610c9c32d3ab172ddbecf665d"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:01:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-25 12:15:01.000000000","tz":-240},"subject":"runner: refactor image_exist","message":"runner: refactor image_exist\n\niamge_exist only makes sense in podman where the inspect command\nrequires the image to be here.\n\nAlso, rather than returning a return code, return a bool to make the use\nof image_exist simpler in the code.\n\nThis patch is preparing the code to have container_exist and re-use the\nsame logic later.\n\nChange-Id: I9b7f1df0f466f8e5cd44f0c8e4465618fe4fc8a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/bd32de1f209e13fd275ae8b96b6489c6a1102f66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/bd32de1f209e13fd275ae8b96b6489c6a1102f66"}]},"branch":"refs/heads/master"},"a7a6d823abf7ae6112684e3253f5849059aaf855":{"kind":"REWORK","_number":4,"created":"2019-06-25 17:10:57.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/39/665739/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/39/665739/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/39/665739/4"}}},"commit":{"parents":[{"commit":"d1eef2930ea2ac3610c9c32d3ab172ddbecf665d","subject":"Use k8s-file log driver with podman","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/d1eef2930ea2ac3610c9c32d3ab172ddbecf665d"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:01:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-25 17:10:53.000000000","tz":-240},"subject":"runner: refactor image_exist","message":"runner: refactor image_exist\n\niamge_exist only makes sense in podman where the inspect command\nrequires the image to be here.\n\nAlso, rather than returning a return code, return a bool to make the use\nof image_exist simpler in the code.\n\nThis patch is preparing the code to have container_exist and re-use the\nsame logic later.\n\nChange-Id: I9b7f1df0f466f8e5cd44f0c8e4465618fe4fc8a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/a7a6d823abf7ae6112684e3253f5849059aaf855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/a7a6d823abf7ae6112684e3253f5849059aaf855"}]},"branch":"refs/heads/master"},"747a90ac21540e35d18e8cd70472a9b103901c64":{"kind":"REWORK","_number":5,"created":"2019-06-26 13:41:52.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/39/665739/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/39/665739/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/39/665739/5"}}},"commit":{"parents":[{"commit":"d1eef2930ea2ac3610c9c32d3ab172ddbecf665d","subject":"Use k8s-file log driver with podman","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/d1eef2930ea2ac3610c9c32d3ab172ddbecf665d"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:01:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-26 13:41:50.000000000","tz":0},"subject":"runner: refactor image_exist","message":"runner: refactor image_exist\n\niamge_exist only makes sense in podman where the inspect command\nrequires the image to be here.\n\nAlso, rather than returning a return code, return a bool to make the use\nof image_exist simpler in the code.\n\nThis patch is preparing the code to have container_exist and re-use the\nsame logic later.\n\nChange-Id: I9b7f1df0f466f8e5cd44f0c8e4465618fe4fc8a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/747a90ac21540e35d18e8cd70472a9b103901c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/747a90ac21540e35d18e8cd70472a9b103901c64"}]},"branch":"refs/heads/master"},"43468616c89e695f65befd4d49b8c1c7bdcd0b2c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-07-04 19:38:38.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/39/665739/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/39/665739/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/665739/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/39/665739/6"}}},"commit":{"parents":[{"commit":"40d951047e33805ecf9990e75240c64d027d9ef0","subject":"Merge \"Quote the health check command argument during a debug with print-cmd.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/40d951047e33805ecf9990e75240c64d027d9ef0"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:01:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-07-04 19:38:38.000000000","tz":0},"subject":"runner: refactor image_exist","message":"runner: refactor image_exist\n\niamge_exist only makes sense in podman where the inspect command\nrequires the image to be here.\n\nAlso, rather than returning a return code, return a bool to make the use\nof image_exist simpler in the code.\n\nThis patch is preparing the code to have container_exist and re-use the\nsame logic later.\n\nChange-Id: I9b7f1df0f466f8e5cd44f0c8e4465618fe4fc8a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/43468616c89e695f65befd4d49b8c1c7bdcd0b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/43468616c89e695f65befd4d49b8c1c7bdcd0b2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
