)]}'
{"id":"openstack%2Fpaunch~665743","triplet_id":"openstack%2Fpaunch~master~I6dfbfe9da3d216cadaca77a53a0d5223c9cbd57d","project":"openstack/paunch","branch":"master","topic":"bug/1833081","hashtags":[],"change_id":"I6dfbfe9da3d216cadaca77a53a0d5223c9cbd57d","subject":"runner: implement container_exist","status":"MERGED","created":"2019-06-17 18:15:33.000000000","updated":"2019-07-08 18:03:03.000000000","submitted":"2019-07-08 18:03:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665743-1562608983135-fd40a304","meta_rev_id":"b6b5bfb85f22dd7a50de50c3a6557737f769c5b6","_number":665743,"virtual_id_number":665743,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-07-08 18:03:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-04 22:13:47.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-04 19:38:42.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-08 03:55:35.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":1,"date":"2019-07-04 19:38:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-04 19:38:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-07-08 03:59:55.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 13:43:51.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-26 16:26:55.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-06-28 19:32:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-04 22:13:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-08 03:55:35.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-07-08 18:03:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"358402286cff83b7cd1ff1085185f9e0a4b2177a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-17 18:15:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a94198ba8f56eae9e958357334feeeb3bcc73af","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-17 20:02:27.000000000","message":"Patch Set 1: Workflow-1\n\npodman container exists is too slow we might need to use podman ps","accounts_in_message":[],"_revision_number":1},{"id":"b89545911a181831a93ecaf819c0fbebe5ff7da8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-17 20:38:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/43/665743/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7cfc1da/ : SUCCESS in 2h 16m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/43/665743/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/0cbca11/ : SUCCESS in 2h 21m 18s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/43/665743/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/0fe3b1a/ : SUCCESS in 1h 12m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9f62eca3b758acc4c1d4067fd048c64ea252ca74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 01:13:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-containers-multinode/ac0e899/ : SUCCESS in 2h 19m 41s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-undercloud-containers/0830904/ : SUCCESS in 1h 29m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/cb366a7/ : SUCCESS in 2h 08m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/756a9e0/ : SUCCESS in 2h 33m 50s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-standalone/72c5c9a/ : SUCCESS in 1h 52m 19s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/43/665743/1/check/tripleo-ci-fedora-28-standalone/6205145/ : SUCCESS in 1h 55m 58s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-standalone-upgrade/13468d7/ : SUCCESS in 2h 02m 24s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/389b41b/ : SUCCESS in 2h 09m 42s\n- openstack-tox-py35 http://logs.openstack.org/43/665743/1/check/openstack-tox-py35/7b426e4/ : SUCCESS in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/43/665743/1/check/openstack-tox-py36/b548a0e/ : SUCCESS in 5m 20s\n- openstack-tox-py37 http://logs.openstack.org/43/665743/1/check/openstack-tox-py37/4b74cdf/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/43/665743/1/check/openstack-tox-docs/1ab80c0/html/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/43/665743/1/check/openstack-tox-pep8/beeb412/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/43/665743/1/check/openstack-tox-py27/9042d81/ : SUCCESS in 5m 05s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/b0665c3/ : SUCCESS in 2h 16m 14s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/cc1f2ac/ : SUCCESS in 2h 03m 03s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-scenario001-standalone/3e2871f/ : SUCCESS in 1h 44m 18s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-scenario002-standalone/1a38868/ : SUCCESS in 1h 28m 36s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-scenario003-standalone/879fa48/ : SUCCESS in 1h 18m 42s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-scenario004-standalone/1c80b91/ : SUCCESS in 1h 40m 06s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-scenario012-standalone/52847d4/ : FAILURE in 26m 25s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/43/665743/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d057400/ : FAILURE in 1h 29m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f6df4a58986c1e46f9c23c4793abe959d05528e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-19 15:15:15.000000000","message":"Abandoned\n\nnot in the plans for now","accounts_in_message":[],"_revision_number":1},{"id":"e21b0afa506650651aa08c07059003a546ed038e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-21 15:14:10.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"4dd25ffbf30b2464651d0ab6828e579129587395","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-21 15:14:12.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"00bea2a7793879c976bf65069e8bb2d3e268ef0c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-21 15:14:23.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"9108f229563466185b3b30ba4edc4e5c37fe91b9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-21 18:07:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/43/665743/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/57551ce/ : SUCCESS in 2h 41m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/43/665743/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/51e64b9/ : SUCCESS in 2h 51m 24s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/43/665743/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/00bf38f/ : SUCCESS in 1h 41m 32s","accounts_in_message":[],"_revision_number":2},{"id":"41ed3672b9385fc3ed1ef40dc59854bb61a3f195","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 20:17:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-containers-multinode/dd70c36/ : SUCCESS in 1h 59m 57s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-undercloud-containers/270cb94/ : SUCCESS in 1h 27m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4725503/ : SUCCESS in 2h 11m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0331922/ : SUCCESS in 2h 14m 58s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-standalone/314568d/ : SUCCESS in 2h 01m 11s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/43/665743/2/check/tripleo-ci-fedora-28-standalone/06bc2aa/ : SUCCESS in 1h 58m 19s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-standalone-upgrade/67d9034/ : SUCCESS in 2h 16m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/e562de1/ : SUCCESS in 2h 07m 30s\n- openstack-tox-py35 http://logs.openstack.org/43/665743/2/check/openstack-tox-py35/c255f4e/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/43/665743/2/check/openstack-tox-py36/81d954d/ : SUCCESS in 4m 32s\n- openstack-tox-py37 http://logs.openstack.org/43/665743/2/check/openstack-tox-py37/fc61cbc/ : SUCCESS in 5m 03s\n- openstack-tox-docs http://logs.openstack.org/43/665743/2/check/openstack-tox-docs/3987ac5/html/ : SUCCESS in 6m 23s\n- openstack-tox-pep8 http://logs.openstack.org/43/665743/2/check/openstack-tox-pep8/48596ce/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/43/665743/2/check/openstack-tox-py27/05ddf38/ : SUCCESS in 4m 28s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/70e2f6e/ : SUCCESS in 2h 23m 54s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/fba7e3d/ : SUCCESS in 2h 03m 09s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-scenario001-standalone/36de93c/ : SUCCESS in 1h 34m 11s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-scenario002-standalone/ff6ba64/ : SUCCESS in 1h 36m 50s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-scenario003-standalone/a8cb898/ : SUCCESS in 1h 26m 50s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-scenario004-standalone/8fb36f6/ : SUCCESS in 1h 48m 33s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-scenario012-standalone/1b7bbba/ : FAILURE in 27m 41s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/43/665743/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ecee48e/ : FAILURE in 1h 37m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47ea4a17a279383d244578d472be103b84a199b8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-25 10:11:23.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nHello\n\nSee my comment inline","accounts_in_message":[],"_revision_number":2},{"id":"4846ca931bef30105e521d06668c11184cb963ed","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-25 12:16:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"118f1ff6c94f7cab754fe3ba58c2c01da117266c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-25 12:17:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"68eecf8c5e6d97250e84911dec34d1e81c77fc52","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-25 12:37:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"56cbeff179f10330ddaab3b261b3f05cbbdd16bf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-25 12:51:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"da9f756b6864f55efd493fb7eae633cd3c868a94","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-25 15:17:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/43/665743/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2232f12/ : SUCCESS in 2h 55m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/43/665743/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ab46609/ : SUCCESS in 2h 57m 46s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/43/665743/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/52a2754/ : SUCCESS in 1h 21m 14s","accounts_in_message":[],"_revision_number":3},{"id":"bde2014b540942d0b8dd4240a9ddeda4b3296588","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-25 17:11:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bfd10fc9ab0723555801ec0ff90e05a5cae27c35","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-25 18:12:02.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nThanks for the changes.\n\nSome comments inline:","accounts_in_message":[],"_revision_number":4},{"id":"15887f6173454a891fd3e6a8d1d1f22922c9fe15","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-25 19:43:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/43/665743/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/729518f/ : SUCCESS in 2h 23m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/43/665743/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c1aa464/ : SUCCESS in 2h 29m 52s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/43/665743/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/4ab8e21/ : SUCCESS in 1h 20m 42s","accounts_in_message":[],"_revision_number":4},{"id":"fa83ee3fe7f85b60f2434c9b380abbf34429807f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 04:07:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-containers-multinode/26ddef3/ : SUCCESS in 2h 07m 21s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-undercloud-containers/c668dee/ : SUCCESS in 1h 41m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6eca36a/ : SUCCESS in 2h 20m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1123e4d/ : TIMED_OUT in 3h 17m 23s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-standalone/73e3b99/ : SUCCESS in 2h 09m 37s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/43/665743/4/check/tripleo-ci-fedora-28-standalone/218d114/ : SUCCESS in 1h 43m 09s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-standalone-upgrade/9de184e/ : SUCCESS in 2h 15m 29s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-standalone-upgrade-stein/333103f/ : SUCCESS in 2h 07m 24s\n- openstack-tox-py35 http://logs.openstack.org/43/665743/4/check/openstack-tox-py35/5815945/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/43/665743/4/check/openstack-tox-py36/0aa05ae/ : SUCCESS in 3m 07s\n- openstack-tox-py37 http://logs.openstack.org/43/665743/4/check/openstack-tox-py37/c1198fa/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/43/665743/4/check/openstack-tox-docs/88bc184/html/ : SUCCESS in 2m 27s\n- openstack-tox-pep8 http://logs.openstack.org/43/665743/4/check/openstack-tox-pep8/721d352/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/43/665743/4/check/openstack-tox-py27/da7928b/ : SUCCESS in 3m 14s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/9a9b85f/ : SUCCESS in 2h 33m 20s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/1cc12e0/ : SUCCESS in 2h 19m 29s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-scenario001-standalone/67d92ac/ : SUCCESS in 1h 49m 58s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-scenario002-standalone/62fc61f/ : SUCCESS in 1h 39m 45s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-scenario003-standalone/af7cf65/ : SUCCESS in 1h 16m 42s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-scenario004-standalone/76d3707/ : SUCCESS in 2h 06m 26s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-scenario012-standalone/d11a5b3/ : FAILURE in 28m 29s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/43/665743/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d53eb9a/ : FAILURE in 1h 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"84494de9315b27449717eb1a6723e82a978bfad8","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-26 13:42:15.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"bf2a507e71d1d82368ce2685c51d82b50c43b3f8","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-26 13:42:37.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"c1d749adbeb2405bd6877c22b3200d183cd37e50","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-26 13:43:51.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"4301f42a1908a147b30865cbf73b0e5d405cac7d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-26 13:44:08.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"b54f2c811e22389fef3c914278af828a5273ea4f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-26 16:26:55.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"6968309a0fc3e8b8046cd7952cc0ce882ad0a16c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 16:36:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/32a0e69/ : SUCCESS in 2h 30m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/aaca31a/ : SUCCESS in 2h 43m 54s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/a3376a3/ : SUCCESS in 1h 32m 22s","accounts_in_message":[],"_revision_number":7},{"id":"d432126b25dd5ec0d304ea5a75cf673b6d9fda9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 22:52:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-containers-multinode/1a5dca7/ : SUCCESS in 2h 12m 17s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-undercloud-containers/0a6f824/ : SUCCESS in 1h 28m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9af23e9/ : SUCCESS in 2h 23m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/370a472/ : SUCCESS in 2h 48m 51s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-standalone/f5e57c4/ : FAILURE in 2h 26m 55s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-fedora-28-standalone/bfcd5de/ : SUCCESS in 2h 03m 54s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-standalone-upgrade/781ff3e/ : SUCCESS in 2h 27m 38s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/02586fc/ : SUCCESS in 2h 17m 25s\n- openstack-tox-py35 http://logs.openstack.org/43/665743/7/check/openstack-tox-py35/3f7a1e4/ : SUCCESS in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/43/665743/7/check/openstack-tox-py36/02a990c/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/43/665743/7/check/openstack-tox-py37/7245064/ : SUCCESS in 4m 28s\n- openstack-tox-docs http://logs.openstack.org/43/665743/7/check/openstack-tox-docs/c34a6cd/html/ : SUCCESS in 3m 14s\n- openstack-tox-pep8 http://logs.openstack.org/43/665743/7/check/openstack-tox-pep8/649127a/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/43/665743/7/check/openstack-tox-py27/04eff3e/ : SUCCESS in 3m 25s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/36fcae7/ : SUCCESS in 2h 28m 53s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/d74233e/ : SUCCESS in 2h 29m 44s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario001-standalone/bba2b73/ : SUCCESS in 1h 47m 30s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario002-standalone/7c8ea4c/ : SUCCESS in 1h 47m 00s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario003-standalone/1c05278/ : SUCCESS in 1h 23m 46s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario004-standalone/617d928/ : SUCCESS in 1h 56m 09s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario012-standalone/a16e776/ : FAILURE in 26m 34s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/774554f/ : FAILURE in 1h 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"71d15306ef3fead21a1fdeb8065d7b4dfa24829d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-27 11:21:41.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"24eefaf4f952fc63577335e17454af91bb53f5d4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 14:18:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6b7e431/ : FAILURE in 2h 03m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c0cff0e/ : SUCCESS in 2h 54m 51s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 finger://ze06.softwarefactory-project.io:7979/9afabf07693b43a894c319c92768fefb : POST_FAILURE in 2h 18m 00s","accounts_in_message":[],"_revision_number":7},{"id":"3a7f84272786f2eb7627b579dd86e6a9d41a3a16","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-27 16:14:36.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"99f80768786b765e6722293822cc5bcd456c33b4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 18:56:05.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e4de270/ : FAILURE in 1h 37m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/51635f0/ : SUCCESS in 2h 38m 05s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/80374f7/ : SUCCESS in 1h 23m 56s","accounts_in_message":[],"_revision_number":7},{"id":"8b7b0826dbc56fb86d4b339ff7d3cec1a40b5c4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 21:27:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-containers-multinode/5bd723a/ : SUCCESS in 2h 13m 55s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-undercloud-containers/b1aae83/ : SUCCESS in 1h 32m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6dfe603/ : SUCCESS in 2h 28m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/74e7481/ : SUCCESS in 2h 33m 15s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-standalone/e040059/ : SUCCESS in 2h 04m 31s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-fedora-28-standalone/fd51aa3/ : SUCCESS in 2h 02m 42s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-standalone-upgrade/79d09a0/ : SUCCESS in 2h 21m 34s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/7c7f779/ : SUCCESS in 2h 31m 54s\n- openstack-tox-py35 http://logs.openstack.org/43/665743/7/check/openstack-tox-py35/98f6fa1/ : SUCCESS in 3m 42s\n- openstack-tox-py36 http://logs.openstack.org/43/665743/7/check/openstack-tox-py36/a202d0b/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/43/665743/7/check/openstack-tox-py37/986a829/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/43/665743/7/check/openstack-tox-docs/4f097f1/html/ : SUCCESS in 2m 56s\n- openstack-tox-pep8 http://logs.openstack.org/43/665743/7/check/openstack-tox-pep8/c80f7c1/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/43/665743/7/check/openstack-tox-py27/5eaac6f/ : SUCCESS in 4m 02s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/8b1b0ed/ : SUCCESS in 2h 32m 07s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/60f1876/ : SUCCESS in 2h 22m 52s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario001-standalone/3786b4a/ : SUCCESS in 1h 41m 48s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario002-standalone/0e5d491/ : SUCCESS in 1h 42m 22s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario003-standalone/e916a1c/ : SUCCESS in 1h 26m 29s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario004-standalone/c615cb6/ : SUCCESS in 1h 56m 00s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-scenario012-standalone/d4f9fd8/ : FAILURE in 21m 41s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/43/665743/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/43b550a/ : FAILURE in 1h 31m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a541f0b84e6de23f0420aa109c0ccb33ea42c0ac","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-28 15:01:15.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dd852222c89032e49be2ec5a3989ec270b9fafb3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-28 18:51:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c144c960f4e812912847805a26b42ca2ec5e98c6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-28 19:32:57.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"09fc0bf982d3840017c8cf2cb855a24a956aed13","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-02 12:49:06.000000000","message":"Patch Set 7:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":7},{"id":"4887471f31fd90a9e622de5d9db3520a40040b5d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-02 14:09:38.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/645d8ff/ : FAILURE in 1h 09m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2ccbe3a/ : FAILURE in 1h 14m 11s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/74279e8/ : SUCCESS in 1h 18m 30s","accounts_in_message":[],"_revision_number":7},{"id":"2d672676ad3926d3df2a65aaa23a23bf8f0d12cf","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-04 13:52:46.000000000","message":"Patch Set 7:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":7},{"id":"22be726aa17dca43beeb1e44728ae648ac21aadb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-04 17:09:42.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c534fe0/ : RETRY_LIMIT in 4m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1e5bc9d/ : SUCCESS in 3h 15m 21s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/43/665743/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/d6eb8e2/ : SUCCESS in 1h 27m 59s","accounts_in_message":[],"_revision_number":7},{"id":"99dd6af9d77664811d48e7b06db8ce2c95370cb2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-04 19:38:42.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"595d7b19ba5f62e72d4422f26c9c633992b97e8b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-04 22:13:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/43/665743/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4ab0c6a/ : SUCCESS in 2h 30m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/43/665743/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5900082/ : SUCCESS in 2h 33m 25s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/43/665743/8/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/e3a01df/ : SUCCESS in 1h 16m 27s","accounts_in_message":[],"_revision_number":8},{"id":"75b5cfa5fd9b0f9aee65e3ebdf9344ebc38feef7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 22:47:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-containers-multinode/7c67120/ : SUCCESS in 2h 09m 02s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-undercloud-containers/03f57e0/ : SUCCESS in 1h 35m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/0f30ebb/ : SUCCESS in 2h 33m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1e319da/ : SUCCESS in 2h 45m 01s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-standalone/cf935f3/ : SUCCESS in 1h 59m 03s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-fedora-28-standalone/a98b048/ : FAILURE in 30m 54s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-standalone-upgrade/703ba65/ : SUCCESS in 2h 05m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-standalone-upgrade-stein/5bf5aba/ : SUCCESS in 2h 18m 40s\n- openstack-tox-py35 http://logs.openstack.org/43/665743/8/check/openstack-tox-py35/e6720b5/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/43/665743/8/check/openstack-tox-py36/8ef5d7f/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/43/665743/8/check/openstack-tox-py37/39f44b8/ : SUCCESS in 3m 49s\n- openstack-tox-docs http://logs.openstack.org/43/665743/8/check/openstack-tox-docs/d9b453c/html/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/43/665743/8/check/openstack-tox-pep8/70f39a2/ : SUCCESS in 2m 31s\n- openstack-tox-py27 http://logs.openstack.org/43/665743/8/check/openstack-tox-py27/1d43c1f/ : SUCCESS in 3m 20s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/f41f2c5/ : SUCCESS in 2h 17m 11s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/90dbddc/ : SUCCESS in 2h 09m 53s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario001-standalone/f1126bc/ : SUCCESS in 1h 48m 06s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario002-standalone/39f89e0/ : SUCCESS in 1h 52m 06s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario003-standalone/9fde929/ : SUCCESS in 1h 15m 47s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario004-standalone/d79ae7a/ : SUCCESS in 1h 38m 10s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario012-standalone/9b6e894/ : SUCCESS in 1h 46m 01s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/983b777/ : FAILURE in 2h 23m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"45bb381b29089d226c2de7970080750077a8dd47","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-08 03:55:35.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"ce8552df99b4e2a382a92c6922e47e8f940ac95c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-08 03:59:44.000000000","message":"Topic changed from container/exists to bug/1833081","accounts_in_message":[],"_revision_number":8},{"id":"b369a0802c4ec03d1db73b40dcf7f340bd377e85","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-08 03:59:55.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"987db85c649d6cde0fd5942b9def44061f8d66c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 04:00:06.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"7f5831d32fa6c0be2389d2bc4b50a1b233b3d676","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 06:43:39.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/43/665743/8/gate/tripleo-ci-centos-7-containers-multinode/b345cf7/ : SUCCESS in 1h 56m 53s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/43/665743/8/gate/tripleo-ci-centos-7-standalone/2fcaf2d/ : FAILURE in 1h 44m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/43/665743/8/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e645658/ : SUCCESS in 2h 03m 32s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/43/665743/8/gate/tripleo-ci-centos-7-standalone-upgrade-stein/ef6a959/ : SUCCESS in 2h 05m 00s\n- openstack-tox-py35 http://logs.openstack.org/43/665743/8/gate/openstack-tox-py35/56de25b/ : SUCCESS in 4m 51s\n- openstack-tox-py36 http://logs.openstack.org/43/665743/8/gate/openstack-tox-py36/5ccee09/ : SUCCESS in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/43/665743/8/gate/openstack-tox-py37/f5775a2/ : SUCCESS in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/43/665743/8/gate/openstack-tox-docs/438c879/html/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/43/665743/8/gate/openstack-tox-pep8/cacf613/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/43/665743/8/gate/openstack-tox-py27/d54656e/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":8},{"id":"a9b54c3afdc1dcdbdc1a55bad0534557a44fe515","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-08 07:58:15.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"488e289c6311a6d03d43392a00d57a2983bd695a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 10:45:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/43/665743/8/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/8f08499/ : SUCCESS in 2h 46m 25s","accounts_in_message":[],"_revision_number":8},{"id":"6adaabc32e0a2a79379a5096983f3d3cb9c9ed19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 12:30:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-containers-multinode/cea032d/ : SUCCESS in 2h 22m 54s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-undercloud-containers/57226b0/ : SUCCESS in 1h 33m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b45fb4d/ : SUCCESS in 2h 22m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8c498c6/ : SUCCESS in 2h 39m 45s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-standalone/5b3d5fe/ : SUCCESS in 2h 00m 36s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-fedora-28-standalone/d8b61bd/ : SUCCESS in 1h 53m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-standalone-upgrade/f83f9eb/ : SUCCESS in 2h 22m 17s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-standalone-upgrade-stein/f356cfa/ : SUCCESS in 2h 29m 40s\n- openstack-tox-py35 http://logs.openstack.org/43/665743/8/check/openstack-tox-py35/4754a1a/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/43/665743/8/check/openstack-tox-py36/67f3c68/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/43/665743/8/check/openstack-tox-py37/b6ad6d5/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/43/665743/8/check/openstack-tox-docs/9c8cbdf/html/ : SUCCESS in 2m 26s\n- openstack-tox-pep8 http://logs.openstack.org/43/665743/8/check/openstack-tox-pep8/98a1ea1/ : SUCCESS in 2m 35s\n- openstack-tox-py27 http://logs.openstack.org/43/665743/8/check/openstack-tox-py27/7b5b198/ : SUCCESS in 3m 33s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/5105e22/ : SUCCESS in 2h 25m 44s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/013f9d5/ : SUCCESS in 2h 08m 29s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario001-standalone/9731708/ : SUCCESS in 1h 57m 41s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario002-standalone/e304bc2/ : SUCCESS in 1h 38m 20s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario003-standalone/b646089/ : SUCCESS in 1h 21m 21s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario004-standalone/0a197fe/ : SUCCESS in 1h 45m 28s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-scenario012-standalone/df17882/ : SUCCESS in 2h 04m 21s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/43/665743/8/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/2078437/ : FAILURE in 2h 12m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7b191e3239f8edf3340a9c57283ee4178c32c152","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 12:30:32.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"ce5062091a8d944b1be7ae5f8c5d4729e4378ba9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 14:07:41.000000000","message":"Patch Set 8:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c0087bb722d3134e858f9e018b330f6176482684","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 18:03:02.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/43/665743/8/gate/tripleo-ci-centos-7-containers-multinode/9e6f5f4/ : SUCCESS in 2h 05m 16s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/43/665743/8/gate/tripleo-ci-centos-7-standalone/62172e6/ : SUCCESS in 2h 17m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/43/665743/8/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/cae17a9/ : SUCCESS in 2h 19m 44s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/43/665743/8/gate/tripleo-ci-centos-7-standalone-upgrade-stein/491040e/ : SUCCESS in 2h 26m 15s\n- openstack-tox-py35 http://logs.openstack.org/43/665743/8/gate/openstack-tox-py35/6f00927/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/43/665743/8/gate/openstack-tox-py36/c95ac14/ : SUCCESS in 4m 09s\n- openstack-tox-py37 http://logs.openstack.org/43/665743/8/gate/openstack-tox-py37/870163f/ : SUCCESS in 4m 17s\n- openstack-tox-docs http://logs.openstack.org/43/665743/8/gate/openstack-tox-docs/8c0a05b/html/ : SUCCESS in 3m 24s\n- openstack-tox-pep8 http://logs.openstack.org/43/665743/8/gate/openstack-tox-pep8/3ad0d90/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/43/665743/8/gate/openstack-tox-py27/b79ef39/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":8},{"id":"b6b5bfb85f22dd7a50de50c3a6557737f769c5b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 18:03:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"277eb726b97a0bb31137e33fd3e5d1a7a4784188","revisions":{"f51cc76c8d1bfa0744801012e4e6956ce9c40387":{"kind":"REWORK","_number":1,"created":"2019-06-17 18:15:33.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/43/665743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/43/665743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/43/665743/1"}}},"commit":{"parents":[{"commit":"29bb794a0f5796113ccc737f3aaef9d4581f5bf0","subject":"runner: refactor image_exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/29bb794a0f5796113ccc737f3aaef9d4581f5bf0"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:13:49.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:13:49.000000000","tz":-240},"subject":"runner: implement container_exist","message":"runner: implement container_exist\n\nOn podman, we can run \"podman container exists\" command to check if\nwhether or not a container does exist.\n\nThis will be used later when we\u0027ll reduce our usage of \"podman inspect\".\n\nChange-Id: I6dfbfe9da3d216cadaca77a53a0d5223c9cbd57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f51cc76c8d1bfa0744801012e4e6956ce9c40387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f51cc76c8d1bfa0744801012e4e6956ce9c40387"}]},"branch":"refs/heads/master"},"240a441e238e0774a2289339fcff11e257eb5b0d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-21 15:14:12.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/43/665743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/43/665743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/43/665743/2"}}},"commit":{"parents":[{"commit":"ed21d4f6276a2cd892ab70bcefddd9d2e0f774f0","subject":"runner: refactor image_exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/ed21d4f6276a2cd892ab70bcefddd9d2e0f774f0"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:13:49.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-21 15:14:12.000000000","tz":0},"subject":"runner: implement container_exist","message":"runner: implement container_exist\n\nOn podman, we can run \"podman container exists\" command to check if\nwhether or not a container does exist.\n\nThis will be used later when we\u0027ll reduce our usage of \"podman inspect\".\n\nChange-Id: I6dfbfe9da3d216cadaca77a53a0d5223c9cbd57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/240a441e238e0774a2289339fcff11e257eb5b0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/240a441e238e0774a2289339fcff11e257eb5b0d"}]},"branch":"refs/heads/master"},"335dee3098287f02d9f871bf89ce057684e425d9":{"kind":"REWORK","_number":3,"created":"2019-06-25 12:17:43.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/43/665743/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/43/665743/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/43/665743/3"}}},"commit":{"parents":[{"commit":"bd32de1f209e13fd275ae8b96b6489c6a1102f66","subject":"runner: refactor image_exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/bd32de1f209e13fd275ae8b96b6489c6a1102f66"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:13:49.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-25 12:16:39.000000000","tz":-240},"subject":"runner: implement container_exist","message":"runner: implement container_exist\n\nOn podman, we can run \"podman container exists\" command to check if\nwhether or not a container does exist.\n\nThis will be used later when we\u0027ll reduce our usage of \"podman inspect\".\n\nChange-Id: I6dfbfe9da3d216cadaca77a53a0d5223c9cbd57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/335dee3098287f02d9f871bf89ce057684e425d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/335dee3098287f02d9f871bf89ce057684e425d9"}]},"branch":"refs/heads/master"},"8ae1c5e4b3d3f3651cd3e625898a5e19a4ba6937":{"kind":"REWORK","_number":4,"created":"2019-06-25 17:11:42.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/43/665743/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/43/665743/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/43/665743/4"}}},"commit":{"parents":[{"commit":"a7a6d823abf7ae6112684e3253f5849059aaf855","subject":"runner: refactor image_exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/a7a6d823abf7ae6112684e3253f5849059aaf855"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:13:49.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-25 17:11:37.000000000","tz":-240},"subject":"runner: implement container_exist","message":"runner: implement container_exist\n\nOn podman, we can run \"podman container exists\" command to check if\nwhether or not a container does exist.\n\nThis will be used later when we\u0027ll reduce our usage of \"podman inspect\".\n\nChange-Id: I6dfbfe9da3d216cadaca77a53a0d5223c9cbd57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/8ae1c5e4b3d3f3651cd3e625898a5e19a4ba6937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/8ae1c5e4b3d3f3651cd3e625898a5e19a4ba6937"}]},"branch":"refs/heads/master"},"837ea6725bf7e72e6f323a45fd45a0f943f090af":{"kind":"REWORK","_number":5,"created":"2019-06-26 13:42:15.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/43/665743/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/43/665743/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/43/665743/5"}}},"commit":{"parents":[{"commit":"a7a6d823abf7ae6112684e3253f5849059aaf855","subject":"runner: refactor image_exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/a7a6d823abf7ae6112684e3253f5849059aaf855"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:13:49.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-26 13:42:13.000000000","tz":0},"subject":"runner: implement container_exist","message":"runner: implement container_exist\n\nOn podman, we can run \"podman container exists\" command to check if\nwhether or not a container does exist.\n\nThis will be used later when we\u0027ll reduce our usage of \"podman inspect\".\n\nChange-Id: I6dfbfe9da3d216cadaca77a53a0d5223c9cbd57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/837ea6725bf7e72e6f323a45fd45a0f943f090af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/837ea6725bf7e72e6f323a45fd45a0f943f090af"}]},"branch":"refs/heads/master"},"dbfaf34e0906bb22e03f55abab0f3f4517e7cf8f":{"kind":"REWORK","_number":6,"created":"2019-06-26 13:42:37.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/43/665743/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/43/665743/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/43/665743/6"}}},"commit":{"parents":[{"commit":"a7a6d823abf7ae6112684e3253f5849059aaf855","subject":"runner: refactor image_exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/a7a6d823abf7ae6112684e3253f5849059aaf855"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:13:49.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-26 13:42:35.000000000","tz":0},"subject":"runner: implement container_exist","message":"runner: implement container_exist\n\nOn podman, we can run \"podman container exists\" command to check if\nwhether or not a container does exist.\n\nThis will be used later when we\u0027ll reduce our usage of \"podman inspect\".\n\nChange-Id: I6dfbfe9da3d216cadaca77a53a0d5223c9cbd57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/dbfaf34e0906bb22e03f55abab0f3f4517e7cf8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/dbfaf34e0906bb22e03f55abab0f3f4517e7cf8f"}]},"branch":"refs/heads/master"},"5fbe083237d5096416d7f085401f891875997c22":{"kind":"REWORK","_number":7,"created":"2019-06-26 13:44:08.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/43/665743/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/43/665743/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/43/665743/7"}}},"commit":{"parents":[{"commit":"747a90ac21540e35d18e8cd70472a9b103901c64","subject":"runner: refactor image_exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/747a90ac21540e35d18e8cd70472a9b103901c64"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:13:49.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-26 13:43:58.000000000","tz":-240},"subject":"runner: implement container_exist","message":"runner: implement container_exist\n\nOn podman, we can run \"podman container exists\" command to check if\nwhether or not a container does exist.\n\nThis will be used later when we\u0027ll reduce our usage of \"podman inspect\".\n\nChange-Id: I6dfbfe9da3d216cadaca77a53a0d5223c9cbd57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5fbe083237d5096416d7f085401f891875997c22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5fbe083237d5096416d7f085401f891875997c22"}]},"branch":"refs/heads/master"},"277eb726b97a0bb31137e33fd3e5d1a7a4784188":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-07-04 19:38:42.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/43/665743/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/43/665743/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/43/665743/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/43/665743/8"}}},"commit":{"parents":[{"commit":"43468616c89e695f65befd4d49b8c1c7bdcd0b2c","subject":"runner: refactor image_exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/43468616c89e695f65befd4d49b8c1c7bdcd0b2c"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 18:13:49.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-07-04 19:38:42.000000000","tz":0},"subject":"runner: implement container_exist","message":"runner: implement container_exist\n\nOn podman, we can run \"podman container exists\" command to check if\nwhether or not a container does exist.\n\nThis will be used later when we\u0027ll reduce our usage of \"podman inspect\".\n\nChange-Id: I6dfbfe9da3d216cadaca77a53a0d5223c9cbd57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/277eb726b97a0bb31137e33fd3e5d1a7a4784188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/277eb726b97a0bb31137e33fd3e5d1a7a4784188"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
