)]}'
{"id":"openstack%2Fpaunch~670983","triplet_id":"openstack%2Fpaunch~master~Ifb3f595210b673c33956ec10b3576e9b23641451","project":"openstack/paunch","branch":"master","hashtags":[],"change_id":"Ifb3f595210b673c33956ec10b3576e9b23641451","subject":"Bump openstackdocstheme to 1.20.0","status":"ABANDONED","created":"2019-07-16 08:50:44.000000000","updated":"2019-07-16 09:24:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3dac89fcf6f29f24e136009aa297218c27bd6486","_number":670983,"virtual_id_number":670983,"owner":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-07-16 09:15:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 09:15:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3451e3627bca3d6b6fb82a971a29a24291106dbf","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2019-07-16 08:50:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"acf38feac17d5a5a5d6bcb0178a027bf8d4a2649","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 09:15:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/83/670983/1/check/openstack-tox-docs/0072b9e/ : POST_FAILURE in 2m 25s\n- openstack-tox-pep8 http://logs.openstack.org/83/670983/1/check/openstack-tox-pep8/927ab58/ : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3dac89fcf6f29f24e136009aa297218c27bd6486","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2019-07-16 09:24:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"93ff0d61b00b5772d5b693eb6ca60278b5d9c9df","revisions":{"93ff0d61b00b5772d5b693eb6ca60278b5d9c9df":{"kind":"REWORK","_number":1,"created":"2019-07-16 08:50:44.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/83/670983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/83/670983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/83/670983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/83/670983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/83/670983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/83/670983/1"}}},"commit":{"parents":[{"commit":"a43eeb616e69b13ba95bf5b1222de7cd5e48b7c9","subject":"Bump openstackdocstheme to 1.20.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/a43eeb616e69b13ba95bf5b1222de7cd5e48b7c9"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2019-07-16 08:49:44.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2019-07-16 08:49:52.000000000","tz":-240},"subject":"Bump openstackdocstheme to 1.20.0","message":"Bump openstackdocstheme to 1.20.0\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Ifb3f595210b673c33956ec10b3576e9b23641451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/93ff0d61b00b5772d5b693eb6ca60278b5d9c9df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/93ff0d61b00b5772d5b693eb6ca60278b5d9c9df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
