)]}'
{"id":"openstack%2Fpaunch~672239","triplet_id":"openstack%2Fpaunch~master~Ibc458d3badb807a0deef5a799db6682d7128dbc1","project":"openstack/paunch","branch":"master","hashtags":[],"change_id":"Ibc458d3badb807a0deef5a799db6682d7128dbc1","subject":"Check if volume paths exist before executing Docker.","status":"MERGED","created":"2019-07-23 09:17:10.000000000","updated":"2019-07-30 14:55:45.000000000","submitted":"2019-07-29 18:47:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672239-1564426060183-66f2565c","meta_rev_id":"323196b0cc379f3989ae57a4d0f85312638ad709","_number":672239,"virtual_id_number":672239,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"date":"2019-07-24 12:37:04.000000000","_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"value":2,"date":"2019-07-29 18:47:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-23 14:07:43.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8958,"name":"Jeremy Agee","email":"jagee@redhat.com"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-24 15:31:17.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-29 14:56:49.000000000","_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8958,"name":"Jeremy Agee","email":"jagee@redhat.com"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-29 15:00:14.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"date":"2019-07-23 18:03:15.000000000","_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8958,"name":"Jeremy Agee","email":"jagee@redhat.com"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8958,"name":"Jeremy Agee","email":"jagee@redhat.com"},{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 09:17:53.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2019-07-23 09:18:10.000000000","updated_by":{"_account_id":8958,"name":"Jeremy Agee","email":"jagee@redhat.com"},"reviewer":{"_account_id":8958,"name":"Jeremy Agee","email":"jagee@redhat.com"},"state":"REVIEWER"},{"updated":"2019-07-23 09:20:26.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-23 09:37:37.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-23 14:07:43.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-23 18:03:15.000000000","updated_by":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"reviewer":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"state":"REVIEWER"},{"updated":"2019-07-29 14:56:49.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2019-07-29 15:00:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-29 18:47:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d79c58bae1a695aa0e6c00deaeec78f6f13020ee","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-23 09:17:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a40a4e8709718271db65d3dd29a8b2e4f811be8","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-23 09:30:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"49ea4f469dbbbd51cf28da8ae82d9874dcef3245","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-23 09:32:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a2e0e3d04c28d77e36b09bc7194c0d2f963606df","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 11:40:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/39/672239/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/93db461/ : FAILURE in 2h 06m 03s","accounts_in_message":[],"_revision_number":3},{"id":"fb16a7965a224ef5e96a3a0ba303d36b84be56f4","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-23 11:51:55.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"10b434d300210a3bb36fee1727904ab3fad631ec","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-23 11:53:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a99d8a4ac5263e800f29bd9f1bd1acffc8e473b7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 14:07:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/39/672239/4/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/4669ddb/ : SUCCESS in 2h 12m 39s","accounts_in_message":[],"_revision_number":4},{"id":"a936a45fea4a00d2486dd6cbb08b75b5f2c96529","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:48:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-containers-multinode/25175c1/ : SUCCESS in 2h 22m 38s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-undercloud-containers/dc3fa94/ : SUCCESS in 1h 34m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/48dc615/ : SUCCESS in 2h 14m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/89987c2/ : SUCCESS in 2h 27m 03s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-standalone/bb93124/ : SUCCESS in 2h 28m 07s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-fedora-28-standalone/64334f5/ : SUCCESS in 1h 43m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-standalone-upgrade/237ddf0/ : FAILURE in 1h 48m 01s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/39/672239/4/check/openstack-tox-py35/9f64f03/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/39/672239/4/check/openstack-tox-py36/27ad4b3/ : SUCCESS in 4m 08s\n- openstack-tox-py37 http://logs.openstack.org/39/672239/4/check/openstack-tox-py37/e03978b/ : SUCCESS in 4m 06s\n- openstack-tox-docs http://logs.openstack.org/39/672239/4/check/openstack-tox-docs/d23811c/html/ : SUCCESS in 2m 31s\n- openstack-tox-pep8 http://logs.openstack.org/39/672239/4/check/openstack-tox-pep8/7f191a4/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/39/672239/4/check/openstack-tox-py27/053be51/ : SUCCESS in 4m 14s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/e6e9e06/ : SUCCESS in 2h 20m 05s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/d6ad60c/ : FAILURE in 1h 27m 30s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario001-standalone/776af78/ : FAILURE in 1h 15m 33s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario002-standalone/bce5f33/ : SUCCESS in 1h 46m 58s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario003-standalone/8d35e00/ : SUCCESS in 1h 22m 03s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario004-standalone/82665b6/ : SUCCESS in 1h 45m 27s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario012-standalone/3082667/ : SUCCESS in 1h 53m 33s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/f779df2/ : FAILURE in 2h 42m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"955d1f258eafa7392e6339cc099b6433a14f54c6","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-24 12:37:04.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"33517c554abdec4953b1aa7b8edcf1bf091408ef","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-24 15:00:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/39/672239/4/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/5db3852/ : SUCCESS in 2h 21m 10s","accounts_in_message":[],"_revision_number":4},{"id":"c8b02b212b135ede2315bc1cd18f6e7110dddecc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 15:30:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-containers-multinode/9579b57/ : SUCCESS in 2h 26m 19s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-undercloud-containers/6e73335/ : SUCCESS in 1h 51m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ca34201/ : SUCCESS in 2h 11m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9e57e5b/ : SUCCESS in 2h 35m 31s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-standalone/a51261f/ : SUCCESS in 1h 42m 10s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-fedora-28-standalone/7c1f93b/ : FAILURE in 1h 39m 24s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-standalone-upgrade/ee78472/ : FAILURE in 2h 08m 04s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/39/672239/4/check/openstack-tox-py35/2adfe4e/ : SUCCESS in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/39/672239/4/check/openstack-tox-py36/6988897/ : SUCCESS in 5m 03s\n- openstack-tox-py37 http://logs.openstack.org/39/672239/4/check/openstack-tox-py37/6860c1e/ : SUCCESS in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/39/672239/4/check/openstack-tox-docs/331b0d8/html/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/39/672239/4/check/openstack-tox-pep8/6749914/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/39/672239/4/check/openstack-tox-py27/5801c17/ : SUCCESS in 3m 47s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/3a1b0fd/ : SUCCESS in 2h 29m 43s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/2d5dfad/ : FAILURE in 1h 43m 30s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario001-standalone/9e5c273/ : SUCCESS in 2h 21m 16s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario002-standalone/163d1f6/ : SUCCESS in 1h 36m 22s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario003-standalone/375f9d4/ : SUCCESS in 1h 19m 48s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario004-standalone/451ab44/ : SUCCESS in 1h 58m 26s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-scenario012-standalone/1978c18/ : SUCCESS in 1h 53m 12s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/39/672239/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/e64ca8a/ : FAILURE in 2h 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"578ada92832c0ce2a32bbc16f9f469cd63f848b4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-24 15:31:17.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"68178e94d1eeb19a4e33e4f5d5c1e740accd8e29","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2019-07-29 14:56:49.000000000","message":"Patch Set 4: Code-Review+2\n\nnice!","accounts_in_message":[],"_revision_number":4},{"id":"276d1dd228dfa3b90eae0f278f12391c153bb41f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-29 15:00:14.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b858e9ad75c91dab9028c35b35cf3441d7f95546","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 15:00:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b8bb0a6a70b5a568f4cee1e044eef3e5eba0cf47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 18:47:39.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/672239/4/gate/tripleo-ci-centos-7-containers-multinode/4c178ad/ : SUCCESS in 2h 25m 50s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/39/672239/4/gate/tripleo-ci-centos-7-standalone/026ad1d/ : SUCCESS in 2h 04m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/39/672239/4/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e0e56eb/ : SUCCESS in 2h 29m 45s\n- openstack-tox-py35 http://logs.openstack.org/39/672239/4/gate/openstack-tox-py35/9798df7/ : SUCCESS in 5m 40s\n- openstack-tox-py36 http://logs.openstack.org/39/672239/4/gate/openstack-tox-py36/f3e6919/ : SUCCESS in 4m 48s\n- openstack-tox-py37 http://logs.openstack.org/39/672239/4/gate/openstack-tox-py37/89c36e7/ : SUCCESS in 6m 40s\n- openstack-tox-docs http://logs.openstack.org/39/672239/4/gate/openstack-tox-docs/c31aaa4/html/ : SUCCESS in 6m 02s\n- openstack-tox-pep8 http://logs.openstack.org/39/672239/4/gate/openstack-tox-pep8/f918a71/ : SUCCESS in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/39/672239/4/gate/openstack-tox-py27/202501a/ : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":4},{"id":"9dcb841cd43d805dc16e88a71822a01e3b3de8ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 18:47:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"323196b0cc379f3989ae57a4d0f85312638ad709","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-30 14:55:45.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit a85515a89395fbc6c20ff8a0bc4015777750f773","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce","revisions":{"645554bb5fb2d8162d4e1cc2b7a4508426264ce6":{"kind":"REWORK","_number":1,"created":"2019-07-23 09:17:10.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/39/672239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/39/672239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/672239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/672239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/672239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/39/672239/1"}}},"commit":{"parents":[{"commit":"6e2ef79af303866384f8dc4de83967dbfbb704a4","subject":"Reduce the usage of \"podman inspect\" command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/6e2ef79af303866384f8dc4de83967dbfbb704a4"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:17:08.000000000","tz":120},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version\n1.13.1 creates a directory in it\u0027s place. This behavior is different\nfrom Podman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors which\nare very difficult to diagnose.\n\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/645554bb5fb2d8162d4e1cc2b7a4508426264ce6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/645554bb5fb2d8162d4e1cc2b7a4508426264ce6"}]},"branch":"refs/heads/master"},"4923d2669f4ca296eadaf7e144ec053b3526d61d":{"kind":"REWORK","_number":2,"created":"2019-07-23 09:30:10.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/39/672239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/39/672239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/672239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/672239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/672239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/39/672239/2"}}},"commit":{"parents":[{"commit":"6e2ef79af303866384f8dc4de83967dbfbb704a4","subject":"Reduce the usage of \"podman inspect\" command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/6e2ef79af303866384f8dc4de83967dbfbb704a4"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:29:31.000000000","tz":120},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/4923d2669f4ca296eadaf7e144ec053b3526d61d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/4923d2669f4ca296eadaf7e144ec053b3526d61d"}]},"branch":"refs/heads/master"},"8c184db5b67979cc4ed12e9443014481579c988c":{"kind":"REWORK","_number":3,"created":"2019-07-23 09:32:42.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/39/672239/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/39/672239/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/672239/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/672239/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/672239/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/39/672239/3"}}},"commit":{"parents":[{"commit":"6e2ef79af303866384f8dc4de83967dbfbb704a4","subject":"Reduce the usage of \"podman inspect\" command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/6e2ef79af303866384f8dc4de83967dbfbb704a4"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:32:22.000000000","tz":120},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/8c184db5b67979cc4ed12e9443014481579c988c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/8c184db5b67979cc4ed12e9443014481579c988c"}]},"branch":"refs/heads/master"},"6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce":{"kind":"REWORK","_number":4,"created":"2019-07-23 11:53:43.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/39/672239/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/39/672239/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/672239/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/672239/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/39/672239/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/39/672239/4"}}},"commit":{"parents":[{"commit":"6e2ef79af303866384f8dc4de83967dbfbb704a4","subject":"Reduce the usage of \"podman inspect\" command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/6e2ef79af303866384f8dc4de83967dbfbb704a4"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 11:53:28.000000000","tz":120},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
