)]}'
{"id":"openstack%2Fpaunch~673561","triplet_id":"openstack%2Fpaunch~stable%2Frocky~Ibc458d3badb807a0deef5a799db6682d7128dbc1","project":"openstack/paunch","branch":"stable/rocky","topic":"bug/1830734","hashtags":[],"change_id":"Ibc458d3badb807a0deef5a799db6682d7128dbc1","subject":"Check if volume paths exist before executing Docker.","status":"MERGED","created":"2019-07-30 15:31:47.000000000","updated":"2019-09-12 15:56:07.000000000","submitted":"2019-08-12 16:52:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673561-1565628728779-d21f4b70","meta_rev_id":"8bdde183351b9c7c05aaa862b62ce7f45baa40ff","_number":673561,"virtual_id_number":673561,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2019-08-12 16:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":1,"date":"2019-08-09 19:01:09.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"recommended":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-12 13:22:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-08-12 13:19:36.000000000","_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-12 13:25:21.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-12 13:25:21.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-09 00:04:48.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-09 19:01:09.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-12 13:19:36.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2019-08-12 13:22:46.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-08-12 13:25:21.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-08-12 16:52:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2a986f3c7a599f6e3101e70b38786ea20188987","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-30 15:31:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5556b922f56d59d5666b8b09f238f1c6add3b850","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-30 16:08:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/61/673561/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/179d71f/ : FAILURE in 20m 21s","accounts_in_message":[],"_revision_number":1},{"id":"88503e1dfb3036510fb180d16a89a147bca53a12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 21:54:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/61/673561/1/check/tripleo-ci-centos-7-containers-multinode/f560972/ : FAILURE in 28m 35s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/61/673561/1/check/tripleo-ci-centos-7-undercloud-containers/8ae2fb1/ : FAILURE in 23m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/61/673561/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e69eb24/ : FAILURE in 30m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/61/673561/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ee24f4d/ : FAILURE in 24m 41s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/61/673561/1/check/tripleo-ci-centos-7-standalone-rocky/14a8e5d/ : RETRY_LIMIT in 5m 41s\n- openstack-tox-py35 http://logs.openstack.org/61/673561/1/check/openstack-tox-py35/de4caf1/ : FAILURE in 32m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/673561/1/check/build-openstack-sphinx-docs/f3f8d99/html/ : SUCCESS in 3m 24s\n- openstack-tox-pep8 http://logs.openstack.org/61/673561/1/check/openstack-tox-pep8/ddb0b19/ : FAILURE in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/61/673561/1/check/openstack-tox-py27/5aa4f14/ : FAILURE in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"84cf57a28c0443121712574b4d538f95dca62991","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-31 07:24:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe9ea6c2b0078dac45a5f00d44ad2fbd2be7ae29","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-31 08:34:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/61/673561/2/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/bd8bfbd/ : FAILURE in 1h 09m 34s","accounts_in_message":[],"_revision_number":2},{"id":"652c25bafbfb821c31a6365b76bb23d281ba868b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 09:49:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/61/673561/2/check/tripleo-ci-centos-7-containers-multinode/6ff79b2/ : FAILURE in 1h 19m 15s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/61/673561/2/check/tripleo-ci-centos-7-undercloud-containers/53c61db/ : FAILURE in 1h 25m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/61/673561/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4f3e843/ : FAILURE in 1h 26m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/61/673561/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/819c34a/ : FAILURE in 1h 21m 09s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/61/673561/2/check/tripleo-ci-centos-7-standalone-rocky/dbffcb4/ : FAILURE in 1h 11m 13s\n- openstack-tox-py35 http://logs.openstack.org/61/673561/2/check/openstack-tox-py35/3354571/ : SUCCESS in 5m 49s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/673561/2/check/build-openstack-sphinx-docs/38c58cc/html/ : SUCCESS in 4m 08s\n- openstack-tox-pep8 http://logs.openstack.org/61/673561/2/check/openstack-tox-pep8/027e7a2/ : FAILURE in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/61/673561/2/check/openstack-tox-py27/5e10c84/ : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"72af794d4ee9e5d4df9a8105377ae30e44e5c800","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-31 10:44:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"317b2ec533db4ee700bf152d2f12c578916b4d91","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-31 10:44:56.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"10d161f6546698d402a66562e12e991d6f3b0687","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-31 11:55:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/61/673561/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/bd5ad0e/ : FAILURE in 1h 08m 24s","accounts_in_message":[],"_revision_number":3},{"id":"1058e5b4cffed50ff456cb4379530729075d1db7","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-31 12:09:16.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"b5518da6feaf4aeb4fb03fd5961a271736605e91","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-31 13:30:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/61/673561/4/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/6e83334/ : FAILURE in 1h 18m 08s","accounts_in_message":[],"_revision_number":4},{"id":"d8fa6f1c4c010cd42c0df73d09ca41d7c4ed1519","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 16:02:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/61/673561/4/check/tripleo-ci-centos-7-containers-multinode/7058970/ : FAILURE in 1h 26m 13s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/61/673561/4/check/tripleo-ci-centos-7-undercloud-containers/781c158/ : FAILURE in 1h 22m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/61/673561/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/103e727/ : FAILURE in 1h 36m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/61/673561/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/64640e2/ : FAILURE in 1h 23m 16s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/61/673561/4/check/tripleo-ci-centos-7-standalone-rocky/90e894d/ : FAILURE in 1h 13m 39s\n- openstack-tox-py35 http://logs.openstack.org/61/673561/4/check/openstack-tox-py35/af532f6/ : SUCCESS in 4m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/673561/4/check/build-openstack-sphinx-docs/240c359/html/ : SUCCESS in 5m 37s\n- openstack-tox-pep8 http://logs.openstack.org/61/673561/4/check/openstack-tox-pep8/d9f6419/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/61/673561/4/check/openstack-tox-py27/32d312c/ : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":4},{"id":"82ab71aaee8f05bb527772ad2ef1e19da9b47a99","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-08-05 16:34:19.000000000","message":"Topic set to bug/1830734","accounts_in_message":[],"_revision_number":4},{"id":"0662649dcc4fde0604ca18d54e3691f74fbcdba5","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-08-06 08:18:20.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"75d354fbd0faee1ada489918d5702c688bc898fd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 09:42:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/61/673561/5/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/3894973/ : FAILURE in 1h 21m 45s","accounts_in_message":[],"_revision_number":5},{"id":"3e90b59502b300a5490c2d5a436e4c05b0e2fa6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 12:33:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/61/673561/5/check/tripleo-ci-centos-7-containers-multinode/71a921d/ : FAILURE in 1h 31m 20s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/61/673561/5/check/tripleo-ci-centos-7-undercloud-containers/52a7e4d/ : FAILURE in 1h 32m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/61/673561/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8209140/ : FAILURE in 1h 26m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/61/673561/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1df296e/ : FAILURE in 1h 22m 22s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky https://logs.opendev.org/61/673561/5/check/tripleo-ci-centos-7-standalone-rocky/af919c9/ : FAILURE in 1h 25m 46s\n- openstack-tox-py35 https://logs.opendev.org/61/673561/5/check/openstack-tox-py35/907f33d/ : SUCCESS in 4m 26s\n- build-openstack-sphinx-docs https://logs.opendev.org/61/673561/5/check/build-openstack-sphinx-docs/2ec33f7/html/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://logs.opendev.org/61/673561/5/check/openstack-tox-pep8/33dcfb4/ : SUCCESS in 4m 52s\n- openstack-tox-py27 https://logs.opendev.org/61/673561/5/check/openstack-tox-py27/ece97e5/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":5},{"id":"dacc802df9a3d464ca1bd39bfb595d11f5e1a6d4","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-08-07 10:13:46.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"41495fdc9febaa9fd1744489bf9882461f4a02bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 13:46:37.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/61/673561/5/check/tripleo-ci-centos-7-containers-multinode/cfe2815/ : FAILURE in 1h 20m 57s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/61/673561/5/check/tripleo-ci-centos-7-undercloud-containers/3825345/ : FAILURE in 1h 26m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/61/673561/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/17c4305/ : FAILURE in 1h 25m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/61/673561/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8da1333/ : FAILURE in 1h 22m 23s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky https://logs.opendev.org/61/673561/5/check/tripleo-ci-centos-7-standalone-rocky/b8fcb0e/ : FAILURE in 1h 12m 48s\n- openstack-tox-py35 https://logs.opendev.org/61/673561/5/check/openstack-tox-py35/640f6c0/ : SUCCESS in 3m 28s\n- build-openstack-sphinx-docs https://logs.opendev.org/61/673561/5/check/build-openstack-sphinx-docs/1d3b605/html/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://logs.opendev.org/61/673561/5/check/openstack-tox-pep8/852c89e/ : SUCCESS in 4m 33s\n- openstack-tox-py27 https://logs.opendev.org/61/673561/5/check/openstack-tox-py27/115d458/ : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":5},{"id":"61549c4edaddb531c8c97ffb0248f43b95f86b71","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 14:03:36.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/61/673561/5/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/daa3f42/ : TIMED_OUT in 3h 48m 41s","accounts_in_message":[],"_revision_number":5},{"id":"9e4907e87a8a8d93d459e67b0c4d1c0b3170860c","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-08-07 17:39:07.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"e32ecd122c4f1a30e5fd8cfee74d59acd6e6c30b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 20:03:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/61/673561/6/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/0278756/ : SUCCESS in 2h 22m 50s","accounts_in_message":[],"_revision_number":6},{"id":"816d0ed1b8980a0157e7cef4d2a82467241872e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 23:11:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/61/673561/6/check/tripleo-ci-centos-7-containers-multinode/0fd58b9/ : FAILURE in 2h 21m 50s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/61/673561/6/check/tripleo-ci-centos-7-undercloud-containers/e6641e3/ : SUCCESS in 1h 51m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/61/673561/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3029f78/ : SUCCESS in 2h 42m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/61/673561/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/72a9660/ : SUCCESS in 3h 14m 16s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky https://logs.opendev.org/61/673561/6/check/tripleo-ci-centos-7-standalone-rocky/0089cc6/ : FAILURE in 1h 20m 58s\n- openstack-tox-py35 https://logs.opendev.org/61/673561/6/check/openstack-tox-py35/d83eaf7/ : SUCCESS in 3m 57s\n- build-openstack-sphinx-docs https://logs.opendev.org/61/673561/6/check/build-openstack-sphinx-docs/b5caa43/html/ : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://logs.opendev.org/61/673561/6/check/openstack-tox-pep8/1990f02/ : SUCCESS in 3m 49s\n- openstack-tox-py27 https://logs.opendev.org/61/673561/6/check/openstack-tox-py27/d137006/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":6},{"id":"947f61d969dbe4cb689c8c7bd34b411d8ebdc630","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-08-08 07:03:56.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ec4a177d1c0ec2f68c7c99e30535cbba2c6d80a5","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-08-08 07:52:33.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"9592f0e04e6c5c7751f2e394e6f5caf6fda48089","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-08-08 08:04:44.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"4d720c82157e77c1d1353be305e767d99ee73f6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 08:07:39.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"40368b941a2282ab196c42e73051a8eb7febb57b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 08:07:39.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"a3e53fd16802d11c1d5ed3611bfca2ef7cb83ce9","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-08-08 08:16:02.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"1c20cf2d08e897928825e8f633a55d8e549b3457","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 12:28:07.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/61/673561/8/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/115b69b/ : TIMED_OUT in 4h 07m 49s","accounts_in_message":[],"_revision_number":8},{"id":"4c37fc3e9e07c02239a13926e7dc10f9c1489799","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 13:05:11.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/61/673561/8/check/tripleo-ci-centos-7-containers-multinode/5250c4d/ : SUCCESS in 2h 22m 21s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/61/673561/8/check/tripleo-ci-centos-7-undercloud-containers/ca0f25e/ : SUCCESS in 1h 38m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/61/673561/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c8e98f7/ : SUCCESS in 2h 50m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/61/673561/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/08dbf02/ : SUCCESS in 2h 39m 36s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky https://logs.opendev.org/61/673561/8/check/tripleo-ci-centos-7-standalone-rocky/ea20837/ : SUCCESS in 1h 55m 26s\n- openstack-tox-py35 https://logs.opendev.org/61/673561/8/check/openstack-tox-py35/6f40e87/ : SUCCESS in 6m 11s\n- build-openstack-sphinx-docs https://logs.opendev.org/61/673561/8/check/build-openstack-sphinx-docs/679681f/html/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://logs.opendev.org/61/673561/8/check/openstack-tox-pep8/07025b5/ : SUCCESS in 4m 08s\n- openstack-tox-py27 https://logs.opendev.org/61/673561/8/check/openstack-tox-py27/5b56119/ : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":8},{"id":"341f9086d5e4d8d9d1964a34afaa523c2822203e","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-08-08 13:46:43.000000000","message":"Patch Set 8:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":8},{"id":"962cec493b3cc7181ed59d33e072840b42873480","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 16:14:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/61/673561/8/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/5043ce3/ : SUCCESS in 2h 25m 44s","accounts_in_message":[],"_revision_number":8},{"id":"b406383db0e92c592e1328e1a4069ddd4ef81263","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-09 00:04:48.000000000","message":"Patch Set 8: Code-Review-1\n\nPlease look https://review.opendev.org/#/c/675494/ - i\u0027m not sure we want to backport all of this tbh.","accounts_in_message":[],"_revision_number":8},{"id":"17f4d7bd39f16745fb685aacdc55f3b3ea99335b","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2019-08-09 16:43:20.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"448cb78ec2c6865dc05985e51a649109b6b768ec","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-09 19:01:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/61/673561/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/40130d9/ : SUCCESS in 2h 16m 06s","accounts_in_message":[],"_revision_number":9},{"id":"6e348fa84db5e60cb0c8c732b5343545c7b6d4aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 03:03:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/61/673561/9/check/tripleo-ci-centos-7-containers-multinode/54b9c6b/ : SUCCESS in 2h 31m 53s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/61/673561/9/check/tripleo-ci-centos-7-undercloud-containers/b2f343c/ : SUCCESS in 1h 46m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/61/673561/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a23a82b/ : SUCCESS in 2h 44m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/61/673561/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3615f31/ : SUCCESS in 3h 00m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky https://logs.opendev.org/61/673561/9/check/tripleo-ci-centos-7-standalone-rocky/a915498/ : SUCCESS in 1h 48m 10s\n- openstack-tox-py35 https://logs.opendev.org/61/673561/9/check/openstack-tox-py35/19b023e/ : SUCCESS in 3m 06s\n- build-openstack-sphinx-docs https://logs.opendev.org/61/673561/9/check/build-openstack-sphinx-docs/414af3c/html/ : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://logs.opendev.org/61/673561/9/check/openstack-tox-pep8/c9305cc/ : SUCCESS in 4m 46s\n- openstack-tox-py27 https://logs.opendev.org/61/673561/9/check/openstack-tox-py27/3d65e25/ : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":9},{"id":"6bc6a9188c07ab4fb732a158a24985d38f75f075","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2019-08-12 13:19:36.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"1298c331e7ca7a62b205b24e6a8184329b337ec7","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-12 13:22:46.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"a3a5a77b73695568b0cb2c2b130c38c6345d2a92","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-12 13:25:21.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"340289ead26de3b52961b802897cdc318b62e9a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 13:27:03.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"b1a3ff8b3c21eaf297f89949b99c2216c76466e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 16:52:08.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/721e47d42e984d20b0771a86b100d65c : SUCCESS in 2h 38m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/48c75c23e81f4db183465f879008c2b6 : SUCCESS in 2h 52m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e98fb5784ca149aa9be4a8c8ad66bc67 : SUCCESS in 4m 15s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ca2208f69f88403aa7721aebc2bad754 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/521dfbbf041943b5b049c0cb196d588e : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/299e6a0f598c4c138ba5e6eb2ded949c : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":9},{"id":"12fe759ae969aadb680cfabaae7825619a23f6f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 16:52:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"8bdde183351b9c7c05aaa862b62ce7f45baa40ff","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-12 15:56:07.000000000","message":"Created a revert of this change as I8e1c753c22ba1304ab64b64a40cbfc758ac1861f","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"23154b26d57e4f25456b1f43bc5dff5fb373fc9e","revisions":{"61f67fcb479f576cfb5854340c77b494f18b8492":{"kind":"REWORK","_number":1,"created":"2019-07-30 15:31:47.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/61/673561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/61/673561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/61/673561/1"}}},"commit":{"parents":[{"commit":"b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8","subject":"Quote the health check command argument during a debug with print-cmd.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-30 15:31:05.000000000","tz":120},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/61f67fcb479f576cfb5854340c77b494f18b8492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/61f67fcb479f576cfb5854340c77b494f18b8492"}]},"branch":"refs/heads/stable/rocky"},"db1f98531125d0a6454686992475e0ce572d913e":{"kind":"REWORK","_number":2,"created":"2019-07-31 07:24:07.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/61/673561/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/61/673561/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/61/673561/2"}}},"commit":{"parents":[{"commit":"b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8","subject":"Quote the health check command argument during a debug with print-cmd.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-31 07:23:51.000000000","tz":120},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/db1f98531125d0a6454686992475e0ce572d913e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/db1f98531125d0a6454686992475e0ce572d913e"}]},"branch":"refs/heads/stable/rocky"},"dcde389736cb9a775fea9a5d09fe2a7e072fdf52":{"kind":"REWORK","_number":3,"created":"2019-07-31 10:44:45.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/61/673561/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/61/673561/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/61/673561/3"}}},"commit":{"parents":[{"commit":"b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8","subject":"Quote the health check command argument during a debug with print-cmd.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-31 10:33:54.000000000","tz":120},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/dcde389736cb9a775fea9a5d09fe2a7e072fdf52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/dcde389736cb9a775fea9a5d09fe2a7e072fdf52"}]},"branch":"refs/heads/stable/rocky"},"86c1661e0f6558ad47452caf4e5e6bf10e1e9c5a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-07-31 12:09:16.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/61/673561/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/61/673561/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/61/673561/4"}}},"commit":{"parents":[{"commit":"b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8","subject":"Quote the health check command argument during a debug with print-cmd.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-31 11:56:45.000000000","tz":120},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nDepends-On: https://review.opendev.org/673798\nDepends-On: https://review.opendev.org/673797\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/86c1661e0f6558ad47452caf4e5e6bf10e1e9c5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/86c1661e0f6558ad47452caf4e5e6bf10e1e9c5a"}]},"branch":"refs/heads/stable/rocky"},"2012fe021d817cd3ccda02b32ebb5b50c923aabc":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-08-06 08:18:20.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/61/673561/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/61/673561/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/61/673561/5"}}},"commit":{"parents":[{"commit":"b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8","subject":"Quote the health check command argument during a debug with print-cmd.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-08-06 08:18:00.000000000","tz":0},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nDepends-On: https://review.opendev.org/673798\nDepends-On: https://review.opendev.org/673797\nDepends-On: https://review.opendev.org/674648\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2012fe021d817cd3ccda02b32ebb5b50c923aabc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2012fe021d817cd3ccda02b32ebb5b50c923aabc"}]},"branch":"refs/heads/stable/rocky"},"5ff382c15196afa42aed42cfa0d11ea3acb7b216":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-08-07 17:39:07.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/61/673561/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/61/673561/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/61/673561/6"}}},"commit":{"parents":[{"commit":"b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8","subject":"Quote the health check command argument during a debug with print-cmd.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-08-07 17:38:51.000000000","tz":0},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nDepends-On: https://review.opendev.org/675162\nDepends-On: https://review.opendev.org/673798\nDepends-On: https://review.opendev.org/673797\nDepends-On: https://review.opendev.org/674648\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5ff382c15196afa42aed42cfa0d11ea3acb7b216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5ff382c15196afa42aed42cfa0d11ea3acb7b216"}]},"branch":"refs/heads/stable/rocky"},"eb624c935c21ae394daf2cfc76d315a3966e0d08":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-08-08 07:52:33.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/61/673561/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/61/673561/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/61/673561/7"}}},"commit":{"parents":[{"commit":"b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8","subject":"Quote the health check command argument during a debug with print-cmd.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-08-08 07:52:26.000000000","tz":0},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nDepends-On: https://review.opendev.org/675162\nDepends-On: https://review.opendev.org/675262\nDepends-On: https://review.opendev.org/673798\nDepends-On: https://review.opendev.org/673797\nDepends-On: https://review.opendev.org/674648\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/eb624c935c21ae394daf2cfc76d315a3966e0d08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/eb624c935c21ae394daf2cfc76d315a3966e0d08"}]},"branch":"refs/heads/stable/rocky"},"ba5587c92e364c72dcf28a92d2746347d158c22b":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-08-08 08:04:44.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/61/673561/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/61/673561/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/61/673561/8"}}},"commit":{"parents":[{"commit":"b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8","subject":"Quote the health check command argument during a debug with print-cmd.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-08-08 08:04:36.000000000","tz":0},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nDepends-On: https://review.opendev.org/675162\nDepends-On: https://review.opendev.org/675262\nDepends-On: https://review.opendev.org/673798\nDepends-On: https://review.opendev.org/673797\nDepends-On: https://review.opendev.org/674648\nDepends-On: https://review.opendev.org/675265\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/ba5587c92e364c72dcf28a92d2746347d158c22b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/ba5587c92e364c72dcf28a92d2746347d158c22b"}]},"branch":"refs/heads/stable/rocky"},"23154b26d57e4f25456b1f43bc5dff5fb373fc9e":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-08-09 16:43:20.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/61/673561/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/61/673561/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/61/673561/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/61/673561/9"}}},"commit":{"parents":[{"commit":"b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8","subject":"Quote the health check command argument during a debug with print-cmd.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/b1ed7cc2366b8108f0a8f2fd1a08fcabf1525ec8"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-08-09 16:43:11.000000000","tz":0},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nDepends-On: https://review.opendev.org/675162\nDepends-On: https://review.opendev.org/675262\nDepends-On: https://review.opendev.org/673798\nDepends-On: https://review.opendev.org/673797\nDepends-On: https://review.opendev.org/675666\nDepends-On: https://review.opendev.org/675265\n\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/23154b26d57e4f25456b1f43bc5dff5fb373fc9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/23154b26d57e4f25456b1f43bc5dff5fb373fc9e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
