)]}'
{"id":"openstack%2Fpaunch~673565","triplet_id":"openstack%2Fpaunch~stable%2Fqueens~Ibc458d3badb807a0deef5a799db6682d7128dbc1","project":"openstack/paunch","branch":"stable/queens","topic":"bug/1830734-stable/queens","hashtags":[],"change_id":"Ibc458d3badb807a0deef5a799db6682d7128dbc1","subject":"Check if volume paths exist before executing Docker.","status":"MERGED","created":"2019-07-30 15:42:42.000000000","updated":"2019-09-12 15:55:50.000000000","submitted":"2019-08-13 14:59:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673565-1565708367994-e4a4d76b","meta_rev_id":"be36d340785ee713048a7e4eec89f70c694745f5","_number":673565,"virtual_id_number":673565,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2019-08-13 14:59:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-09 19:18:43.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-12 13:22:34.000000000","_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":1,"date":"2019-08-12 18:03:59.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-09 00:05:04.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-09 19:18:43.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-12 13:22:34.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2019-08-12 18:03:59.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-13 14:59:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"646622f55d77682dc753a3c5a8ca87976c8b39e1","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-30 15:42:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a388febb485f7be0545d5e70cf29d7a8c4625b82","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-30 18:33:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/65/673565/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/0aefbd9/ : FAILURE in 17m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ef8329965770c401f13e5dac8d0dc1a7d395b874","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 21:50:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/65/673565/1/check/tripleo-ci-centos-7-containers-multinode/34e3746/ : FAILURE in 31m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/65/673565/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d4b15df/ : FAILURE in 28m 02s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/65/673565/1/check/build-openstack-sphinx-docs/4a90c6a/html/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/65/673565/1/check/openstack-tox-pep8/5cfb0e0/ : FAILURE in 28m 40s\n- openstack-tox-py27 http://logs.openstack.org/65/673565/1/check/openstack-tox-py27/a34e6ee/ : FAILURE in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"eeae99e1223caa6b9d1673ee068bfd9e236435d3","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-31 07:24:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c65e77247a98082762554bafa5d3a64c7c9a8ea2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-31 08:12:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/65/673565/2/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/0ffde90/ : FAILURE in 46m 25s","accounts_in_message":[],"_revision_number":2},{"id":"8fecbfff6d8731ec388922421935311a7433db57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 10:25:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/65/673565/2/check/tripleo-ci-centos-7-containers-multinode/02f1223/ : FAILURE in 1h 56m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/65/673565/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c8d0271/ : FAILURE in 1h 51m 57s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/65/673565/2/check/build-openstack-sphinx-docs/85d1956/html/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/65/673565/2/check/openstack-tox-pep8/2c232ba/ : FAILURE in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/65/673565/2/check/openstack-tox-py27/3e86e56/ : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"60483e2c14f08bdd17955c674a2dbbc106eec60a","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-31 10:44:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1049350d272a47592f39201034b598f497ebdb19","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-31 10:44:19.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"744365fedc62d8708441ad5e9dd86becc76600db","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-31 12:39:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/65/673565/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/d886808/ : FAILURE in 1h 54m 19s","accounts_in_message":[],"_revision_number":3},{"id":"0eaa65e8ed3893881ffc1dd1dc14b1a20233247d","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-07-31 13:07:55.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"a969743e612f6bac7e6be860de8f9b102b8361cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 14:12:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/65/673565/3/check/tripleo-ci-centos-7-containers-multinode/270d799/ : FAILURE in 2h 13m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/65/673565/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/63c7697/ : FAILURE in 2h 00m 03s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/65/673565/3/check/build-openstack-sphinx-docs/0be445a/html/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/65/673565/3/check/openstack-tox-pep8/b32f2c2/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/65/673565/3/check/openstack-tox-py27/a86a0a8/ : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":3},{"id":"a78ecf8c1af398309fea55f7254793fd5e151f2b","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-08-08 14:41:54.000000000","message":"Topic set to bug/1830734-stable/queens","accounts_in_message":[],"_revision_number":3},{"id":"7af7d4665f8a8f1467e2884ce0f38de3ea38dc0c","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-08-08 14:46:37.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"865582898762b6960c152920007f8d6105dcf756","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 17:37:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/65/673565/4/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/b4e2a19/ : SUCCESS in 2h 17m 26s","accounts_in_message":[],"_revision_number":4},{"id":"2575415256b1560f2bcce233ae09c8c71759e201","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 23:02:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/65/673565/4/check/tripleo-ci-centos-7-containers-multinode/3ba8100/ : SUCCESS in 2h 23m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/65/673565/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a40d561/ : FAILURE in 2h 08m 59s (non-voting)\n- build-openstack-sphinx-docs https://logs.opendev.org/65/673565/4/check/build-openstack-sphinx-docs/c274ab1/html/ : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://logs.opendev.org/65/673565/4/check/openstack-tox-pep8/1c7700a/ : SUCCESS in 3m 00s\n- openstack-tox-py27 https://logs.opendev.org/65/673565/4/check/openstack-tox-py27/11505cc/ : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":4},{"id":"55d320bf5ac5dba332e6561ea478d52cbabd4b7f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-09 00:05:04.000000000","message":"Patch Set 4: Code-Review-1\n\nI\u0027m not sure we want to backport it yet, check https://review.opendev.org/#/c/675494/","accounts_in_message":[],"_revision_number":4},{"id":"88dc169bca7bb57e37db605518c97b9aef674dd5","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-08-09 09:19:53.000000000","message":"Patch Set 4:\n\nEmilien, it looks to me like the patch you mention changes paunch behavior to stop executing commands for other containers when one of them fails. I think its unrelated to this patch, which streamlines docker behavior with podman.","accounts_in_message":[],"_revision_number":4},{"id":"ede4d31cbf474721629cf8bb826f3b97aea4e666","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2019-08-09 17:05:57.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f342ae044dde764a13609a7432299f1414717a7a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-09 19:18:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/65/673565/5/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/35a45d8/ : SUCCESS in 2h 10m 30s","accounts_in_message":[],"_revision_number":5},{"id":"267384d80d05fe32ed9d0f3e55efcf3b5e005652","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 02:40:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/65/673565/5/check/tripleo-ci-centos-7-containers-multinode/250a8aa/ : SUCCESS in 2h 34m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/65/673565/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/17d01aa/ : FAILURE in 2h 09m 05s (non-voting)\n- build-openstack-sphinx-docs https://logs.opendev.org/65/673565/5/check/build-openstack-sphinx-docs/2b548c0/html/ : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://logs.opendev.org/65/673565/5/check/openstack-tox-pep8/aa2e458/ : SUCCESS in 4m 06s\n- openstack-tox-py27 https://logs.opendev.org/65/673565/5/check/openstack-tox-py27/7435854/ : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":5},{"id":"8c2f418dee68c92e7b2bca9a49bbe27ec9297c53","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2019-08-12 13:22:34.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7aebb66276913a4252abefd4dd48b56ca7220c98","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-12 18:03:59.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f8feafdb94e6d97dafca253d378d771c972937e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 18:04:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e2a2501bb09d0e08bc902bbfcc3c15c2d5d5bced","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 22:43:09.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"fd4cc876f68d25945c4392799d98200e589df3e1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-13 00:06:52.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"579c1df0c1506115ecc70e5a993a65943a9e3b80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 02:19:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e5e3aa6d2bfb4114a471c6653bd3f8ce : SUCCESS in 1h 59m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6e3dd296db65424d86b8777e90714f73 : FAILURE in 1h 59m 43s (non-voting)\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b50a8abf175f40c597a4eeaeebc45d0f : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/317459fb68c741e3a2b3892a2a6918a8 : SUCCESS in 2m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2abf98dabb5e4993a9b2263f47af45eb : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":5},{"id":"68bba9efad96ca78a6aafa97166d18ddc55cbb96","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 02:33:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/65/673565/5/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/e069e09/ : SUCCESS in 2h 24m 52s","accounts_in_message":[],"_revision_number":5},{"id":"36f980ebed5a1d53750432b0e8c436ab2972ec75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 11:24:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c486d00339c6a3486f3731c02ce40eb9020c89b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 14:59:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"796daa16b976cdec53626eb523a3463cc6eb92ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 14:59:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/88c9e904fbb1452983546f4bdd720a8b : SUCCESS in 2h 03m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/90cd58c58647434ba84380284f808ffc : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63e98c85dc974089858f6fceb94b325f : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2743738431c44c585eb326bc3c0988b : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":5},{"id":"be36d340785ee713048a7e4eec89f70c694745f5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-12 15:55:50.000000000","message":"Created a revert of this change as Iff17da926ca5bab78bec3ab7245109e9b1dca748","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"85c1bc9d15acaaf9156991ea57b17d0922329af5","revisions":{"230405b84ab8e433260974c6b2d7a037e8cdad0c":{"kind":"REWORK","_number":1,"created":"2019-07-30 15:42:42.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/65/673565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/65/673565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/65/673565/1"}}},"commit":{"parents":[{"commit":"90cf8bbb3e0a5baf4dbee519ffacfddf04cdb7b0","subject":"Revert \"docker/compose: quote health-cmd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/90cf8bbb3e0a5baf4dbee519ffacfddf04cdb7b0"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-30 15:41:58.000000000","tz":120},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/230405b84ab8e433260974c6b2d7a037e8cdad0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/230405b84ab8e433260974c6b2d7a037e8cdad0c"}]},"branch":"refs/heads/stable/queens"},"53d0e42c88ef3206415532b376579e2d78ec6e8c":{"kind":"REWORK","_number":2,"created":"2019-07-31 07:24:23.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/65/673565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/65/673565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/65/673565/2"}}},"commit":{"parents":[{"commit":"90cf8bbb3e0a5baf4dbee519ffacfddf04cdb7b0","subject":"Revert \"docker/compose: quote health-cmd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/90cf8bbb3e0a5baf4dbee519ffacfddf04cdb7b0"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-31 07:23:24.000000000","tz":120},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/53d0e42c88ef3206415532b376579e2d78ec6e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/53d0e42c88ef3206415532b376579e2d78ec6e8c"}]},"branch":"refs/heads/stable/queens"},"ef6680fe03960d1f3860a5f1dc29b4a09d1ec586":{"kind":"REWORK","_number":3,"created":"2019-07-31 10:44:07.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/65/673565/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/65/673565/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/65/673565/3"}}},"commit":{"parents":[{"commit":"90cf8bbb3e0a5baf4dbee519ffacfddf04cdb7b0","subject":"Revert \"docker/compose: quote health-cmd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/90cf8bbb3e0a5baf4dbee519ffacfddf04cdb7b0"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-31 10:34:12.000000000","tz":120},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/ef6680fe03960d1f3860a5f1dc29b4a09d1ec586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/ef6680fe03960d1f3860a5f1dc29b4a09d1ec586"}]},"branch":"refs/heads/stable/queens"},"96bc6ad9d080ed37c7da2029270ffc25f41a83b8":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-08-08 14:46:37.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/65/673565/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/65/673565/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/65/673565/4"}}},"commit":{"parents":[{"commit":"90cf8bbb3e0a5baf4dbee519ffacfddf04cdb7b0","subject":"Revert \"docker/compose: quote health-cmd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/90cf8bbb3e0a5baf4dbee519ffacfddf04cdb7b0"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-08-08 14:46:30.000000000","tz":0},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nDepends-On: https://review.opendev.org/675356\nDepends-On: https://review.opendev.org/675357\nDepends-On: https://review.opendev.org/675358\nDepends-On: https://review.opendev.org/675359\nDepends-On: https://review.opendev.org/675354\nDepends-On: https://review.opendev.org/675360 \nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/96bc6ad9d080ed37c7da2029270ffc25f41a83b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/96bc6ad9d080ed37c7da2029270ffc25f41a83b8"}]},"branch":"refs/heads/stable/queens"},"85c1bc9d15acaaf9156991ea57b17d0922329af5":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-08-09 17:05:57.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/65/673565/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/65/673565/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/65/673565/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/65/673565/5"}}},"commit":{"parents":[{"commit":"90cf8bbb3e0a5baf4dbee519ffacfddf04cdb7b0","subject":"Revert \"docker/compose: quote health-cmd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/90cf8bbb3e0a5baf4dbee519ffacfddf04cdb7b0"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-23 09:13:44.000000000","tz":120},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-08-09 17:05:52.000000000","tz":0},"subject":"Check if volume paths exist before executing Docker.","message":"Check if volume paths exist before executing Docker.\n\nIf the file to be mounted doesn\u0027t exist, Docker version 1.13.1\ncreates a directory in it\u0027s place. This behavior is different from\nPodman\u0027s, which throws an error. We want to replicate the Podman\nbehavior. Creating directories in place of files causes errors\nwhich are very difficult to diagnose.\n\nDepends-On: https://review.opendev.org/675356\nDepends-On: https://review.opendev.org/675357\nDepends-On: https://review.opendev.org/675358\nDepends-On: https://review.opendev.org/675359\nDepends-On: https://review.opendev.org/675360\nDepends-On: https://review.opendev.org/675668\nRelated-Bug #1830734\n\nChange-Id: Ibc458d3badb807a0deef5a799db6682d7128dbc1\n(cherry picked from commit 6ca0170eebfad694d6d9e5e0d25f9f1b2fa971ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/85c1bc9d15acaaf9156991ea57b17d0922329af5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/85c1bc9d15acaaf9156991ea57b17d0922329af5"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
