)]}'
{"id":"openstack%2Fpaunch~674090","triplet_id":"openstack%2Fpaunch~stable%2Fstein~Idaf5d4871ad1231f2592238a7925857af8f40548","project":"openstack/paunch","branch":"stable/stein","topic":"podman-drop-in","hashtags":[],"change_id":"Idaf5d4871ad1231f2592238a7925857af8f40548","subject":"Generate addition drop-in dependencies for podman containers","status":"MERGED","created":"2019-08-01 16:53:27.000000000","updated":"2019-08-02 20:27:36.000000000","submitted":"2019-08-02 15:15:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674090-1564758935806-c33df9d0","meta_rev_id":"b8a8465ec7de091098a64dfcc96725a64a394b54","_number":674090,"virtual_id_number":674090,"owner":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":2,"date":"2019-08-02 15:15:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-01 19:26:38.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-02 07:26:12.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-02 12:24:51.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 19:26:38.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-02 07:26:12.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-08-02 12:24:51.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-02 15:15:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae91cd99824ac6bfe6c3f3ca580e5299e832c2f9","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2019-08-01 16:53:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45854f02e844bc6f7dc92d53b46429e2558cbd65","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2019-08-01 17:13:29.000000000","message":"Topic changed from podman-drop-in-stable/stein to podman-drop-in","accounts_in_message":[],"_revision_number":1},{"id":"ac59cc2804311a7e462d2a30bced08475aebd824","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 19:26:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/90/674090/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/b8e2d74/ : SUCCESS in 2h 32m 23s","accounts_in_message":[],"_revision_number":1},{"id":"bbcf2b0c9371cdc7b90ea1cf3ce543b5554f9f19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 03:04:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/90/674090/1/check/tripleo-ci-centos-7-containers-multinode/0a813e0/ : SUCCESS in 2h 17m 26s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/90/674090/1/check/tripleo-ci-centos-7-undercloud-containers/a4a3c96/ : SUCCESS in 1h 38m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/90/674090/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2b2ec81/ : SUCCESS in 2h 16m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/90/674090/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/09f23c2/ : SUCCESS in 2h 34m 08s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/90/674090/1/check/tripleo-ci-centos-7-standalone/cae07c5/ : SUCCESS in 1h 53m 09s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/90/674090/1/check/tripleo-ci-fedora-28-standalone/9118887/ : SUCCESS in 1h 42m 23s (non-voting)\n- openstack-tox-py35 https://logs.opendev.org/90/674090/1/check/openstack-tox-py35/b5189a1/ : SUCCESS in 4m 12s\n- openstack-tox-py36 https://logs.opendev.org/90/674090/1/check/openstack-tox-py36/dfa3f24/ : SUCCESS in 3m 58s\n- openstack-tox-py37 https://logs.opendev.org/90/674090/1/check/openstack-tox-py37/3772226/ : SUCCESS in 3m 31s\n- openstack-tox-docs https://logs.opendev.org/90/674090/1/check/openstack-tox-docs/8f8d083/html/ : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://logs.opendev.org/90/674090/1/check/openstack-tox-pep8/70b596a/ : SUCCESS in 3m 28s\n- openstack-tox-py27 https://logs.opendev.org/90/674090/1/check/openstack-tox-py27/fd3c752/ : SUCCESS in 3m 05s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/90/674090/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/6f5a015/ : SUCCESS in 2h 27m 08s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/90/674090/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/863d8b1/ : SUCCESS in 2h 18m 48s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/90/674090/1/check/tripleo-ci-centos-7-scenario001-standalone/a253861/ : SUCCESS in 1h 52m 08s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/90/674090/1/check/tripleo-ci-centos-7-scenario002-standalone/adc026b/ : SUCCESS in 1h 52m 12s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/90/674090/1/check/tripleo-ci-centos-7-scenario003-standalone/821592a/ : SUCCESS in 1h 26m 10s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/90/674090/1/check/tripleo-ci-centos-7-scenario004-standalone/d99ec21/ : SUCCESS in 1h 37m 37s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/90/674090/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9670cf6/ : FAILURE in 1h 25m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45be3aeecd6d4e8a43e090c08222bbf43c920cda","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-02 07:26:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"09b02aed38d1d4fa59117dcbc177bab763b4c506","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-02 12:24:51.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a173149ccf574ff2cc635af03293e504280882dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 12:25:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbbaf9f11b90d88b84a05b03920a519bcdc82570","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 15:15:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f9738a02eb3012dccfeb2f9a392d0ce403ed0bb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 15:15:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/90/674090/1/gate/tripleo-ci-centos-7-containers-multinode/9bb9c11/ : SUCCESS in 2h 23m 25s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/90/674090/1/gate/tripleo-ci-centos-7-standalone/d0714d4/ : SUCCESS in 2h 03m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/90/674090/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/99efbf9/ : SUCCESS in 2h 40m 33s\n- openstack-tox-py35 https://logs.opendev.org/90/674090/1/gate/openstack-tox-py35/ccf14b3/ : SUCCESS in 4m 40s\n- openstack-tox-py36 https://logs.opendev.org/90/674090/1/gate/openstack-tox-py36/7d63031/ : SUCCESS in 3m 53s\n- openstack-tox-py37 https://logs.opendev.org/90/674090/1/gate/openstack-tox-py37/9fc92c4/ : SUCCESS in 3m 59s\n- openstack-tox-docs https://logs.opendev.org/90/674090/1/gate/openstack-tox-docs/bcda45d/html/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://logs.opendev.org/90/674090/1/gate/openstack-tox-pep8/66c6c60/ : SUCCESS in 3m 40s\n- openstack-tox-py27 https://logs.opendev.org/90/674090/1/gate/openstack-tox-py27/4f972da/ : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b8a8465ec7de091098a64dfcc96725a64a394b54","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-02 20:27:36.000000000","message":"Created a revert of this change as Ic26c6400590f86d5fd172066f8d8a87802094e2d","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d105c6e9a21fc70bc5bfdf5a9d9796badb019704","revisions":{"d105c6e9a21fc70bc5bfdf5a9d9796badb019704":{"kind":"REWORK","_number":1,"created":"2019-08-01 16:53:27.000000000","uploader":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"ref":"refs/changes/90/674090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/90/674090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/90/674090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/90/674090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/90/674090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/90/674090/1"}}},"commit":{"parents":[{"commit":"a85515a89395fbc6c20ff8a0bc4015777750f773","subject":"Check if volume paths exist before executing Docker.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/a85515a89395fbc6c20ff8a0bc4015777750f773"}]}],"author":{"name":"Damien Ciabrini","email":"dciabrin@redhat.com","date":"2019-07-09 20:23:28.000000000","tz":120},"committer":{"name":"Damien Ciabrini","email":"dciabrin@redhat.com","date":"2019-08-01 16:52:33.000000000","tz":120},"subject":"Generate addition drop-in dependencies for podman containers","message":"Generate addition drop-in dependencies for podman containers\n\nIf a container managed by Paunch and has the drop-in file enabled, we\nwant to start the containers with the script provided by the Paunch rpm:\npaunch-start-podman-container\n\nThe script will take care of proper ordering when creating systemd scope\nfiles.\n\nPartial-Bug: #1838668\nChange-Id: Idaf5d4871ad1231f2592238a7925857af8f40548\n(cherry picked from commit b33aeea9728233aca852a3e132f23fca71ac42df)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/d105c6e9a21fc70bc5bfdf5a9d9796badb019704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/d105c6e9a21fc70bc5bfdf5a9d9796badb019704"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
