)]}'
{"id":"openstack%2Fpaunch~674600","triplet_id":"openstack%2Fpaunch~stable%2Fqueens~Ife65f082f389ad4a730b3f9cf07d32ac4be32b45","project":"openstack/paunch","branch":"stable/queens","hashtags":[],"change_id":"Ife65f082f389ad4a730b3f9cf07d32ac4be32b45","subject":"Optimize container CLI for getting unique names","status":"MERGED","created":"2019-08-05 12:53:50.000000000","updated":"2019-08-12 20:43:26.000000000","submitted":"2019-08-12 20:43:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674600-1565642606607-ff059232","meta_rev_id":"c20cfa0ca9f17176ba5a75d39d19bdbe66ff50e6","_number":674600,"virtual_id_number":674600,"owner":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-12 20:43:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-05 15:02:59.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-12 15:04:08.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-08-09 08:52:27.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-12 15:04:08.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 15:02:59.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-09 08:52:27.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-12 15:04:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-12 20:43:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f459d96b021e72a580a3ccf817248f416c1acce6","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-05 12:53:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9424b21a7be264b930ca304b86f335185c389d48","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 15:02:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/00/674600/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/714500e/ : SUCCESS in 2h 08m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e2398e52f0f56a4637f83ae70c991d11da544562","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:13:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/00/674600/1/check/tripleo-ci-centos-7-containers-multinode/1f2c6bc/ : SUCCESS in 2h 14m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/00/674600/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/695046e/ : FAILURE in 1h 53m 27s (non-voting)\n- build-openstack-sphinx-docs https://logs.opendev.org/00/674600/1/check/build-openstack-sphinx-docs/25cbf37/html/ : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://logs.opendev.org/00/674600/1/check/openstack-tox-pep8/c77209e/ : SUCCESS in 4m 30s\n- openstack-tox-py27 https://logs.opendev.org/00/674600/1/check/openstack-tox-py27/6c10fe7/ : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e7c367e0383ede1e0898a4b3919e6f5e700dc1df","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-12 15:04:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1bc4cf70ac0fb17c743e967fd5a046508d8b394c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:04:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c8b60bbad80dec6cc87485d369e5cb65ac658cdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 20:43:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6c2250417c1c4b39be0382911300490b : SUCCESS in 2h 17m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/932a399034a04d12ab8b1a99b17f1b1d : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4dba7e7bcc45ecab4edb6b9c8bb0c1 : SUCCESS in 2m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f39e53c220724df89f3f0998bc9c5388 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c20cfa0ca9f17176ba5a75d39d19bdbe66ff50e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 20:43:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17d23079ccc48f5c2f11143733736fd1c37d7628","revisions":{"17d23079ccc48f5c2f11143733736fd1c37d7628":{"kind":"REWORK","_number":1,"created":"2019-08-05 12:53:50.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/00/674600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/00/674600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/00/674600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/00/674600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/00/674600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/00/674600/1"}}},"commit":{"parents":[{"commit":"90cf8bbb3e0a5baf4dbee519ffacfddf04cdb7b0","subject":"Revert \"docker/compose: quote health-cmd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/90cf8bbb3e0a5baf4dbee519ffacfddf04cdb7b0"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-07-30 13:19:42.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 12:53:35.000000000","tz":120},"subject":"Optimize container CLI for getting unique names","message":"Optimize container CLI for getting unique names\n\nDo not call it twice in the while{} loop. It is pointless to\ndouble-check for a container existence after we have determined weither\nits name is in use or not.\n\nThis decreases the pressure on the container system.\n\nChange-Id: Ife65f082f389ad4a730b3f9cf07d32ac4be32b45\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n(cherry picked from commit a6df749249b36d2c4aa1621892e778eb9d8248ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/17d23079ccc48f5c2f11143733736fd1c37d7628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/17d23079ccc48f5c2f11143733736fd1c37d7628"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
