)]}'
{"id":"openstack%2Fpaunch~675637","triplet_id":"openstack%2Fpaunch~master~I68f806981f3f52890be5d4d73ef5f67868be4ab5","project":"openstack/paunch","branch":"master","topic":"bug/1839559","hashtags":[],"change_id":"I68f806981f3f52890be5d4d73ef5f67868be4ab5","subject":"runner: do not run exec if container not running","status":"ABANDONED","created":"2019-08-09 14:58:09.000000000","updated":"2019-08-19 22:40:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"09693163bf5acf7de4f5a8adead7184c080cd602","_number":675637,"virtual_id_number":675637,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":-1,"date":"2019-08-13 20:20:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-14 16:28:40.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":-1,"date":"2019-08-13 14:46:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-08-14 13:55:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":1,"date":"2019-08-13 14:46:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-13 14:46:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-09 16:58:08.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-08-13 07:44:34.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-08-13 08:03:49.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2019-08-13 20:20:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-14 13:55:49.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-08-14 16:28:40.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a6eb05af1ee2b0556f5f68f3ac165b884af186e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-09 14:58:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bda08c841f51dc04d4744cfdc50302528c346625","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-09 15:09:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f7154b855e877565d27940bca5389902b6515bef","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-09 16:58:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9d588e85b0dd454c9aa31055fd702add7c592324","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-09 17:23:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/37/675637/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/b739108/ : SUCCESS in 2h 24m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5f818c4658918a04391c68f7730a67267cec0fc1","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-08-09 22:27:35.000000000","message":"Patch Set 1: Code-Review+1\n\nNice, re-using an existing variable/logic to detect if the container is actually running before executing.","accounts_in_message":[],"_revision_number":1},{"id":"e8f6edaee35d1684e2ebc9130730e6f207b1cf11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 00:42:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-containers-multinode/32a3b83/ : SUCCESS in 2h 15m 57s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-undercloud-containers/6495320/ : SUCCESS in 1h 34m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/bd3a038/ : FAILURE in 1h 09m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/49fc795/ : FAILURE in 1h 10m 13s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-standalone/e52b47c/ : SUCCESS in 1h 47m 11s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/37/675637/1/check/tripleo-ci-fedora-28-standalone/eeccbea/ : SUCCESS in 1h 43m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-standalone-upgrade/2b500d7/ : SUCCESS in 2h 09m 08s (non-voting)\n- openstack-tox-py35 https://logs.opendev.org/37/675637/1/check/openstack-tox-py35/1edbf08/ : SUCCESS in 5m 27s\n- openstack-tox-py36 https://logs.opendev.org/37/675637/1/check/openstack-tox-py36/f97d35d/ : SUCCESS in 6m 15s\n- openstack-tox-py37 https://logs.opendev.org/37/675637/1/check/openstack-tox-py37/0b02e68/ : SUCCESS in 4m 55s\n- openstack-tox-docs https://logs.opendev.org/37/675637/1/check/openstack-tox-docs/26f19af/html/ : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://logs.opendev.org/37/675637/1/check/openstack-tox-pep8/96d831b/ : SUCCESS in 5m 36s\n- openstack-tox-py27 https://logs.opendev.org/37/675637/1/check/openstack-tox-py27/96e4440/ : SUCCESS in 5m 25s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/0faf754/ : SUCCESS in 2h 38m 08s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/672f547/ : SUCCESS in 2h 01m 19s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-scenario001-standalone/8e985aa/ : SUCCESS in 1h 34m 47s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-scenario002-standalone/cb045ce/ : SUCCESS in 1h 30m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-scenario003-standalone/ca0175d/ : SUCCESS in 1h 11m 46s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-scenario004-standalone/d916f99/ : SUCCESS in 1h 39m 07s\n- tripleo-ci-centos-7-scenario012-standalone https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-scenario012-standalone/ed5cb13/ : SUCCESS in 1h 54m 01s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/37/675637/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/840e83b/ : FAILURE in 1h 43m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ea57afc25f834951eb02bb14fdddc597690894f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-10 07:42:09.000000000","message":"Patch Set 1: Code-Review-1\n\ntested internally by QE, it doesnt seem to work. I\u0027m leaving on PTO btw, so feel free to take over.","accounts_in_message":[],"_revision_number":1},{"id":"ca378468dd0486d3f372aea270cc438c0ef7234a","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 15:42:48.000000000","message":"Patch Set 1: -Code-Review\n\nthis seems causing https://bugs.launchpad.net/tripleo/+bug/1839841","accounts_in_message":[],"_revision_number":1},{"id":"726f9236b34327715a7e6f571fb20138f7ec695c","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 15:43:57.000000000","message":"Patch Set 1:\n\nBut unlocking reset counter does not help to unblock the local testing of this patch, the problem seems connected to why the unit cannot start:\nNo such file or directory error for its PID file","accounts_in_message":[],"_revision_number":1},{"id":"f3dcf71cb687e7b3f2680670b1faab09237c5376","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-13 07:44:34.000000000","message":"Patch Set 1:\n\nthe real issue what blocks testing is https://bugs.launchpad.net/tripleo/+bug/1839929","accounts_in_message":[],"_revision_number":1},{"id":"8758389953a208ded28a80ea529004ded2db0214","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-13 14:38:57.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"1f02f3e1cb9603f456b2f75ea93bb3a5311cb545","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-08-13 14:46:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"711db0e03ab5e9acf58518105d1de95383abed35","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 16:29:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/37/675637/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/7a18fca/ : FAILURE in 1h 17m 55s","accounts_in_message":[],"_revision_number":3},{"id":"2d522d974ac4a40d1765f78871a6a2c5f06cb251","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 20:20:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/64870808692c48ef963a293fb4b6b919 : POST_FAILURE in 2h 51m 23s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/dfa9200dc4e04092a8e69763285c94c5 : SUCCESS in 1h 37m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a20d25634b4f4d648020868c972b7006 : SUCCESS in 2h 26m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/48419cc5ce354168855f8c71253ea669 : SUCCESS in 2h 50m 55s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/1c5a15631995458f8c1c5eb80aeccfb5 : SUCCESS in 2h 03m 41s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/79d4104f159c4e6ea74330e6b95e2980 : FAILURE in 53m 55s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/6366b089c0c049188a8f24aca2217bb2 : SUCCESS in 2h 18m 37s (non-voting)\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d8bf84e9c1c3459fa030dcfe2afd140f : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c71b6587f2a24ed69b1df20f8752d303 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ea6b450b3714e828bc88202352684c8 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fd96fd73be1490997fa5ffdf17e29fd : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b8bff5648ca4acc97de7c3a95b7bd36 : SUCCESS in 2m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/074c948737504ad2b3649ef1cb4bdfaf : SUCCESS in 3m 50s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4878c9bb9e56429db23b02db9caa090c : SUCCESS in 2h 11m 32s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a802979864f14a7e99ffa86ee94f6809 : FAILURE in 1h 24m 28s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/3dd225d8aff84ce09dfacebb5268c5dd : SUCCESS in 1h 45m 54s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4b9f76d051ae4475b6ce343fe0da0bc9 : SUCCESS in 1h 41m 35s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5007c97c9efc4a6c8c3ed70b9af5f540 : FAILURE in 56m 18s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/b6c79f29b54a41ddbdb45d3f5d607cc8 : SUCCESS in 1h 44m 33s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/60b1a0b77dc947c5a8eaa964601f86b9 : SUCCESS in 2h 04m 37s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/68aba32446ad4c77a67ecce7c34822c5 : FAILURE in 1h 59m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"94d6021c356df04712b818862a047e5308dd9fef","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-08-14 13:55:49.000000000","message":"Patch Set 3: -Code-Review\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"80ea33f7956df8dd3cc108bd5fec7376e0d897cd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-14 16:28:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/37/675637/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/03ae2ba/ : SUCCESS in 2h 30m 41s","accounts_in_message":[],"_revision_number":3},{"id":"cf1031a2ddc8983e5ba1b161ca41d53ebfe4ce82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 16:59:05.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c0f2d554dfd54482a1a1a5b56be579d3 : SUCCESS in 2h 34m 13s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c3305f41462142e8a62b5145071a7359 : SUCCESS in 1h 31m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2474ad1e673b4c45995b5a8a9572f5f3 : SUCCESS in 2h 20m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/31a20a670e0541af98114e442f7c0f50 : SUCCESS in 2h 42m 17s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/6fca3234b14a4514ba1a73306d940fee : SUCCESS in 1h 55m 29s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/63e49c3e6b4e424b8d3ca71b0fe8eef6 : SUCCESS in 2h 11m 54s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/27d07225465347e2bed5749990a2c7c5 : SUCCESS in 2h 03m 47s (non-voting)\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1f6617c402a643e1aeef1175b2a5f031 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de8e8cce5a5a4e1fb2dbc21a51ad07f6 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86300c2b949d4f96b2d08df48e129569 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/327abb584b4241a591c52debad4f4f51 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1a5e8bfbc5d493fb5f764a506fa775d : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f800e272d61140aca6ab95c572ce472f : SUCCESS in 3m 29s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/76c279cd191f414a826dc60e1e824320 : SUCCESS in 2h 29m 30s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/605642ebf4294d578252592cf20edf75 : FAILURE in 1h 28m 16s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/9f8e0606ddcd410bbcc33ec37c64aafc : SUCCESS in 1h 58m 50s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/111c382628d640068b73089d0987693a : FAILURE in 1h 02m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/187ef52edec24fbab51c5ae877b5f8bb : FAILURE in 55m 59s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/171f92c1550644dbbb143f31be70dc35 : SUCCESS in 1h 45m 48s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/db1d67acfacf4e7490a0dfc0891c2cb8 : FAILURE in 1h 29m 48s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/5e8df4408e234784bcc8d827e653a2fa : FAILURE in 1h 55m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e766dd992c09d628acb22c979c0d190ed7ce04ef","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-19 22:40:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ef874b82b490a6a36046a5443a370f872514cccc","revisions":{"b2f055f234966729aec274d26932ba8df7091d9b":{"kind":"REWORK","_number":1,"created":"2019-08-09 14:58:09.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/37/675637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/37/675637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/37/675637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/37/675637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/37/675637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/37/675637/1"}}},"commit":{"parents":[{"commit":"aa0e32b004c8f053444f9214af6a0e0019a03908","subject":"Merge \"Handle defined containers that are stopped.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/aa0e32b004c8f053444f9214af6a0e0019a03908"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-09 14:54:21.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-09 14:54:21.000000000","tz":-240},"subject":"runner: do not run exec if container not running","message":"runner: do not run exec if container not running\n\nUse \"podman\" inspect to check if whether or not the container is\nrunning before doing \"podman exec\"; and raise if not running.\n\nIt re-uses what was done for \"podman run\", where we run the \"podman\ninspect\" by moving a block our of the conditions to always give the\ncontainer state for each container in the loop, and send a warning if\nthe state can\u0027t be found.\n\nCloses-Bug: #1839559\nrhbz#1739224\n\nChange-Id: I68f806981f3f52890be5d4d73ef5f67868be4ab5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/b2f055f234966729aec274d26932ba8df7091d9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/b2f055f234966729aec274d26932ba8df7091d9b"}]},"branch":"refs/heads/master"},"52aa70c79a77d8d8da2a3c4c117f7c83c42c7ca1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-13 14:38:57.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/37/675637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/37/675637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/37/675637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/37/675637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/37/675637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/37/675637/2"}}},"commit":{"parents":[{"commit":"5d174c1bea474a34da6f4f68173323a8243c4fc5","subject":"Fix mismatching fixed vs unique container names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5d174c1bea474a34da6f4f68173323a8243c4fc5"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-09 14:54:21.000000000","tz":-240},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-13 14:38:26.000000000","tz":120},"subject":"runner: do not run exec if container not running","message":"runner: do not run exec if container not running\n\nUse \"podman\" inspect to check if whether or not the container is\nrunning before doing \"podman exec\"; and raise if not running.\n\nIt re-uses what was done for \"podman run\", where we run the \"podman\ninspect\" by moving a block our of the conditions to always give the\ncontainer state for each container in the loop, and send a warning if\nthe state can\u0027t be found.\n\nCloses-Bug: #1839559\nrhbz#1739224\n\nChange-Id: I68f806981f3f52890be5d4d73ef5f67868be4ab5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/52aa70c79a77d8d8da2a3c4c117f7c83c42c7ca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/52aa70c79a77d8d8da2a3c4c117f7c83c42c7ca1"}]},"branch":"refs/heads/master"},"ef874b82b490a6a36046a5443a370f872514cccc":{"kind":"NO_CHANGE","_number":3,"created":"2019-08-13 14:46:52.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/37/675637/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/37/675637/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/37/675637/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/37/675637/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/37/675637/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/37/675637/3"}}},"commit":{"parents":[{"commit":"5d174c1bea474a34da6f4f68173323a8243c4fc5","subject":"Fix mismatching fixed vs unique container names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5d174c1bea474a34da6f4f68173323a8243c4fc5"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-09 14:54:21.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2019-08-13 14:08:56.000000000","tz":-240},"subject":"runner: do not run exec if container not running","message":"runner: do not run exec if container not running\n\nUse \"podman\" inspect to check if whether or not the container is\nrunning before doing \"podman exec\"; and raise if not running.\n\nIt re-uses what was done for \"podman run\", where we run the \"podman\ninspect\" by moving a block our of the conditions to always give the\ncontainer state for each container in the loop, and send a warning if\nthe state can\u0027t be found.\n\nCloses-Bug: #1839559\nrhbz#1739224\n\nChange-Id: I68f806981f3f52890be5d4d73ef5f67868be4ab5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/ef874b82b490a6a36046a5443a370f872514cccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/ef874b82b490a6a36046a5443a370f872514cccc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
