)]}'
{"id":"openstack%2Fpaunch~675895","triplet_id":"openstack%2Fpaunch~master~I537fbf9933f2cbe6e1c2f627ba77da645bd55f25","project":"openstack/paunch","branch":"master","topic":"bug/1839841","hashtags":[],"change_id":"I537fbf9933f2cbe6e1c2f627ba77da645bd55f25","subject":"Fix systemd service start rate limiting","status":"MERGED","created":"2019-08-12 13:25:52.000000000","updated":"2019-08-14 11:39:55.000000000","submitted":"2019-08-14 11:39:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675895-1565782795278-92a0a85e","meta_rev_id":"46a35998aeb56c547da3e9f8a89586662c4c5212","_number":675895,"virtual_id_number":675895,"owner":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18851,"name":"Marius Cornea","email":"mcornea@redhat.com","username":"mcornea"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-08-14 11:39:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-13 15:08:07.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":18851,"name":"Marius Cornea","email":"mcornea@redhat.com","username":"mcornea"},{"value":2,"date":"2019-08-13 16:55:45.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-08-13 12:52:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":2,"date":"2019-08-14 07:12:38.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"_account_id":18851,"name":"Marius Cornea","email":"mcornea@redhat.com","username":"mcornea"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-14 07:56:34.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18851,"name":"Marius Cornea","email":"mcornea@redhat.com","username":"mcornea"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 14:15:52.000000000","updated_by":{"_account_id":18851,"name":"Marius Cornea","email":"mcornea@redhat.com","username":"mcornea"},"reviewer":{"_account_id":18851,"name":"Marius Cornea","email":"mcornea@redhat.com","username":"mcornea"},"state":"REVIEWER"},{"updated":"2019-08-13 12:52:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-08-13 15:08:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-13 16:55:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-14 07:12:38.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-14 11:39:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"603816961efa322556d44b7ce7ac1db66f01a80f","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 13:25:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c94b84d515a7add666e2e56d1ff3e9d190dc42f1","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 13:30:54.000000000","message":"Patch Set 1:\n\n15:27:19 - bogdando: well, the alternative is to evaluate some value based of the restart value\n15:27:27 - bogdando: and the default 10s interval\n15:27:42 - bogdando: and to set StartLimitBurst to that\n15:28:39 - bogdando: I mean we need to make sure all possible ways to define the time interval are supported\n15:29:07 - bogdando: I think paunch already contains something like that\n15:29:22 - bogdando: Steve wrote that for docker-compose IIRC","accounts_in_message":[],"_revision_number":1},{"id":"c28a7e51acddcca20b79062fb004f38bf1a4aa13","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 13:34:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec6c2e217330767047ae0f1d3f18439194756aa1","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 13:37:32.000000000","message":"Patch Set 1:\n\nbut I think the problem is paunch wants --enable-now too fast for the configured services. No config changes would help here.","accounts_in_message":[],"_revision_number":1},{"id":"b79236c9f694ccf90b53166f472fe041d3926735","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 13:38:05.000000000","message":"Patch Set 1:\n\n(but only disabling rate limiting or learning paunch to rate limit itself)","accounts_in_message":[],"_revision_number":1},{"id":"e406c3c361126dea39ded04b9b3ed7a9aaa6d675","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-12 13:43:46.000000000","message":"Patch Set 1:\n\nthere\u0027s also the \"systemctl reset-failed \u003cservice\u003e\" that can be used in order to reset the fail-count.\n\nDisabling rate limit smells like a bad idea, since it might create some instability on the node (load, for instance). Not sure this is the right approach :/","accounts_in_message":[],"_revision_number":1},{"id":"4fc2f5db3837243d8439a8ca200e83d02fa16a2f","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 13:49:51.000000000","message":"Patch Set 1:\n\nIt is possible that the issue is not present in the master branch, but only comes after https://review.opendev.org/#/c/675692/1/paunch/builder/base.py","accounts_in_message":[],"_revision_number":1},{"id":"c59087c050a8c15c9866fe11dce3982e700adb00","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 14:15:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b1f9577ef2e8a96a446d09e97b27a1b98c716fd1","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 15:18:34.000000000","message":"Patch Set 2:\n\nlocal testing shows this does not help (nor did the revision #1 of the patch, which is disabling rate limiting):\n\n2019-08-12 15:09:40.722 833524 DEBUG paunch [  ] Creating systemd unit file: /etc/systemd/system/tripleo_memcached-a9hb0cp7.service\n2019-08-12 15:09:40.723 833524 DEBUG paunch.utils.systemctl [  ] Executing: systemctl daemon-reload\n2019-08-12 15:09:41.109 833524 DEBUG paunch.utils.systemctl [  ] Executing: systemctl enable --now tripleo_memcached-a9hb0cp7\n2019-08-12 15:09:41.766 833524 DEBUG paunch.utils.systemctl [  ] Executing: systemctl reset-failed tripleo_memcached-a9hb0cp7\n2019-08-12 15:09:43.721 833524 DEBUG paunch.utils.systemctl [  ] Executing: systemctl enable --now tripleo_memcached-a9hb0cp7\n2019-08-12 15:09:44.219 833524 DEBUG paunch.utils.systemctl [  ] Executing: systemctl reset-failed tripleo_memcached-a9hb0cp7\n2019-08-12 15:09:48.139 833524 DEBUG paunch.utils.systemctl [  ] Executing: systemctl enable --now tripleo_memcached-a9hb0cp7\n2019-08-12 15:09:48.575 833524 DEBUG paunch.utils.systemctl [  ] Executing: systemctl reset-failed tripleo_memcached-a9hb0cp7\n2019-08-12 15:09:55.695 833524 DEBUG paunch.utils.systemctl [  ] Executing: systemctl enable --now tripleo_memcached-a9hb0cp7\n2019-08-12 15:09:56.186 833524 DEBUG paunch.utils.systemctl [  ] Executing: systemctl reset-failed tripleo_memcached-a9hb0cp7\n2019-08-12 15:09:59.026 833524 DEBUG paunch.utils.systemctl [  ] Executing: systemctl enable --now tripleo_memcached-a9hb0cp7\n2019-08-12 15:09:59.467 833524 DEBUG paunch.utils.systemctl [  ] Executing: systemctl reset-failed tripleo_memcached-a9hb0cp7\n2019-08-12 15:09:59.483 833524 ERROR paunch [  ] systemctl failed\nTraceback (most recent call last):\n  File \"/usr/lib/python3.6/site-packages/paunch/utils/systemctl.py\", line 32, in systemctl\n    subprocess.check_call(cmd)\n  File \"/usr/lib64/python3.6/subprocess.py\", line 311, in check_call\n    raise CalledProcessError(retcode, cmd)\nsubprocess.CalledProcessError: Command \u0027[\u0027systemctl\u0027, \u0027enable\u0027, \u0027--now\u0027, \u0027tripleo_memcached-a9hb0cp7\u0027]\u0027 returned non-zero exit status 1.","accounts_in_message":[],"_revision_number":2},{"id":"924a037d5d8e767172b12c8e01a40d001198fc5c","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 15:24:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"348ba9e5cb3f16d7748ce9d9f33406388ac1321f","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 15:41:53.000000000","message":"Patch Set 2:\n\nunlocking reset counter does not help (although this patch might be nice to have anyway...)\n\nthe problem seems connected to why the unit cannot start:\nNo such file or directory error for its PID file","accounts_in_message":[],"_revision_number":2},{"id":"5c7def0fb525f0bea9d8f664ab42084216d89db6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-08-12 16:29:34.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nHello,\n\nThanks for these changes.\n\nI think that using tenacity to manage this it\u0027s a good idea, but I had introduce the `systemctl` module in the past to wrap the systemctl CLI and I think that we are out of the scope of this module with these changes.\n\nSee my comments inline and let me know your thoughts:","accounts_in_message":[],"_revision_number":2},{"id":"445658192805d5ee5794fb1cf1e715b17a52b38c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-12 17:50:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 finger://ze02.softwarefactory-project.io:7979/c1588d580822417e9ac1617372ef1440 : POST_FAILURE in 3h 26m 28s","accounts_in_message":[],"_revision_number":2},{"id":"3443e2460d5f318255504956e5acf807a2466df9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 18:53:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d16735f75f7642918cc3221e788289ad : SUCCESS in 1h 33m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/91b5cd989be24cd38bd6b43b6a32a5e9 : SUCCESS in 1h 47m 24s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/d7d8a784cacc48d488db37b74288a9f4 : SUCCESS in 1h 50m 43s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/02c605aab5aa470883aa5e91017b3c40 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e19cb666d60b4cf6840b89d395520591 : SUCCESS in 3m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee09dff84cad4760b0181f6826afc660 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f395138d199c4cb99c4b6836881b6870 : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8339b9b7a8024371aed96ca14336ee9c : FAILURE in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/94978ba107754f708cf97820f15df159 : SUCCESS in 3m 12s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2bf4d5e4b71c7f5d684b66c70c8b5f94b672e968","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-13 12:26:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b58dc18de316d209adf06bb33b3bf818898cc39","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-13 12:29:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eb619122256cf5fbc233771a730cd1c329597920","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-08-13 12:52:05.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"11ba72870449d570b8c3783f807f6ee76c1d077c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 15:08:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/95/675895/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/ca20236/ : SUCCESS in 2h 38m 11s","accounts_in_message":[],"_revision_number":3},{"id":"c7aad06bce5e6b33e44f3e5ecd2654ba942849d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 16:49:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d11651bd70f64f8e8fd9803cbfc9244c : SUCCESS in 2h 44m 49s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/5926bea93dfa4905a2ab651be4f31413 : SUCCESS in 1h 35m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/76f6eabe4dc54d468e61eadebda24153 : SUCCESS in 2h 23m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4aa2062e6c4d478b9c9cde2152e015e9 : SUCCESS in 2h 30m 04s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/90ebfa7c0a1b430b920fa33ff8dd03ee : SUCCESS in 1h 59m 12s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/d8ff74e4a1634a3cb7c31b2e4c7d38db : SUCCESS in 2h 03m 54s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/caea1960fc9d447cae85ed8764958460 : SUCCESS in 2h 30m 05s (non-voting)\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/48f8165fe1ae41549e020d6c840abf28 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30736ef022004b1fb424e08b5b39a8dd : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d978f8062ba474888cc2f18acc4f477 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe690d479be14bb08b9355bf0e70813f : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c88670d28ff4006892523a8ed831d5a : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/881cf881f0a948ef9e724f5d838017bf : SUCCESS in 4m 12s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/34c2ec6ed5d84b7ea44073fc4a65e1a3 : SUCCESS in 2h 44m 46s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/fc9b712658f34272a90d85a6077accf9 : FAILURE in 1h 38m 58s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/941437dff284419bb9391afff9ca74f1 : SUCCESS in 2h 07m 32s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/b80d2d7395754266938848179610b1bc : SUCCESS in 2h 28m 43s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/72d5ec17446348168837a08cdea4ce26 : SUCCESS in 1h 17m 43s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/94ea2f34005a4615992c4c659be408a3 : SUCCESS in 1h 52m 58s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/06510a5ad6aa402f96e12fb5f4c41e43 : SUCCESS in 2h 14m 27s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/b898c897c6a741a884302b79a8518ce2 : FAILURE in 1h 58m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"721dd0bce4d4e4236780675b9a3e2d52ec451858","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-13 16:55:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9414682b6397e86940b69cd2750ed0a13da71245","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-14 07:12:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4662d88307fd1a5a1a930fb78173b48d7f5b2a2f","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-14 07:56:34.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6bc0a377f68cd0d00efccefbe32eb637596d3e3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 07:56:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4b9f9bc229fc25337216bfef98a7e4b417eab5a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 11:39:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5166ce9dd23cd47f091d65f0b1426cba6fa92af1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 11:39:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a61d3258bc7740e994edb0e1282ce181 : SUCCESS in 2h 27m 23s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/51900eacd99b446591a4f9a803afdcc7 : SUCCESS in 1h 54m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1ac6de842f054ce5b3c48a493f874f6d : SUCCESS in 2h 37m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2a34b0a90a014f1288aeafa9c8f0a439 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dffe8eb528eb477aa971190bd86d6ea2 : SUCCESS in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbd5192030804e4183af522d74db748d : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c5dee757db8470dae6181cba73e468b : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ac3796eaa8b4a738f8ecc96f0dad090 : SUCCESS in 2m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/400815714f2049718d9e45f3143e8feb : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2eaebe2cd96aa1e121d02f6998e5addd7c52ef86","revisions":{"f60da748adaf5a086349b1a793798e097838dd80":{"kind":"REWORK","_number":1,"created":"2019-08-12 13:25:52.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/95/675895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/95/675895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/95/675895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/95/675895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/95/675895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/95/675895/1"}}},"commit":{"parents":[{"commit":"aa0e32b004c8f053444f9214af6a0e0019a03908","subject":"Merge \"Handle defined containers that are stopped.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/aa0e32b004c8f053444f9214af6a0e0019a03908"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-12 13:24:33.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-12 13:24:33.000000000","tz":120},"subject":"Disable systemd service start rate limiting","message":"Disable systemd service start rate limiting\n\nThe default limit is to allow 5 restarts in a 10sec period. If a\nservice goes over that threshold due to the Restart\u003d config option in\nthe service definition, it will not attempt to restart any further.\n\nSet StartLimitIntervalSec to 0 to disable any kind of rate limiting.\n\nCloses-bug: #1839841\n\nChange-Id: I537fbf9933f2cbe6e1c2f627ba77da645bd55f25\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f60da748adaf5a086349b1a793798e097838dd80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f60da748adaf5a086349b1a793798e097838dd80"}]},"branch":"refs/heads/master"},"8060f2bef0609a104e3ecd55b8278fb93dda6802":{"kind":"REWORK","_number":2,"created":"2019-08-12 14:15:36.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/95/675895/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/95/675895/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/95/675895/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/95/675895/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/95/675895/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/95/675895/2"}}},"commit":{"parents":[{"commit":"aa0e32b004c8f053444f9214af6a0e0019a03908","subject":"Merge \"Handle defined containers that are stopped.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/aa0e32b004c8f053444f9214af6a0e0019a03908"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-12 13:24:33.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-12 14:13:30.000000000","tz":120},"subject":"Fix systemd service start rate limiting","message":"Fix systemd service start rate limiting\n\nThe default limit is to allow 5 restarts in a 10sec period. If a\nservice goes over that threshold due to the Restart\u003d config option in\nthe service definition, it will not attempt to restart any further.\n\nWe should not set StartLimitIntervalSec to 0 to disable any kind of\nrate limiting as that may end up impacting the node load.\n\nInstead, use tenacity to retry with an exponential backoff, when the\nservice unit enablement fails. Before to retry it, reset the unit\u0027s\nfailure counters with the systemctl wrapper.\n\nCloses-bug: #1839841\n\nChange-Id: I537fbf9933f2cbe6e1c2f627ba77da645bd55f25\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/8060f2bef0609a104e3ecd55b8278fb93dda6802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/8060f2bef0609a104e3ecd55b8278fb93dda6802"}]},"branch":"refs/heads/master"},"2eaebe2cd96aa1e121d02f6998e5addd7c52ef86":{"kind":"REWORK","_number":3,"created":"2019-08-13 12:29:14.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/95/675895/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/95/675895/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/95/675895/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/95/675895/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/95/675895/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/95/675895/3"}}},"commit":{"parents":[{"commit":"aa0e32b004c8f053444f9214af6a0e0019a03908","subject":"Merge \"Handle defined containers that are stopped.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/aa0e32b004c8f053444f9214af6a0e0019a03908"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-12 13:24:33.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-13 12:29:08.000000000","tz":120},"subject":"Fix systemd service start rate limiting","message":"Fix systemd service start rate limiting\n\nThe default limit is to allow 5 restarts in a 10sec period. If a\nservice goes over that threshold due to the Restart\u003d config option in\nthe service definition, it will not attempt to restart any further.\n\nWe should not set StartLimitIntervalSec to 0 to disable any kind of\nrate limiting as that may end up impacting the node load.\n\nInstead, use tenacity to retry with an exponential backoff, when the\nservice unit enablement fails. Before to retry it, reset the unit\u0027s\nfailure counters with the systemctl wrapper. This is a crash-loop\napproach that provides an efficient feature parity to the classic\nrate limiting, shall we want to implement that for the systemctl\ncommand wrapper instead.\n\nCloses-bug: #1839841\n\nChange-Id: I537fbf9933f2cbe6e1c2f627ba77da645bd55f25\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2eaebe2cd96aa1e121d02f6998e5addd7c52ef86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2eaebe2cd96aa1e121d02f6998e5addd7c52ef86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
