)]}'
{"id":"openstack%2Fpaunch~676156","triplet_id":"openstack%2Fpaunch~master~I5617e11f5d315f408d818e1ce47aa68f4a0d777a","project":"openstack/paunch","branch":"master","topic":"bug/1839929","hashtags":[],"change_id":"I5617e11f5d315f408d818e1ce47aa68f4a0d777a","subject":"Fix mismatching fixed vs unique container names","status":"MERGED","created":"2019-08-13 11:00:30.000000000","updated":"2019-08-26 13:05:51.000000000","submitted":"2019-08-15 21:36:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676156-1565904981192-5d7c1590","meta_rev_id":"41b957e77757353737335d1500133a6b6123c2d3","_number":676156,"virtual_id_number":676156,"owner":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":2,"date":"2019-08-15 21:36:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-13 16:31:33.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-14 07:50:39.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"all":[{"value":1,"date":"2019-08-13 14:32:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":1,"date":"2019-08-14 12:39:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-08-15 18:53:38.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":1,"date":"2019-08-13 14:49:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-08-15 18:53:38.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-08-13 11:59:19.000000000","_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 11:59:19.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2019-08-13 14:32:29.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-08-13 14:49:42.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-13 16:31:33.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-14 12:39:35.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-08-15 18:53:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-15 21:36:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3d007f9048dc58ee58729cc3f8e116d48a98176","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-13 11:00:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"967fd220b3d931833daa8ee80bd04d6811d42688","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 12:31:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/56/676156/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/658b84c/ : FAILURE in 1h 30m 28s","accounts_in_message":[],"_revision_number":1},{"id":"da073f69d520bcb8275522003e78927908e54ddd","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-13 13:39:40.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"316c4ba01b0e8846101db23258b1427e67c85cd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 14:25:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/246e2623a08846f698297405900819e9 : SUCCESS in 2h 29m 44s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/70a4fd8f764d464c877d96635ba7e8bd : SUCCESS in 1h 49m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5ee5f797b786421da7be97a7458a5d65 : SUCCESS in 2h 11m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/bde8f4ba112b46f69729e5a900470595 : SUCCESS in 2h 30m 25s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/53ce9729d3b04443ac8f0647f21bf85f : FAILURE in 54m 32s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/f9953dd9768d4712a70e68142040a855 : FAILURE in 1h 55m 14s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/344d2b0807f84e37839ef090b8363745 : SUCCESS in 2h 17m 31s (non-voting)\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5868c6b11d5a44f0ae2244dc89c58687 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64563083232d479ebafcdf50894cc841 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f178d0d9f6a94960b233f5b5a466c598 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f48415a91e439a816ce61213788fff : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/070e35f28bcd407ba3fddfe7309840b7 : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c5081a3b8faa4a4b889c6ed469749c3b : SUCCESS in 2m 40s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e214c806e0a842d78ec6830331a653b4 : SUCCESS in 2h 31m 36s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9e037dcc46844bbdb7fef97e1db9397f : FAILURE in 1h 26m 41s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/2c44cea81e554301bda388fd655b4ff1 : SUCCESS in 1h 51m 39s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/1fc9dace26f1472b871a9c99ed5a2b63 : SUCCESS in 1h 46m 05s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5c2612d25a4b4ac781725197701d9c5a : SUCCESS in 1h 16m 30s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/8f7f5c1755474fb8acda03652d2fc3e6 : SUCCESS in 1h 45m 08s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/4abca47b2e1b4aa8ad99b04a30dc0735 : SUCCESS in 2h 04m 46s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/5a52742a069141f1845fa656330241e0 : FAILURE in 1h 57m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2dda5fc84119aee25146b4868fd1d821d07959b1","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-13 14:30:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ddfd63a68a7a800e5c5070c33bc4374b74219b54","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-08-13 14:32:29.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM. The CI failures look to be false-positives. https://logs.opendev.org/56/676156/1/check/tripleo-ci-centos-7-standalone/53ce972/logs/undercloud/home/zuul/standalone_deploy.log.txt.gz#_2019-08-13_11_29_09 and http://logs.rdoproject.org/56/676156/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/658b84c/logs/undercloud/home/zuul/overcloud_prep_images.log.txt.gz","accounts_in_message":[],"_revision_number":1},{"id":"59a0d8eb32c8813ea1cb0c9a9ec2215bc6be3bd6","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-13 14:49:42.000000000","message":"Patch Set 1: Code-Review+1\n\nCI false-positive indeed.\nPredictability is the key to success :).","accounts_in_message":[],"_revision_number":1},{"id":"07107ac9b8a709d648e9f89c088339ac09b47190","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 16:31:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/56/676156/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/45b47dd/ : SUCCESS in 2h 37m 07s","accounts_in_message":[],"_revision_number":1},{"id":"8dffd15e52d9a7ee38de81145d77617c33bdcdcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 20:15:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5db8f745f35c4452abb1120d916a40e4 : SUCCESS in 2h 30m 40s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0991d472cf0849a3a8b99d6c84c0582c : SUCCESS in 2h 23m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1b376c0049774780ab8be6ed879a57c0 : SUCCESS in 2h 49m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ad0bfc18164a425cab967788350c4c75 : SUCCESS in 2h 39m 37s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/aa87e6d31696463aa97b49aba7758c4b : SUCCESS in 1h 53m 16s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/0d17191380084a9dafd8d430360e42f5 : SUCCESS in 1h 56m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/ba74173d72f44bc5ae0e460439905566 : SUCCESS in 2h 16m 16s (non-voting)\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9102a6e564da4eb4a128df32bbda9f1d : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7a71140f5c249fab9a146bdbc098298 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ecd84ab8eb04d6cbc747e69d467f90e : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a71dfe04272c494eb263a62330dd56a8 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0709f0a6fa104d4aade58425e859b7b4 : SUCCESS in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/107cc799aef545f6a9b8ef1aa198725c : SUCCESS in 3m 20s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2a3e0bd8634a4415a5d92eac7f0da638 : SUCCESS in 2h 39m 47s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/010fc26ccf6947d9a20eeced97eb213c : FAILURE in 1h 30m 31s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d1aeabc7a71246d6b522138c756fff80 : SUCCESS in 1h 46m 13s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/54bb5246200a4cb98814f1c0fd03bdeb : SUCCESS in 1h 32m 10s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b893db9d5f9e4a4e89cd323bffa6ddae : FAILURE in 1h 02m 23s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/617d12fb091847c58a7601085dc4b6e8 : FAILURE in 57m 40s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/b3433913896c45c798a809621be808a2 : SUCCESS in 2h 08m 33s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/5cc7ac1e30694121a4c370d2b843818b : FAILURE in 1h 57m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f935edb46562e6e32eb0ac3b009339d22b56a0f4","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-14 07:50:30.000000000","message":"Patch Set 1:\n\nthis worked with my local testing\nThe conmon-pidfile matches the PIDFile in the systemd unit","accounts_in_message":[],"_revision_number":1},{"id":"8cdf0268e1735771d8e267ef57adfe9bfaa5ed96","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-14 07:50:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bf993699c21b1b114308e5c6ae5664654ae6c7ca","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-14 10:24:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/56/676156/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/a94db18/ : SUCCESS in 2h 33m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d1a34dab8ec7a2be00a3cb852d36228171a358e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 10:27:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9926d49b5d83491885c2568db00cba9b : SUCCESS in 2h 18m 54s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d5854852355a4bda8b334cb83f783097 : SUCCESS in 1h 29m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a13694ec97df44dba1ec461d3f6631ea : SUCCESS in 2h 19m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4ae3152533674241a2c3f9f7cdcecb99 : SUCCESS in 2h 22m 19s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d7e7115e84ad47c6bae7a2abf3ef743f : SUCCESS in 2h 02m 09s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/850d346184544f9db5d03ebb435ab149 : SUCCESS in 1h 41m 00s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/0736de9c780549f69f8f7ba91fd58d43 : SUCCESS in 2h 04m 55s (non-voting)\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eece9bf160d045fdafd2a4e6ce02df59 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69a0e37a790642f1b9668475b62ad9f8 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f69c715a856431a8bc2e85567d7c6b7 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cc952424fbd4bf5b66445ab1e3fb2f8 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4a63b6671749e8a9d8903f8c508b2a : SUCCESS in 2m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04e9a581c848434591bcdf255ebee2c8 : SUCCESS in 3m 23s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3db2fb93d6724b6c92bddf19cd987da3 : SUCCESS in 2h 19m 06s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/d2e9aecf5d444a62abaefa6fcdbd2b31 : FAILURE in 1h 41m 29s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7570027e42b848ad8eb2de1e357820a0 : SUCCESS in 1h 47m 09s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/2d8938c3b2884570b2be60a47bd3bd34 : SUCCESS in 1h 41m 50s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/3e71264488bf4220a13ff470d0fe58f1 : SUCCESS in 1h 23m 17s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/8a45c167186241e18951919ca9bdf687 : SUCCESS in 1h 50m 17s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/5523e26bee50465698147a7c52655780 : SUCCESS in 2h 01m 20s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/5ca62f3bd94c4cf895060be66ac60b28 : FAILURE in 2h 19m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64c6a30f09a2e6649b022f13d99980a73601b181","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-08-14 12:39:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5b6ebaf6bea16b983f7ab68da4d23f0e3e0bf9df","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-15 18:53:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7eb0e3822543206446277928281db6e16c85e9b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 18:53:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a42555a7f0a0feb7214fb129b3d68e87a8f154da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 21:36:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f0fa77326c5a4ff5a88fd6d001cdace0 : SUCCESS in 2h 31m 36s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/78002776dcfa485d8502aa98209f06ae : SUCCESS in 1h 59m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2036b9824fc842908df1f7d78094a5ae : SUCCESS in 2h 38m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7d1a33f25b4f499f89caeffa80e41d44 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc5d33ebe2e74e7d8b2651588739e03d : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6755fd1ad16244d984773fc7a631520c : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f427407fa50841cdb68b45dfa585f450 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18ad8d69089e4d349a13710c1052cba6 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f654f8220c90447a9beabad22dbfcedf : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4a911aa6df3c324a2877067d063420b63c9d56ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 21:36:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f0bfb0f2116a647ffaab3eda548eb2b61c7a3143","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-21 14:53:27.000000000","message":"Created a revert of this change as I1fd76970ee4168f4d4457bbd9da524565dfc0680","accounts_in_message":[],"_revision_number":1},{"id":"d252ce200cac7b981e8500247597859e17530a00","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-21 22:54:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f59f715bca71b72257c06509e883638ee9165a0e","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-26 10:29:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f11d6297a31fab0b151cb835f93137d0ebd76136","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-26 13:03:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1834cd59ba15b881e2b74536ddce97060b71aa36","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-26 13:05:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d174c1bea474a34da6f4f68173323a8243c4fc5","revisions":{"5d174c1bea474a34da6f4f68173323a8243c4fc5":{"kind":"REWORK","_number":1,"created":"2019-08-13 11:00:30.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/56/676156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/56/676156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/56/676156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/56/676156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/56/676156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/56/676156/1"}}},"commit":{"parents":[{"commit":"aa0e32b004c8f053444f9214af6a0e0019a03908","subject":"Merge \"Handle defined containers that are stopped.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/aa0e32b004c8f053444f9214af6a0e0019a03908"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-13 09:42:26.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-13 10:56:41.000000000","tz":120},"subject":"Fix mismatching fixed vs unique container names","message":"Fix mismatching fixed vs unique container names\n\nSometimes, like when redeploying in-place, containers for the \"fixed\"\nservice name might already existed, thus get the prefixed names. That\nmight create mismatches. For example the pidfile names may diverge by\nthe \"fixzed\" container service name vs its predictable prefixed unique\nname.\n\nFix that by using the predictable unique names instead of the service\ncontainer names for the builder and paunch actions run,\ndebug/print-cmd that rely on it. This is achieved via a new parameter\nfor the real container name (a delegate) used for the \"fixed\" service\ncontainer name.\n\nFor podman builder, we use that delegate for conmon pidfile and logging\nsetup. So that now it always matches the PIDFile specified in the\nsystemd unit generated for that container.\n\nFor docker builder, we have no special uses for delegates, but we\nsupport that parameter to simplify the code around (so that there will\nbe no need to wrap things with \u0027if cli \u003d\u003d podman else...\u0027).\n\nCloses-bug: #1839929\n\nChange-Id: I5617e11f5d315f408d818e1ce47aa68f4a0d777a\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5d174c1bea474a34da6f4f68173323a8243c4fc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5d174c1bea474a34da6f4f68173323a8243c4fc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
