)]}'
{"id":"openstack%2Fpaunch~677860","triplet_id":"openstack%2Fpaunch~master~If4d8c268218bf83abed877a699fc583fb55ccbed","project":"openstack/paunch","branch":"master","topic":"bug/1839559","hashtags":[],"change_id":"If4d8c268218bf83abed877a699fc583fb55ccbed","subject":"Check if container is running before doing an exec","status":"MERGED","created":"2019-08-22 02:07:22.000000000","updated":"2019-08-28 11:12:00.000000000","submitted":"2019-08-28 11:09:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677860-1566990585452-2a78a105","meta_rev_id":"163ff6b3e25702ee53264560081662771b1b20fe","_number":677860,"virtual_id_number":677860,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":2,"date":"2019-08-28 11:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-28 02:27:34.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"all":[{"value":0,"date":"2019-08-27 23:19:14.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-28 05:48:49.000000000","_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-28 08:04:50.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":2,"date":"2019-08-28 06:11:57.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-28 08:06:15.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 16:43:09.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-08-22 22:51:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-28 02:27:34.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-28 05:48:49.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-08-28 06:11:57.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-28 08:06:15.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-08-28 11:09:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4418159d89166885c92e93e4d9265192edc89a92","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 02:07:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e96e445a45192cb2c71d9f665df2ab63abdc9efe","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 03:51:26.000000000","message":"Patch Set 1: Workflow-1\n\nstill wip","accounts_in_message":[],"_revision_number":1},{"id":"6d95c4dfe817d50a36eb2f55429aa81566c5b5ae","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 04:00:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"448b123bbaf2094758c20f75a0870ee571f70fce","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 04:01:00.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"f96e0af4de5ee38d6880197298d2c4ee0b5e158a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 04:23:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/2/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/180e3e0/ : FAILURE in 21m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8db14b63e5099e81fe787eff823fbba5f5ca25ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 04:25:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b4db411e0361477cbd9898d3e5b728f7 : FAILURE in 22m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/92af22506ff04afb80806cc31f6ad9b1 : FAILURE in 23m 08s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2d56345bfbf74bd781f19b4879dd24fe : FAILURE in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/925131c756bb406897e0bcf57b20ac34 : FAILURE in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f83b843634f4f159d34eb3260c11921 : FAILURE in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd6915958161445b969cf1d66c6c1eaf : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c75a64e7aa24410929fa2606ec236a1 : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f9191292f764afd8e4acf6df67abb90 : FAILURE in 3m 41s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab791cb8b5b11ade1eb69b4b0fabd86b5928377c","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-22 05:14:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"da95900a1c07f8e8515f7f3b14f8057b9d983b3c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 14:40:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"644955bb19afea149811729b25f6614decbedbf4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 15:37:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/64aaf84/ : FAILURE in 56m 13s","accounts_in_message":[],"_revision_number":3},{"id":"5854cbc44433f4a56bc996652e71ebf6631f0fbf","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 16:08:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ff8fb4ed3aad4de165bfc062500859ac1d4752be","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 16:24:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"db68c85a8fd0007226c8623652d279958b6d9225","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 16:42:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a7c8058c4b2abed4417ce03e99004ce31aa697e9","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 17:31:14.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3bd04767b74ef1cf51d56487d2aa7cb59489c207","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 17:49:11.000000000","message":"Patch Set 7: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1bd1b25f35b81a179f8171fcd75830f66f1c73e0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 17:49:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"06767cf2beb58817aeaf8ffca4cf20e58a2c6fd7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 18:25:09.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/7/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/ce312c1/ : FAILURE in 53m 17s","accounts_in_message":[],"_revision_number":7},{"id":"168a07f795809012c03626a39028912381ab4a01","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 19:05:00.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5890d95d68c0c412b48e383bb29c6d9863addd58","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 19:09:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"38ccd6ac59d52e64e017751edba0d95ed2bea4be","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 19:37:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/b48582a/ : FAILURE in 25m 36s","accounts_in_message":[],"_revision_number":9},{"id":"05a256218b0f12684ea7a716a6a26c4ed44e2ba1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 19:40:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a911d318953e4b3f875f82d95b80670a : FAILURE in 21m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/16866a0071ed47f9a7a31753c433295a : FAILURE in 25m 40s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5de5915ae5de4afc8d2bd58c66e4af6f : FAILURE in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f609a0531004800b7b799717f3d8220 : FAILURE in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aaf1508cde524d22b58a32a3f510847b : FAILURE in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e507092c506c4e7894c5f2e63bb5c716 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f52428a2c07443399272621398ae214b : FAILURE in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e5d7a3a553394623a32db1e5e46a4db9 : FAILURE in 4m 09s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8f0bc8edbddcf04388c9e7eae8ade7374fbb6ce8","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 19:52:36.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"9fe14bf887d49d09ee0b5c3738690366a3503edc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 20:17:06.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/10/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/76c096d/ : FAILURE in 23m 03s","accounts_in_message":[],"_revision_number":10},{"id":"6fca582f61275be3c2760f8a823795138b078b21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:29:41.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f6c7db7603ec409b9ec68816cc9093b1 : FAILURE in 21m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/cf003f5d873f445299f24746c65ae4df : FAILURE in 34m 48s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/16991c10c9564ca9b3394da1a007288a : FAILURE in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f76f86185b6421fa8e1cc99f9c0f3f5 : FAILURE in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f831edf7cf134338854bbe0c7bf83ede : FAILURE in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4111ed47eec046a796f4bc7078182824 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fd39fa83d374d8e885b3fe1399303e0 : FAILURE in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dcecc0dcd949484e9ee8d50386062bf8 : FAILURE in 3m 23s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0c33bdf928bac564af321c418e944da863d370e6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 20:34:01.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"3aa66d397b93b2e032cdf979a521cd4d8b521003","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 21:29:12.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/11/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/6a71bc3/ : FAILURE in 54m 35s","accounts_in_message":[],"_revision_number":11},{"id":"471e0fde9223d16c049b75f2a62f8b378a5e2904","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 22:46:36.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0ef2c211ce374053bfe552ea54ae9270 : FAILURE in 1h 14m 15s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/784d540422a94ff3856c9eaaa7bfce03 : FAILURE in 1h 05m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6e6e1e2e021d4a20a6d6483d3bd68178 : FAILURE in 1h 11m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/da2b03ab20eb40158652ddb85c8c7a2b : FAILURE in 1h 13m 25s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/cbe25010b80f4cb3ae2b40d89da528e3 : FAILURE in 1h 11m 05s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/2a750110d4e14fe6abd6d75b5b7a1df3 : FAILURE in 1h 41m 49s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/babab05b75954ceb8d6e5b52e06fe4a7 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5e9fbc72605407fa4add215553300cf : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d03ef35607d40c48c8363d278a00c8b : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddc4d19ee29d4e10baa84d9983c37692 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/31f70e1e77ab49a7a62d255bee16aede : SUCCESS in 5m 34s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6cca1579565444c08341f52c637523b1 : FAILURE in 1h 04m 36s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4056ea7ecf0a4bdd9cb26be031b793a3 : FAILURE in 1h 14m 38s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/3db79795caf74d0990bb143152d1960f : FAILURE in 1h 20m 27s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/6457b2a3e35f4910ac9a795a4cbc24f4 : FAILURE in 1h 07m 13s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/516be30627444ec393460f89a9e227d0 : FAILURE in 1h 02m 00s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/f1f0602f5cb7425791ad7463b7360ad6 : SUCCESS in 1h 49m 55s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/7222086329fa4dc2b130e3f01986b0c2 : FAILURE in 1h 31m 54s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/bc8869c27397419385740c3d46ded8d9 : FAILURE in 1h 37m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"200341d2499735229ca3c1b2239e691821ad75c9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-22 22:51:21.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"06f84141da5736ba8867f64e4bf44307cd8a52f6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 23:09:18.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"36ef3aca1cd4dc4280fa80d11d1fdddac8a70c0c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 23:11:32.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"ac3f9c8920597e937e097c08d6891bcdeb6ea56b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 23:12:15.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"fb687cd7ded3efb6ce972a0d507ad6ad7e56a30c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 00:03:32.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/12/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/57a14b2/ : FAILURE in 52m 51s","accounts_in_message":[],"_revision_number":12},{"id":"cb8fd5135747a9615f1fc33cb8a6861f95273bcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 01:24:42.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b61c6d35906140d58aff7dcf6a14a643 : FAILURE in 1h 15m 55s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/dff574973b164371bbda9d7450cd671d : FAILURE in 1h 07m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/608aaa74d37f4c998a6963777c4fcac6 : FAILURE in 1h 12m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/7a22b095fbdd45b7977c85bc20669cd5 : FAILURE in 1h 21m 09s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/7c34d8baecfd4a23bb45baad37575558 : FAILURE in 1h 08m 24s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/d7553ba814604491bf024789cb8b4434 : FAILURE in 1h 38m 33s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/000323a3c3274dec9aa4ceb2d88f48f0 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64394ad783184efc96398f6bcc3fb321 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0cd11e05821459984a6a4b62d7bc759 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/addc6118c591471bba46d5a74d545eae : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1564ad824e85401bb1aacbe0eac97aab : SUCCESS in 4m 05s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4b2b56de781047ea8fa73aeb7e78f58a : FAILURE in 1h 26m 00s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/d773da9451f94ff18c7ba3b7330db89d : FAILURE in 1h 20m 45s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/79a75e280c5048989aed7b8188b54573 : FAILURE in 1h 29m 58s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/b7af5e6e95094d78b282c339147b7cf8 : FAILURE in 1h 14m 09s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/643ed746dc474ba0a9502f28ad7f8e93 : FAILURE in 1h 04m 54s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/8677b1803d5c48f1a3a9b1c359c4e0d9 : SUCCESS in 1h 47m 53s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/f07e66ddc81247be9c2b24cc62302189 : FAILURE in 1h 48m 30s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/2743a4eda61e4ab3a12855a57b7d822a : FAILURE in 1h 58m 51s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2bef1a3a9f55c9a41b2060b150a6e12cb71226c4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-23 02:46:05.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"4a01f6efdba52d0e883e4970f3e477c7081063f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 03:43:29.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/13/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/5638c3a/ : FAILURE in 56m 21s","accounts_in_message":[],"_revision_number":13},{"id":"68924f8e785033c3ad20699585eb2e8791a4c423","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 04:53:04.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b0565b0fa05d4f11adde3f13bdaacff1 : FAILURE in 1h 08m 57s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e6fb1b51f6644089b438a9df7dbf838d : FAILURE in 1h 01m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a0eaa456931043a5b635d4b3c4be08c7 : FAILURE in 1h 19m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6878c3f961494fe8a2c7fae7dd8d2257 : FAILURE in 1h 14m 09s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/0bbc458d0bbc4c36b4663d639326066c : FAILURE in 1h 04m 03s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7ee4e60eb7854b9087d43fc97b151abb : FAILURE in 1h 42m 47s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c0da721d5d5425fad1c194d3c7c1a66 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11532bd562de4a2abf8a5994381c7513 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51d178c5dea4123912caf11615b5ba7 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c28a259304498c89fd6b8a4782ac10 : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c19927c68e84eeb8cf938100002684b : SUCCESS in 3m 30s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0fdf439040d442e49bbdb17555213b25 : FAILURE in 1h 02m 31s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9b9a9f2cb2cf4d6e999e2fa97171f827 : FAILURE in 1h 14m 56s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/68ee3edfb7844dc484cfa9b4bd3d02cc : FAILURE in 1h 20m 38s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/2a498001c751482cbc3a8258cc2ebc3b : FAILURE in 1h 20m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/4f2e8f44c3d4488f8c92f934637cdf0c : FAILURE in 1h 01m 34s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1905d04a866c41d080bc9e20587387f4 : SUCCESS in 1h 55m 18s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/e50001fc29434da2a8fcbdcda2a5163a : FAILURE in 1h 19m 24s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4c8413747c5148ecbe2e6ec3b18b2f02 : FAILURE in 1h 47m 29s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"1b14fcde9fc085877074580cd8f4dbdb43c7ea41","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-23 12:33:50.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"a54e01fff0bfa60d5f6aecf242f9acc3b0874733","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 14:16:12.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/14/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/090d13c/ : FAILURE in 1h 36m 28s","accounts_in_message":[],"_revision_number":14},{"id":"3e8959ef839b215c5e2a1a832de62e0694caf769","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 14:43:39.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/218ffaa3a4b240a6a1e0d29f21728376 : FAILURE in 1h 16m 35s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/64b586ff59934575a91cc5ec90c6a7ad : FAILURE in 1h 24m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ace98edfb857467c80be963960861e02 : FAILURE in 1h 18m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/3588f39272ac47c6a74d602179024f88 : FAILURE in 1h 15m 43s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/fce15f0f31fc4c5e8f8c0a38ea2b832a : FAILURE in 1h 20m 47s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/fcfaa45130ab421896942ba6f7e5f07e : FAILURE in 1h 51m 16s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7a97c3033164f5d9a05098871f13c45 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d805ef121e54338a35638eff34abe41 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b22d86de3cba40dea0927f5c6c74be38 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37d42b55117e4b2f9b1663ece6a16f03 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08378599242a47b9ae3313792411ca8d : SUCCESS in 4m 45s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/10a7d9da8127474aa25f0fce4a8a52f5 : FAILURE in 1h 24m 41s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b78d3341521a4a00b3a5273d6e607f71 : FAILURE in 1h 21m 14s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/5b620a6c39d3487592b9c05cb4ec2219 : FAILURE in 1h 20m 43s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/174cde26accd4d4aafea9304f2636c45 : FAILURE in 1h 27m 47s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/10ed0e4fce934445a7854b1cb9c815ce : FAILURE in 1h 05m 58s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/faec0c53c23d4aa2b46f661bc0be1923 : SUCCESS in 1h 53m 17s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/d139dc7c40a14c23a28bbca25adcbb0a : FAILURE in 1h 44m 05s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/b5e1030ae2b142ab8d787a0263c80f70 : FAILURE in 1h 53m 32s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"3575c08e3c0261e5c90e69da957981ecec0ef7a1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-23 15:54:11.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"d8b34713a95ad5561b539775237c594240d46104","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-23 15:59:15.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"191f21242b093649dd10e8390416cfed78d1c940","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-23 16:36:25.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"763200d1d650c568195ff29cd3e3ca379f480abb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 19:22:34.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/17/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/36eda30/ : SUCCESS in 2h 45m 19s","accounts_in_message":[],"_revision_number":17},{"id":"40c72ed673f4b12e97e32ffe8f02c695de6540bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 20:03:57.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/575c6772e94f460e9b9096502913bde2 : SUCCESS in 2h 47m 19s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/37d751cb40694f18a79eb07cbbbdce44 : SUCCESS in 2h 02m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a158ac54094d467e862b5b1bc097b760 : SUCCESS in 2h 52m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/78346eae0b90497dbd5fb0fd9a602793 : FAILURE in 2h 44m 32s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/7dc216c31179472093228d3668a35734 : SUCCESS in 2h 07m 47s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/fd4a505b4cf545769c8e5b06804adf1f : SUCCESS in 2h 19m 45s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7defb6d091145f7a19b771cc74b9a78 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb27cb3268474484a94b875da206aafa : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/560886c687bc4ed0af84d5060df45ae8 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c07f8a5b274418a8f9d0fd53e73d3b : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/98089117d0c34780bcead7a772c6341e : SUCCESS in 4m 37s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/94dc4b285dcb48abb0d335af3721dccb : SUCCESS in 3h 10m 05s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0f5f6e6fc0d446f0a250a87af955e040 : FAILURE in 1h 28m 53s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/40d4e8a2cd1147c88bce669a522e16ed : SUCCESS in 2h 10m 36s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/8d38814cbddd463e94dff13bc37e317a : SUCCESS in 2h 07m 30s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/6756387fef26477a9ba2f66b7df1e69c : SUCCESS in 1h 34m 50s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/96926ea4cac14eca9d9818b65346c457 : SUCCESS in 2h 36m 36s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/8aca1efb873b42909333615b1203fd4a : FAILURE in 1h 39m 31s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/6a4f0630325540e883db5f21403823de : FAILURE in 2h 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"bb4c616072b7d31fa577adabdaaf8b5d57255d03","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-25 15:39:01.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"6f71711300ab6874fbc7b54a6944810d573d2db3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 16:02:14.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/dd1c38515f3d4c3dbeded6c9e2db1ab7 : FAILURE in 21m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/215fe9c10e6f4ee5922a872f9d25f7b7 : FAILURE in 21m 45s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d5528fae8be4846b17307b644b66754 : FAILURE in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80756108cd33481a8f41dc12780be0a2 : FAILURE in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2871e6e66fa34582b6f1f4d1b62d54e0 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03ad37e1120c4da59464a0c8c68249c5 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd3f1ed6d00d41d48ac662ae0afd39a1 : FAILURE in 4m 35s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"c8833cf64b2b303e5999068bbe06c433bea8e214","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-25 16:03:06.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/18/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/4eef0d1/ : FAILURE in 23m 14s","accounts_in_message":[],"_revision_number":18},{"id":"b76575060f5fb8daeeb60a8948d6c4729427bf44","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-25 16:35:10.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"bbebeebe7b24dbfc3178324b4fc9161f5fad70ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 19:04:40.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/095352b0ebaf40eabc0bf1c1278d9820 : SUCCESS in 2h 22m 27s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4b514dab2fa3498f87b1fad727043a55 : SUCCESS in 1h 21m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/da506c8e4881463391ba05ca6559cf20 : SUCCESS in 2h 01m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/881d6e73bd0644029de60e7642d8ae21 : SUCCESS in 2h 16m 16s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/5db695129dc948729d079c31ae8357ba : SUCCESS in 1h 46m 24s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/488bf2dc71e6471fb2d2a2c1b0158be7 : SUCCESS in 1h 48m 18s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43301a3ebe6a4a33bc4a9021b878ecff : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c27146d60cd14f78aad819c5f1d32842 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb8087654104a0880df06514d9ba00e : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23072756fed14c29a4a94c53d13251a7 : SUCCESS in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b6436d3ac714016bc28da20ad1913ad : SUCCESS in 3m 39s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7836b5d733c64cd38d515e4258191167 : SUCCESS in 2h 04m 11s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/19470bddf730458ab8e5644c70b5e545 : FAILURE in 1h 16m 25s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/333a4462489b456fba1198af4c6e83e9 : SUCCESS in 1h 33m 47s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/1fbe76b273f94b20994bb69121aee1e2 : SUCCESS in 1h 31m 14s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/bcd275b96f0a4721a9c9369454680c1a : SUCCESS in 1h 08m 18s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/05317ec437704f2cb445138189aef6a2 : SUCCESS in 1h 35m 53s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/28e8de9f78f7440d8134b37b0f20bb83 : FAILURE in 1h 08m 34s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/72aacd2522e341a387da1aeec451020f : FAILURE in 1h 40m 37s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"4e77227b2f2855c02ca31677a86dc768b188c46d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-25 19:07:44.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/19/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/daab2ce/ : SUCCESS in 2h 31m 29s","accounts_in_message":[],"_revision_number":19},{"id":"fa643bfcbe7244fa682923527771723f27f66cc4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-26 01:28:34.000000000","message":"Patch Set 19: Code-Review-2","accounts_in_message":[],"_revision_number":19},{"id":"7e08fb03839519eedebe8eb615ac0e0bd5b76e5b","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-26 13:18:30.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"e7d35ece432a1a0520783a3d304cd83f4af3a058","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-26 14:29:43.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"a8b83cf58895fdd4b8bfaab8b21fdcc39c787c3a","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-26 15:38:51.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"b55f57b1069aa5fb484dee3454b2eebc0b6804d8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-26 17:39:55.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/20/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/b612e2f/ : FAILURE in 2h 00m 04s","accounts_in_message":[],"_revision_number":20},{"id":"45b4d86baca8f0354f100697fd68c3019ef80cad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 20:16:34.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3906596036714f9c8c5edea20d8cc379 : SUCCESS in 2h 49m 35s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/74ad583bbb164d05a76c111eb396a7b7 : FAILURE in 18m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5753fc277a6248d297d1eb2a22624e8a : SUCCESS in 2h 26m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/cacb991c821346ccbdee677854328add : FAILURE in 25m 00s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/0400c623fc154000ad46bac6c85e6019 : FAILURE in 22m 46s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/a31ed300d5ad425dbedd14382386105b : FAILURE in 1h 17m 39s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3def1e6b0bf343eaa9f0cd1c5923eb3f : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74135d04337248fb8d76b9a3ddcf1cc8 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f151b80f6e2c446d85a66f964f64ea85 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abb07c7b9bbe4a7c8db18c6bf2ea3df4 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1524c04e68f44f680674c7debd82a34 : SUCCESS in 3m 22s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a56bdc97d61a4c6eac1e09c47ed60294 : FAILURE in 23m 13s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a81b3831f6df4630963ece9a846de102 : FAILURE in 1h 38m 28s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/72550440a2d9469c90058e75c2fc2a78 : SUCCESS in 2h 16m 49s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/645bd3a0187e4cbd9bf70c293c30bbb6 : FAILURE in 17m 49s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/bae6b1d4ccb74d5e9635fb5fa1f4f196 : SUCCESS in 1h 26m 12s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/11d5b59f402a4e42b2db4fa2e380803f : SUCCESS in 2h 15m 19s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/38d4c59628a546bfab3371f31bd73d54 : FAILURE in 1h 44m 17s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/0921b4023b7d4dc4992efafc83dddd44 : FAILURE in 2h 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"40d8a4f62321a19be8bf2ca24bbbe53fe38f05ef","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-27 06:19:12.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"5e97479c91e27d3d232a26e9174e91a3776f1085","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-27 14:48:27.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"ec75d5001bcf5caea3ac420ded380e6929b2568e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-27 17:17:03.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"96e000eee25c1e5adb232fe6b195b539f8228961","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-27 17:53:50.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"9473c3926de0a7792a6714a8e3c67a6557494023","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-27 21:27:16.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/21/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/7c9abd5/ : FAILURE in 2h 12m 14s","accounts_in_message":[],"_revision_number":21},{"id":"a37a6ee9885c7eac448556b7c5afc9382703d816","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 23:08:59.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/15f948d9e98542ad9390c45bf8c8f4e5 : SUCCESS in 2h 58m 14s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a8168050d5114c27a8c401176a59b59a : SUCCESS in 1h 56m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3149863a63ba4fe78ea10434cd935142 : SUCCESS in 2h 30m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d44dc6c9aba24d5f98daa590bd28fbe1 : SUCCESS in 3h 01m 24s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/7d38807b288b494e92a33ec58e31d16f : SUCCESS in 2h 10m 52s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/305d7e8226124673bab87f5ab2646c59 : SUCCESS in 2h 38m 00s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3a65f9b1a0f43e9b6a56c8e1acffc17 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d234e8862d3941bbb5704eb9674abc9e : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a53e9275aa2a4f6292f07b57975e9521 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e12d4c728d24543bd75c4a8f42e874c : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92641567243c4f67b189865f6844e1d1 : SUCCESS in 3m 47s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/baac31efefbe4505b619ab29b6695488 : SUCCESS in 3h 01m 14s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c59a5de4930b4724b837c99785d606d2 : FAILURE in 1h 59m 36s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b4adba4917364c9f8129f4472ebb82f7 : SUCCESS in 2h 09m 00s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/b8c976c3047645d2828816f961607beb : SUCCESS in 1h 51m 29s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/d0fb3d482b5c4a6495173e268add155d : SUCCESS in 1h 30m 53s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ff2c0a7c83d344b48b61690e07620a23 : SUCCESS in 2h 06m 18s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/e0fadfb87df04ad387685e9ebbc8f123 : FAILURE in 1h 35m 57s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d606c9e3738840a199f9b503561b3a44 : FAILURE in 2h 08m 03s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"171d3a9dc3356149c30e81d8e736f8eddba9a802","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-27 23:19:05.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"67defba2357b50ddb42f2e78b51b98d4630f16ff","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-27 23:19:14.000000000","message":"Patch Set 22: -Code-Review\n\nthis one is ready for review I think.","accounts_in_message":[],"_revision_number":22},{"id":"5f091da5f6eb028e6548a50c8b8ec6eae5b46db6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-28 02:27:34.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/60/677860/22/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/4fd8ca3/ : SUCCESS in 3h 07m 32s","accounts_in_message":[],"_revision_number":22},{"id":"fe3727a2d95c4f5e3abd3535df1de30ea78fc6d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 03:08:28.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9fb9c6133d2a4f1c800266852c9de93c : SUCCESS in 3h 07m 34s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/53bbda1dd4974b4a82210649bcbeb544 : SUCCESS in 1h 50m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c1537cdadad34846bdced3534e9fac52 : SUCCESS in 2h 45m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/97dbd265ddf54fc8b5c6eb112150b001 : SUCCESS in 2h 45m 18s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/cae022d78ccf45db93d1e75a078ceb5e : SUCCESS in 2h 13m 06s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7b8de75f8a0b4fd89361e452d13c0528 : SUCCESS in 2h 00m 14s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04366cab9d144da8abd80f66aa8feffd : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/312348e219214ace9f8a05a0492fee3b : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21038cdaab364e08abc118c9958ec2e0 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d9267e801574837925405119c42ec7f : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f37339f33b524ac89d6e3ed042812525 : SUCCESS in 3m 46s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/74bd4788100541efa25762ea53dad996 : SUCCESS in 2h 46m 09s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2b76652e6a9644c9a19d020879863d6f : FAILURE in 1h 29m 15s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/469eed8a8669408fa131f29a900a815e : SUCCESS in 1h 43m 51s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/706744734ba54be58e97e212e5d13ca1 : SUCCESS in 1h 47m 35s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/f8c2aef98f9242378c7ce02a53a7c175 : SUCCESS in 1h 31m 09s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/01821d73602a4c4ab98f2930e61ee521 : SUCCESS in 2h 01m 44s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/b077881f59cf42d5aa5dc303f0a5a00f : FAILURE in 1h 26m 40s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d12bef56ae704c1382aafd31c934dd33 : FAILURE in 2h 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"86474d68a3fdddd93d25aadf5785292713efecfd","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-08-28 05:48:49.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"25827284bfa2dcf26cc36fc582c7824abefb219e","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-28 06:11:57.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"ece7f8092d4e863aa66f0fda5388af12dd054152","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-28 07:51:06.000000000","message":"Patch Set 22: Code-Review-1\n\nAs I noted earlier, we don\u0027t need to use any additional checks.","accounts_in_message":[],"_revision_number":22},{"id":"fd055f0c083f81bb37022b174abd504b405b0bfd","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-28 08:04:50.000000000","message":"Patch Set 22: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"d53104418ed5ad641bcf992cb052f015c1ae9024","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-28 08:06:15.000000000","message":"Patch Set 22: Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"8b7fde754612f4f4b7d0aacb408471820b4af22d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 08:06:32.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"f4e40ff65dd46434534351ab5c37c6c0a9eb84e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 11:09:45.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1e9684155b7d40e4a6a79196dae78e80 : SUCCESS in 2h 59m 21s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/b6f6d181e97c4efda5ee0650087415bf : SUCCESS in 2h 17m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a9a6229b4d9f480f91e21d6fb6f53c5c : SUCCESS in 2h 35m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a27a2324bd5e41a7b548254a83daf717 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b5dc5a93c324d3e994d1b81c71ed79e : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f402eba67d3d45e39e0646c4614b1c4a : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06a31ca84e64c3295e22bf55222e18c : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a12734e4788a46eaac15b08e155cf64a : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":22},{"id":"7d1b92d1626d377014fa05e3f326b0590ea97305","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 11:09:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":22},{"id":"88375271303a0bd675b8029b6bbcb4eec3536d89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 11:12:00.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d06d2d37a7749ad9c31ef854770642c : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"983ab98f61f530eec7cac31e3a0556bf6722e2a0","revisions":{"dd78c6d14cbd2294d9b4a66a5c8258cc7646f5c0":{"kind":"REWORK","_number":1,"created":"2019-08-22 02:07:22.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/1"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:57.000000000","tz":-240},"subject":"runner/discover_container_name: force ps --sync when podman","message":"runner/discover_container_name: force ps --sync when podman\n\nRunning \"podman ps --sync (...)\" is expensive but it is needed to\navoid race conditions when the containers are started.\nThis flag forces a sync of Podman\u0027s state against the OCI\nruntime, resolving some state desync errors\n\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/dd78c6d14cbd2294d9b4a66a5c8258cc7646f5c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/dd78c6d14cbd2294d9b4a66a5c8258cc7646f5c0"}]},"branch":"refs/heads/master"},"61d6fb0e18321383ed19918f4b4a4245459d0d7c":{"kind":"REWORK","_number":2,"created":"2019-08-22 04:00:56.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/2"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 04:00:05.000000000","tz":-240},"subject":"WIP - runner/discover_container_name: drop -a and force -sync for podman ps","message":"WIP - runner/discover_container_name: drop -a and force -sync for podman ps\n\nRunning \"podman ps --sync (...)\" is expensive but it is needed to\navoid race conditions when the containers are started.\nThis flag forces a sync of Podman\u0027s state against the OCI\nruntime, resolving some state desync errors.\n\nAlso, removing the -a to only list the running containers.\n\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/61d6fb0e18321383ed19918f4b4a4245459d0d7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/61d6fb0e18321383ed19918f4b4a4245459d0d7c"}]},"branch":"refs/heads/master"},"f03e144afe79439510fe6bbfdf3c91d5420be27b":{"kind":"REWORK","_number":3,"created":"2019-08-22 14:40:13.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/3"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 14:39:43.000000000","tz":-240},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- add systemd support\n- figure out if we need a sleep in the while between each iterations (in ms)\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f03e144afe79439510fe6bbfdf3c91d5420be27b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f03e144afe79439510fe6bbfdf3c91d5420be27b"}]},"branch":"refs/heads/master"},"645999d233c357e7bd15d0e6612dec862b91cc18":{"kind":"REWORK","_number":4,"created":"2019-08-22 16:08:42.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/4"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 16:08:05.000000000","tz":-240},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- add systemd support\n- figure out if we need a sleep in the while between each iterations (in ms)\n- refactor the usage of discover_container_name in builder/base.py\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/645999d233c357e7bd15d0e6612dec862b91cc18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/645999d233c357e7bd15d0e6612dec862b91cc18"}]},"branch":"refs/heads/master"},"c2c21a8a31827bcd6a964149796f0eadebe96fb9":{"kind":"REWORK","_number":5,"created":"2019-08-22 16:24:14.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/5"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 16:24:10.000000000","tz":-240},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- add systemd support\n- figure out if we need a sleep in the while between each iterations (in ms)\n- refactor the usage of discover_container_name in builder/base.py\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/c2c21a8a31827bcd6a964149796f0eadebe96fb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/c2c21a8a31827bcd6a964149796f0eadebe96fb9"}]},"branch":"refs/heads/master"},"3102d2ee280e2f265d3374bad7c9783b00e42497":{"kind":"REWORK","_number":6,"created":"2019-08-22 16:42:14.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/6"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 16:37:00.000000000","tz":-240},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- add systemd support\n- figure out if we need a sleep in the while between each iterations (in ms)\n- refactor the usage of discover_container_name in builder/base.py\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/3102d2ee280e2f265d3374bad7c9783b00e42497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/3102d2ee280e2f265d3374bad7c9783b00e42497"}]},"branch":"refs/heads/master"},"7b78c2c5125b0271200f40627276ff52d02a67fa":{"kind":"REWORK","_number":7,"created":"2019-08-22 17:31:14.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/7"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 17:30:50.000000000","tz":-240},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- figure out if we need a sleep in the while between each iterations (in ms)\n- more unit tests?\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/7b78c2c5125b0271200f40627276ff52d02a67fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/7b78c2c5125b0271200f40627276ff52d02a67fa"}]},"branch":"refs/heads/master"},"baee8ecbb01b7cc86ca1c0efa6bc003772598c8a":{"kind":"REWORK","_number":8,"created":"2019-08-22 19:05:00.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/8"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 19:04:53.000000000","tz":-240},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- figure out if we need a sleep in the while between each iterations (in ms)\n- more unit tests?\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/baee8ecbb01b7cc86ca1c0efa6bc003772598c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/baee8ecbb01b7cc86ca1c0efa6bc003772598c8a"}]},"branch":"refs/heads/master"},"9b394b76d03de47afb2f5a51f16eb45d44b1060e":{"kind":"REWORK","_number":9,"created":"2019-08-22 19:09:10.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/9"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 19:09:06.000000000","tz":-240},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- figure out if we need a sleep in the while between each iterations (in ms)\n- more unit tests?\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/9b394b76d03de47afb2f5a51f16eb45d44b1060e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/9b394b76d03de47afb2f5a51f16eb45d44b1060e"}]},"branch":"refs/heads/master"},"179452685cd4df2dd27fd8424eeec9d3ea7a860d":{"kind":"REWORK","_number":10,"created":"2019-08-22 19:52:36.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/10"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 19:52:33.000000000","tz":0},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- figure out if we need a sleep in the while between each iterations (in ms)\n- more unit tests?\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/179452685cd4df2dd27fd8424eeec9d3ea7a860d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/179452685cd4df2dd27fd8424eeec9d3ea7a860d"}]},"branch":"refs/heads/master"},"53be95b95d935616195f35d27c8cc2e0b271f62c":{"kind":"REWORK","_number":11,"created":"2019-08-22 20:34:01.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/11"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 20:33:27.000000000","tz":-240},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- figure out if we need a sleep in the while between each iterations (in ms)\n- more unit tests?\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/53be95b95d935616195f35d27c8cc2e0b271f62c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/53be95b95d935616195f35d27c8cc2e0b271f62c"}]},"branch":"refs/heads/master"},"e8a90d8e359ae3a67327de7e49e7d3840db92a22":{"kind":"REWORK","_number":12,"created":"2019-08-22 23:09:18.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/12"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 23:09:14.000000000","tz":-240},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- figure out if we need a sleep in the while between each iterations (in ms)\n- more unit tests?\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/e8a90d8e359ae3a67327de7e49e7d3840db92a22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/e8a90d8e359ae3a67327de7e49e7d3840db92a22"}]},"branch":"refs/heads/master"},"1d7abcbedf4f7e535d65b1939ce044712c2b0db7":{"kind":"REWORK","_number":13,"created":"2019-08-23 02:46:05.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/13"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-23 02:46:00.000000000","tz":-240},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- figure out if we need a sleep in the while between each iterations (in ms)\n- more unit tests?\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/1d7abcbedf4f7e535d65b1939ce044712c2b0db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/1d7abcbedf4f7e535d65b1939ce044712c2b0db7"}]},"branch":"refs/heads/master"},"59d90a6dd94e1c020e3816f0dcb0cbd2208f7f95":{"kind":"REWORK","_number":14,"created":"2019-08-23 12:33:50.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/14"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-23 12:33:44.000000000","tz":-240},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- figure out if we need a sleep in the while between each iterations (in ms)\n- more unit tests?\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/59d90a6dd94e1c020e3816f0dcb0cbd2208f7f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/59d90a6dd94e1c020e3816f0dcb0cbd2208f7f95"}]},"branch":"refs/heads/master"},"72208cb0afc4f80a9340786799ad85768cb0ad82":{"kind":"REWORK","_number":15,"created":"2019-08-23 15:54:11.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/15"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-23 15:54:03.000000000","tz":-240},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- more unit tests?\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/72208cb0afc4f80a9340786799ad85768cb0ad82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/72208cb0afc4f80a9340786799ad85768cb0ad82"}]},"branch":"refs/heads/master"},"c8f34a637557435a277b62c7c5ecb59f71974488":{"kind":"REWORK","_number":16,"created":"2019-08-23 15:59:15.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/60/677860/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/16"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-08-23 15:59:06.000000000","tz":120},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- more unit tests?\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/c8f34a637557435a277b62c7c5ecb59f71974488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/c8f34a637557435a277b62c7c5ecb59f71974488"}]},"branch":"refs/heads/master"},"5317569ee65bd1e2db89d9d74b646d3de839017a":{"kind":"REWORK","_number":17,"created":"2019-08-23 16:36:25.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/17"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-23 16:36:20.000000000","tz":-240},"subject":"WIP - introduce container_running() and use it before running podman exec","message":"WIP - introduce container_running() and use it before running podman exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime.\n\nIt will be first used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error.\n\nTODO:\n- more unit tests?\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5317569ee65bd1e2db89d9d74b646d3de839017a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5317569ee65bd1e2db89d9d74b646d3de839017a"}]},"branch":"refs/heads/master"},"a8660b36680452f471996add85c2832ab0db2a78":{"kind":"REWORK","_number":18,"created":"2019-08-25 15:39:01.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/18"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-25 15:38:53.000000000","tz":-240},"subject":"Check if container is running before doing an exec","message":"Check if container is running before doing an exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime. Before doing a\n\"podman ps\", we try to check if the service is running in systemd.\nThere is a very short sleep between the retries to give a chance to\npodman to find the container if it takes a bit of time to start or seen\nas started.\n\nIt will be used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error and stop the deployment.\n\nThis patch is mainly a workaround against a race condition where in\nheavy-loaded environments, an exec can be run too early in a step where\nthe container is still starting.\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/a8660b36680452f471996add85c2832ab0db2a78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/a8660b36680452f471996add85c2832ab0db2a78"}]},"branch":"refs/heads/master"},"15d8e02afdf26f5087c7bcfd6f7d1207570de339":{"kind":"REWORK","_number":19,"created":"2019-08-25 16:35:10.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/19"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-25 16:35:01.000000000","tz":-240},"subject":"Check if container is running before doing an exec","message":"Check if container is running before doing an exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime. Before doing a\n\"podman ps\", we try to check if the service is running in systemd.\nThere is a very short sleep between the retries to give a chance to\npodman to find the container if it takes a bit of time to start or seen\nas started.\n\nIt will be used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error and stop the deployment.\n\nThis patch is mainly a workaround against a race condition where in\nheavy-loaded environments, an exec can be run too early in a step where\nthe container is still starting.\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/15d8e02afdf26f5087c7bcfd6f7d1207570de339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/15d8e02afdf26f5087c7bcfd6f7d1207570de339"}]},"branch":"refs/heads/master"},"9f22ce6986ed203988a4605ea1afea6ced7db719":{"kind":"REWORK","_number":20,"created":"2019-08-26 15:38:51.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/60/677860/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/20"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-08-26 15:15:33.000000000","tz":120},"subject":"Check if container is running before doing an exec","message":"Check if container is running before doing an exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime. Before doing a\n\"podman ps\", we try to check if the service is running in systemd.\nThere is a very short sleep between the retries to give a chance to\npodman to find the container if it takes a bit of time to start or seen\nas started.\n\nIt will be used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error and stop the deployment.\n\nThis patch is mainly a workaround against a race condition where in\nheavy-loaded environments, an exec can be run too early in a step where\nthe container is still starting.\n\nIt also consolidate the discover_container_name method in order to get\nmore chance to actually get a name.\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/9f22ce6986ed203988a4605ea1afea6ced7db719"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/9f22ce6986ed203988a4605ea1afea6ced7db719"}]},"branch":"refs/heads/master"},"362391d3daec9749489d21a66a3b2087edc0a451":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2019-08-27 17:53:50.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/21"}}},"commit":{"parents":[{"commit":"d8a1c044e9c7e6015c4d569514244faabadf3855","subject":"Revert \"Fix mismatching fixed vs unique container names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/d8a1c044e9c7e6015c4d569514244faabadf3855"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-27 17:53:50.000000000","tz":0},"subject":"Check if container is running before doing an exec","message":"Check if container is running before doing an exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime. Before doing a\n\"podman ps\", we try to check if the service is running in systemd.\nThere is a very short sleep between the retries to give a chance to\npodman to find the container if it takes a bit of time to start or seen\nas started.\n\nIt will be used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error and stop the deployment.\n\nThis patch is mainly a workaround against a race condition where in\nheavy-loaded environments, an exec can be run too early in a step where\nthe container is still starting.\n\nIt also consolidate the discover_container_name method in order to get\nmore chance to actually get a name.\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/362391d3daec9749489d21a66a3b2087edc0a451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/362391d3daec9749489d21a66a3b2087edc0a451"}]},"branch":"refs/heads/master"},"983ab98f61f530eec7cac31e3a0556bf6722e2a0":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2019-08-27 23:19:05.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/60/677860/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/60/677860/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/60/677860/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/60/677860/22"}}},"commit":{"parents":[{"commit":"b03bdce32213091145b6256de182a7459ad3c84d","subject":"[Configuration] Switch to stestr.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/b03bdce32213091145b6256de182a7459ad3c84d"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-22 02:06:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-27 23:19:05.000000000","tz":0},"subject":"Check if container is running before doing an exec","message":"Check if container is running before doing an exec\n\ncontainer_running is a new method which will allow to return True if a\ncontainer is detected as running or False if not running.\nThere is a retry mechanism which if \"podman ps\" is used, will add \"--sync\" to\nthe command so we synchronize the state of OCI runtime. Before doing a\n\"podman ps\", we try to check if the service is running in systemd.\nThere is a very short sleep between the retries to give a chance to\npodman to find the container if it takes a bit of time to start or seen\nas started.\n\nIt will be used by the builder when a container is configured to\nrun \"podman exec\"; we\u0027ll first verify that the container exist otherwise\nreturn an error and stop the deployment.\n\nThis patch is mainly a workaround against a race condition where in\nheavy-loaded environments, an exec can be run too early in a step where\nthe container is still starting.\n\nIt also consolidate the discover_container_name method in order to get\nmore chance to actually get a name.\n\nCo-Authored-By: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1839559\nChange-Id: If4d8c268218bf83abed877a699fc583fb55ccbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/983ab98f61f530eec7cac31e3a0556bf6722e2a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/983ab98f61f530eec7cac31e3a0556bf6722e2a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
