)]}'
{"id":"openstack%2Fpaunch~678117","triplet_id":"openstack%2Fpaunch~master~I6b39714e2b8e19d5439d1eb7629e061cf986a7b1","project":"openstack/paunch","branch":"master","topic":"story/2006430","hashtags":[],"change_id":"I6b39714e2b8e19d5439d1eb7629e061cf986a7b1","subject":"Change cmd checking order for backward compatibility","status":"ABANDONED","created":"2019-08-23 01:24:54.000000000","updated":"2019-09-04 05:34:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"71b5a03cdf2c9e27d5f8a3ddbd969630b9ec53ee","_number":678117,"virtual_id_number":678117,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":1,"date":"2019-08-23 04:44:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-23 04:08:12.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"date":"2019-08-26 04:35:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":-1,"date":"2019-08-25 21:48:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"date":"2019-08-23 01:50:36.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 01:50:36.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-23 04:08:12.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-23 04:44:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-25 21:48:34.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2019-08-26 04:35:28.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"}],"messages":[{"id":"a54bf835261f1e617cf64d1cf023597bb7f15ae1","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-08-23 01:24:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"252d7375808d143ed08ab614f97e18dc1db01928","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 04:08:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/17/678117/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/2e6a749/ : SUCCESS in 2h 42m 26s","accounts_in_message":[],"_revision_number":1},{"id":"580ce0307df535b9fcb7491a659c5cddf30df65e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 04:44:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b01e9472e63f48f99648ca6016d62706 : SUCCESS in 3h 01m 23s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c0942e2bb0604aad84c718d63122405e : SUCCESS in 1h 44m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/110b04da545f41958249942abec746de : SUCCESS in 2h 43m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d9fa8a9ae0e045e28fe61292b95e8985 : SUCCESS in 2h 52m 19s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/2994a893ed604021ae4f6743007c47bf : SUCCESS in 2h 19m 28s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/2893c5d7a5b34a9c8bb3e197b329c167 : SUCCESS in 2h 06m 23s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e197365d29e468cb73014981e4f7e04 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d6c4b1cb8e14bfb879f8d94e5e56f8a : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c1ed058a13b4878b149599d2e8c09a5 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42020e954f404146a5e11f11e31197d1 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5260934dc6345e2b98698b6658ee082 : SUCCESS in 3m 45s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/806f8a17836f479ea9a5e0c1670c27ee : SUCCESS in 2h 50m 34s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/19a2b30452a24fd88114d1c7cde66987 : FAILURE in 1h 47m 04s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/0a0cc970b06648a69049c53bb347b086 : SUCCESS in 2h 11m 58s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/e77a93efb75c4fcd9f3ac0c8345258d1 : SUCCESS in 1h 51m 36s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/7f37f339645f42e09c9497c75ac1d493 : SUCCESS in 1h 34m 33s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/3930f69817d94689b79ae8c4e39409cc : SUCCESS in 2h 04m 49s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/7515038169c14a3bbf36fd667a2afa2a : FAILURE in 1h 50m 39s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/8ba249afa9ac45498717cd812c531c86 : FAILURE in 2h 06m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62e94f1eda7bff6bf8db313df505354aa2df49d9","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-23 17:48:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f707d940063eac4f30a0ee05a92e2c8fd28a0984","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-25 21:48:34.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027m proposing https://review.opendev.org/#/c/678434 as a way of preventing this sort of breakage from ever happening again","accounts_in_message":[],"_revision_number":1},{"id":"f74efbcd2fa8dc4f35c6d98746f13c5b20fd02c4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-26 04:35:28.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"d9661d985ede82b8ead38a13e526a1d9e2b3d2b7","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-09-04 05:34:42.000000000","message":"Abandoned\n\nIn favor of https://review.opendev.org/#/c/678434/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2d7e4b377be11a6ed9e2ee1028c36148f26401b","revisions":{"e2d7e4b377be11a6ed9e2ee1028c36148f26401b":{"kind":"REWORK","_number":1,"created":"2019-08-23 01:24:54.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/17/678117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/17/678117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/17/678117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/17/678117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/17/678117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/17/678117/1"}}},"commit":{"parents":[{"commit":"2de73e1ebbafb00a43f62a167b2130d288706e65","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/2de73e1ebbafb00a43f62a167b2130d288706e65"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2019-08-23 01:22:46.000000000","tz":480},"committer":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2019-08-23 01:22:46.000000000","tz":480},"subject":"Change cmd checking order for backward compatibility","message":"Change cmd checking order for backward compatibility\n\nChange-Id: I6b39714e2b8e19d5439d1eb7629e061cf986a7b1\nTask: 36339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/e2d7e4b377be11a6ed9e2ee1028c36148f26401b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/e2d7e4b377be11a6ed9e2ee1028c36148f26401b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
