)]}'
{"id":"openstack%2Fpaunch~681257","triplet_id":"openstack%2Fpaunch~master~I9443ad60affe9c7823b17daa259efee542c6fe22","project":"openstack/paunch","branch":"master","topic":"cpuset","hashtags":[],"change_id":"I9443ad60affe9c7823b17daa259efee542c6fe22","subject":"Add --cpuset-cpus support for both Docker and Podman","status":"MERGED","created":"2019-09-10 13:39:41.000000000","updated":"2019-10-01 17:11:18.000000000","submitted":"2019-10-01 17:10:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681257-1569949809725-aecd3d01","meta_rev_id":"4df5bf76e4a3e6b9eff67f5c85baf39d70e04e7c","_number":681257,"virtual_id_number":681257,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":23618,"name":"Christophe Fontaine","email":"cfontain@redhat.com","username":"christophefontaine"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 17:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-24 06:53:48.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-09-24 03:48:54.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":23618,"name":"Christophe Fontaine","email":"cfontain@redhat.com","username":"christophefontaine"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-01 13:01:32.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":23618,"name":"Christophe Fontaine","email":"cfontain@redhat.com","username":"christophefontaine"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23618,"name":"Christophe Fontaine","email":"cfontain@redhat.com","username":"christophefontaine"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 06:09:14.000000000","updated_by":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"reviewer":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"state":"REVIEWER"},{"updated":"2019-09-12 06:41:29.000000000","updated_by":{"_account_id":23618,"name":"Christophe Fontaine","email":"cfontain@redhat.com","username":"christophefontaine"},"reviewer":{"_account_id":23618,"name":"Christophe Fontaine","email":"cfontain@redhat.com","username":"christophefontaine"},"state":"REVIEWER"},{"updated":"2019-09-14 08:59:03.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-09-16 13:37:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-23 13:20:03.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2019-09-24 06:53:48.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-01 17:10:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d7c11867aed86b1ccfe0b3f1a3122d6c5fe48c99","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-10 13:39:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a81da47a25fa2676546b4289e859aa70d02d777c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-10 14:59:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"12e1f2387406c84bc670ef1d4abbf52d93a99ea1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 17:19:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8e1eee35b39147a986f272eb26471e85 : SUCCESS in 3h 07m 10s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/012871dec60243f39768daea70e446ae : SUCCESS in 2h 02m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ff26ffd271754cf4bd682a5b72bf39de : SUCCESS in 2h 50m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5c76b993c8a64709a9396647912c0c3e : SUCCESS in 2h 31m 00s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/cba783044b5c4925b2d82e929fb50378 : SUCCESS in 2h 01m 24s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/aca56cb85ff04f4aa63dcd96359ce20b : SUCCESS in 2h 16m 00s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47a2051bef7145ae91e57d1c9b160ff3 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e257af2fbf8345afb764483bd3bc7d61 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0aa507e8bdf48908aebde8d3f5b92e4 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80cd7f0d93c2423ab71b6b598eedd23c : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/05220525d9d8476f8197a9cd89978d44 : SUCCESS in 3m 03s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1dd46a5dd5094fe7b959ce29055a9bd7 : SUCCESS in 3h 04m 19s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/84698f129e4d44f8803dc8bfc4eb91ec : FAILURE in 1h 51m 53s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e65b7d8814fc48b9b91f638bd141d0a9 : SUCCESS in 2h 19m 53s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/e39cea2f402141b1ab6b5a5d7918bee9 : SUCCESS in 1h 44m 26s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/86678c77bd2243239888e59dab357237 : SUCCESS in 1h 37m 30s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/2963f4fc58cc430085213be5547dc53b : SUCCESS in 2h 15m 22s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/fab6e177d0124a8b98f0c50a968fd9b8 : FAILURE in 2h 00m 14s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/0f3a63fe77c84d4d973d0402739e0969 : FAILURE in 1h 39m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c76f484ca73afb6b28697653e96ed1ce10c215a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-10 18:09:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d676ba55278e7f6cd91a4a4ce060aa8e621e922d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-10 18:13:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8cd9fad375de2c87038c7002a6a1efd5a2131565","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-10 18:15:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1cd3d48d47e55f714556d837ffd80ee1a7d747eb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-10 18:29:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"448449cc6e30d2b353a62a8806449ac0ecc7f7a2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-10 18:32:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/2/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/ee5cc55/ : FAILURE in 21m 31s","accounts_in_message":[],"_revision_number":2},{"id":"01634a39d47447e72570302cad557ffce7db4986","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-10 18:33:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3650772e197e1b994bc47ba54e1557ea103158a0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-10 18:52:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cbb532e50482e60c53af6f8baaa6bcbd83fc2cc0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-10 18:52:34.000000000","message":"Patch Set 3: Workflow-1\n\nI haven\u0027t addressed Alex\u0027s comment yet, I\u0027m waiting for Christophe\u0027s input on the needs.","accounts_in_message":[],"_revision_number":3},{"id":"e6849382b4823287bd4bd1dd7092e63bc589d1d8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-10 19:14:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/5e9fef5/ : FAILURE in 22m 01s","accounts_in_message":[],"_revision_number":3},{"id":"40ef8dc626b01c11e582dda4cd49ba417a96e53e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 20:40:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/798942c782cd4c869695e9eb5f8bbe76 : FAILURE in 28m 48s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/874a77722ec34e0eb26f1d10a565226b : FAILURE in 21m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/fb8a45de53be4f349ed98010df1c17a3 : FAILURE in 23m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/894e505d0dbb449f9ca79c5bea411d05 : FAILURE in 24m 04s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/82d22c4f0847445da11413a48d6ba80c : FAILURE in 21m 27s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/509595cd582743b59bf0880114c059c5 : FAILURE in 1h 32m 21s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f501dcfb528a425297d2e91a889f7212 : SUCCESS in 2m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75e5491f58a4419ab365c1e0ff5edb8a : SUCCESS in 3m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/29d391141eec409fb66d8045df9d3131 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b4fbf90602f43f09526750595c9267a : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/192b05b1129843898974bbe8f325ccad : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79267006ae194c5ba04aef95dbaa3cd1 : SUCCESS in 3m 00s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/04ac412022b14fdd934ef21fac9aecfb : FAILURE in 33m 48s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a49bfeb84cfe440fa4dd0ce0e351820e : FAILURE in 29m 47s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e5db9e99f5c044cbb7ade76e9e14a0f9 : FAILURE in 25m 04s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/ed4bc34047fc478490786dc2487f5973 : FAILURE in 26m 07s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/defbc2dfc39948129a5fefb35ec15a6e : FAILURE in 25m 20s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/2f3c2d048f914c8989e1f02d2a3aae29 : FAILURE in 27m 35s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/f8d03a564ae1470abe99849994c629fc : FAILURE in 17m 31s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ff851a0a87bc4acbb164657b745b52b2 : FAILURE in 1h 08m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"04ee44888040b9f10db32321ab53f40339c72e9c","author":{"_account_id":23618,"name":"Christophe Fontaine","email":"cfontain@redhat.com","username":"christophefontaine"},"date":"2019-09-11 08:52:30.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"67095a0eb6bb81fc6d896b81e08567f1691fcce2","author":{"_account_id":23618,"name":"Christophe Fontaine","email":"cfontain@redhat.com","username":"christophefontaine"},"date":"2019-09-11 08:58:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"680af6f3ed55f37e4bd2be35f784f45c181a2289","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-11 16:59:59.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"aa769615461a06ca54d0c30dccf13ef23ed1537e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-11 17:00:02.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"57ecc3e74a15ea07f0402771d6cb68386ce98bd0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-11 17:01:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a31df010dce4fe62e573372c0d52d3f424036ae2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-11 17:04:48.000000000","message":"Patch Set 4:\n\nNote that the rpm will need the new psutil dependency: https://review.rdoproject.org/r/#/c/22213/","accounts_in_message":[],"_revision_number":4},{"id":"42a9ec22e4f8c34421f133914874c7230605a0d7","author":{"_account_id":23618,"name":"Christophe Fontaine","email":"cfontain@redhat.com","username":"christophefontaine"},"date":"2019-09-11 17:10:56.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1d5bbcf842861235530a5aa7f751b77fa07219cd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-11 17:25:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/4/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/409d3f3/ : FAILURE in 22m 40s","accounts_in_message":[],"_revision_number":4},{"id":"40fba31e240d41661fd1027f2289c5e0a1cd6a5f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-11 17:30:38.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7b6823495c5ccdee67929a743ffe0b50870bab46","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-11 17:35:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"28fb69a0e7bb5cfed174934fb44414a6ef01b0c9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-11 17:57:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/5/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/10b68fb/ : FAILURE in 21m 02s","accounts_in_message":[],"_revision_number":5},{"id":"67e608252131a604946c94d7f8a9939b72e525b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 22:36:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6858a63958b24fdcb242957cf71d366b : FAILURE in 30m 49s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/844eaa33515b445dac7b705389554e05 : FAILURE in 24m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/59f0220ce354484e9512cea060ae3bb4 : FAILURE in 36m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/12265729ccbf44e28ec5a6c4929f4cce : FAILURE in 33m 01s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ae0607cc62824cd5b2c591dda47fcee3 : FAILURE in 29m 58s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/716235a3358d4c03906f55b0be294233 : FAILURE in 1h 33m 04s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb00e4fa783a4fc7865d84b192a9c901 : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2ea162a9d3346e2a29580f39fdbf700 : SUCCESS in 4m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/023ea95c8219451281c9e5a5ef3010df : SUCCESS in 2m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fbe37c664f649f9892a822f69166b91 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4892af42aeeb437d94692710b68eeec3 : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/20a5cb599dcc42beb339939158835b39 : SUCCESS in 3m 41s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b3ead8ae99284802bee601d1985a958b : FAILURE in 26m 14s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0f71ddbab3434bbfa190c50db841db2f : FAILURE in 27m 50s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/78078dacdc9b4fd289470cf815fc455d : FAILURE in 26m 01s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/5e6e48e40de947fc85afc0cfdcb992c8 : FAILURE in 24m 50s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/4c8d976468294905af8d3cfa50f94621 : FAILURE in 24m 52s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1e556e250e5e418db2335014fda5eaa5 : FAILURE in 24m 57s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/72d93617c80f4cfeb800be0f2db5958a : FAILURE in 22m 49s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d1ee640589cf431fb746a43481c615ed : FAILURE in 1h 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5a8257a3a8c6ee936aceadf77817f045745b50a5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-11 22:39:19.000000000","message":"Patch Set 5:\n\nI think this one needs https://review.rdoproject.org/r/#/c/22213/\n\nI\u0027ll recheck tomorrow","accounts_in_message":[],"_revision_number":5},{"id":"f91dd1b16ed7a9ee55c14144c15ef74d25854d3a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-11 22:48:21.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"035cd78b1744073521e68292e60b603d65c834f9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-11 23:11:39.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/5/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/c7bc062/ : FAILURE in 22m 24s","accounts_in_message":[],"_revision_number":5},{"id":"56e3f7ce2f7ae33f88ff81978e8850637b936249","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 00:34:42.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d5e3749b2b4046bd8d1da8753b498db8 : FAILURE in 26m 41s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/3bbb0c7cda804003a042e9b82e316285 : FAILURE in 25m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/dab6f77687ad4dab81289f336fbc97c9 : FAILURE in 29m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/253c33def9594de99f7a9ba47535ce5a : FAILURE in 32m 10s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/11dd10610d0648afabc7655c30e26616 : FAILURE in 27m 42s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/322b511024004a689915d23cefeb8873 : FAILURE in 1h 27m 05s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd960ffb99494a4dbd8c3e25b564c84f : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/937de6ffa1b349cd81b1d70dc31fc77c : SUCCESS in 3m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1fa72044a5ff41a180e671dbd18c9e4b : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b60cc3c6a6764bfa960aedce531acaae : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a97b54efa4f84a8e97ab2f948d6d8c89 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bd4d3d55fb24f46a58cc158887bdd03 : SUCCESS in 3m 33s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/914a7f6f96ad457fb074e1cd33783be2 : FAILURE in 26m 55s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9636731fa4dd40ab8367b7d90c559da0 : FAILURE in 26m 59s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/a3a43d8c7f6e4a5fbc85d97e38157def : FAILURE in 18m 58s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/42a9d4d0635248a0a9df4bb9e0a4fd4a : FAILURE in 24m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/26bc7dfbb2844288873e136e6ca30763 : FAILURE in 25m 43s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/51f248597afe4f70ae2763a35c8a0126 : FAILURE in 23m 36s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/faf6c5cc636443dda964da48025cd8bc : FAILURE in 27m 34s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/53b7292163d4494e9ae98e762bbdc2ad : FAILURE in 1h 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c419e5afc5593bae69fe99a2a47c592ef5e9ae85","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-12 03:01:32.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a17a877667c349fb67752978ef55ed55109b851a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-12 03:24:13.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/5/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/45eec53/ : FAILURE in 21m 26s","accounts_in_message":[],"_revision_number":5},{"id":"bfa7b245a18f5148b0ea6632742d2ef7b6fc0846","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 04:52:49.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/570e06bc7d9a4a279be75a6c4ab3903b : FAILURE in 26m 01s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/eb00b7fb2dc24dcaa6c0f1a01a77f869 : FAILURE in 23m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b9f463ca4d884c7b83746d51467577d3 : FAILURE in 28m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e161ca7963a64d1fafa067468f7c849d : FAILURE in 33m 16s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/2f267319de7f4682aa065bdcfd05e98a : FAILURE in 24m 35s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/6ffd2f35662242bb87cd9597920b81e1 : FAILURE in 1h 28m 08s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/beb4f180b4eb46139ec4606d1e24fa77 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb43af5abc3a4c3cb03c8fce8e7b7ae9 : SUCCESS in 3m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/22fb12fc2cce4c0e8eaf266c967c5d65 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b43cdd6d742f4511b70322f0c26edf19 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ccd15164f2c48558f64010583c8de82 : SUCCESS in 2m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e0d53ab79bc343f1a20a05e922e5b4b5 : SUCCESS in 2m 47s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/aa8c9f679ce4448ea7d752d1da43f444 : FAILURE in 27m 19s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/03464953a80947269f9b0f3b45b73f82 : FAILURE in 30m 33s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/a1b6be13163741b5a3e0d30cba42aaea : FAILURE in 25m 51s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/174b2d39c1f34a1f971c7ee0914c2ea0 : FAILURE in 29m 14s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/59da14d13b4b437fa3475afaff73cc2d : FAILURE in 26m 11s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/053826eab4df466da9623086eba0822d : FAILURE in 26m 08s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/1ac42fd732404294a9189d3a71bcc351 : FAILURE in 29m 54s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/20eb5c30771446f180e46a1c94d51a7f : FAILURE in 1h 18m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bf88a8cd7f295d601cacbe6d6fd1c2af98ac6a18","author":{"_account_id":23618,"name":"Christophe Fontaine","email":"cfontain@redhat.com","username":"christophefontaine"},"date":"2019-09-12 06:41:29.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"03fb2d01400d28ebbb4ae55dc5b4a3b350a0f854","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-12 11:56:17.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e53c47f9f8cb6afc262bfe7e5d94637579cd8924","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-12 12:19:06.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/5/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/f7fc2f0/ : FAILURE in 21m 29s","accounts_in_message":[],"_revision_number":5},{"id":"e5f3b7781bde42c1880c24eae3c1714ed80e57cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 15:09:13.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fd0f7fd8b8e243ecad8515c43acf5b27 : FAILURE in 26m 15s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/3741f912b03446c29e1ea41274890320 : FAILURE in 20m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/527ea78c82894ff7a14b7840acbdcce6 : FAILURE in 33m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/0896c9f5c96b48e9b2c3e616c8118ab3 : FAILURE in 27m 15s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/864b4d43d9d74975a86f4e3c69dbd791 : FAILURE in 28m 17s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/e9b35a8e6ff04ea3b9599ca40487b42a : FAILURE in 1h 23m 21s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93090a6c8d2647d8af7e00a2f440ef81 : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32eb61d7548b4345b7d30de31d475ee2 : SUCCESS in 5m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/54b2450c04274ad69499f3a999352242 : SUCCESS in 1m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a74828f55af48878c7eb15b460407c2 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36cea6f904e64e4b841d3d3e3bb8e9ec : SUCCESS in 2m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/beef09bb11044b80827a0693c910bf32 : SUCCESS in 3m 10s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9887dbb67fa94254aa317db6fc53a996 : FAILURE in 26m 51s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/54c11aa263a2407e83506972ed262f97 : FAILURE in 33m 47s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/8b74ba6ecdb2459db412ecffdb6ec28c : FAILURE in 30m 14s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/af9a9bce11a74cfd93871c520399f302 : FAILURE in 26m 30s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0cafd8eaa3474e8c891b55bcc0e9d8e0 : FAILURE in 22m 46s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/503db175cfaf45ee91d480c4dc010dfb : FAILURE in 23m 55s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/ee13df9fd5d84c17995c86ab21a06294 : FAILURE in 27m 34s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/6410272574ac4cf4a04ffa181b6b2e41 : FAILURE in 1h 22m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cee0012d8861252f0fb92063703f7855f4bd9b3d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-12 15:23:39.000000000","message":"Patch Set 5:\n\nwe\u0027ll need https://review.rdoproject.org/r/#/c/22254/","accounts_in_message":[],"_revision_number":5},{"id":"da139db6b7353aae5e8205d8cb8af7ddbe9f9a60","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-12 17:52:51.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"febd52e0913ce6447a7ac58f42ad7d8dd577febc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-12 19:52:31.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/5/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/7464f97/ : FAILURE in 1h 58m 41s","accounts_in_message":[],"_revision_number":5},{"id":"fc96ffcab7f81f49887c58c69e8a4b0ffd99b141","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 23:00:16.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/39c9192cb7434331b72c327ab85238ad : SUCCESS in 3h 22m 17s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/261d295775be4964987f053acaf00cbc : SUCCESS in 1h 58m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e46817c0ee694ce6b59ea0ebc1a71c6b : SUCCESS in 2h 30m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b40ac8d45a2242bbbfc2970e7542fc3a : SUCCESS in 2h 57m 34s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f049b507e607491f948715635baeb1fc : SUCCESS in 1h 58m 26s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/956c5355857946e2bb0f036a98f566c6 : FAILURE in 2h 31m 50s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31cb9790b3cc4f28ab8074fe00e795f2 : SUCCESS in 2m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14a48e91ce9245a3a8ce3ca241dd3f2b : SUCCESS in 3m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2486e70f763b4614a57a622397af6148 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6f8078b59074b56ac8e683a0d8c47ea : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2422e71d06640949a75fcd48272bcbe : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af8f274ce0dd479da159052ae5cbe75b : SUCCESS in 3m 09s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/860450fcb80f4b1e82b28a20d4cc85b6 : SUCCESS in 3h 30m 46s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/57e99595571d4bf99370cc2d962db6bb : FAILURE in 1h 29m 57s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/4e92476e6f8d4a7b8ac6554dab59d720 : SUCCESS in 2h 23m 09s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4109ca95137e4538a4357a63193ed87d : SUCCESS in 2h 06m 32s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/8fd20cb9f7294e16a104c8bc93f3bb2a : FAILURE in 1h 52m 40s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/58e74f6d26ff4ec8968fa71dca721ae7 : SUCCESS in 2h 07m 23s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/874ed5be6f7e4577bd558f909c9c5649 : SUCCESS in 2h 37m 27s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/eb80426ed5354e278d1acff94ae84627 : FAILURE in 2h 00m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"94cc4d89eebd1744491e422cdde738d64b84595f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-13 12:55:43.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"fc59a94cd869c7269ee136e6c694e38ebe898ecd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-13 15:50:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/5/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/80a0a10/ : SUCCESS in 2h 53m 10s","accounts_in_message":[],"_revision_number":5},{"id":"ed272dbdb57de668b43598e1bb4f215ad3402919","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 20:21:49.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f768d02fc2544dd7b546ce69f3c32d93 : SUCCESS in 3h 06m 26s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a9a9ea7302964d0b99852754e043c193 : SUCCESS in 2h 10m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/283f4b5bddde4275860e7ffef70a3efe : SUCCESS in 2h 22m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/9fc963c337cf4c4388c17eea9a81cf82 : SUCCESS in 2h 55m 04s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/2c34dfa5cc104e529817ba184c734ccb : SUCCESS in 2h 16m 19s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/593592b62ab84181beba73ed00bc6edf : FAILURE in 2h 11m 03s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1edebf87667442888d7fbced1934656 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a5b4d7194194b74ab448df842cd5d4d : SUCCESS in 4m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a5982fa4d689494e8d721079c6175971 : RETRY_LIMIT in 1m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e3ca834da44553bfd0930779a90975 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b57b54646f40492ca0804f61ce41a95e : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a05274144078469c88cc5939e32b1bfd : SUCCESS in 4m 53s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/bf50ec9813c94fa0bc26d9b1d9ec9f8f : SUCCESS in 3h 38m 46s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/458452e49a634eeda9ff4fa2c69fa285 : FAILURE in 1h 50m 24s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/a2b9a00cc41542c683553005c8c4217a : SUCCESS in 2h 37m 22s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/93ed55984dea459f99fc63f54018c51f : SUCCESS in 2h 00m 53s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/606052e1540c4f3fbaf7714a539caed6 : SUCCESS in 1h 34m 45s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/21ea999a83844269a5d51a07d0bcdb15 : SUCCESS in 2h 34m 00s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/718953d458224501a8a37b3312fb556b : FAILURE in 1h 58m 52s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/55a6394dec6f44c296f74835589534e7 : FAILURE in 2h 08m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d8c7a826278f78322fbbd3e4fff69c765a496db7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-14 02:15:50.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e63404a570f54b6dcf7b51946cee869e61a2b612","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-14 04:55:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/5/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/10bb05d/ : SUCCESS in 2h 38m 05s","accounts_in_message":[],"_revision_number":5},{"id":"9ae2dd1b5c9662fcb64e9a4ecbd22b0bc5cc2ecf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 05:51:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f1f2e523bb4d40b0a0c24d44a3b5c601 : SUCCESS in 3h 16m 57s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/92f4d144d43540fb8b60c35711f3c68f : SUCCESS in 2h 04m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d1cbd5118c0c4461961c7d2661c321d0 : SUCCESS in 2h 22m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a016c6397f154594850c2f67c477bcc0 : SUCCESS in 2h 30m 02s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/0825776be0094f21a6cd943bb174636f : SUCCESS in 2h 09m 12s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/1b8cc0d8f60a4a3588a38d8e80860446 : FAILURE in 2h 20m 33s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04ecee8021b74598972a256d4c2ee4cf : SUCCESS in 2m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d8542dca6ab449fa21b5ae85bbd967a : SUCCESS in 3m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/748a20c6b769475aaeca7e60fb2c78e1 : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9645d10de0954f67b0af7e3161c80fe7 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01800ed4aa354bb2bade218f29094300 : SUCCESS in 2m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb2ec79efcf744dcbb0e93320afc1b26 : SUCCESS in 4m 21s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9401b9e164d549a086dfb128d1e82299 : SUCCESS in 2h 51m 54s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8368150a4b424d7b8b88155986a32009 : FAILURE in 1h 47m 32s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/62de330b9e9c42278db748a3369e2160 : SUCCESS in 2h 20m 41s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/2cb9e45d404d40c281600dbc1dc411f3 : SUCCESS in 2h 05m 29s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/6400a500d70744f0b8f63674218ffbbe : SUCCESS in 1h 59m 44s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/42c0a80e719a42449e904d66d87941e8 : SUCCESS in 2h 32m 16s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/4f74a35301f64e85b897fae71e5bc10e : SUCCESS in 2h 09m 09s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/b81e749445134037bf852124111af43e : FAILURE in 2h 14m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c8d947f10c7b9bad814849e9e74bd6a623420cbe","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-09-14 08:59:03.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d45f4edc844c6749e2ca676dc02a0a371437db51","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-16 13:37:18.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"634af7634ecfb24f160f61a5fedb4d94bc0ecb72","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-09-16 20:41:57.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"3abcdb46788a3c79f05d044ec0c4e1c3ef8c6b82","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-16 20:42:49.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"31dfe2b52c5659f18edfb0682bb2ac4fb2266e9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 20:43:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6afdc7ee0f6d644456ee6973e8deb1e420dcd007","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:36:28.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f280efe9de104c7ea65b003f807846f9 : TIMED_OUT in 3h 27m 15s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/fe3577c24c1a432c8e2278f0dabc2f6b : SUCCESS in 2h 14m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/04ca157e9a434b5f87bc20f039c8c327 : SUCCESS in 2h 37m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20bfa9c42452437bbda656f9aa8177bb : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f164e248e4d94a189b54c1db89cb61b0 : SUCCESS in 3m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/754441e80a4a4b8e958f67fd3e82a68c : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7140111dcfff46c1b94887b8fe47063f : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92ea27ad8e194da3b246e70d4459299c : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/37c4408b1e114472aaad7cd6e59d0b50 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":5},{"id":"fbaf7c58bb096f1ae1128f4f6f59a3b49c6429e1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-19 12:01:49.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0f541ea68c666f4907f60fe4696372292482273c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-19 15:07:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/5/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/b87b031/ : FAILURE in 3h 05m 00s","accounts_in_message":[],"_revision_number":5},{"id":"c54c3fad01030524bb8e6f969fc2cf02e08fa4ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 03:12:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/87d1472aa4284c1895b0022376cc89d9 : SUCCESS in 3h 36m 25s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f11c216cd3644316b70067afacadff05 : SUCCESS in 2h 22m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c53ba35586d0434dab2a3600b3e6130f : SUCCESS in 3h 00m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d10438c75f8148bfadc4372ee4029f97 : SUCCESS in 3h 15m 08s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/4099ad7659b746d48bebbe7bf4e1ec8a : SUCCESS in 2h 30m 36s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/bbded4a5db9a40caa2d9b3cbb3641c66 : SUCCESS in 3h 13m 21s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b046f69fdf340a88b149bdaf40d08f1 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f4747676221240f4a1d1efbe8e4a4409 : SUCCESS in 5m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/51b7ba35b27e4d6a87317ba2b3fa10a8 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e0a368176d64820a86fadb5d339c171 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f732902ba0754af583a635a2b3895339 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce34d239b4c84dc6870098ee022ca877 : SUCCESS in 4m 21s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4c67146b152a4fb19f715f422738ec7f : SUCCESS in 3h 29m 40s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b0d49843a5644acb8197ade988d928fe : SUCCESS in 3h 11m 16s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/6fcdfc6b2bcc416b9f0170778579dac5 : SUCCESS in 2h 27m 20s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/202f05ea16e440bbbe71da1556be9069 : SUCCESS in 2h 04m 03s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/295d7286284f4fd2a2b64f44f435088c : SUCCESS in 2h 25m 03s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/04b80866bb5542f08351a44db284676b : SUCCESS in 2h 28m 35s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/5f6ff8f14116485f8783757e24645cd5 : FAILURE in 2h 45m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d83ac43a28562972e1920e5c5a70d89f4becb07e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 06:45:40.000000000","message":"Patch Set 5: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"3521e6ebc4cb64515f3a4dde700b889c43528b27","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-20 13:38:18.000000000","message":"Patch Set 5:\n\nwe shouldn\u0027t have landed this in the stables until master merged. now it\u0027s all conflicty","accounts_in_message":[],"_revision_number":5},{"id":"fbe10935cd04e5f23de9a7aaaa9fef234f21a8aa","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-20 13:41:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5105d2a0d326f8b874e5afb612ee62507f5abcba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 14:05:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/6/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/7d6e73c/ : FAILURE in 22m 12s","accounts_in_message":[],"_revision_number":6},{"id":"06db84e91b8a166f4cca35034bb6495b1ed1dc69","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-20 14:32:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ba1ad94e6cd913703f77cab05a4c5ca0edd910ed","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-20 14:40:04.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7a7bdb2f33f1b1c2d50deb426bf46de616788cf0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 17:31:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/8/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/d7da615/ : SUCCESS in 2h 50m 15s","accounts_in_message":[],"_revision_number":8},{"id":"57f19df4875fff69c4131d257a5de870b824fc53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 03:54:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d63c7c8505054ee5857305f791e8db5a : SUCCESS in 3h 25m 54s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/49f375dc18484954be189df245d66921 : SUCCESS in 2h 19m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f890bf9c005c4cfba4764293a71fc18c : SUCCESS in 2h 54m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/f59bca99fa114a77b206695efe507f2f : SUCCESS in 3h 38m 00s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e0ac08a234a9443ab4a64777ad2868a9 : SUCCESS in 2h 40m 57s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/694ec2eeaf0d44999f8bc345135d94a8 : SUCCESS in 2h 57m 36s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/444e983ff6104fefa8b7eff5eee3ff4d : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3dc88c4b5e884a43b24400ab2de117f8 : SUCCESS in 3m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0c25c1b43f3e4126a643b612b27c3f7a : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5f8c6d002d94ff79cc60bff389d8fb2 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20d5a420c4124bbf9bb03e561363a48a : SUCCESS in 2m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a58094528f44f6b82295bdd1a4f5d3d : SUCCESS in 2m 53s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9217da9404794917bb658823b6841448 : SUCCESS in 3h 11m 35s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b1e0c391900f43349dea4afa5c523080 : SUCCESS in 3h 21m 33s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/8124005f4c6344fe8d7123bcd5446519 : SUCCESS in 2h 30m 36s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/60bdab7298d14494add4fb28231de54c : SUCCESS in 2h 06m 03s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/0e1814a5be4645008740d52ec90bd7ab : SUCCESS in 2h 34m 48s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/44779d8070a440b295b478d36808d459 : FAILURE in 2h 13m 23s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/11146487adbd482cb0a65e177d5b8788 : FAILURE in 2h 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"871f986865ae570b2c614a2c5b55ccdea3cf5ea9","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-09-23 13:20:03.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"1fd381384a15a75c0a11c40a3fad095e9f674b80","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-24 03:32:06.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"b50134d35541c3d5af6a2acd5d0a146791b30140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 03:32:20.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"0a2a50eec07186878c02ab0a91d5b29a54a81237","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-24 03:48:54.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"4ed06e158c9627e26533df881892616014fe5bc1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-24 06:53:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/b6d922c/ : SUCCESS in 3h 03m 05s","accounts_in_message":[],"_revision_number":9},{"id":"3e8cf863e2f07eec429aa35e7fcabc50c9409fa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 07:55:53.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0c5afbdb249d47a8a3f10c688d411536 : TIMED_OUT in 3h 46m 20s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/5a6eb7315d5947b58822147c1ce67750 : SUCCESS in 2h 24m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/73d5582888e2403d92c090123bce0518 : SUCCESS in 3h 28m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a9411a2660384788be2bd2bbd5c2b85b : SUCCESS in 3h 37m 58s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/19495bfcc52e48f199e874a3e253a7d8 : SUCCESS in 2h 41m 44s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/47284a90505a41eaa0a8080610c4f6e9 : FAILURE in 2h 36m 06s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7c9cefa160e4e2a8199635fc60c5197 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7353c81492c34d76b40889b75d433b12 : SUCCESS in 3m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/22c6bf54c63144218e651bf2693415a3 : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e76b1680c6495a8694110420feeb4f : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e286094c1eba4bb99d25ee33022b9349 : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f3e7af0b2eb477c881cd4644a264608 : SUCCESS in 3m 18s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/acc45b695b414e23ae95fae790270c87 : SUCCESS in 3h 36m 11s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/f9ca173b9a3742598bcbc449ce4c6c8c : SUCCESS in 2h 56m 40s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/c8a84f7c3e92407eaf114753312279c6 : FAILURE in 2h 34m 24s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a2c207060fdd4cab9119cce1199c226a : SUCCESS in 2h 02m 27s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/02e28e6415774735a44d648df15bbc2b : SUCCESS in 2h 34m 42s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/f3b24eff79d9420795dcadaf44bff84a : FAILURE in 2h 23m 33s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/cdf1b93c5bbb451585242ad036682e40 : FAILURE in 2h 17m 52s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4ed4222213403f1f660ac4f7629d1ce2c5f34808","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-30 15:43:33.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"5371fa8c9df6fb5e389a7ef19da0e0c127f81985","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-30 18:36:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/c99218c/ : SUCCESS in 2h 51m 41s","accounts_in_message":[],"_revision_number":9},{"id":"a0f6051181d2f61813c24b70d1d86fecddbd00d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 19:43:51.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3be5146c29cf4c45aa6407da5e32db1c : SUCCESS in 2h 58m 44s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/592cb70b4a9446aaa8e422461de29c39 : SUCCESS in 2h 10m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0b578c62333c440a9da33e1b5602285d : SUCCESS in 2h 35m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/0b3844bf24244fb29bab79c638b953f6 : SUCCESS in 2h 35m 05s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/46d76f1252be402daaaaacc4f0cb4b2f : SUCCESS in 2h 40m 52s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/549bf8054c1349a1810343f9f1b212d6 : FAILURE in 1h 58m 09s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96a8f23e14e74e1d8fee988564a838f5 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/158159c57b6c4e72903158643ce528b8 : SUCCESS in 4m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/307b7eed272e410d820c80bda1787457 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b962c3e5124064b710c38651debb58 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/347168d8afa8434aadfbde63f1ff5deb : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c99b3b3c6b04b9392989aa64ff4060e : SUCCESS in 3m 28s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a297be87d4994e03a7ab3bdcd0ab8bcc : SUCCESS in 3h 05m 18s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/6534316478944a59b051d157fa08eaad : SUCCESS in 2h 34m 29s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d98b55f9f2b74569b68bd464ee22b2d4 : FAILURE in 1h 52m 33s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/f0d9167d843c4e3a9639272dbe08efea : SUCCESS in 1h 44m 42s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/e4f86011a2724ac9bb7372b85160a8a3 : SUCCESS in 2h 38m 54s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/eca086d38b6644a8829efa4038efc891 : FAILURE in 2h 16m 57s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/764fd35aebaa4786a974c6376df09834 : FAILURE in 2h 06m 16s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8abeeb56768d27a1dfd8835f4277073eea82a8d0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-01 01:30:11.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"ec8865733142be338af0d7523f3ac73a472f8a98","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 04:19:16.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/57/681257/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/cad279d/ : SUCCESS in 2h 48m 01s","accounts_in_message":[],"_revision_number":9},{"id":"903aaba7dbbd4b6cf04fa1b07fb40c058529922a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 05:20:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/81354fa3e31047a1b79374a380daee47 : SUCCESS in 3h 10m 39s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8d4d939eec7a42d8b8e4da279b09c504 : SUCCESS in 2h 08m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2c26eda6e2de47d3b3c291e96b8f7151 : SUCCESS in 2h 34m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/260ba3d125d14f778af5dd279a02173c : SUCCESS in 2h 41m 14s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/9863feb9dd2b4452b070d56cf1593733 : SUCCESS in 2h 16m 08s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/2ee120db5cf845ffa412f700068b49fa : FAILURE in 2h 06m 22s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a8d6a78dd9247dc92d1fb85468f5de9 : SUCCESS in 3m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87c627342c95440c93cca8be9a2e0367 : SUCCESS in 3m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/54359171f8bd4f8cb7b58e2a5d99dba7 : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78fca4a5ac084fdbbcc1d1dc39723c3a : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0099c884bf3c4ac988fdcf4325ff64b3 : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ef6945df30c4ac28051134abeee4345 : SUCCESS in 3m 10s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/301d0e59c97a42c182bde493d7ebe05f : SUCCESS in 3h 33m 05s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/3e2ad336db1246e6a4d95a4a69213275 : SUCCESS in 2h 30m 07s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/a2d0fdb86b27446788154c967335c55c : SUCCESS in 2h 15m 35s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/63fd0a72ce3846069f900227b7961e2f : SUCCESS in 1h 54m 45s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/6acd6aa7af844278b644877e5e5bc863 : SUCCESS in 2h 31m 06s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/74cdb107afea4971bfeaa0ee5141197a : FAILURE in 2h 25m 11s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/058ef492b59f44aeba8e9c68bb8cf5c5 : FAILURE in 2h 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4298f3f744883aa548e22cbe232911e7fca48dbb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-01 13:01:32.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"ec525d05e8dab366b77616cb6274cbc2aa1187d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 13:01:47.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4e5d7d5fbf886681b339a0f0b48cc1541f921341","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 17:10:09.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b0f0f712846b44bbba3bb1881a71eba7 : SUCCESS in 3h 03m 27s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ccf5fa213a5f491885c45c8c8b5bcb6e : SUCCESS in 2h 23m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/43d19a8a0ae44f5b9e292f6af2ecbb34 : SUCCESS in 2h 35m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8242d5b9c894d9c81d5ed6687dde650 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57922ff5c6884c04a5d348ee1fcce7ed : SUCCESS in 3m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f7c6d00ace04487984bf7164e58d03e6 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa6e0bcf2b64de0a7220d6bafa0361e : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0619a4dc67684479b9bdd2d90f58fd45 : SUCCESS in 3m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9551896837864061a8c7a4edf01b3f17 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":9},{"id":"e73037c0c33b827ff87313c7afcd396257a757eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 17:10:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"47d0b65a5dce3050738e5162172a9227967c1c0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 17:11:18.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b403db20c2d54c7a914323a2831ca8a0 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"5e88de63f5763da94d26ffcd87c609527986e954","revisions":{"f29c3b6f1c27620a46418d59f6bc387f40f9d057":{"kind":"REWORK","_number":1,"created":"2019-09-10 13:39:41.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/57/681257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/57/681257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/57/681257/1"}}},"commit":{"parents":[{"commit":"f20c8f8158008e0582651509fef0648ab6a76512","subject":"Merge \"Revert \"Handle defined containers that are stopped.\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f20c8f8158008e0582651509fef0648ab6a76512"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-10 13:36:28.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-10 13:36:28.000000000","tz":-240},"subject":"Add --cpuset-cpus support for both Docker and Podman","message":"Add --cpuset-cpus support for both Docker and Podman\n\nLimit the specific CPUs or cores a container can use.\nA comma-separated list or hyphen-separated range of CPUs a container can use,\nif you have more than one CPU. The first CPU is numbered 0.\nA valid value might be 0-3 (to use the first, second, third, and fourth CPU)\nor 1,3 (to use the second and fourth CPU).\n\nThis parameter is particulary useful for NFV:\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1750781\nIndeed, for NFV workloads, in order to achieve 0 packet loss, linux processes,\novs-dpdk (if applicable) and VMs are isolated thanks to kernel args (isolcpus)\nand tuned profiles (cpu-partitioning).\n\nChange-Id: I9443ad60affe9c7823b17daa259efee542c6fe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f29c3b6f1c27620a46418d59f6bc387f40f9d057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f29c3b6f1c27620a46418d59f6bc387f40f9d057"}]},"branch":"refs/heads/master"},"e98306aa452aae2f9ce7f403297f29272c9bfc53":{"kind":"REWORK","_number":2,"created":"2019-09-10 18:09:14.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/57/681257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/57/681257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/57/681257/2"}}},"commit":{"parents":[{"commit":"f20c8f8158008e0582651509fef0648ab6a76512","subject":"Merge \"Revert \"Handle defined containers that are stopped.\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f20c8f8158008e0582651509fef0648ab6a76512"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-10 13:36:28.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-10 18:08:49.000000000","tz":-240},"subject":"Add --cpuset-cpus support for both Docker and Podman","message":"Add --cpuset-cpus support for both Docker and Podman\n\nLimit the specific CPUs or cores a container can use.\nThe cpuset-cpus value is computed by using psutil with a new function which\nreturns a comma-separated list range of CPUs that a container can use.\nThe parameter is not tunable on purpose, and is opinionated for now, as\nit seems to not be needed. The returned value from psutil is good\nenough.\n\nThis parameter is particulary useful for NFV:\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1750781\nIndeed, for NFV workloads, in order to achieve 0 packet loss, linux processes,\novs-dpdk (if applicable) and VMs are isolated thanks to kernel args (isolcpus)\nand tuned profiles (cpu-partitioning).\n\nChange-Id: I9443ad60affe9c7823b17daa259efee542c6fe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/e98306aa452aae2f9ce7f403297f29272c9bfc53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/e98306aa452aae2f9ce7f403297f29272c9bfc53"}]},"branch":"refs/heads/master"},"a735258af41f9aebd591ed98a3929b8f0ea649d0":{"kind":"REWORK","_number":3,"created":"2019-09-10 18:52:04.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/57/681257/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/57/681257/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/57/681257/3"}}},"commit":{"parents":[{"commit":"f20c8f8158008e0582651509fef0648ab6a76512","subject":"Merge \"Revert \"Handle defined containers that are stopped.\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f20c8f8158008e0582651509fef0648ab6a76512"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-10 13:36:28.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-10 18:33:37.000000000","tz":-240},"subject":"Add --cpuset-cpus support for both Docker and Podman","message":"Add --cpuset-cpus support for both Docker and Podman\n\nLimit the specific CPUs or cores a container can use.\nThe cpuset-cpus value is computed by using psutil with a new function which\nreturns a comma-separated list range of CPUs that a container can use.\nThe parameter is not tunable on purpose, and is opinionated for now, as\nit seems to not be needed. The returned value from psutil is good\nenough.\n\nThis parameter is particulary useful for NFV:\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1750781\nIndeed, for NFV workloads, in order to achieve 0 packet loss, linux processes,\novs-dpdk (if applicable) and VMs are isolated thanks to kernel args (isolcpus)\nand tuned profiles (cpu-partitioning).\n\nChange-Id: I9443ad60affe9c7823b17daa259efee542c6fe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/a735258af41f9aebd591ed98a3929b8f0ea649d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/a735258af41f9aebd591ed98a3929b8f0ea649d0"}]},"branch":"refs/heads/master"},"7342a2f9ca1fb11bfd2d1ef908dd9c696bbc16ae":{"kind":"REWORK","_number":4,"created":"2019-09-11 17:01:32.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/57/681257/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/57/681257/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/57/681257/4"}}},"commit":{"parents":[{"commit":"f20c8f8158008e0582651509fef0648ab6a76512","subject":"Merge \"Revert \"Handle defined containers that are stopped.\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f20c8f8158008e0582651509fef0648ab6a76512"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-10 13:36:28.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-11 17:00:44.000000000","tz":-240},"subject":"Add --cpuset-cpus support for both Docker and Podman","message":"Add --cpuset-cpus support for both Docker and Podman\n\nLimit the specific CPUs or cores a container can use.\n\nIf cpuset-cpus is configured in the container layout, then the value\nwill be used when running the container cli with --cpuset-cpus.\nIf \u0027all\u0027 is used as a value, we\u0027ll then take all available cpus,\ncomputed by: \"0-\" + str(psutil.cpu_count()-1)\n\nIf unset (default), the cpuset-cpus value is computed by using psutil with a\nnew function which returns a comma-separated list range of CPUs that a\ncontainer can use.\n\nThis parameter is particulary useful for NFV:\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1750781\nIndeed, for NFV workloads, in order to achieve 0 packet loss, linux processes,\novs-dpdk (if applicable) and VMs are isolated thanks to kernel args (isolcpus)\nand tuned profiles (cpu-partitioning).\n\nChange-Id: I9443ad60affe9c7823b17daa259efee542c6fe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/7342a2f9ca1fb11bfd2d1ef908dd9c696bbc16ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/7342a2f9ca1fb11bfd2d1ef908dd9c696bbc16ae"}]},"branch":"refs/heads/master"},"d6ae074d2567b638fccd32868f6421de4044fbbe":{"kind":"REWORK","_number":5,"created":"2019-09-11 17:35:18.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/57/681257/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/57/681257/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/57/681257/5"}}},"commit":{"parents":[{"commit":"f20c8f8158008e0582651509fef0648ab6a76512","subject":"Merge \"Revert \"Handle defined containers that are stopped.\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f20c8f8158008e0582651509fef0648ab6a76512"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-10 13:36:28.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-11 17:35:05.000000000","tz":-240},"subject":"Add --cpuset-cpus support for both Docker and Podman","message":"Add --cpuset-cpus support for both Docker and Podman\n\nLimit the specific CPUs or cores a container can use.\n\nIf cpuset-cpus is configured in the container layout, then the value\nwill be used when running the container cli with --cpuset-cpus.\nIf \u0027all\u0027 is used as a value, we\u0027ll then take all available cpus,\ncomputed by: \"0-\" + str(psutil.cpu_count()-1)\n\nIf unset (default), the cpuset-cpus value is computed by using psutil with a\nnew function which returns a comma-separated list range of CPUs that a\ncontainer can use.\n\nThis parameter is particulary useful for NFV:\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1750781\nIndeed, for NFV workloads, in order to achieve 0 packet loss, linux processes,\novs-dpdk (if applicable) and VMs are isolated thanks to kernel args (isolcpus)\nand tuned profiles (cpu-partitioning).\n\nChange-Id: I9443ad60affe9c7823b17daa259efee542c6fe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/d6ae074d2567b638fccd32868f6421de4044fbbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/d6ae074d2567b638fccd32868f6421de4044fbbe"}]},"branch":"refs/heads/master"},"ebc972274bf6bc88235bd334add88dd47a18ff70":{"kind":"REWORK","_number":6,"created":"2019-09-20 13:41:11.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/57/681257/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/57/681257/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/57/681257/6"}}},"commit":{"parents":[{"commit":"e7a12e883cea717c219f579462b66c0abde04f1a","subject":"Merge \"Improve volume validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/e7a12e883cea717c219f579462b66c0abde04f1a"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-10 13:36:28.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-09-20 13:41:07.000000000","tz":-360},"subject":"Add --cpuset-cpus support for both Docker and Podman","message":"Add --cpuset-cpus support for both Docker and Podman\n\nLimit the specific CPUs or cores a container can use.\n\nIf cpuset-cpus is configured in the container layout, then the value\nwill be used when running the container cli with --cpuset-cpus.\nIf \u0027all\u0027 is used as a value, we\u0027ll then take all available cpus,\ncomputed by: \"0-\" + str(psutil.cpu_count()-1)\n\nIf unset (default), the cpuset-cpus value is computed by using psutil with a\nnew function which returns a comma-separated list range of CPUs that a\ncontainer can use.\n\nThis parameter is particulary useful for NFV:\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1750781\nIndeed, for NFV workloads, in order to achieve 0 packet loss, linux processes,\novs-dpdk (if applicable) and VMs are isolated thanks to kernel args (isolcpus)\nand tuned profiles (cpu-partitioning).\n\nChange-Id: I9443ad60affe9c7823b17daa259efee542c6fe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/ebc972274bf6bc88235bd334add88dd47a18ff70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/ebc972274bf6bc88235bd334add88dd47a18ff70"}]},"branch":"refs/heads/master"},"31507f6914161e0a9cee692645e251e47ea8c7e3":{"kind":"REWORK","_number":7,"created":"2019-09-20 14:32:40.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/57/681257/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/57/681257/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/57/681257/7"}}},"commit":{"parents":[{"commit":"e7a12e883cea717c219f579462b66c0abde04f1a","subject":"Merge \"Improve volume validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/e7a12e883cea717c219f579462b66c0abde04f1a"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-10 13:36:28.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-09-20 14:32:36.000000000","tz":-360},"subject":"Add --cpuset-cpus support for both Docker and Podman","message":"Add --cpuset-cpus support for both Docker and Podman\n\nLimit the specific CPUs or cores a container can use.\n\nIf cpuset-cpus is configured in the container layout, then the value\nwill be used when running the container cli with --cpuset-cpus.\nIf \u0027all\u0027 is used as a value, we\u0027ll then take all available cpus,\ncomputed by: \"0-\" + str(psutil.cpu_count()-1)\n\nIf unset (default), the cpuset-cpus value is computed by using psutil with a\nnew function which returns a comma-separated list range of CPUs that a\ncontainer can use.\n\nThis parameter is particulary useful for NFV:\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1750781\nIndeed, for NFV workloads, in order to achieve 0 packet loss, linux processes,\novs-dpdk (if applicable) and VMs are isolated thanks to kernel args (isolcpus)\nand tuned profiles (cpu-partitioning).\n\nChange-Id: I9443ad60affe9c7823b17daa259efee542c6fe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/31507f6914161e0a9cee692645e251e47ea8c7e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/31507f6914161e0a9cee692645e251e47ea8c7e3"}]},"branch":"refs/heads/master"},"50427129b5ffc2e20948a7334fd5edc49ea69b55":{"kind":"REWORK","_number":8,"created":"2019-09-20 14:40:04.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/57/681257/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/57/681257/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/57/681257/8"}}},"commit":{"parents":[{"commit":"e7a12e883cea717c219f579462b66c0abde04f1a","subject":"Merge \"Improve volume validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/e7a12e883cea717c219f579462b66c0abde04f1a"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-10 13:36:28.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-09-20 14:40:01.000000000","tz":-360},"subject":"Add --cpuset-cpus support for both Docker and Podman","message":"Add --cpuset-cpus support for both Docker and Podman\n\nLimit the specific CPUs or cores a container can use.\n\nIf cpuset-cpus is configured in the container layout, then the value\nwill be used when running the container cli with --cpuset-cpus.\nIf \u0027all\u0027 is used as a value, we\u0027ll then take all available cpus,\ncomputed by: \"0-\" + str(psutil.cpu_count()-1)\n\nIf unset (default), the cpuset-cpus value is computed by using psutil with a\nnew function which returns a comma-separated list range of CPUs that a\ncontainer can use.\n\nThis parameter is particulary useful for NFV:\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1750781\nIndeed, for NFV workloads, in order to achieve 0 packet loss, linux processes,\novs-dpdk (if applicable) and VMs are isolated thanks to kernel args (isolcpus)\nand tuned profiles (cpu-partitioning).\n\nChange-Id: I9443ad60affe9c7823b17daa259efee542c6fe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/50427129b5ffc2e20948a7334fd5edc49ea69b55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/50427129b5ffc2e20948a7334fd5edc49ea69b55"}]},"branch":"refs/heads/master"},"5e88de63f5763da94d26ffcd87c609527986e954":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-09-24 03:48:54.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/57/681257/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/57/681257/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/57/681257/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/57/681257/9"}}},"commit":{"parents":[{"commit":"9ed99000d30cb048d1ca62c5cd9cf85fba672ccd","subject":"Merge \"Fix discovering container names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/9ed99000d30cb048d1ca62c5cd9cf85fba672ccd"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-10 13:36:28.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-24 03:48:54.000000000","tz":0},"subject":"Add --cpuset-cpus support for both Docker and Podman","message":"Add --cpuset-cpus support for both Docker and Podman\n\nLimit the specific CPUs or cores a container can use.\n\nIf cpuset-cpus is configured in the container layout, then the value\nwill be used when running the container cli with --cpuset-cpus.\nIf \u0027all\u0027 is used as a value, we\u0027ll then take all available cpus,\ncomputed by: \"0-\" + str(psutil.cpu_count()-1)\n\nIf unset (default), the cpuset-cpus value is computed by using psutil with a\nnew function which returns a comma-separated list range of CPUs that a\ncontainer can use.\n\nThis parameter is particulary useful for NFV:\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1750781\nIndeed, for NFV workloads, in order to achieve 0 packet loss, linux processes,\novs-dpdk (if applicable) and VMs are isolated thanks to kernel args (isolcpus)\nand tuned profiles (cpu-partitioning).\n\nChange-Id: I9443ad60affe9c7823b17daa259efee542c6fe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5e88de63f5763da94d26ffcd87c609527986e954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/5e88de63f5763da94d26ffcd87c609527986e954"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
