)]}'
{"id":"openstack%2Fpaunch~769186","triplet_id":"openstack%2Fpaunch~stable%2Ftrain~I82172f5a2d4683509c802c0f14d571a40e21e6cc","project":"openstack/paunch","branch":"stable/train","topic":"bug/1907833-stable/train","hashtags":[],"change_id":"I82172f5a2d4683509c802c0f14d571a40e21e6cc","subject":"Actually fail if a stop or remove call fails","status":"MERGED","created":"2021-01-05 02:04:29.000000000","updated":"2021-01-17 18:18:31.000000000","submitted":"2021-01-17 18:17:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769186-bug/1907833-stable/train","cherry_pick_of_change":766779,"cherry_pick_of_patch_set":1,"meta_rev_id":"7824676b76bd820b409e33bdcfa89bbd7b1d873d","_number":769186,"virtual_id_number":769186,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-17 18:17:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-01-11 17:19:35.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-01-11 20:14:16.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2021-01-17 11:44:08.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-01-11 20:14:16.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-05 02:04:29.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-01-05 03:43:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-05 06:19:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"093cd2cafc46b302a6fef2aa469331284335bcb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-05 02:04:29.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"e4c362d895410f83abfdf8f8744cfdc8404cba94","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-05 03:43:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/03f6983701d348dd856af8f7e55e57f7 : FAILURE in 1h 08m 34s","accounts_in_message":[],"_revision_number":1},{"id":"bc756d4ccc55cb54303cd650c8a06915d60725ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 06:19:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5c7bb1e362ab4502b13df16738b2205a : SUCCESS in 2h 43m 31s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/81275b49c37445398f5f6c3f66baa331 : SUCCESS in 1h 07m 23s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/8ba5332f6b704d40bb36260deff3f815 : SUCCESS in 2h 00m 15s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/fcfbc8daad8446fa8b1da019d2ca181b : SUCCESS in 4h 07m 02s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/badfb78f65c147c3a1f7934902a2a9ac : FAILURE in 1h 40m 22s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/7c8826d481dd4054bee93c1de307c4ea : FAILURE in 50m 13s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a636c668ee45416f94d2b5738063b008 : FAILURE in 44m 03s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/cc7dc50bfefc4de4889cf3c889c42904 : FAILURE in 11m 56s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/64c4dd0a160d47b3903b3a8ae1a8b57c : SUCCESS in 2h 29m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/672abdb5d5dc4e3097b9fc6462766602 : FAILURE in 6m 43s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/48998e277cac4a71a4eca597a3fbe30a : FAILURE in 1h 02m 51s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6b47822da564f50bf6d54c7546be022 : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a24e1625f102450398e87407f778f3da : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bfee8db9abd4983b837cd77085ad31f : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91f4f3510feb4ec18835804606c573d6 : SUCCESS in 2m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/63779398d1a24fe1b499dc2dcb5a8646 : SUCCESS in 3m 01s\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5768acb9e3c2435fa23ada4160bb57ee : FAILURE in 1h 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc0906c5860ae0cce867a779f591fb112deb828c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-05 14:09:16.000000000","message":"Patch Set 1: Code-Review-1\n\nI think we\u0027re missing another backport or something. AttributeError: \u0027PodmanRunner\u0027 object has no attribute \u0027container_exists\u0027\"","accounts_in_message":[],"_revision_number":1},{"id":"09f13ad2ec062c27bd43c4e468e066c9f8450308","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-11 12:06:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f208ee2dcb14c10025ee26964db600b9e2756928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-11 14:37:20.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d1611dc83c01569aa487987239a7b83a9c58aa4f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-11 17:19:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8530505126c74397a21b53de5fbae9fc : FAILURE in 2h 41m 11s","accounts_in_message":[],"_revision_number":3},{"id":"5d2a08b14b7fc4b73500dc9d2210a3f2238d8d84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 19:52:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/62aca6dbcc0e451a9622c32d0700e4cc : SUCCESS in 3h 29m 12s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/ce24a831363940138a9fdda848f44f05 : SUCCESS in 1h 08m 52s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/4e3cd77ec48f4f1e86fe56dd705f36df : SUCCESS in 2h 28m 46s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/ccab9728fecc409f918d257a46e02d64 : SUCCESS in 4h 23m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7d3902e36f1e400297275c03932a4a06 : SUCCESS in 2h 09m 25s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/b1cb13ac7a004307b98a1a70baaa30e9 : SUCCESS in 1h 34m 01s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/05254ec61e9c4486a54cafe8989a23ad : SUCCESS in 1h 10m 10s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/6307cfb1c0f14e9e908db951eab84bcc : FAILURE in 1h 46m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9818b19517d64116b93b51540b92ab94 : SUCCESS in 2h 06m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a01a61b866264217b3e9f30559740529 : FAILURE in 1h 59m 14s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/bed7e579879046bbb4b666f0b933753c : FAILURE in 2h 11m 17s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a271c0de4d644489a123612f17d2d9d : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d51d6252d6841f79254ed78b7810340 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d80d3651c3d64bc58e965d58a44f7e3b : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9aed53a2fb44e2a8c6abf1d1caee909 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ee95a3039e646dca3b7747e8808a9fb : SUCCESS in 3m 42s\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/28f440f7126f425d8f45f97824c335d0 : FAILURE in 1h 05m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cb7131d20f89488f323b9e6af85c4ee6cd72cc39","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-11 20:14:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2ac83674b498ee4d7faa264cf38b0581f1c7a26b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 20:14:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a3cc6534c9d4b9590cb65da677739e2cd0c8fa8f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-12 00:38:25.000000000","message":"Patch Set 3: Workflow-1\n\nWe need to merge https://review.opendev.org/c/openstack/paunch/+/770114 first.","accounts_in_message":[],"_revision_number":3},{"id":"cdf498ba2a15ac36fd90e3cfbb125820d0983b02","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-12 00:38:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"dc91f19ab3c81db57a90e49a6560f6fd5b0939e4","tag":"autogenerated:gerrit:restore","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-12 00:38:39.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"6376acc5986ca3957cf0932aa17bf2d91eb14f82","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-12 03:09:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7bbccdfe384d4b32a1ab6d560cd4965e : FAILURE in 2h 23m 06s","accounts_in_message":[],"_revision_number":3},{"id":"e0395765443d73b2cab0bacc60d27560e0d5e461","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 05:08:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5dbd7bc3ad09450699421743ef9f53b5 : SUCCESS in 3h 19m 05s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/ee93273697b2462f88dc4e7597da4788 : SUCCESS in 1h 01m 35s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/03f071489b5a48ec86924b73c4db40cd : SUCCESS in 2h 47m 42s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/7d8ca003457745de9742a75c0b8d33c1 : SUCCESS in 4h 17m 32s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c788b9185b3a43a7a8d0883afa2b35ae : SUCCESS in 2h 01m 25s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/64ce3fdd871a4a7880dc3849ebeb4d67 : SUCCESS in 1h 38m 51s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ed55bd502af24d38ac95ec28f54d8c46 : SUCCESS in 1h 20m 15s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/df336e969ecd4a738076291966fbf3c5 : FAILURE in 1h 55m 12s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e37c9ec2f7f142fa9d39c3d056ce8656 : SUCCESS in 2h 27m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8e179eef6a144b839f8ec7ea8db50545 : SUCCESS in 2h 12m 13s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c7cebfe6340a46638707e86da619f374 : TIMED_OUT in 3h 26m 41s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9360d8e20d540648672b72afc5e7b05 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f53611335b24f8fa564193bab20e306 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc3850f7390c4ce5bbd4946ae86fdb16 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30cbcc99392d423a9843e6fabb40feb7 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9389de1495084d08bb1741c67c3505ac : SUCCESS in 2m 57s\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/806f2a493ad84eacaa7e4767d46d531b : FAILURE in 1h 18m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e923c0142638cffcacef9d704dd549ecfb0094c5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-15 12:11:56.000000000","message":"Patch Set 3: -Workflow\n\nThe subsequent fix was merged into ussuri.","accounts_in_message":[],"_revision_number":3},{"id":"a1154983218095ec3a8590c45b306c3015ef6c5f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-15 12:12:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c2aff9fec225d9755e74f0adedab2e820af70ef6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-15 14:59:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/965542187d184062b7119a8e99aa57c4 : FAILURE in 2h 35m 01s","accounts_in_message":[],"_revision_number":3},{"id":"541ed53de4f5f4bc53d5d698b850da61024cb25c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 18:48:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b8dd9e0634304e4b923cdd3dea031fd5 : SUCCESS in 3h 18m 45s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/b56147fb0b924a90b5a97d0e5588cc40 : SUCCESS in 1h 11m 05s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/d2dca92e8ff84fd69fc343a6dfe44943 : SUCCESS in 5h 10m 37s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/421a79e8f528486a9617c2a7c0e0b0ca : SUCCESS in 4h 33m 39s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/cd5867503dbd43309755958223bd9f01 : SUCCESS in 2h 09m 24s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/68d07ef336d240f2a173252a897e6d31 : TIMED_OUT in 3h 59m 26s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/dbe63ca9aa814448ac65ff9ff3d8da5b : SUCCESS in 1h 16m 02s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/17d3c23be30f4e1c91b53de223abbe06 : FAILURE in 1h 46m 56s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/875c8ace04df41e180999d2c2886f12a : SUCCESS in 2h 36m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e92f4f4beae3405ab21dae784c1422a7 : TIMED_OUT in 3h 15m 49s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/53e5ec48b2464c1994217abf77decda5 : SUCCESS in 2h 10m 11s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aea38b349df74f9f8ef191c41c32a5c8 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3db1d3eeb0054265aaed0a4b647a9d95 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a03535629de940d8bf8a5ea0985f187e : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/836a4ec6a7934cc7a33f6153470875be : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fc3b70a0391489494ce354c384e9f0a : SUCCESS in 2m 54s\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6f79486dafc3481dac687fb0ef8cd3b8 : FAILURE in 1h 16m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"472baccae59c8f871e65ea02c62f7de47ee01a81","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-17 11:44:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1e326158a3e17a7718668d4a21113038d1eb86fa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-17 13:56:02.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e2f4087adb4841c6bff3bb672e4a8b5a : FAILURE in 2h 10m 57s","accounts_in_message":[],"_revision_number":3},{"id":"661546c81a7c78767b6f5d2eb7c8ade75ab40d19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-17 15:11:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/270b5d93df5142038af407f53ad23b4d : SUCCESS in 2h 52m 40s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/0637c401c6064a12a186307c152e37b6 : SUCCESS in 50m 32s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/f7229763b3f347d4a90150fb0d598233 : SUCCESS in 2h 28m 48s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/13613b01e18148c5ab7b7ff50461a28c : SUCCESS in 3h 17m 39s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/dc671ac882b545dfb67c9fc8e5ea41a0 : SUCCESS in 1h 56m 31s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/256d30ff1ef7486990191f33571bce0c : SUCCESS in 1h 34m 08s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/489ad669bb7447d9a0ece7a998788b13 : SUCCESS in 1h 06m 57s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/63281ed950f04704a45f8157871c6b1f : FAILURE in 1h 40m 52s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/791ed7bd0a604c768afb526aae703dca : SUCCESS in 1h 59m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/776bbd394dee4aa6b42b7623b3f3813e : SUCCESS in 1h 58m 22s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d3eee524fd7d42ddb637055eee480c20 : SUCCESS in 2h 04m 42s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9a8bfd6fe71478187c1dac6d6d8ae5a : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2e58fbd89e24e15ac6a3cadf02aadcb : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac63c654b74f4a26a6ece81297621bea : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c0c82f152e749a0b6f56238081dd1a1 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb738568c3134c8bb7320e43e4dbd0d7 : SUCCESS in 3m 57s\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/79da7de2b4e249de9d7fe9e21ca8233c : FAILURE in 59m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"49f9768c7cf558b6cdd398546a269b07400aa105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-17 15:11:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0a1bea209a4f9a1ca30385f1a0fb26b044316564","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-17 18:17:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9c3cafb4304b4fe98e7db2cb5f46e32e : SUCCESS in 2h 48m 24s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/9cfb10fa2b144de7ae26dbc56a9e1cfd : SUCCESS in 51m 15s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/9885ac96e4d647e8a004396f84df7849 : SUCCESS in 2h 27m 22s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/02aecb1944bd41f99fa62cb0b36d50d7 : SUCCESS in 3h 04m 48s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ba4a72ccea2945cf993fbfc4797353f1 : SUCCESS in 1h 59m 57s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/fff884f2d9aa42bba767de407247c3c9 : SUCCESS in 1h 03m 29s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/a8f8891fc5fc40f2a8e086e07a1792ba : SUCCESS in 1h 35m 27s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/86ea03fa2aec4d5cbd5995785b49f857 : SUCCESS in 1h 51m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1410951b3a24c46b502d9c6212dcc6e : SUCCESS in 2m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bce25c7c77c45cba792078c6e9c1393 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ad4f008ee5420cb1c2047f949c45c9 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c85bb74ea8a24de29a426ecf5492cb40 : SUCCESS in 2m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c988acb59bfb42df94af3e895af2eb1a : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":3},{"id":"f485a50476a820339d8a5d204f9fa00f6b37886e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-17 18:17:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7824676b76bd820b409e33bdcfa89bbd7b1d873d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-17 18:18:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a05d8fcb9134382b76d48bd33842b6d : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f1a0879feac318db490bcfd831be507ae208363d","revisions":{"c0eed440281bf74f448c7a6dc2e4fc20ad0e9187":{"kind":"REWORK","_number":1,"created":"2021-01-05 02:04:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/769186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/86/769186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/86/769186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/86/769186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/86/769186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/86/769186/1"}}},"commit":{"parents":[{"commit":"7e496911451b5d6fca9766fe1e294c7e5e5523ff","subject":"Fix inspect images calls to not use the index format keyword","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/7e496911451b5d6fca9766fe1e294c7e5e5523ff"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-12-11 17:27:21.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-05 02:04:29.000000000","tz":0},"subject":"Actually fail if a stop or remove call fails","message":"Actually fail if a stop or remove call fails\n\nAs part of the container rename actions, attempt to stop and remove\ncontainers without actually checking if they exist. Additionally we\ndon\u0027t actually fail if these command fail which leads to other errors\nlater that may be unrelated to the core issue. This change adds\nadditional checks around the stop and remove container actions to ensure\nthat the container is running before stopping and exists before\nremoving.\n\nChange-Id: I82172f5a2d4683509c802c0f14d571a40e21e6cc\nCloses-Bug: #1907833\n(cherry picked from commit 523465fe8cd845186e6b3095f6fd35fcc9712ace)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/c0eed440281bf74f448c7a6dc2e4fc20ad0e9187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/c0eed440281bf74f448c7a6dc2e4fc20ad0e9187"}]},"branch":"refs/heads/stable/train"},"ef40fcab84642dba232d3fd1f07dcceee832a122":{"kind":"REWORK","_number":2,"created":"2021-01-11 12:06:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/769186/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/86/769186/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/86/769186/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/86/769186/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/86/769186/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/86/769186/2"}}},"commit":{"parents":[{"commit":"7e496911451b5d6fca9766fe1e294c7e5e5523ff","subject":"Fix inspect images calls to not use the index format keyword","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/7e496911451b5d6fca9766fe1e294c7e5e5523ff"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-12-11 17:27:21.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-11 12:06:39.000000000","tz":540},"subject":"Actually fail if a stop or remove call fails","message":"Actually fail if a stop or remove call fails\n\nAs part of the container rename actions, attempt to stop and remove\ncontainers without actually checking if they exist. Additionally we\ndon\u0027t actually fail if these command fail which leads to other errors\nlater that may be unrelated to the core issue. This change adds\nadditional checks around the stop and remove container actions to ensure\nthat the container is running before stopping and exists before\nremoving.\n\nBackport note:\nThis change includes the subsequent change[1] to fix the wrong method\nname used.\n\n[1] 523465fe8cd845186e6b3095f6fd35fcc9712ace\n\nChange-Id: I82172f5a2d4683509c802c0f14d571a40e21e6cc\nCloses-Bug: #1907833\n(cherry picked from commit 523465fe8cd845186e6b3095f6fd35fcc9712ace)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/ef40fcab84642dba232d3fd1f07dcceee832a122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/ef40fcab84642dba232d3fd1f07dcceee832a122"}]},"branch":"refs/heads/stable/train"},"f1a0879feac318db490bcfd831be507ae208363d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-01-11 14:37:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/769186/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/86/769186/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/86/769186/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/86/769186/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/86/769186/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/86/769186/3"}}},"commit":{"parents":[{"commit":"7e496911451b5d6fca9766fe1e294c7e5e5523ff","subject":"Fix inspect images calls to not use the index format keyword","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/7e496911451b5d6fca9766fe1e294c7e5e5523ff"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-12-11 17:27:21.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-11 14:37:17.000000000","tz":0},"subject":"Actually fail if a stop or remove call fails","message":"Actually fail if a stop or remove call fails\n\nAs part of the container rename actions, attempt to stop and remove\ncontainers without actually checking if they exist. Additionally we\ndon\u0027t actually fail if these command fail which leads to other errors\nlater that may be unrelated to the core issue. This change adds\nadditional checks around the stop and remove container actions to ensure\nthat the container is running before stopping and exists before\nremoving.\n\nBackport note:\nThis change includes the subsequent change[1] to fix the wrong method\nname used.\n\n[1] Ie0462cd89f0beb2a2adc82768dea7d42dc343192\n\nChange-Id: I82172f5a2d4683509c802c0f14d571a40e21e6cc\nCloses-Bug: #1907833\n(cherry picked from commit 523465fe8cd845186e6b3095f6fd35fcc9712ace)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f1a0879feac318db490bcfd831be507ae208363d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f1a0879feac318db490bcfd831be507ae208363d"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
