)]}'
{"id":"openstack%2Fpaunch~825249","triplet_id":"openstack%2Fpaunch~stable%2Fussuri~Idc04b6a4276f0ac9e5a07af03c317e9f665eb6ce","project":"openstack/paunch","branch":"stable/ussuri","topic":"bug/1955056","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-04-12 14:50:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idc04b6a4276f0ac9e5a07af03c317e9f665eb6ce","subject":"Ensure systemd services are activated","status":"ABANDONED","created":"2022-01-19 04:19:37.000000000","updated":"2022-04-12 14:50:54.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"17935516b5aa2efc59c1cb4dd0fbea981134d239","_number":825249,"virtual_id_number":825249,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-04 08:18:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-19 04:51:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-19 04:55:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 09:30:12.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"CC"}],"messages":[{"id":"ba753fe68fdcd7f79eeecca348ccd29902b77114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-19 04:19:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f69ce2d360d052c1ae6948da05ef22a07f98c043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 04:51:17.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abe20fe5756548e6aa4fa016795d77eb : FAILURE in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb6c2b61ae34450bb905595d99f9d981 : FAILURE in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14985bbf8bee4a389c66d49cabdf3555 : FAILURE in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18e1bdabaf514922b6a87e301e52e546 : SUCCESS in 4m 10s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/009f7b2ef4f841c38379da52fbf24645 : FAILURE in 28m 46s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/9ed3e745a67c4e2883e86c2f6174c44c : SKIPPED","accounts_in_message":[],"_revision_number":1},{"id":"30275f1b617901c3982306965012b900d48cb62d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-19 04:55:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5b152e1dfec04a7994be7e334b4c4555 : FAILURE in 35m 48s","accounts_in_message":[],"_revision_number":1},{"id":"231f26ca9ea0d782564b645adc105761293124d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-26 09:29:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9a236a14dc75727e9686b9f2edf8bea435684422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 13:07:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55caf048c86f4352a1a67759417c93cb : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9dc44f3f7a7d490ea9fe203a135424af : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91804f76fdea49148e0a075eb7d03f3b : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/012c4342ebb04599b106674d8c6ebe33 : SUCCESS in 4m 35s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c7ca74b31a464c43a632c8489dbf0a1f : SUCCESS in 3h 30m 04s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/af433f65aebd421b8026ef40c137ee9b : SUCCESS in 1h 46m 38s","accounts_in_message":[],"_revision_number":2},{"id":"6de036ee491eda5ba32a3c537e3e7f6240e18bcf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-26 14:50:56.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"527e410d7247577f83b923e87f48bd12744fe29c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-04 00:44:06.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"0f75873c7ea18746a716e449df4c389454d3ddee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-04 04:32:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c1c0feb8a2572efc1ec4c3e471090aec0fec8b58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 08:18:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/954e8270820d4428bdd5640b8cc5ca5a : SUCCESS in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddfb5839dd0543fdb8f6815774fe368e : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01d0e5ba56514869808fb8da3105a721 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c421b4ed8a4fc081e668ec203dc16d : SUCCESS in 3m 09s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2f59c402907b4fdd95a4f8df34401d32 : SUCCESS in 3h 39m 49s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ad24ad87e9904cbb9371699b62576918 : SUCCESS in 1h 55m 57s","accounts_in_message":[],"_revision_number":3},{"id":"095efcc4eaf1e0aa6b077c2a6926150995e171a0","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-05 09:30:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"17935516b5aa2efc59c1cb4dd0fbea981134d239","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-12 14:50:54.000000000","message":"Abandoned\n\nabandoning per http://lists.openstack.org/pipermail/openstack-discuss/2022-April/028026.html - so we can move EOL https://review.opendev.org/c/openstack/releases/+/834049","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"483c0a37333b2fa239b163ebb2e26706710a15f5","revisions":{"eb1d690f6183721da961f882054bf98406db506c":{"kind":"REWORK","_number":1,"created":"2022-01-19 04:19:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/825249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/49/825249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/49/825249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/49/825249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/49/825249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/49/825249/1"}}},"commit":{"parents":[{"commit":"f546658bdf5991a5f02393f120b449141daa55f0","subject":"Revert \"Ensure systemd services are activated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f546658bdf5991a5f02393f120b449141daa55f0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-19 04:16:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-19 04:19:12.000000000","tz":540},"subject":"Ensure systemd services are activated","message":"Ensure systemd services are activated\n\nWhen deployment times out during running the paunch command, containers\nmight be left in an incomplete status. Especially when podman is used\nand containers are managed by systemd, sometimes paunch succeeds to\ncreate a container but it is aborted before creating and enabling\nthe systemd service for that container.\n\nThis change makes paunch check status of systemd services, to ensure\nany disabled/stopped/missing services and remove containers with\nincomplete systemd services so that the containers and the associated\nsystemd services are recreated and reconfigured.\n\nThis reimplements commit 6ddd429488d0e0043b9d6c46c56023c126807cfa which\nhad several issues.\n\nCloses-Bug: #1955056\nChange-Id: Idc04b6a4276f0ac9e5a07af03c317e9f665eb6ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/eb1d690f6183721da961f882054bf98406db506c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/eb1d690f6183721da961f882054bf98406db506c"}]},"branch":"refs/heads/stable/ussuri"},"097d80d81bc6846001e5940fc72832401ca0c674":{"kind":"REWORK","_number":2,"created":"2022-01-26 09:29:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/825249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/49/825249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/49/825249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/49/825249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/49/825249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/49/825249/2"}}},"commit":{"parents":[{"commit":"f546658bdf5991a5f02393f120b449141daa55f0","subject":"Revert \"Ensure systemd services are activated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f546658bdf5991a5f02393f120b449141daa55f0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-19 04:16:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-26 09:29:10.000000000","tz":540},"subject":"Ensure systemd services are activated","message":"Ensure systemd services are activated\n\nWhen deployment times out during running the paunch command, containers\nmight be left in an incomplete status. Especially when podman is used\nand containers are managed by systemd, sometimes paunch succeeds to\ncreate a container but it is aborted before creating and enabling\nthe systemd service for that container.\n\nThis change makes paunch check status of systemd services, to ensure\nany disabled/stopped/missing services and remove containers with\nincomplete systemd services so that the containers and the associated\nsystemd services are recreated and reconfigured.\n\nThis reimplements commit 6ddd429488d0e0043b9d6c46c56023c126807cfa which\nhad several issues.\n\nCloses-Bug: #1955056\nChange-Id: Idc04b6a4276f0ac9e5a07af03c317e9f665eb6ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/097d80d81bc6846001e5940fc72832401ca0c674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/097d80d81bc6846001e5940fc72832401ca0c674"}]},"branch":"refs/heads/stable/ussuri"},"483c0a37333b2fa239b163ebb2e26706710a15f5":{"kind":"REWORK","_number":3,"created":"2022-04-04 04:32:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/825249/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/paunch","ref":"refs/changes/49/825249/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/paunch refs/changes/49/825249/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/paunch refs/changes/49/825249/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/paunch refs/changes/49/825249/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/paunch refs/changes/49/825249/3"}}},"commit":{"parents":[{"commit":"f546658bdf5991a5f02393f120b449141daa55f0","subject":"Revert \"Ensure systemd services are activated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/f546658bdf5991a5f02393f120b449141daa55f0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-19 04:16:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-04 04:31:50.000000000","tz":540},"subject":"Ensure systemd services are activated","message":"Ensure systemd services are activated\n\nWhen deployment times out during running the paunch command, containers\nmight be left in an incomplete status. Especially when podman is used\nand containers are managed by systemd, sometimes paunch succeeds to\ncreate a container but it is aborted before creating and enabling\nthe systemd service for that container.\n\nThis change makes paunch check status of systemd services, to ensure\nany disabled/stopped/missing services and remove containers with\nincomplete systemd services so that the containers and the associated\nsystemd services are recreated and reconfigured.\n\nThis reimplements commit 6ddd429488d0e0043b9d6c46c56023c126807cfa which\nhad several issues.\n\nCloses-Bug: #1955056\nChange-Id: Idc04b6a4276f0ac9e5a07af03c317e9f665eb6ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/483c0a37333b2fa239b163ebb2e26706710a15f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/paunch/commit/483c0a37333b2fa239b163ebb2e26706710a15f5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
