)]}'
{"id":"openstack%2Fpbr~1000951","triplet_id":"openstack%2Fpbr~master~I0e900cfccf9022b1350602784d293e8a95c4ecfe","project":"openstack/pbr","branch":"master","hashtags":[],"change_id":"I0e900cfccf9022b1350602784d293e8a95c4ecfe","subject":"fixtures: Allow Venv to be called manually","status":"NEW","created":"2026-08-14 09:59:19.000000000","updated":"2026-08-14 10:53:18.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5717a337a00f00768503685503b38e9b2269f86c","_number":1000951,"virtual_id_number":1000951,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-14 10:53:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-14 10:53:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c5bef946799ae0b0a197fda0d1d9c837f5ec3f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-14 09:59:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5717a337a00f00768503685503b38e9b2269f86c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 10:53:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cce814710d7428fb5cd7683e933720d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2867f71741b547c193658e0d68e774f6 : SUCCESS in 51m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76f4a55460c94775bac9d97d4c4aa474 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d233a32778fb4fb49eb3d0a30d811dfb : SUCCESS in 4m 33s\n- build-python-release https://zuul.opendev.org/t/openstack/build/92779f604af441bcbebd60f7dd311897 : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79ace2655b104248850ff57ce3e6fca0 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f273bf66319491bb18e1972bb335409 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a35463455a443449f78ef689c479269 : SUCCESS in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/539c409b50e84954a8bc0f8f81930e3e : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76e5facf45c444f9976a7bae1eba5427 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37b2ef90dee043a09edf818efc1e9f67 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b85f98470f64dc9bc046def26179c54 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1cfcdb29e8974fc3acd6b5561e186776 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/afb1e495c5964231bd895daaf3a2cec7 : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1aaf9c594c74f63a474deaf5f19a916 : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0feb5576ca29443c9f994f4882dd1f68 : SUCCESS in 5m 47s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/4128012c52b64894b25b6fab704baf55 : SUCCESS in 4m 16s (non-voting)\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/656e3a6929af4ceab0325dc560c264c9 : SUCCESS in 35m 53s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/2dcfd390d9ee455fbe95a8382c25cacb : SUCCESS in 20m 28s\n- pbr-installation-openstack-resolute https://zuul.opendev.org/t/openstack/build/ecdc8dd6fdd6429e87d7cdd8cf61d46e : SUCCESS in 28m 15s\n- pbr-installation-openstack-pip-dev-resolute https://zuul.opendev.org/t/openstack/build/8105770cd2e2484d9414fb6508806577 : SUCCESS in 18m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19785a130cdaec8dfe9a7c4dc645e1b0a0c81c95","revisions":{"19785a130cdaec8dfe9a7c4dc645e1b0a0c81c95":{"kind":"REWORK","_number":1,"created":"2026-08-14 09:59:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/1000951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/1000951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/1000951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/1000951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/1000951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/1000951/1"}}},"commit":{"parents":[{"commit":"2bc3327add794086b10ead859cad6b5447739c4f","subject":"tests: Handle projects without setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2bc3327add794086b10ead859cad6b5447739c4f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-13 13:51:11.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-14 09:58:53.000000000","tz":60},"subject":"fixtures: Allow Venv to be called manually","message":"fixtures: Allow Venv to be called manually\n\nThis avoid mega indentation for longer uses of a given venv.\n\nChange-Id: I0e900cfccf9022b1350602784d293e8a95c4ecfe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/19785a130cdaec8dfe9a7c4dc645e1b0a0c81c95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/19785a130cdaec8dfe9a7c4dc645e1b0a0c81c95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
