)]}'
{"id":"openstack%2Fpbr~1000953","triplet_id":"openstack%2Fpbr~master~Ifbb45a82d67565e13602facf21cf090c7002eb4b","project":"openstack/pbr","branch":"master","hashtags":[],"change_id":"Ifbb45a82d67565e13602facf21cf090c7002eb4b","subject":"tests: Combine two tests","status":"NEW","created":"2026-08-14 09:59:19.000000000","updated":"2026-08-14 11:21:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f84ba259b223f95ccef01249eb04bfdfe50ac48c","_number":1000953,"virtual_id_number":1000953,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-14 11:21:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-14 11:21:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab78872f013f0069996f6f210aa079da3572fdac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-14 09:59:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f84ba259b223f95ccef01249eb04bfdfe50ac48c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 11:21:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7863813144fc4c5db06d3eb58b1c568a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5417e4ca9874628bd1b3def401800a3 : SUCCESS in 1h 17m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15f4d0de902c415aa84568403fee9689 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c98b6e8b6f3a495a89144a2873254730 : SUCCESS in 4m 16s\n- build-python-release https://zuul.opendev.org/t/openstack/build/b246e3ed4ae542d5880cb9ddcef6abf3 : SUCCESS in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fb81f1794c4496ba735f0ae3880b885 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7769ebdccd2046ddaa5b6a604b4b4a86 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec97130e997a4f54917f244f6a436871 : SUCCESS in 8m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/baa340122b08421cba289a8d00df0c0c : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5bfb8c965a945a8bf7a07d5db434957 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4419a46f64784bf291666d60149ecbef : SUCCESS in 8m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a694127dad046c496708017894aefd8 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f13ed9f0b0b46a5939d05278b3b71c5 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/24a1eb5188ac43c29ed47b75dde5ec1d : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d62725433626498d8f718a97a260d842 : SUCCESS in 4m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/276c8a3f72c842ea89c36c57b3eafa5f : SUCCESS in 5m 09s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/e0147016d8b948acb8f2f1a014f0ee98 : SUCCESS in 11m 00s (non-voting)\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/26824a572bf74ce18fc3324458e5aed8 : SUCCESS in 33m 40s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/7391337069db45d38647668201702d65 : SUCCESS in 30m 00s\n- pbr-installation-openstack-resolute https://zuul.opendev.org/t/openstack/build/a8884fea748d446da6adf271f79910c1 : SUCCESS in 37m 30s\n- pbr-installation-openstack-pip-dev-resolute https://zuul.opendev.org/t/openstack/build/08ccdc408ef2410a8ee1b46aa7d38e42 : SUCCESS in 35m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1eedb43b0bd90ac7807dd11e21a90c80a05c9207","revisions":{"1eedb43b0bd90ac7807dd11e21a90c80a05c9207":{"kind":"REWORK","_number":1,"created":"2026-08-14 09:59:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/1000953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/1000953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/1000953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/1000953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/1000953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/1000953/1"}}},"commit":{"parents":[{"commit":"71504421be436a2fd666cf9eb13f878d6ae17bcd","subject":"tests: Trivial pre-rework changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/71504421be436a2fd666cf9eb13f878d6ae17bcd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-14 08:59:16.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-14 09:58:53.000000000","tz":60},"subject":"tests: Combine two tests","message":"tests: Combine two tests\n\nThis also resolves an outstanding TODO. Note that we also move\ninstallation of the setuptools and pip packages from a loop into the\nfixture, to avoid the double installation (install then downgrade).\n\nChange-Id: Ifbb45a82d67565e13602facf21cf090c7002eb4b\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1eedb43b0bd90ac7807dd11e21a90c80a05c9207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1eedb43b0bd90ac7807dd11e21a90c80a05c9207"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
