)]}'
{"id":"openstack%2Fpbr~1002356","triplet_id":"openstack%2Fpbr~master~Icd37a8e104e33af10939bb482fbd9c7c91efc61e","project":"openstack/pbr","branch":"master","attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-02 01:50:17.000000000","reason":"\u003cGERRIT_ACCOUNT_12404\u003e replied on the change","reason_account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-01 13:34:36.000000000","reason":"\u003cGERRIT_ACCOUNT_12404\u003e replied on the change","reason_account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-09-02 01:50:17.000000000","reason":"\u003cGERRIT_ACCOUNT_12404\u003e replied on the change","reason_account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Icd37a8e104e33af10939bb482fbd9c7c91efc61e","subject":"Only provide PEP-660 entry points on setuptools \u003e\u003d 64.0.0","status":"NEW","created":"2026-08-25 17:25:03.000000000","updated":"2026-09-02 01:50:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"25c23c0569f61ded9df770806d8e3f16f22926c7","_number":1002356,"virtual_id_number":1002356,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-27 12:24:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2026-08-25 17:34:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2026-09-01 13:34:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-25 17:34:33.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2026-08-25 18:42:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 13:34:36.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2026-09-02 01:50:17.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-09-02 01:50:17.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"569f9921ae52035b6c9b924bb96575d0f5946f77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-25 17:25:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"076efb8fe259b7edb677a89bdad8af8d31fbdadc","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-08-25 17:34:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e6be26844565c77a4299f3b900c7893bb457bdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 18:42:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/894f27df551e4b2aaba33247e4962ced\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd5606cd64424c6a95ea5365490e92a4 : SUCCESS in 48m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b1a68253824151833cbc3d680c105f : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17028f7429d342f28a5de3f8f113735f : SUCCESS in 4m 23s\n- build-python-release https://zuul.opendev.org/t/openstack/build/e7e6c0b5a6b34266986635bfc9e932ed : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bb2f78c02024da5a2c00b609dc47e29 : RETRY_LIMIT in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f154945aafcd47ce9da310c249796883 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba6535cf797b45fda1314d3e996d96c3 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dba409f97c494cb38f6155015f1fd4e8 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc784707a8024f82a20ab03a8d687d33 : SUCCESS in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1361702a79a74054ae332f8fc5ae7177 : SUCCESS in 7m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70173335c4ef474583c2628420d3162c : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/387c3329b4e24c22b43cb799dce94952 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a01e5bc681794b129de11c81be98506c : SUCCESS in 5m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b869dca06cf042179a0cd1ccf325ae17 : SUCCESS in 5m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/59cf995f642e4cbcaa2a8f6833eb454b : SUCCESS in 6m 59s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/b6276517192e47d38acac8ce57a77fc4 : SUCCESS in 5m 09s (non-voting)\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9ba6803dcf1a49dd98930466b67deac6 : SUCCESS in 21m 32s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/c5fbd22477ac408e99b50a7fbd2dc7ff : SUCCESS in 41m 11s\n- pbr-installation-openstack-resolute https://zuul.opendev.org/t/openstack/build/e8031fba535742d780c811d0a0099caf : SUCCESS in 18m 28s\n- pbr-installation-openstack-pip-dev-resolute https://zuul.opendev.org/t/openstack/build/0d16387d529f41e7b13e799603d2ab04 : SUCCESS in 36m 27s","accounts_in_message":[],"_revision_number":1},{"id":"04a8bb4653326eaeb2a7b35d1e8587b08d3d5b7c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-27 11:21:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28b3b42f82b7c0a09244e1b1b4bdd6dc2b0950e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 12:24:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/810d8862e6dd4718882c4f339ff3ee77\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf27f3daf2a14082976dfbee029fd9ec : SUCCESS in 1h 00m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2315b378d13442af9325078c7ef4ec2a : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1da24ddbe7ab4300bfa720d4a4edda64 : SUCCESS in 4m 26s\n- build-python-release https://zuul.opendev.org/t/openstack/build/dcdcfa5e47ae4a7dad83a61e054ab26c : SUCCESS in 1m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/681e1081d41e42af98ab6740bb1f1537 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c83b6664996344f1a2734300714db087 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50cd56ee4be448c58605c13bcde00eb3 : SUCCESS in 10m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c10eeb0c640a49968474e76f2f5b0cf6 : SUCCESS in 9m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e20004fc7b747499e97d3731741ef0f : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cb352add3d44fa1802695397311127d : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03134aadd9654ca894faf510b928ff50 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f34a90c7c32c4b6288ec9e45484397b0 : SUCCESS in 6m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26c2bc1d34ee4af9a5c05e2e705d8d8b : SUCCESS in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21266930f7954d5795f754b6aad0c075 : SUCCESS in 6m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a18a823d58ef412ebc8dc18c7027c657 : SUCCESS in 4m 37s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/7f1a2919730c4527ac206677e57d8ed2 : SUCCESS in 4m 59s (non-voting)\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/ca6051e86d7f4103bfcad53a59f19ea5 : SUCCESS in 34m 23s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/83d1168ca8434733b330e066bba48bb1 : SUCCESS in 44m 21s\n- pbr-installation-openstack-resolute https://zuul.opendev.org/t/openstack/build/52e389291e94468bb3698b158e631862 : SUCCESS in 37m 11s\n- pbr-installation-openstack-pip-dev-resolute https://zuul.opendev.org/t/openstack/build/6204e08e1035479095b28e0bd88c41c2 : SUCCESS in 16m 58s","accounts_in_message":[],"_revision_number":1},{"id":"efb9b514f903f54c45fa35a4703b404ee60420aa","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2026-09-01 13:34:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"db2ed34be1ef3f65e78bff31fdc0e678d3cfedeb","revisions":{"db2ed34be1ef3f65e78bff31fdc0e678d3cfedeb":{"kind":"REWORK","_number":1,"created":"2026-08-25 17:25:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/1002356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/56/1002356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/56/1002356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/56/1002356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/56/1002356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/56/1002356/1"}}},"commit":{"parents":[{"commit":"15d52512c913aa52d757858438ac30ddaa47b633","subject":"Revert \"Set setuptools min in requirements.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/15d52512c913aa52d757858438ac30ddaa47b633"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-25 17:15:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-25 17:17:35.000000000","tz":60},"subject":"Only provide PEP-660 entry points on setuptools \u003e\u003d 64.0.0","message":"Only provide PEP-660 entry points on setuptools \u003e\u003d 64.0.0\n\nThis is a latent issue that appears to have been triggered by the (since\nyanked) 7.1.0 release. I haven\u0027t been able to determine *why*, but we\nalready attempted a fix in I7da60b2ceabc8a5ba3152c93e2622369c1f42185\nwhich we later had to revert. The approach is more correct and avoid\nmeddling with requirements files, which is a whole \"thing\" when\neasy_install gets in the mix.\n\nChange-Id: Icd37a8e104e33af10939bb482fbd9c7c91efc61e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/db2ed34be1ef3f65e78bff31fdc0e678d3cfedeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/db2ed34be1ef3f65e78bff31fdc0e678d3cfedeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
