)]}'
{"id":"openstack%2Fpbr~1002357","triplet_id":"openstack%2Fpbr~master~I85d3d589272859c285666324646c115f749aa92c","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-27 11:21:25.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I85d3d589272859c285666324646c115f749aa92c","subject":"Insist on Python 3.11 for pyproject.toml-only projects","status":"NEW","created":"2026-08-25 17:25:03.000000000","updated":"2026-08-27 12:10:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"0ec8d181444f2fc48be858351f9d5518ca89642e","_number":1002357,"virtual_id_number":1002357,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-27 12:10:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2026-08-25 17:34:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-25 17:34:24.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2026-08-25 18:11:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26f8cf4bc8a95e14f9780eab370b8cbc7946fa98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-25 17:25:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbd586494976eea4f40d8962f9c6fcb58e1c3be3","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-08-25 17:34:24.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b80e33d6f242f25daaf55a1e197926e3e8f239b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 18:11:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/441cd5f9df3445419fa81b1350a636a0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a21e2f8bbddb44ec80019f36b5a6c6c5 : FAILURE in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1879f5b7324e4728b9689c7052dea8b8 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd69d315b43a428c82458ddfeb99eba2 : SUCCESS in 3m 52s\n- build-python-release https://zuul.opendev.org/t/openstack/build/844a8e3111cd42f9a2d76b9afed65a74 : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/831e5f8098704ee5af68d1aa69d6097b : SUCCESS in 7m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/247a0a13f08f44a6b819fcea82054de8 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba754e87a6ec4571bd386938c7bc3804 : SUCCESS in 12m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15764aee7120462fa299c80ea8669a47 : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f49dca9e010c456ea3cf3b8efae81838 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dbc8ba4abfc459f8f8d2e4f4d1ac864 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f1f03f7dcad46e9a8f093294ad5612c : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84c5bf586996458d96249e6b1bbee1a2 : SUCCESS in 6m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35f95e71b78c44bebc9df36f52503fd6 : SUCCESS in 6m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b3def684bce49b296abaafa029b1774 : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6b77e027705a433fa52eb00e41fd0b6d : SUCCESS in 7m 16s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/459629ae3ff4424badefe0b84fcdbb3f : SUCCESS in 8m 25s (non-voting)\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/66019746273f4697a607b0d4395f24b4 : SUCCESS in 22m 36s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/472b731f725145eebfef648182eaf9cb : SUCCESS in 37m 25s\n- pbr-installation-openstack-resolute https://zuul.opendev.org/t/openstack/build/089d9c32b80c4c8da4622f523daba706 : SUCCESS in 19m 33s\n- pbr-installation-openstack-pip-dev-resolute https://zuul.opendev.org/t/openstack/build/59041e73a5f34c1097c4fbb833be9fcd : SUCCESS in 39m 20s","accounts_in_message":[],"_revision_number":1},{"id":"3a4a8f0d51ebd1e766a9e87a0e0a5ac429c0e94d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-27 11:21:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ec8d181444f2fc48be858351f9d5518ca89642e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 12:10:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd731a7ce3ea498099403d9fce6cdee7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39f22598693047a8b0ba29060d280b0c : SUCCESS in 43m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5295a61035ec4b0597441a1cd54a7cc9 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15f19d45c3d24709be4ad3b3ed57b05d : SUCCESS in 3m 11s\n- build-python-release https://zuul.opendev.org/t/openstack/build/a2610f9d1c474be7b0437fc60e03df1d : SUCCESS in 1m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/161d6858f48b43acb67b2d8439f65248 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95cc67faf83b48cb9a904ecddd336f4d : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef8a7a3c40764349a794aeed6d6b5ad8 : SUCCESS in 10m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c438b43a2c4844758e5f4e81ee6862d1 : SUCCESS in 10m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c14b623b4e4146d9ac72871ab13bf3e3 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d1ac7c26c754c82a06fde1d950f8fbe : SUCCESS in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24dbe0e7491c432e996ab03e51c9829d : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34648041fe634d0b802aa2c7343e5181 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3792dd2a6e4f495a968c3b367e6c3bd0 : SUCCESS in 7m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0cd69ab9bca4412b9d0d5ded781dfc93 : SUCCESS in 4m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6679b910fd91403ba8595bdf185d8037 : SUCCESS in 6m 54s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/02a985fbddbd4b05baba51d93828d3e0 : SUCCESS in 8m 20s (non-voting)\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/7e310d8881fe4cad84c5bbfe1e83fdb6 : SUCCESS in 35m 28s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/fac535323c3d42f6a06ca9936eacc836 : SUCCESS in 35m 07s\n- pbr-installation-openstack-resolute https://zuul.opendev.org/t/openstack/build/ea62f29aa2774850a6e0b603c22958d6 : SUCCESS in 18m 22s\n- pbr-installation-openstack-pip-dev-resolute https://zuul.opendev.org/t/openstack/build/22ddc5b5bb7640f0a86ac15b74c0ac5c : SUCCESS in 34m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34cd3b86d793b24edc67d97396f953975fc3c842","revisions":{"34cd3b86d793b24edc67d97396f953975fc3c842":{"kind":"REWORK","_number":1,"created":"2026-08-25 17:25:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/57/1002357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/57/1002357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/57/1002357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/57/1002357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/57/1002357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/57/1002357/1"}}},"commit":{"parents":[{"commit":"db2ed34be1ef3f65e78bff31fdc0e678d3cfedeb","subject":"Only provide PEP-660 entry points on setuptools \u003e\u003d 64.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/db2ed34be1ef3f65e78bff31fdc0e678d3cfedeb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-25 13:29:27.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-25 17:24:14.000000000","tz":60},"subject":"Insist on Python 3.11 for pyproject.toml-only projects","message":"Insist on Python 3.11 for pyproject.toml-only projects\n\nThe current tomli fallback only works on certain setuptools versions or\nrequires users pre-install that package themselves. As nice as it would\nbe to provide this functionality to projects supporting older Python\nversions, it is proving hard to test and arguably is not worth the\nbenefit. Limiting ourselves to Python 3.11 allows us to simplify our\ncode and hopefully avoid breaking older environments.\n\nChange-Id: I85d3d589272859c285666324646c115f749aa92c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/34cd3b86d793b24edc67d97396f953975fc3c842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/34cd3b86d793b24edc67d97396f953975fc3c842"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
