)]}'
{"id":"openstack%2Fpbr~217733","triplet_id":"openstack%2Fpbr~master~Ib9e2207b0d1d9cee656736e94865fb404b6a868d","project":"openstack/pbr","branch":"master","topic":"bug/1482271","hashtags":[],"change_id":"Ib9e2207b0d1d9cee656736e94865fb404b6a868d","subject":"Protect WSGI application with a critical section","status":"MERGED","created":"2015-08-27 14:27:08.000000000","updated":"2015-09-10 23:39:54.000000000","submitted":"2015-09-10 23:39:52.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"13eb2158407ae1e7e1760d41673c27feab1c39b2","_number":217733,"virtual_id_number":217733,"owner":{"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2015-09-10 23:39:52.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},{"value":0,"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"recommended":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"all":[{"value":1,"date":"2015-08-31 18:15:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":2,"date":"2015-08-30 15:25:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2015-08-28 16:08:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":2,"date":"2015-09-10 22:21:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},{"value":0,"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"all":[{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":1,"date":"2015-09-10 22:21:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},{"value":0,"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},{"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-08-28 16:08:56.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2015-08-30 15:25:01.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2015-08-31 18:15:15.000000000","updated_by":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"reviewer":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"state":"REVIEWER"},{"updated":"2015-09-10 22:21:29.000000000","updated_by":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"reviewer":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"state":"REVIEWER"},{"updated":"2015-09-10 23:39:52.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"1c622440873c16344eeacc51eb076bb5058c6fc0","author":{"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"},"date":"2015-08-27 14:27:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd2ad1ae233e1fa488fcf17b8fed3d9c002eb674","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2015-08-27 14:58:57.000000000","message":"Patch Set 1: Code-Review+1\n\nit would be really good to explain the bug more in the commit message here. The fix is probably good, but a detailed explaination would make this more likely to get people to review it.","accounts_in_message":[],"_revision_number":1},{"id":"76018e86628281a5b024eef7d6973b91753aa777","author":{"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"},"date":"2015-08-27 15:08:12.000000000","message":"Patch Set 2: Commit message was updated","accounts_in_message":[],"_revision_number":2},{"id":"b6602ece25c72bb1494697ae0d5e8d0cb106cfea","author":{"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"},"date":"2015-08-27 15:09:13.000000000","message":"Patch Set 2:\n\n@Sean Dague, thanks, I\u0027ve added bug description in the message (w/o links)","accounts_in_message":[],"_revision_number":2},{"id":"2ae6aa8cbd46a2a6325dea1ba19931c17086b148","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2015-08-27 15:11:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a8f4ba289727cf0546602e6c47249fff1e662812","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2015-08-27 15:11:50.000000000","message":"Patch Set 2:\n\nthanks!","accounts_in_message":[],"_revision_number":2},{"id":"0f39cc004c3456d5ff659ea2d5c38bc623ef3778","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-27 17:23:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-pbr-python26 http://logs.openstack.org/33/217733/2/check/gate-pbr-python26/f37a885/ : FAILURE in 44s\n- gate-pbr-pep8 http://logs.openstack.org/33/217733/2/check/gate-pbr-pep8/f89819b/ : FAILURE in 25s\n- gate-pbr-docs http://docs-draft.openstack.org/33/217733/2/check/gate-pbr-docs/f57449c//doc/build/html/ : SUCCESS in 32s\n- gate-pbr-python27 http://logs.openstack.org/33/217733/2/check/gate-pbr-python27/95d9c71/ : FAILURE in 45s\n- gate-pbr-python34 http://logs.openstack.org/33/217733/2/check/gate-pbr-python34/c3f3205/ : FAILURE in 47s\n- gate-tempest-dsvm-neutron-src-pbr http://logs.openstack.org/33/217733/2/check/gate-tempest-dsvm-neutron-src-pbr/04abf33/ : SUCCESS in 1h 14m 33s\n- gate-pbr-pypy http://logs.openstack.org/33/217733/2/check/gate-pbr-pypy/5718abc/ : FAILURE in 1m 27s (non-voting)\n- gate-pbr-installation-dsvm http://logs.openstack.org/33/217733/2/check/gate-pbr-installation-dsvm/7be3bcb/ : SUCCESS in 27m 40s\n- gate-pbr-installation-dsvm-upstream http://logs.openstack.org/33/217733/2/check/gate-pbr-installation-dsvm-upstream/f6eb7eb/ : SUCCESS in 26m 57s","accounts_in_message":[],"_revision_number":2},{"id":"9c19db3ca63c713e7c34bb13005d110ee648f7ec","author":{"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"},"date":"2015-08-28 10:20:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0b0a9a60aa57f5e8e897fcb406ed2ee4b622d1cc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-28 11:52:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-pbr-python26 http://logs.openstack.org/33/217733/3/check/gate-pbr-python26/ab03265/ : SUCCESS in 48s\n- gate-pbr-pep8 http://logs.openstack.org/33/217733/3/check/gate-pbr-pep8/e515049/ : FAILURE in 35s\n- gate-pbr-docs http://docs-draft.openstack.org/33/217733/3/check/gate-pbr-docs/8d60945//doc/build/html/ : SUCCESS in 42s\n- gate-pbr-python27 http://logs.openstack.org/33/217733/3/check/gate-pbr-python27/5021d2b/ : SUCCESS in 1m 07s\n- gate-pbr-python34 http://logs.openstack.org/33/217733/3/check/gate-pbr-python34/802e6a8/ : SUCCESS in 1m 24s\n- gate-tempest-dsvm-neutron-src-pbr http://logs.openstack.org/33/217733/3/check/gate-tempest-dsvm-neutron-src-pbr/932d85d/ : SUCCESS in 1h 16m 39s\n- gate-pbr-pypy http://logs.openstack.org/33/217733/3/check/gate-pbr-pypy/78f4111/ : SUCCESS in 1m 32s (non-voting)\n- gate-pbr-installation-dsvm http://logs.openstack.org/33/217733/3/check/gate-pbr-installation-dsvm/726ed57/ : SUCCESS in 27m 27s\n- gate-pbr-installation-dsvm-upstream http://logs.openstack.org/33/217733/3/check/gate-pbr-installation-dsvm-upstream/db4487d/ : SUCCESS in 24m 28s","accounts_in_message":[],"_revision_number":3},{"id":"4343f98e8743991022df2c812865a0e55bafc132","author":{"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"},"date":"2015-08-28 12:04:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5e42b5133b71ce0f624724caa83662420fb11ebc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-28 13:14:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-pbr-python26 http://logs.openstack.org/33/217733/4/check/gate-pbr-python26/3e83507/ : SUCCESS in 37s\n- gate-pbr-pep8 http://logs.openstack.org/33/217733/4/check/gate-pbr-pep8/11fcad0/ : SUCCESS in 20s\n- gate-pbr-docs http://docs-draft.openstack.org/33/217733/4/check/gate-pbr-docs/e6dacb9//doc/build/html/ : SUCCESS in 33s\n- gate-pbr-python27 http://logs.openstack.org/33/217733/4/check/gate-pbr-python27/576d9f4/ : SUCCESS in 37s\n- gate-pbr-python34 http://logs.openstack.org/33/217733/4/check/gate-pbr-python34/a21ca36/ : SUCCESS in 52s\n- gate-tempest-dsvm-neutron-src-pbr http://logs.openstack.org/33/217733/4/check/gate-tempest-dsvm-neutron-src-pbr/c057f84/ : SUCCESS in 1h 01m 55s\n- gate-pbr-pypy http://logs.openstack.org/33/217733/4/check/gate-pbr-pypy/0d17432/ : SUCCESS in 57s (non-voting)\n- gate-pbr-installation-dsvm http://logs.openstack.org/33/217733/4/check/gate-pbr-installation-dsvm/1ff780a/ : SUCCESS in 21m 10s\n- gate-pbr-installation-dsvm-upstream http://logs.openstack.org/33/217733/4/check/gate-pbr-installation-dsvm-upstream/31d5fd5/ : SUCCESS in 31m 12s","accounts_in_message":[],"_revision_number":4},{"id":"f5ea7a008e8908f0de8739f65e0499a53a3d8d9f","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2015-08-28 16:08:56.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9ba94deaa32221ad84fe032848d7227c78bd035e","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2015-08-30 15:25:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4c8ed36e751da3f44836b682c997c76405558de6","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2015-08-31 18:15:15.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"bf15c4d0f467bcbba4be635d428a36adbfcb90b7","author":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"date":"2015-09-10 22:21:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nI think we need to remove the innards of the wsgi stuff from here: revving the build system every time a deployment/operational bug occurs is an anti-pattern. \n\nWe should instead have a library which encapsulates this, and just write an entry point that uses that.","accounts_in_message":[],"_revision_number":4},{"id":"4cdbad8e7289892e91996cfc8af438abfc1eb8a1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-10 22:21:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"5d369ed23943c2f9d523ebe534bf263ce63f75b4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-10 23:39:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-pbr-python26 http://logs.openstack.org/33/217733/4/gate/gate-pbr-python26/b27400e/ : SUCCESS in 32s\n- gate-pbr-docs http://docs-draft.openstack.org/33/217733/4/gate/gate-pbr-docs/e9a937d//doc/build/html/ : SUCCESS in 39s\n- gate-pbr-pep8 http://logs.openstack.org/33/217733/4/gate/gate-pbr-pep8/5638f2b/ : SUCCESS in 31s\n- gate-pbr-python27 http://logs.openstack.org/33/217733/4/gate/gate-pbr-python27/cf9462d/ : SUCCESS in 55s\n- gate-pbr-python34 http://logs.openstack.org/33/217733/4/gate/gate-pbr-python34/efe47c7/ : SUCCESS in 58s\n- gate-tempest-dsvm-neutron-src-pbr http://logs.openstack.org/33/217733/4/gate/gate-tempest-dsvm-neutron-src-pbr/fb33375/ : SUCCESS in 1h 17m 51s\n- gate-pbr-installation-dsvm http://logs.openstack.org/33/217733/4/gate/gate-pbr-installation-dsvm/d1760bb/ : SUCCESS in 24m 33s","accounts_in_message":[],"_revision_number":4},{"id":"d1d0494caeb53a27280ed7ca2b54bedadd7eb8b3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-10 23:39:54.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6757913f0f7bf36c845a222cd8e5691a4d07c503","revisions":{"85e28aa2a41aeca3392a6ba84fb2cff2e6d0693d":{"kind":"REWORK","_number":1,"created":"2015-08-27 14:27:08.000000000","uploader":{"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"},"ref":"refs/changes/33/217733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/33/217733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/217733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/217733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/217733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/33/217733/1"}}},"commit":{"parents":[{"commit":"1e05037c2ae02a4d52d0f50abf00d61021971508","subject":"Strip comments in requirements files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1e05037c2ae02a4d52d0f50abf00d61021971508"}]}],"author":{"name":"Alexander Makarov","email":"amakarov@mirantis.com","date":"2015-08-27 14:25:52.000000000","tz":180},"committer":{"name":"Alexander Makarov","email":"amakarov@mirantis.com","date":"2015-08-27 14:27:50.000000000","tz":180},"subject":"Protect WSGI application with a critical section","message":"Protect WSGI application with a critical section\n\nChange-Id: Ib9e2207b0d1d9cee656736e94865fb404b6a868d\nRelated-Bug: 1482271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/85e28aa2a41aeca3392a6ba84fb2cff2e6d0693d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/85e28aa2a41aeca3392a6ba84fb2cff2e6d0693d"}]},"branch":"refs/heads/master"},"fed80fd511f1fe0fea9cb187a84df3c8036117c9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2015-08-27 15:08:12.000000000","uploader":{"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"},"ref":"refs/changes/33/217733/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/33/217733/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/217733/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/217733/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/217733/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/33/217733/2"}}},"commit":{"parents":[{"commit":"1e05037c2ae02a4d52d0f50abf00d61021971508","subject":"Strip comments in requirements files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1e05037c2ae02a4d52d0f50abf00d61021971508"}]}],"author":{"name":"Alexander Makarov","email":"amakarov@mirantis.com","date":"2015-08-27 14:25:52.000000000","tz":180},"committer":{"name":"Alexander Makarov","email":"amakarov@mirantis.com","date":"2015-08-27 15:08:12.000000000","tz":0},"subject":"Protect WSGI application with a critical section","message":"Protect WSGI application with a critical section\n\nWhen configured as an Apache WSGI module a race condition\nis possible during keystone cache initialization.\nThe operation raises exception region.RegionAlreadyConfigured.\nThis is a result of the race condition involving global\n\u0027application\u0027 variable being initialized several times\n(1 per thread).\nApache modwsgi documentation suggests protecting global objects\nwith thread locks.\n\nChange-Id: Ib9e2207b0d1d9cee656736e94865fb404b6a868d\nRelated-Bug: 1482271","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fed80fd511f1fe0fea9cb187a84df3c8036117c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fed80fd511f1fe0fea9cb187a84df3c8036117c9"}]},"branch":"refs/heads/master"},"b35f9d8b72b3b4bd4526e95502bec02fae69a6a1":{"kind":"REWORK","_number":3,"created":"2015-08-28 10:20:53.000000000","uploader":{"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"},"ref":"refs/changes/33/217733/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/33/217733/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/217733/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/217733/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/217733/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/33/217733/3"}}},"commit":{"parents":[{"commit":"1e05037c2ae02a4d52d0f50abf00d61021971508","subject":"Strip comments in requirements files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1e05037c2ae02a4d52d0f50abf00d61021971508"}]}],"author":{"name":"Alexander Makarov","email":"amakarov@mirantis.com","date":"2015-08-27 14:25:52.000000000","tz":180},"committer":{"name":"Alexander Makarov","email":"amakarov@mirantis.com","date":"2015-08-28 10:21:08.000000000","tz":180},"subject":"Protect WSGI application with a critical section","message":"Protect WSGI application with a critical section\n\nWhen configured as an Apache WSGI module a race condition\nis possible during keystone cache initialization.\nThe operation raises exception region.RegionAlreadyConfigured.\nThis is a result of the race condition involving global\n\u0027application\u0027 variable being initialized several times\n(1 per thread).\nApache modwsgi documentation suggests protecting global objects\nwith thread locks.\n\nChange-Id: Ib9e2207b0d1d9cee656736e94865fb404b6a868d\nRelated-Bug: 1482271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b35f9d8b72b3b4bd4526e95502bec02fae69a6a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b35f9d8b72b3b4bd4526e95502bec02fae69a6a1"}]},"branch":"refs/heads/master"},"6757913f0f7bf36c845a222cd8e5691a4d07c503":{"kind":"REWORK","_number":4,"created":"2015-08-28 12:04:13.000000000","uploader":{"_account_id":13055,"name":"Alexander Makarov","email":"ajiekcahdp.makapob@gmail.com","username":"amakarov"},"ref":"refs/changes/33/217733/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/33/217733/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/217733/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/217733/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/217733/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/33/217733/4"}}},"commit":{"parents":[{"commit":"1e05037c2ae02a4d52d0f50abf00d61021971508","subject":"Strip comments in requirements files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1e05037c2ae02a4d52d0f50abf00d61021971508"}]}],"author":{"name":"Alexander Makarov","email":"amakarov@mirantis.com","date":"2015-08-27 14:25:52.000000000","tz":180},"committer":{"name":"Alexander Makarov","email":"amakarov@mirantis.com","date":"2015-08-28 12:04:44.000000000","tz":180},"subject":"Protect WSGI application with a critical section","message":"Protect WSGI application with a critical section\n\nWhen configured as an Apache WSGI module a race condition\nis possible during keystone cache initialization.\nThe operation raises exception region.RegionAlreadyConfigured.\nThis is a result of the race condition involving global\n\u0027application\u0027 variable being initialized several times\n(1 per thread).\nApache modwsgi documentation suggests protecting global objects\nwith thread locks.\n\nChange-Id: Ib9e2207b0d1d9cee656736e94865fb404b6a868d\nRelated-Bug: 1482271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6757913f0f7bf36c845a222cd8e5691a4d07c503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6757913f0f7bf36c845a222cd8e5691a4d07c503"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
