)]}'
{"id":"openstack%2Fpbr~577614","triplet_id":"openstack%2Fpbr~master~Ic66848dbe44ed946800979fb73776ab6d00869f6","project":"openstack/pbr","branch":"master","hashtags":[],"change_id":"Ic66848dbe44ed946800979fb73776ab6d00869f6","subject":"option to print only the version of a package","status":"MERGED","created":"2018-06-23 09:46:47.000000000","updated":"2019-06-24 17:24:34.000000000","submitted":"2019-06-24 17:24:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"577614-1561397074996-1a2baadc","meta_rev_id":"d1ba80f102e96df0a1bfb4e39bfdef4f4c48b13b","_number":577614,"virtual_id_number":577614,"owner":{"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":2,"date":"2019-06-24 17:24:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-06-24 15:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-06-24 09:58:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},{"value":2,"date":"2019-06-24 15:32:22.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":2,"date":"2019-06-24 09:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-06-24 15:31:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-18 00:13:51.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2019-06-24 09:58:49.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-06-24 15:31:17.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-06-24 15:32:22.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2019-06-24 17:24:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62a60a68c4ccee5a356b7d897ee74d01e78083de","author":{"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},"date":"2018-06-23 09:46:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"851587b571494e93af0f08147682c9c7caa95350","author":{"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},"date":"2018-06-23 09:47:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0508b18e7647f977922c73e297ee2247aeb8010f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-23 10:59:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/14/577614/1/check/openstack-tox-pep8/cc1798c/ : FAILURE in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/14/577614/1/check/openstack-tox-py27/81ef22c/ : SUCCESS in 3m 47s\n- openstack-tox-py35 http://logs.openstack.org/14/577614/1/check/openstack-tox-py35/8b61537/ : SUCCESS in 3m 39s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/14/577614/1/check/legacy-tempest-dsvm-neutron-src/f6f309c/ : FAILURE in 1h 07m 23s\n- openstack-tox-pypy http://logs.openstack.org/14/577614/1/check/openstack-tox-pypy/d76f221/ : SUCCESS in 4m 10s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/14/577614/1/check/build-openstack-sphinx-docs/7c53b9b/html/ : SUCCESS in 3m 50s\n- legacy-pbr-installation-dsvm http://logs.openstack.org/14/577614/1/check/legacy-pbr-installation-dsvm/5537c2a/ : SUCCESS in 29m 58s\n- legacy-pbr-installation-dsvm-upstream http://logs.openstack.org/14/577614/1/check/legacy-pbr-installation-dsvm-upstream/f28e89f/ : FAILURE in 1h 04m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/577614/1/check/openstack-tox-lower-constraints/8ef7a97/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f7b0fa6c7c7c82174457db913742a88ebecf3e91","author":{"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},"date":"2018-06-23 11:02:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"24c0e75525cff21b41388b677d982e6a51a1ceac","author":{"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},"date":"2018-06-23 11:02:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ca8001d5fa8ce4546f7039c93df9231d14efc85b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-23 12:26:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/14/577614/2/check/openstack-tox-pep8/8909de9/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/14/577614/2/check/openstack-tox-py27/3f4b8cc/ : SUCCESS in 3m 46s\n- openstack-tox-py35 http://logs.openstack.org/14/577614/2/check/openstack-tox-py35/95588fa/ : SUCCESS in 4m 04s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/14/577614/2/check/legacy-tempest-dsvm-neutron-src/3430672/ : SUCCESS in 1h 22m 04s\n- openstack-tox-pypy http://logs.openstack.org/14/577614/2/check/openstack-tox-pypy/cd68408/ : SUCCESS in 4m 32s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/14/577614/2/check/build-openstack-sphinx-docs/6ca2dd4/html/ : SUCCESS in 3m 40s\n- legacy-pbr-installation-dsvm http://logs.openstack.org/14/577614/2/check/legacy-pbr-installation-dsvm/ad1f0c0/ : SUCCESS in 44m 26s\n- legacy-pbr-installation-dsvm-upstream http://logs.openstack.org/14/577614/2/check/legacy-pbr-installation-dsvm-upstream/6b0f98c/ : SUCCESS in 30m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/577614/2/check/openstack-tox-lower-constraints/4ce515a/ : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"0886446254dbc49e5e9fd4fd4e4774ecc1537fdf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-09-17 15:49:29.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks like this is a dupe of [1] so I\u0027ve abandoned that. This change itself looks good. We might want a reno but I\u0027m not certain so not blocking on it\n\n[1] https://review.openstack.org/#/c/579822/","accounts_in_message":[],"_revision_number":2},{"id":"1e7f954ed6b150f67daf132766f71bce3117772f","author":{"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},"date":"2018-09-17 22:40:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a716a2782a40ec7c5ef6c6d27f93e0d40cb2d35b","author":{"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},"date":"2018-09-17 22:41:02.000000000","message":"Patch Set 3: Code-Review+1\n\nadded a reno slug :)","accounts_in_message":[],"_revision_number":3},{"id":"c5def927856b9481d73c6eb0499fb725c506c2a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-18 00:13:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/14/577614/3/check/openstack-tox-pep8/5b07eac/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/14/577614/3/check/openstack-tox-py27/d1b3a34/ : FAILURE in 5m 27s\n- openstack-tox-py35 http://logs.openstack.org/14/577614/3/check/openstack-tox-py35/14fefa7/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/14/577614/3/check/openstack-tox-py36/8aad613/ : SUCCESS in 4m 51s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/14/577614/3/check/legacy-tempest-dsvm-neutron-src/f2d2485/ : FAILURE in 1h 16m 10s\n- tempest-full-py3 http://logs.openstack.org/14/577614/3/check/tempest-full-py3/c041533/ : SUCCESS in 1h 21m 07s\n- openstack-tox-pypy http://logs.openstack.org/14/577614/3/check/openstack-tox-pypy/788b472/ : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/14/577614/3/check/openstack-tox-docs/41786fe/html/ : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/577614/3/check/openstack-tox-lower-constraints/6379ef1/ : SUCCESS in 5m 24s\n- pbr-installation-devstack http://logs.openstack.org/14/577614/3/check/pbr-installation-devstack/21022a7/ : SUCCESS in 25m 27s\n- pbr-installation-upstream-devstack http://logs.openstack.org/14/577614/3/check/pbr-installation-upstream-devstack/3913e82/ : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":3},{"id":"ce4764858a5bc54876f9edbd4fc09d6a8e837b77","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-09-18 00:13:51.000000000","message":"Patch Set 3:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- legacy-tempest-dsvm-neutron-src: https://bugs.launchpad.net/bugs/1770179\n- openstack-tox-py27: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":3},{"id":"b9eeb01c863663d192c448f1ccde54ce7a2dd4c0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-24 09:58:26.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"91a3610887f8baf2155df55fa27caf574bee4c97","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-24 09:58:49.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3ae8eca6281b6bf1679d4e145046352579a3bfa7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 12:04:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/14/577614/4/check/tempest-full/5208b71/ : SUCCESS in 1h 59m 37s\n- tempest-full-py3 http://logs.openstack.org/14/577614/4/check/tempest-full-py3/43a031a/ : SUCCESS in 1h 33m 36s\n- openstack-tox-cover http://logs.openstack.org/14/577614/4/check/openstack-tox-cover/0ffca91/cover/ : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/577614/4/check/openstack-tox-lower-constraints/3338521/ : SUCCESS in 7m 07s\n- openstack-tox-pep8 http://logs.openstack.org/14/577614/4/check/openstack-tox-pep8/5d55344/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/14/577614/4/check/openstack-tox-py27/25c02e5/ : SUCCESS in 6m 01s\n- openstack-tox-py35 http://logs.openstack.org/14/577614/4/check/openstack-tox-py35/771d95e/ : SUCCESS in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/14/577614/4/check/openstack-tox-py36/85de272/ : SUCCESS in 6m 21s\n- openstack-tox-py37 http://logs.openstack.org/14/577614/4/check/openstack-tox-py37/72387dc/ : SUCCESS in 6m 32s\n- openstack-tox-docs http://logs.openstack.org/14/577614/4/check/openstack-tox-docs/b1cc9c4/html/ : SUCCESS in 4m 35s\n- pbr-installation-devstack http://logs.openstack.org/14/577614/4/check/pbr-installation-devstack/2e81878/ : SUCCESS in 34m 15s\n- pbr-installation-upstream-devstack http://logs.openstack.org/14/577614/4/check/pbr-installation-upstream-devstack/898a1a0/ : SUCCESS in 38m 54s","accounts_in_message":[],"_revision_number":4},{"id":"5b30921a51970badfd68fbde40a0dc9557ec9801","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-24 15:31:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nNice, thanks.","accounts_in_message":[],"_revision_number":4},{"id":"e85cc982783bcc8d2e6f17bf0d1e028a45e6270f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 15:31:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"905b2aff445801c7d0b17fe380f6644db8398573","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-06-24 15:32:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"51794561b341136c735a1a423e8d08bc11f1a86c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 17:24:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d1ba80f102e96df0a1bfb4e39bfdef4f4c48b13b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 17:24:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/14/577614/4/gate/tempest-full/97c38e7/ : SUCCESS in 1h 45m 24s\n- tempest-full-py3 http://logs.openstack.org/14/577614/4/gate/tempest-full-py3/b0c0984/ : SUCCESS in 1h 29m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/577614/4/gate/openstack-tox-lower-constraints/8b8abf0/ : SUCCESS in 6m 21s\n- openstack-tox-pep8 http://logs.openstack.org/14/577614/4/gate/openstack-tox-pep8/90dd0a9/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/14/577614/4/gate/openstack-tox-py27/2e90ac1/ : SUCCESS in 5m 13s\n- openstack-tox-py35 http://logs.openstack.org/14/577614/4/gate/openstack-tox-py35/721386d/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/14/577614/4/gate/openstack-tox-py36/752310b/ : SUCCESS in 5m 05s\n- openstack-tox-py37 http://logs.openstack.org/14/577614/4/gate/openstack-tox-py37/1a4edd5/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/14/577614/4/gate/openstack-tox-docs/4f66322/html/ : SUCCESS in 4m 03s\n- pbr-installation-devstack http://logs.openstack.org/14/577614/4/gate/pbr-installation-devstack/ba18f8c/ : SUCCESS in 32m 52s\n- pbr-installation-upstream-devstack http://logs.openstack.org/14/577614/4/gate/pbr-installation-upstream-devstack/941d6f7/ : SUCCESS in 48m 22s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c715db6df9d47b56cd31a3c65f934d1d220563bb","revisions":{"051664e7fef9980fe5bb595a0a988068e09ebf31":{"kind":"REWORK","_number":1,"created":"2018-06-23 09:46:47.000000000","uploader":{"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},"ref":"refs/changes/14/577614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/577614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/577614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/577614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/577614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/577614/1"}}},"commit":{"parents":[{"commit":"6c546bc9abb2b40c5b5d6ffe57dfd5c4983d1d07","subject":"Update python3 versions in tox.ini envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6c546bc9abb2b40c5b5d6ffe57dfd5c4983d1d07"}]}],"author":{"name":"Gaetan Semet","email":"gaetan@xeberon.net","date":"2018-06-23 09:46:39.000000000","tz":120},"committer":{"name":"Gaetan Semet","email":"gaetan@xeberon.net","date":"2018-06-23 09:46:39.000000000","tz":120},"subject":"option to print only the version of a package","message":"option to print only the version of a package\n\nExample:\n    $ pbr info -s pkgname\n    1.2.3\n\nChange-Id: Ic66848dbe44ed946800979fb73776ab6d00869f6\nSigned-off-by: Gaetan Semet \u003cgaetan@xeberon.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/051664e7fef9980fe5bb595a0a988068e09ebf31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/051664e7fef9980fe5bb595a0a988068e09ebf31"}]},"branch":"refs/heads/master"},"4ad8c1ad0e6cc14341639ee20249733cec9fe0b4":{"kind":"REWORK","_number":2,"created":"2018-06-23 11:02:30.000000000","uploader":{"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},"ref":"refs/changes/14/577614/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/577614/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/577614/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/577614/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/577614/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/577614/2"}}},"commit":{"parents":[{"commit":"6c546bc9abb2b40c5b5d6ffe57dfd5c4983d1d07","subject":"Update python3 versions in tox.ini envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6c546bc9abb2b40c5b5d6ffe57dfd5c4983d1d07"}]}],"author":{"name":"Gaetan Semet","email":"gaetan@xeberon.net","date":"2018-06-23 09:46:39.000000000","tz":120},"committer":{"name":"Gaetan Semet","email":"gaetan@xeberon.net","date":"2018-06-23 11:02:25.000000000","tz":120},"subject":"option to print only the version of a package","message":"option to print only the version of a package\n\nExample:\n    $ pbr info -s pkgname\n    1.2.3\n\nChange-Id: Ic66848dbe44ed946800979fb73776ab6d00869f6\nSigned-off-by: Gaetan Semet \u003cgaetan@xeberon.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4ad8c1ad0e6cc14341639ee20249733cec9fe0b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4ad8c1ad0e6cc14341639ee20249733cec9fe0b4"}]},"branch":"refs/heads/master"},"11c05946239b02496e4db3f85ffa1c1b229131b3":{"kind":"REWORK","_number":3,"created":"2018-09-17 22:40:42.000000000","uploader":{"_account_id":27201,"name":"Gaetan Semet","email":"gaetan@xeberon.net","username":"stibbons"},"ref":"refs/changes/14/577614/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/577614/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/577614/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/577614/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/577614/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/577614/3"}}},"commit":{"parents":[{"commit":"d01abc82d081c98748ff95d7bb47abd9c1ff678c","subject":"Add release note for fix to bug 1786306","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d01abc82d081c98748ff95d7bb47abd9c1ff678c"}]}],"author":{"name":"Gaetan Semet","email":"gaetan@xeberon.net","date":"2018-06-23 09:46:39.000000000","tz":120},"committer":{"name":"Gaetan Semet","email":"gaetan@xeberon.net","date":"2018-09-17 22:40:34.000000000","tz":120},"subject":"option to print only the version of a package","message":"option to print only the version of a package\n\nExample:\n    $ pbr info -s pkgname\n    1.2.3\n\nChange-Id: Ic66848dbe44ed946800979fb73776ab6d00869f6\nSigned-off-by: Gaetan Semet \u003cgaetan@xeberon.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/11c05946239b02496e4db3f85ffa1c1b229131b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/11c05946239b02496e4db3f85ffa1c1b229131b3"}]},"branch":"refs/heads/master"},"c715db6df9d47b56cd31a3c65f934d1d220563bb":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-06-24 09:58:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/577614/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/577614/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/577614/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/577614/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/577614/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/577614/4"}}},"commit":{"parents":[{"commit":"e4c7dba0bd9b5fdb8c387ff0062e28a3df5391e5","subject":"Merge \"Fix Windows support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e4c7dba0bd9b5fdb8c387ff0062e28a3df5391e5"}]}],"author":{"name":"Gaetan Semet","email":"gaetan@xeberon.net","date":"2018-06-23 09:46:39.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-06-24 09:58:26.000000000","tz":0},"subject":"option to print only the version of a package","message":"option to print only the version of a package\n\nExample:\n    $ pbr info -s pkgname\n    1.2.3\n\nChange-Id: Ic66848dbe44ed946800979fb73776ab6d00869f6\nSigned-off-by: Gaetan Semet \u003cgaetan@xeberon.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c715db6df9d47b56cd31a3c65f934d1d220563bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c715db6df9d47b56cd31a3c65f934d1d220563bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
