)]}'
{"id":"openstack%2Fpbr~59532","triplet_id":"openstack%2Fpbr~master~I43dac981fadd31e00010b8073fc5f503aaadfce8","project":"openstack/pbr","branch":"master","hashtags":[],"change_id":"I43dac981fadd31e00010b8073fc5f503aaadfce8","subject":"Trap failures due to old easy_install","status":"ABANDONED","created":"2013-12-02 20:27:26.000000000","updated":"2013-12-10 06:03:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bb2bfb94e1d5216ad703992f81df1910af7ba6be","_number":59532,"virtual_id_number":59532,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":-1,"date":"2013-12-02 22:19:21.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":-1,"date":"2013-12-03 04:34:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2013-12-02 20:32:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-12-03 04:34:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-12-02 22:19:21.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2013-12-03 04:34:01.000000000","updated_by":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"reviewer":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"state":"REVIEWER"}],"messages":[{"id":"5934ce8f5e4caa8e4d941b69e6dd70852ae158d2","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2013-12-02 20:32:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"524039c779cd53de8bc07921fe5f43490629d046","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-02 22:19:21.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-pbr-pep8 http://logs.openstack.org/32/59532/2/check/gate-pbr-pep8/93d039d : SUCCESS in 47s\n- gate-pbr-docs http://docs-draft.openstack.org/32/59532/2/check/gate-pbr-docs/59e42d3/doc/build/html/ : SUCCESS in 42s\n- gate-pbr-python26 http://logs.openstack.org/32/59532/2/check/gate-pbr-python26/363c6c5 : SUCCESS in 56s\n- gate-pbr-python27 http://logs.openstack.org/32/59532/2/check/gate-pbr-python27/9256fa1 : SUCCESS in 1m 01s\n- gate-pbr-python33 http://logs.openstack.org/32/59532/2/check/gate-pbr-python33/0fc260a : SUCCESS in 4m 09s\n- gate-pbr-pypy http://logs.openstack.org/32/59532/2/check/gate-pbr-pypy/05350a3 : SUCCESS in 1m 33s\n- check-requirements-integration http://logs.openstack.org/32/59532/2/check/check-requirements-integration/c851418 : SUCCESS in 1h 09m 53s\n- check-tempest-devstack-vm-full-default http://logs.openstack.org/32/59532/2/check/check-tempest-devstack-vm-full-default/0538b88 : FAILURE in 41m 31s\n- check-tempest-devstack-vm-postgres-full-default http://logs.openstack.org/32/59532/2/check/check-tempest-devstack-vm-postgres-full-default/5922aba : SUCCESS in 35m 30s\n- check-tempest-devstack-vm-neutron-default http://logs.openstack.org/32/59532/2/check/check-tempest-devstack-vm-neutron-default/1ce2c12 : SUCCESS in 29m 30s\n- gate-tempest-devstack-vm-large-ops-default http://logs.openstack.org/32/59532/2/check/gate-tempest-devstack-vm-large-ops-default/0af230a : SUCCESS in 18m 48s\n- gate-tempest-devstack-vm-neutron-large-ops-default http://logs.openstack.org/32/59532/2/check/gate-tempest-devstack-vm-neutron-large-ops-default/251f69b : SUCCESS in 17m 32s\n- check-grenade-devstack-vm http://logs.openstack.org/32/59532/2/check/check-grenade-devstack-vm/78c58c4 : SUCCESS in 29m 18s\n- check-swift-devstack-vm-functional-default http://logs.openstack.org/32/59532/2/check/check-swift-devstack-vm-functional-default/b30601c : FAILURE in 8m 15s\n","accounts_in_message":[],"_revision_number":2},{"id":"713722a0c26e56570cb0fa0683f7dc032ab4eec2","author":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"date":"2013-12-03 04:34:01.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\nTest please.","accounts_in_message":[],"_revision_number":2},{"id":"a12456ad861e5cca1ff7c9109f3ab3dff4865d1d","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-12-10 06:03:03.000000000","message":"Patch Set 2: Abandoned\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":2},{"id":"c2cdc8f3cb28371ae60b843819bdda1c57f0a63f","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-12-10 06:03:03.000000000","message":"Patch Set 2:\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"13c099a74fc6f715e03e79ef5c48c4e5ce547a05","revisions":{"8d071187d2c411f0b3b08a7f15c39f0673676a04":{"kind":"REWORK","_number":1,"created":"2013-12-02 20:27:26.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/32/59532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/59532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/59532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/59532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/59532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/59532/1"}}},"commit":{"parents":[{"commit":"fc9ea6083714f4a0b29a7827441c6d314d880771","subject":"Merge \"Let git do mailmap mapping for us\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fc9ea6083714f4a0b29a7827441c6d314d880771"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-12-02 20:23:57.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-12-02 20:27:23.000000000","tz":-300},"subject":"Trap failures due to old easy_install","message":"Trap failures due to old easy_install\n\nIn some old setuptools/easy_install installations with setup.py install,\neasy_install winds up being none (please don\u0027t ask me why) which means\nthat the call to get the script header winds up being broken. If you\u0027re\nin that env, then being clever with script headers is a waste.\n\nChange-Id: I43dac981fadd31e00010b8073fc5f503aaadfce8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8d071187d2c411f0b3b08a7f15c39f0673676a04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8d071187d2c411f0b3b08a7f15c39f0673676a04"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fc9ea6083714f4a0b29a7827441c6d314d880771","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"13c099a74fc6f715e03e79ef5c48c4e5ce547a05":{"kind":"REWORK","_number":2,"created":"2013-12-02 20:32:22.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/32/59532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/59532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/59532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/59532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/59532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/59532/2"}}},"commit":{"parents":[{"commit":"fc9ea6083714f4a0b29a7827441c6d314d880771","subject":"Merge \"Let git do mailmap mapping for us\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fc9ea6083714f4a0b29a7827441c6d314d880771"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-12-02 20:23:57.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2013-12-02 20:32:19.000000000","tz":-300},"subject":"Trap failures due to old easy_install","message":"Trap failures due to old easy_install\n\nIn some old setuptools/easy_install installations with setup.py install,\neasy_install winds up being none (please don\u0027t ask me why) which means\nthat the call to get the script header winds up being broken. If you\u0027re\nin that env, then being clever with script headers is a waste.\n\nChange-Id: I43dac981fadd31e00010b8073fc5f503aaadfce8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/13c099a74fc6f715e03e79ef5c48c4e5ce547a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/13c099a74fc6f715e03e79ef5c48c4e5ce547a05"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fc9ea6083714f4a0b29a7827441c6d314d880771","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
