)]}'
{"id":"openstack%2Fpbr~610659","triplet_id":"openstack%2Fpbr~master~Icc6cb4979b2523a0edfc6375c5c032bb8be76b9f","project":"openstack/pbr","branch":"master","topic":"py37-job","hashtags":[],"change_id":"Icc6cb4979b2523a0edfc6375c5c032bb8be76b9f","subject":"Add openstack-tox-py37 job","status":"MERGED","created":"2018-10-15 16:39:26.000000000","updated":"2019-05-23 16:41:35.000000000","submitted":"2019-05-23 16:41:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"610659-1558629695730-22732f97","meta_rev_id":"218b013eeb601bd3b6eb87cd3d802878cf66d3a5","_number":610659,"virtual_id_number":610659,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-05-23 16:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-05-23 14:03:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-05-23 10:44:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":1,"date":"2019-05-23 10:42:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":2,"date":"2019-05-23 14:49:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":1,"date":"2019-05-23 14:49:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-21 15:33:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-04-19 12:02:22.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-05-23 10:44:23.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-05-23 14:03:45.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-05-23 14:49:12.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-23 16:41:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e83d0668e4a27286a3eaabbb740924c67d5c870","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2018-10-15 16:39:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b0cee1528da0dd2ace58aa657b208aad7292e66","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2018-10-15 19:46:00.000000000","message":"Patch Set 1: Workflow-1\n\nLet\u0027s wait for this to settle: https://review.openstack.org/#/c/610708/","accounts_in_message":[],"_revision_number":1},{"id":"df4248fafcbf9bffb4d3f426f78af6fbef4b2c52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-15 20:23:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/59/610659/1/check/tempest-full/c230335/ : SUCCESS in 1h 25m 36s\n- tempest-full-py3 http://logs.openstack.org/59/610659/1/check/tempest-full-py3/bcffff7/ : SUCCESS in 1h 15m 50s\n- openstack-tox-cover http://logs.openstack.org/59/610659/1/check/openstack-tox-cover/0baf692/cover/ : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/610659/1/check/openstack-tox-lower-constraints/c148e42/ : SUCCESS in 5m 46s\n- openstack-tox-pep8 http://logs.openstack.org/59/610659/1/check/openstack-tox-pep8/6297cb7/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/59/610659/1/check/openstack-tox-py27/32c7b49/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/59/610659/1/check/openstack-tox-py36/c2e6f3e/ : SUCCESS in 4m 49s\n- openstack-tox-py37 http://logs.openstack.org/59/610659/1/check/openstack-tox-py37/dcb041f/ : SUCCESS in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/59/610659/1/check/openstack-tox-docs/47b6c6f/html/ : SUCCESS in 3m 55s\n- pbr-installation-devstack http://logs.openstack.org/59/610659/1/check/pbr-installation-devstack/19a4662/ : SUCCESS in 31m 28s\n- pbr-installation-upstream-devstack http://logs.openstack.org/59/610659/1/check/pbr-installation-upstream-devstack/5930228/ : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d49b6000390bcbb1e7ee4f4e5706980e9c8cfe81","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2019-02-14 10:24:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8b27c64f88b5e336de20d44d2cb5a36e2ee7da12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-14 12:29:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/59/610659/2/check/tempest-full/23dc0f1/ : SUCCESS in 2h 01m 18s\n- tempest-full-py3 http://logs.openstack.org/59/610659/2/check/tempest-full-py3/5575a09/ : SUCCESS in 1h 44m 09s\n- openstack-tox-cover http://logs.openstack.org/59/610659/2/check/openstack-tox-cover/5e90854/cover/ : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/610659/2/check/openstack-tox-lower-constraints/afdba18/ : SUCCESS in 5m 57s\n- openstack-tox-pep8 http://logs.openstack.org/59/610659/2/check/openstack-tox-pep8/6e8eaae/ : SUCCESS in 5m 50s\n- openstack-tox-py27 http://logs.openstack.org/59/610659/2/check/openstack-tox-py27/4cc36f2/ : FAILURE in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/59/610659/2/check/openstack-tox-py36/4e4e07b/ : SUCCESS in 5m 14s\n- openstack-tox-py37 http://logs.openstack.org/59/610659/2/check/openstack-tox-py37/026aa6d/ : SUCCESS in 5m 45s\n- openstack-tox-docs http://logs.openstack.org/59/610659/2/check/openstack-tox-docs/38ee1b7/html/ : SUCCESS in 4m 43s\n- pbr-installation-devstack http://logs.openstack.org/59/610659/2/check/pbr-installation-devstack/5167ea6/ : SUCCESS in 29m 51s\n- pbr-installation-upstream-devstack http://logs.openstack.org/59/610659/2/check/pbr-installation-upstream-devstack/b54cfe1/ : SUCCESS in 29m 53s","accounts_in_message":[],"_revision_number":2},{"id":"aebf4ff756d5046280e7729c64d466dc6388e80a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-02-15 16:05:26.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"615e5d11122495ded00e33cf85184354ad9a9203","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-15 17:59:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/59/610659/2/check/tempest-full/f81419e/ : SUCCESS in 1h 51m 40s\n- tempest-full-py3 http://logs.openstack.org/59/610659/2/check/tempest-full-py3/c25ff3f/ : SUCCESS in 1h 30m 37s\n- openstack-tox-cover http://logs.openstack.org/59/610659/2/check/openstack-tox-cover/37bf17f/cover/ : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/610659/2/check/openstack-tox-lower-constraints/1772bd4/ : SUCCESS in 6m 23s\n- openstack-tox-pep8 http://logs.openstack.org/59/610659/2/check/openstack-tox-pep8/4d15f5c/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/59/610659/2/check/openstack-tox-py27/561f8b8/ : SUCCESS in 4m 51s\n- openstack-tox-py36 http://logs.openstack.org/59/610659/2/check/openstack-tox-py36/a85adf8/ : SUCCESS in 4m 50s\n- openstack-tox-py37 http://logs.openstack.org/59/610659/2/check/openstack-tox-py37/00f44df/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/59/610659/2/check/openstack-tox-docs/ed7cf85/html/ : SUCCESS in 4m 01s\n- pbr-installation-devstack http://logs.openstack.org/59/610659/2/check/pbr-installation-devstack/fa20593/ : SUCCESS in 33m 11s\n- pbr-installation-upstream-devstack http://logs.openstack.org/59/610659/2/check/pbr-installation-upstream-devstack/de03e16/ : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":2},{"id":"cecb6ffa830fe99aede845752eeda15e1325ed23","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-02-21 10:28:22.000000000","message":"Patch Set 2: Code-Review-1\n\nThis is missing the change in tox, actually","accounts_in_message":[],"_revision_number":2},{"id":"d1287ddf69db718906937fdd559b237bcd3517f5","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-03-19 12:23:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"67ecafa7df233aa53f1709e3b8e0be753a61ce76","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-03-19 12:29:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5c658d6026d0f6fcec7ef3c74e2086a69ae6664e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-19 14:33:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/59/610659/4/check/tempest-full/938542d/ : SUCCESS in 2h 03m 31s\n- tempest-full-py3 http://logs.openstack.org/59/610659/4/check/tempest-full-py3/fb97f99/ : SUCCESS in 1h 40m 53s\n- openstack-tox-cover http://logs.openstack.org/59/610659/4/check/openstack-tox-cover/7ec07b6/cover/ : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/610659/4/check/openstack-tox-lower-constraints/28ba6b8/ : SUCCESS in 5m 46s\n- openstack-tox-pep8 http://logs.openstack.org/59/610659/4/check/openstack-tox-pep8/544b4fa/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/59/610659/4/check/openstack-tox-py27/10f31ff/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/59/610659/4/check/openstack-tox-py36/a07ba94/ : SUCCESS in 6m 44s\n- openstack-tox-py37 http://logs.openstack.org/59/610659/4/check/openstack-tox-py37/406b550/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/59/610659/4/check/openstack-tox-docs/246ffb7/html/ : SUCCESS in 3m 54s\n- pbr-installation-devstack http://logs.openstack.org/59/610659/4/check/pbr-installation-devstack/4fb53ff/ : FAILURE in 11m 23s\n- pbr-installation-upstream-devstack http://logs.openstack.org/59/610659/4/check/pbr-installation-upstream-devstack/628599c/ : FAILURE in 18m 32s","accounts_in_message":[],"_revision_number":4},{"id":"9a1be6daa6d3da1cb434991ff928ed06bb912aa7","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-03-21 12:41:51.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e7e4fb72897f86868e0fa8eaec2bd823e0d60310","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 14:18:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/59/610659/4/check/tempest-full/fd6106a/ : SUCCESS in 1h 36m 07s\n- tempest-full-py3 http://logs.openstack.org/59/610659/4/check/tempest-full-py3/3163596/ : SUCCESS in 1h 29m 34s\n- openstack-tox-cover http://logs.openstack.org/59/610659/4/check/openstack-tox-cover/7a0e31d/cover/ : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/610659/4/check/openstack-tox-lower-constraints/413b8c7/ : SUCCESS in 5m 30s\n- openstack-tox-pep8 http://logs.openstack.org/59/610659/4/check/openstack-tox-pep8/d00039a/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/59/610659/4/check/openstack-tox-py27/9526301/ : SUCCESS in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/59/610659/4/check/openstack-tox-py36/30c7a61/ : SUCCESS in 4m 50s\n- openstack-tox-py37 http://logs.openstack.org/59/610659/4/check/openstack-tox-py37/4dca8bf/ : SUCCESS in 4m 45s\n- openstack-tox-docs http://logs.openstack.org/59/610659/4/check/openstack-tox-docs/f3983c0/html/ : SUCCESS in 4m 12s\n- pbr-installation-devstack http://logs.openstack.org/59/610659/4/check/pbr-installation-devstack/931c00a/ : SUCCESS in 35m 19s\n- pbr-installation-upstream-devstack http://logs.openstack.org/59/610659/4/check/pbr-installation-upstream-devstack/146da93/ : SUCCESS in 37m 54s","accounts_in_message":[],"_revision_number":4},{"id":"8a2cd911713229fb0adf11209f7a38dbcb37c771","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-03-21 15:12:09.000000000","message":"Patch Set 4: Code-Review-1\n\nSorry we don\u0027t want this to be merged as py35 should not be dropped at this point as far as I know. This should probably focus on py37 addition only.","accounts_in_message":[],"_revision_number":4},{"id":"c3f8fd62c3acb3008622f5bdd0818f2ebe4a2c20","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-03-21 15:31:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"073d14112826c58b00ebfee323702b4c2f7af818","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-03-21 15:32:51.000000000","message":"Patch Set 5: Code-Review-1\n\ncommit message needs updating","accounts_in_message":[],"_revision_number":5},{"id":"0b65d9868d361bd8833eeaa3dcff9c55050e28b0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-03-21 15:33:24.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fb7b1871a2e6e6b9febc7afd86d7661291d4b8c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 17:45:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/59/610659/5/check/tempest-full/590d764/ : TIMED_OUT in 2h 10m 00s\n- tempest-full-py3 http://logs.openstack.org/59/610659/5/check/tempest-full-py3/0c53f34/ : SUCCESS in 1h 39m 07s\n- openstack-tox-cover http://logs.openstack.org/59/610659/5/check/openstack-tox-cover/0a55374/ : FAILURE in 7m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/610659/5/check/openstack-tox-lower-constraints/685a684/ : SUCCESS in 7m 10s\n- openstack-tox-pep8 http://logs.openstack.org/59/610659/5/check/openstack-tox-pep8/6dce55f/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/59/610659/5/check/openstack-tox-py27/ed8c59e/ : SUCCESS in 5m 26s\n- openstack-tox-py35 http://logs.openstack.org/59/610659/5/check/openstack-tox-py35/5f923a3/ : SUCCESS in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/59/610659/5/check/openstack-tox-py36/ba041fc/ : SUCCESS in 5m 35s\n- openstack-tox-py37 http://logs.openstack.org/59/610659/5/check/openstack-tox-py37/7f5b80f/ : SUCCESS in 6m 32s\n- openstack-tox-docs http://logs.openstack.org/59/610659/5/check/openstack-tox-docs/9183f4c/html/ : SUCCESS in 6m 00s\n- pbr-installation-devstack http://logs.openstack.org/59/610659/5/check/pbr-installation-devstack/d410bc7/ : SUCCESS in 39m 32s\n- pbr-installation-upstream-devstack http://logs.openstack.org/59/610659/5/check/pbr-installation-upstream-devstack/daf5314/ : SUCCESS in 36m 45s","accounts_in_message":[],"_revision_number":5},{"id":"aea9960ce8612901a2ca5321cda254c50449e3c4","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-03-28 13:14:56.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"c85256a57350d6ba9b12eb9e3af4f260b4fdf0d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 15:03:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/59/610659/6/check/tempest-full/05a1635/ : SUCCESS in 1h 44m 23s\n- tempest-full-py3 http://logs.openstack.org/59/610659/6/check/tempest-full-py3/801dd56/ : SUCCESS in 1h 28m 39s\n- openstack-tox-cover http://logs.openstack.org/59/610659/6/check/openstack-tox-cover/de1259a/cover/ : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/610659/6/check/openstack-tox-lower-constraints/b17dd69/ : SUCCESS in 6m 27s\n- openstack-tox-pep8 http://logs.openstack.org/59/610659/6/check/openstack-tox-pep8/589b4b2/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/59/610659/6/check/openstack-tox-py27/eddf724/ : SUCCESS in 5m 34s\n- openstack-tox-py35 http://logs.openstack.org/59/610659/6/check/openstack-tox-py35/6732484/ : SUCCESS in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/59/610659/6/check/openstack-tox-py36/47aecc5/ : SUCCESS in 5m 58s\n- openstack-tox-py37 http://logs.openstack.org/59/610659/6/check/openstack-tox-py37/b8d57af/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/59/610659/6/check/openstack-tox-docs/60da985/html/ : SUCCESS in 5m 22s\n- pbr-installation-devstack http://logs.openstack.org/59/610659/6/check/pbr-installation-devstack/c8f8b69/ : SUCCESS in 40m 18s\n- pbr-installation-upstream-devstack http://logs.openstack.org/59/610659/6/check/pbr-installation-upstream-devstack/7b7537f/ : SUCCESS in 36m 27s","accounts_in_message":[],"_revision_number":6},{"id":"4652c6541a416c19082a6c1d430ad03a3aeb2aa5","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-04-02 10:42:48.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nfew minor fixes needed.","accounts_in_message":[],"_revision_number":6},{"id":"f6aaae8e52e7354061b62831092e76c521f2da83","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-04-17 15:35:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"dc51f2687ffa1b55d9abf729555adc9aa36d73da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 17:25:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/59/610659/7/check/tempest-full/905660e/ : SUCCESS in 1h 40m 23s\n- tempest-full-py3 http://logs.openstack.org/59/610659/7/check/tempest-full-py3/1a70991/ : SUCCESS in 1h 32m 11s\n- openstack-tox-cover http://logs.openstack.org/59/610659/7/check/openstack-tox-cover/d82f90e/cover/ : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/610659/7/check/openstack-tox-lower-constraints/d2fd48f/ : SUCCESS in 6m 36s\n- openstack-tox-pep8 http://logs.openstack.org/59/610659/7/check/openstack-tox-pep8/f744116/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/59/610659/7/check/openstack-tox-py27/0679a33/ : SUCCESS in 5m 44s\n- openstack-tox-py35 http://logs.openstack.org/59/610659/7/check/openstack-tox-py35/a2580a5/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/59/610659/7/check/openstack-tox-py36/7f8d4fb/ : SUCCESS in 5m 34s\n- openstack-tox-py37 http://logs.openstack.org/59/610659/7/check/openstack-tox-py37/5e38386/ : SUCCESS in 5m 59s\n- openstack-tox-docs http://logs.openstack.org/59/610659/7/check/openstack-tox-docs/8c05b2e/html/ : SUCCESS in 4m 37s\n- pbr-installation-devstack http://logs.openstack.org/59/610659/7/check/pbr-installation-devstack/e004076/ : SUCCESS in 38m 37s\n- pbr-installation-upstream-devstack http://logs.openstack.org/59/610659/7/check/pbr-installation-upstream-devstack/d980ef6/ : SUCCESS in 39m 29s","accounts_in_message":[],"_revision_number":7},{"id":"1f83c9e5cf5f61ef99b0d6ae805d2134b7ed9ea0","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-18 15:00:31.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"fa42517358e62b44ffd3dce8edd59a657107d2c9","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-04-19 12:02:22.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"bd7f62a3d29f6a9043827cee6ccce8e74c41d5fb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-21 08:51:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0f7ba1a9cc7071560b77d15a9211ae7182a1956a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-21 08:56:37.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"d8b792750633d8b1b17311120a85656222039cdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:47:58.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/59/610659/8/check/tempest-full/3fdf589/ : SUCCESS in 1h 43m 37s\n- tempest-full-py3 http://logs.openstack.org/59/610659/8/check/tempest-full-py3/2cd860d/ : SUCCESS in 1h 16m 45s\n- openstack-tox-cover http://logs.openstack.org/59/610659/8/check/openstack-tox-cover/d74d6a9/ : FAILURE in 5m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/610659/8/check/openstack-tox-lower-constraints/c9a0df4/ : FAILURE in 5m 33s\n- openstack-tox-pep8 http://logs.openstack.org/59/610659/8/check/openstack-tox-pep8/131b00a/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/59/610659/8/check/openstack-tox-py27/4e0b90f/ : SUCCESS in 4m 31s\n- openstack-tox-py35 http://logs.openstack.org/59/610659/8/check/openstack-tox-py35/1289d8f/ : SUCCESS in 4m 17s\n- openstack-tox-py36 http://logs.openstack.org/59/610659/8/check/openstack-tox-py36/6d1eb95/ : SUCCESS in 6m 21s\n- openstack-tox-py37 http://logs.openstack.org/59/610659/8/check/openstack-tox-py37/7697abb/ : SUCCESS in 6m 20s\n- openstack-tox-docs http://logs.openstack.org/59/610659/8/check/openstack-tox-docs/dc1dd3a/html/ : SUCCESS in 3m 36s\n- pbr-installation-devstack http://logs.openstack.org/59/610659/8/check/pbr-installation-devstack/a136f1d/ : FAILURE in 32m 23s\n- pbr-installation-upstream-devstack http://logs.openstack.org/59/610659/8/check/pbr-installation-upstream-devstack/77bc90a/ : FAILURE in 22m 07s","accounts_in_message":[],"_revision_number":8},{"id":"7d7ae948d6b0722daf1435bc02bf2c3c64bc3c5b","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-05-23 10:42:30.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"d0083b3b8a8c3f0586d120e32d0295b622bc7dbf","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-05-23 10:44:23.000000000","message":"Patch Set 9: Code-Review+1\n\nI am ok with its current form. There is no need to split it\nin multiple changes as all these 3 lines are inter-dependent.\n\nAFAIK this should be merges as is as it improves testing while\nit does not change what we ship, so no risks.","accounts_in_message":[],"_revision_number":9},{"id":"65e660c2c49ee3f10c9778fe49a28b34b5ede08f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 12:39:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/59/610659/9/check/tempest-full/769be6d/ : SUCCESS in 1h 48m 19s\n- tempest-full-py3 http://logs.openstack.org/59/610659/9/check/tempest-full-py3/e066954/ : SUCCESS in 1h 29m 42s\n- openstack-tox-cover http://logs.openstack.org/59/610659/9/check/openstack-tox-cover/89bd905/cover/ : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/610659/9/check/openstack-tox-lower-constraints/2ce3414/ : SUCCESS in 5m 53s\n- openstack-tox-pep8 http://logs.openstack.org/59/610659/9/check/openstack-tox-pep8/89f42ad/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/59/610659/9/check/openstack-tox-py27/8e4603d/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/59/610659/9/check/openstack-tox-py35/f0323e8/ : SUCCESS in 5m 08s\n- openstack-tox-py36 http://logs.openstack.org/59/610659/9/check/openstack-tox-py36/69c5881/ : SUCCESS in 4m 44s\n- openstack-tox-py37 http://logs.openstack.org/59/610659/9/check/openstack-tox-py37/5d9b7b3/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/59/610659/9/check/openstack-tox-docs/37b410b/html/ : SUCCESS in 4m 42s\n- pbr-installation-devstack http://logs.openstack.org/59/610659/9/check/pbr-installation-devstack/6f2de4d/ : SUCCESS in 28m 51s\n- pbr-installation-upstream-devstack http://logs.openstack.org/59/610659/9/check/pbr-installation-upstream-devstack/2decee0/ : SUCCESS in 36m 40s","accounts_in_message":[],"_revision_number":9},{"id":"c5a8558b713883e29c5b4b9ce863c7c04ef6de50","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-23 14:03:45.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)\n\nAck on the followup. This has been lingering long enough.","accounts_in_message":[],"_revision_number":9},{"id":"0602878fc4553e58be9d5c7319ab08a35bf171fc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-23 14:49:12.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"701b7746d0b6288d5cac7eef6b985136a25e0d3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 14:49:26.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"97c637f828d2dd19f86631264f49aa1a7c8a6ae7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-23 14:49:35.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cfbbedebd6af631d02e51db03035f322ddf79310","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 16:41:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"d8d105e91babd5c969c70ff92cec779033e2d408","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 16:41:35.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/59/610659/9/gate/tempest-full/36d446c/ : SUCCESS in 1h 47m 09s\n- tempest-full-py3 http://logs.openstack.org/59/610659/9/gate/tempest-full-py3/3d48908/ : SUCCESS in 1h 33m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/610659/9/gate/openstack-tox-lower-constraints/bfcb8e7/ : SUCCESS in 6m 41s\n- openstack-tox-pep8 http://logs.openstack.org/59/610659/9/gate/openstack-tox-pep8/57627f6/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/59/610659/9/gate/openstack-tox-py27/1fecc19/ : SUCCESS in 4m 54s\n- openstack-tox-py35 http://logs.openstack.org/59/610659/9/gate/openstack-tox-py35/eee72d6/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/59/610659/9/gate/openstack-tox-py36/ffc6da1/ : SUCCESS in 4m 51s\n- openstack-tox-py37 http://logs.openstack.org/59/610659/9/gate/openstack-tox-py37/131925a/ : SUCCESS in 5m 57s\n- openstack-tox-docs http://logs.openstack.org/59/610659/9/gate/openstack-tox-docs/951cd22/html/ : SUCCESS in 4m 05s\n- pbr-installation-devstack http://logs.openstack.org/59/610659/9/gate/pbr-installation-devstack/ffda952/ : SUCCESS in 31m 08s\n- pbr-installation-upstream-devstack http://logs.openstack.org/59/610659/9/gate/pbr-installation-upstream-devstack/2042d36/ : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"20d95468bcc5df499b17e28b6ce440ad03a9a94f","revisions":{"d060540495995c10a932a3b914df4ef8bd6f0cdc":{"kind":"REWORK","_number":1,"created":"2018-10-15 16:39:26.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/59/610659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/59/610659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/59/610659/1"}}},"commit":{"parents":[{"commit":"32bfe7808f7c988c1d1b9acccd6fd437894c6e3d","subject":"Merge \"packaging: Remove support for pyN requirement files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/32bfe7808f7c988c1d1b9acccd6fd437894c6e3d"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2018-10-15 16:38:53.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2018-10-15 16:38:53.000000000","tz":-240},"subject":"Change python3.5 job to python3.7 job on Stein+","message":"Change python3.5 job to python3.7 job on Stein+\n\nThis is a mechanically generated patch to add a unit test job running\nunder Python 3.7 and drop the Python 3.5 unit test job.\n\npython3.5 was the only supported python3 version on Xenial. Now that\nwe have Bionic nodes supporting python3.6 and python3.7, let\u0027s switch\nto testing with python3.7 in addition to python3.6 in Stein and\nbeyond.\n\nSee ML discussion here [1] for context.\n\n[1] http://lists.openstack.org/pipermail/openstack-dev/2018-October/135626.html\n\nChange-Id: Icc6cb4979b2523a0edfc6375c5c032bb8be76b9f\nStory: #2004073\nTask: #27440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d060540495995c10a932a3b914df4ef8bd6f0cdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d060540495995c10a932a3b914df4ef8bd6f0cdc"}]},"branch":"refs/heads/master"},"3c039dfc1f12185a199b5b4bac4552649d98c4da":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-02-14 10:24:31.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/59/610659/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/59/610659/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/59/610659/2"}}},"commit":{"parents":[{"commit":"7909c7afd31debbc64786c891cacdd39a9932eb8","subject":"Merge \"Ignore --find-links in requirements file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7909c7afd31debbc64786c891cacdd39a9932eb8"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2018-10-15 16:38:53.000000000","tz":-240},"committer":{"name":"sahid","email":"sahid.ferdjaoui@canonical.com","date":"2019-02-14 10:24:31.000000000","tz":0},"subject":"Change python3.5 job to python3.7 job on Stein+","message":"Change python3.5 job to python3.7 job on Stein+\n\nThis is a mechanically generated patch to add a unit test job running\nunder Python 3.7 and drop the Python 3.5 unit test job.\n\npython3.5 was the only supported python3 version on Xenial. Now that\nwe have Bionic nodes supporting python3.6 and python3.7, let\u0027s switch\nto testing with python3.7 in addition to python3.6 in Stein and\nbeyond.\n\nSee ML discussion here [1] for context.\n\n[1] http://lists.openstack.org/pipermail/openstack-dev/2018-October/135626.html\n\nChange-Id: Icc6cb4979b2523a0edfc6375c5c032bb8be76b9f\nStory: #2004073\nTask: #27440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3c039dfc1f12185a199b5b4bac4552649d98c4da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3c039dfc1f12185a199b5b4bac4552649d98c4da"}]},"branch":"refs/heads/master"},"069ea5ce908871c49c43bf30a3491f38f79187f0":{"kind":"REWORK","_number":3,"created":"2019-03-19 12:23:08.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/59/610659/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/59/610659/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/59/610659/3"}}},"commit":{"parents":[{"commit":"7909c7afd31debbc64786c891cacdd39a9932eb8","subject":"Merge \"Ignore --find-links in requirements file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7909c7afd31debbc64786c891cacdd39a9932eb8"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2018-10-15 16:38:53.000000000","tz":-240},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-03-19 12:23:06.000000000","tz":60},"subject":"Change python3.5 job to python3.7 job on Stein+","message":"Change python3.5 job to python3.7 job on Stein+\n\nThis is a mechanically generated patch to add a unit test job running\nunder Python 3.7 and drop the Python 3.5 unit test job.\n\npython3.5 was the only supported python3 version on Xenial. Now that\nwe have Bionic nodes supporting python3.6 and python3.7, let\u0027s switch\nto testing with python3.7 in addition to python3.6 in Stein and\nbeyond.\n\nSee ML discussion here [1] for context.\n\n[1] http://lists.openstack.org/pipermail/openstack-dev/2018-October/135626.html\n\nChange-Id: Icc6cb4979b2523a0edfc6375c5c032bb8be76b9f\nStory: #2004073\nTask: #27440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/069ea5ce908871c49c43bf30a3491f38f79187f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/069ea5ce908871c49c43bf30a3491f38f79187f0"}]},"branch":"refs/heads/master"},"a4637932dfb51d271d787fdd05319e1524d45994":{"kind":"REWORK","_number":4,"created":"2019-03-19 12:29:03.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/59/610659/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/59/610659/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/59/610659/4"}}},"commit":{"parents":[{"commit":"7909c7afd31debbc64786c891cacdd39a9932eb8","subject":"Merge \"Ignore --find-links in requirements file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7909c7afd31debbc64786c891cacdd39a9932eb8"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2018-10-15 16:38:53.000000000","tz":-240},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-03-19 12:28:53.000000000","tz":60},"subject":"Change python3.5 job to python3.7 job on Stein+","message":"Change python3.5 job to python3.7 job on Stein+\n\nThis is a mechanically generated patch to add a unit test job running\nunder Python 3.7 and drop the Python 3.5 unit test job.\n\npython3.5 was the only supported python3 version on Xenial. Now that\nwe have Bionic nodes supporting python3.6 and python3.7, let\u0027s switch\nto testing with python3.7 in addition to python3.6 in Stein and\nbeyond.\n\nSee ML discussion here [1] for context.\n\n[1] http://lists.openstack.org/pipermail/openstack-dev/2018-October/135626.html\n\nChange-Id: Icc6cb4979b2523a0edfc6375c5c032bb8be76b9f\nStory: #2004073\nTask: #27440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a4637932dfb51d271d787fdd05319e1524d45994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a4637932dfb51d271d787fdd05319e1524d45994"}]},"branch":"refs/heads/master"},"5a49eb978bf84108c133d1bbd5ff8011f59bc198":{"kind":"REWORK","_number":5,"created":"2019-03-21 15:31:31.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/59/610659/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/59/610659/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/59/610659/5"}}},"commit":{"parents":[{"commit":"7909c7afd31debbc64786c891cacdd39a9932eb8","subject":"Merge \"Ignore --find-links in requirements file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7909c7afd31debbc64786c891cacdd39a9932eb8"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2018-10-15 16:38:53.000000000","tz":-240},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-03-21 15:31:13.000000000","tz":60},"subject":"Change python3.5 job to python3.7 job on Stein+","message":"Change python3.5 job to python3.7 job on Stein+\n\nThis is a mechanically generated patch to add a unit test job running\nunder Python 3.7 and drop the Python 3.5 unit test job.\n\npython3.5 was the only supported python3 version on Xenial. Now that\nwe have Bionic nodes supporting python3.6 and python3.7, let\u0027s switch\nto testing with python3.7 in addition to python3.6 in Stein and\nbeyond.\n\nSee ML discussion here [1] for context.\n\n[1] http://lists.openstack.org/pipermail/openstack-dev/2018-October/135626.html\n\nChange-Id: Icc6cb4979b2523a0edfc6375c5c032bb8be76b9f\nStory: #2004073\nTask: #27440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5a49eb978bf84108c133d1bbd5ff8011f59bc198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5a49eb978bf84108c133d1bbd5ff8011f59bc198"}]},"branch":"refs/heads/master"},"7db0c2f96ee475e6e1c3ea357e32d39052db9d7e":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-03-28 13:14:56.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/59/610659/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/59/610659/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/59/610659/6"}}},"commit":{"parents":[{"commit":"7909c7afd31debbc64786c891cacdd39a9932eb8","subject":"Merge \"Ignore --find-links in requirements file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7909c7afd31debbc64786c891cacdd39a9932eb8"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2018-10-15 16:38:53.000000000","tz":-240},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-03-28 13:13:08.000000000","tz":60},"subject":"Add python3.7 job on Stein+","message":"Add python3.7 job on Stein+\n\nThis is a mechanically generated patch to add a unit test job running\nunder Python 3.7 job.\n\nChange-Id: Icc6cb4979b2523a0edfc6375c5c032bb8be76b9f\nStory: #2004073\nTask: #27440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7db0c2f96ee475e6e1c3ea357e32d39052db9d7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7db0c2f96ee475e6e1c3ea357e32d39052db9d7e"}]},"branch":"refs/heads/master"},"d371cf497581b38bf5370960a222e91567456e34":{"kind":"REWORK","_number":7,"created":"2019-04-17 15:35:17.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/59/610659/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/59/610659/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/59/610659/7"}}},"commit":{"parents":[{"commit":"fba29ab4f1c165843bc8db08c86dbdbaff9defdc","subject":"Typo fix: s/extract_mesages/extract_messages/","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fba29ab4f1c165843bc8db08c86dbdbaff9defdc"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2018-10-15 16:38:53.000000000","tz":-240},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-04-17 15:34:54.000000000","tz":60},"subject":"Add openstack-tox-py37 job","message":"Add openstack-tox-py37 job\n\nEnables py37 testing while also being more friendly with developer\nenvironments where not all python versions may be available.\n\nRuns the docs jobs last as that one is the least likely to be needed.\n\nTox min version is needed for the new options used inside.\n\nChange-Id: Icc6cb4979b2523a0edfc6375c5c032bb8be76b9f\nStory: #2004073\nTask: #27440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d371cf497581b38bf5370960a222e91567456e34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d371cf497581b38bf5370960a222e91567456e34"}]},"branch":"refs/heads/master"},"4a5fbfa751d2cac786808341857c2d474d5cbc20":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-05-21 08:56:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/59/610659/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/59/610659/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/59/610659/8"}}},"commit":{"parents":[{"commit":"55429ef85692298a6f2dacb38bfa27f2aac5477f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/55429ef85692298a6f2dacb38bfa27f2aac5477f"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2018-10-15 16:38:53.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-21 08:53:24.000000000","tz":60},"subject":"Add openstack-tox-py37 job","message":"Add openstack-tox-py37 job\n\nEnables py37 testing while also being more friendly with developer\nenvironments where not all python versions may be available.\n\nRuns the docs jobs last as that one is the least likely to be needed.\n\nTox min version is needed for the new options used inside.\n\nChange-Id: Icc6cb4979b2523a0edfc6375c5c032bb8be76b9f\nStory: #2004073\nTask: #27440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4a5fbfa751d2cac786808341857c2d474d5cbc20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4a5fbfa751d2cac786808341857c2d474d5cbc20"}]},"branch":"refs/heads/master"},"20d95468bcc5df499b17e28b6ce440ad03a9a94f":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-05-23 10:42:30.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/59/610659/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/59/610659/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/59/610659/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/59/610659/9"}}},"commit":{"parents":[{"commit":"30926f6dbe05996009d407750acb5ea52a039e62","subject":"Update Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/30926f6dbe05996009d407750acb5ea52a039e62"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2018-10-15 16:38:53.000000000","tz":-240},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-05-23 10:42:30.000000000","tz":0},"subject":"Add openstack-tox-py37 job","message":"Add openstack-tox-py37 job\n\nEnables py37 testing while also being more friendly with developer\nenvironments where not all python versions may be available.\n\nRuns the docs jobs last as that one is the least likely to be needed.\n\nTox min version is needed for the new options used inside.\n\nChange-Id: Icc6cb4979b2523a0edfc6375c5c032bb8be76b9f\nStory: #2004073\nTask: #27440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/20d95468bcc5df499b17e28b6ce440ad03a9a94f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/20d95468bcc5df499b17e28b6ce440ad03a9a94f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
