)]}'
{"id":"openstack%2Fpbr~629161","triplet_id":"openstack%2Fpbr~master~Id2cc32e4e40c1f834b19756e922118d8526358d3","project":"openstack/pbr","branch":"master","topic":"bug/1810934","hashtags":[],"change_id":"Id2cc32e4e40c1f834b19756e922118d8526358d3","subject":"Fix white space handling in file names","status":"MERGED","created":"2019-01-08 12:13:33.000000000","updated":"2019-05-23 16:56:42.000000000","submitted":"2019-05-23 16:56:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"629161-1558630602063-95f0d4b7","meta_rev_id":"ebc6218fc6ba5525c914b4d136dba32cd119cc69","_number":629161,"virtual_id_number":629161,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"date":"2019-05-23 12:55:24.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-05-10 09:03:57.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-23 16:56:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-04-25 20:18:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-04-26 10:54:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-04-26 10:54:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-21 16:44:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-04-25 20:18:56.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-04-26 10:54:11.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-10 09:03:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-05-23 12:55:24.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-23 16:56:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2343da726bbb73d36a4ebf0f3b85d39f66a54063","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-01-08 12:13:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e67c1b232bc5a2b84a2402f3befb574bacbd9571","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-01-08 12:59:46.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"8cad24ea89fbd4a3e840d2799fec19c67f24d8e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-08 13:55:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/61/629161/1/check/tempest-full/bdcc754/ : SUCCESS in 1h 17m 29s\n- tempest-full-py3 http://logs.openstack.org/61/629161/1/check/tempest-full-py3/c482928/ : SUCCESS in 1h 38m 35s\n- openstack-tox-cover http://logs.openstack.org/61/629161/1/check/openstack-tox-cover/a42a4fc/cover/ : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/1/check/openstack-tox-lower-constraints/6eeccc0/ : SUCCESS in 5m 54s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/1/check/openstack-tox-pep8/837ce34/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/1/check/openstack-tox-py27/bcb9f29/ : FAILURE in 6m 00s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/1/check/openstack-tox-py35/5b78d7f/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/1/check/openstack-tox-py36/ac0c92b/ : SUCCESS in 5m 46s\n- openstack-tox-docs http://logs.openstack.org/61/629161/1/check/openstack-tox-docs/5e305cb/html/ : SUCCESS in 4m 34s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/1/check/pbr-installation-devstack/8639170/ : SUCCESS in 45m 00s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/1/check/pbr-installation-upstream-devstack/6e5c33a/ : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":1},{"id":"073e81cd038378c39433b57b00a330f8d6c0d810","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-01-08 15:03:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0417809a326048e2430fac23041d363c818d1324","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-08 16:31:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/61/629161/2/check/tempest-full/961db93/ : SUCCESS in 1h 18m 50s\n- tempest-full-py3 http://logs.openstack.org/61/629161/2/check/tempest-full-py3/bc5c8e0/ : SUCCESS in 1h 18m 09s\n- openstack-tox-cover http://logs.openstack.org/61/629161/2/check/openstack-tox-cover/454ea24/ : FAILURE in 4m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/2/check/openstack-tox-lower-constraints/3f698ef/ : FAILURE in 5m 59s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/2/check/openstack-tox-pep8/c8ed5b2/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/2/check/openstack-tox-py27/e3a13f9/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/2/check/openstack-tox-py35/61d95ca/ : FAILURE in 5m 08s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/2/check/openstack-tox-py36/840a471/ : FAILURE in 6m 05s\n- openstack-tox-docs http://logs.openstack.org/61/629161/2/check/openstack-tox-docs/6c9f8c9/html/ : SUCCESS in 4m 30s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/2/check/pbr-installation-devstack/3fd370f/ : SUCCESS in 51m 29s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/2/check/pbr-installation-upstream-devstack/d04666b/ : SUCCESS in 25m 35s","accounts_in_message":[],"_revision_number":2},{"id":"4a6c3e3289ce43478afddd59f27d7460c40e9f5e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-01-08 16:45:18.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74a4b8f708babc78a4dfff9d7696fa2c6c1ea0d6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-01-08 17:35:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"91713cfc2af9460b196754a2387c7b6790f46afc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-01-08 17:49:02.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f59036754e219807617fa9a9fec32057c7cf6e1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-08 18:59:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/61/629161/3/check/tempest-full/33258d7/ : SUCCESS in 1h 17m 40s\n- tempest-full-py3 http://logs.openstack.org/61/629161/3/check/tempest-full-py3/ed25543/ : SUCCESS in 1h 18m 37s\n- openstack-tox-cover http://logs.openstack.org/61/629161/3/check/openstack-tox-cover/b031dcb/cover/ : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/3/check/openstack-tox-lower-constraints/e487319/ : SUCCESS in 6m 10s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/3/check/openstack-tox-pep8/06eb471/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/3/check/openstack-tox-py27/5532ee4/ : SUCCESS in 4m 37s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/3/check/openstack-tox-py35/dcbe238/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/3/check/openstack-tox-py36/28eba8c/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/61/629161/3/check/openstack-tox-docs/1265ab4/html/ : SUCCESS in 4m 15s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/3/check/pbr-installation-devstack/668608a/ : SUCCESS in 49m 26s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/3/check/pbr-installation-upstream-devstack/256e9e0/ : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":3},{"id":"c779a34759d08d066c59c5480754c364e2f8a7f1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-01-09 10:03:58.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"daef0d5636adc03f70442e2b1782997931ada9a6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-01-09 11:03:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"72de6b08bbc9f99ba5c0101dfbd44f1777d88e9c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-01-09 11:47:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b3879790933ba8ae75432f081545f9aa3962e30c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 13:19:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/61/629161/5/check/tempest-full/b57f746/ : SUCCESS in 1h 25m 40s\n- tempest-full-py3 http://logs.openstack.org/61/629161/5/check/tempest-full-py3/8271db6/ : SUCCESS in 1h 14m 51s\n- openstack-tox-cover http://logs.openstack.org/61/629161/5/check/openstack-tox-cover/45231de/cover/ : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/5/check/openstack-tox-lower-constraints/8c994e4/ : SUCCESS in 5m 39s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/5/check/openstack-tox-pep8/16c29a4/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/5/check/openstack-tox-py27/23e0fed/ : SUCCESS in 4m 17s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/5/check/openstack-tox-py35/6244a94/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/5/check/openstack-tox-py36/e81c4f0/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/61/629161/5/check/openstack-tox-docs/38f1a7e/html/ : SUCCESS in 4m 04s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/5/check/pbr-installation-devstack/bdf913f/ : SUCCESS in 55m 55s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/5/check/pbr-installation-upstream-devstack/f5ca907/ : SUCCESS in 32m 35s","accounts_in_message":[],"_revision_number":5},{"id":"de0110e3c759b8e7842279a322909360fd6186a4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-01-09 14:29:34.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)\n\nJust one nit with the reno. Otherwise a-ok. Thanks for the commit message rework","accounts_in_message":[],"_revision_number":5},{"id":"fe673e9c54552b82d894a971600a61ce2234c6d2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-01-11 14:31:30.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cafc6d7b257c1fde26a98507f0ce149c20762620","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 15:58:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/61/629161/6/check/tempest-full/fa7a492/ : SUCCESS in 1h 14m 23s\n- tempest-full-py3 http://logs.openstack.org/61/629161/6/check/tempest-full-py3/ddd747d/ : SUCCESS in 1h 14m 47s\n- openstack-tox-cover http://logs.openstack.org/61/629161/6/check/openstack-tox-cover/c241fc7/cover/ : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/6/check/openstack-tox-lower-constraints/59f4d9d/ : SUCCESS in 6m 02s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/6/check/openstack-tox-pep8/2bd3d79/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/6/check/openstack-tox-py27/ae82d78/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/6/check/openstack-tox-py35/3bc103c/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/6/check/openstack-tox-py36/0ded693/ : SUCCESS in 5m 15s\n- openstack-tox-docs http://logs.openstack.org/61/629161/6/check/openstack-tox-docs/8e41e95/html/ : SUCCESS in 4m 26s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/6/check/pbr-installation-devstack/6d01f6b/ : SUCCESS in 52m 18s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/6/check/pbr-installation-upstream-devstack/9785313/ : SUCCESS in 33m 31s","accounts_in_message":[],"_revision_number":6},{"id":"699f5e8d28e8cd4dcb5500f003e897891c6e4ae8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-01-21 15:54:48.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"a24e368d0e7e8a2333ee8433d1ec24b6ace77390","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-21 16:44:32.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"5ad33a1f9aca2a2e8231d054addf19cf2b1f0538","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-21 16:44:36.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"bf2763392d4a04cfc6c29256ea35df9cb93b7502","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-21 18:47:43.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/61/629161/7/check/tempest-full/7bc2470/ : POST_FAILURE in 1h 55m 50s\n- tempest-full-py3 http://logs.openstack.org/61/629161/7/check/tempest-full-py3/e6ba630/ : POST_FAILURE in 1h 21m 24s\n- openstack-tox-cover http://logs.openstack.org/61/629161/7/check/openstack-tox-cover/ed4a96b/ : POST_FAILURE in 7m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/7/check/openstack-tox-lower-constraints/d2808b3/ : POST_FAILURE in 8m 30s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/7/check/openstack-tox-pep8/8158c09/ : POST_FAILURE in 7m 25s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/7/check/openstack-tox-py27/7734b73/ : POST_FAILURE in 4m 43s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/7/check/openstack-tox-py35/169e8af/ : POST_FAILURE in 5m 56s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/7/check/openstack-tox-py36/784a3cc/ : POST_FAILURE in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/61/629161/7/check/openstack-tox-docs/9ab6f56/ : POST_FAILURE in 4m 16s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/7/check/pbr-installation-devstack/e92a178/ : POST_FAILURE in 52m 20s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/7/check/pbr-installation-upstream-devstack/34c316d/ : POST_FAILURE in 35m 17s","accounts_in_message":[],"_revision_number":7},{"id":"b82b88706390a5ba235f0df111e05710eb570636","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-01-22 09:36:51.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"370e3ce5452a4199a86e5d264a0309dfca47fdf8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-22 11:44:34.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/61/629161/7/check/tempest-full/0f0ad2a/ : TIMED_OUT in 2h 05m 58s\n- tempest-full-py3 http://logs.openstack.org/61/629161/7/check/tempest-full-py3/84796aa/ : SUCCESS in 1h 37m 09s\n- openstack-tox-cover http://logs.openstack.org/61/629161/7/check/openstack-tox-cover/73d09a2/cover/ : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/7/check/openstack-tox-lower-constraints/dcece05/ : SUCCESS in 6m 13s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/7/check/openstack-tox-pep8/589d635/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/7/check/openstack-tox-py27/bc91faa/ : SUCCESS in 4m 42s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/7/check/openstack-tox-py35/816c521/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/7/check/openstack-tox-py36/6702b87/ : SUCCESS in 5m 46s\n- openstack-tox-docs http://logs.openstack.org/61/629161/7/check/openstack-tox-docs/3c492ec/html/ : SUCCESS in 4m 09s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/7/check/pbr-installation-devstack/a9e0d53/ : SUCCESS in 49m 45s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/7/check/pbr-installation-upstream-devstack/8b2b566/ : SUCCESS in 33m 37s","accounts_in_message":[],"_revision_number":7},{"id":"5709969d760797165596289b5e4f95acb6f73433","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-01-22 13:03:54.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"f572e6cee22fc72975f9235761d78a45d2a9bd6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-22 15:16:32.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/61/629161/7/check/tempest-full/56b6871/ : TIMED_OUT in 2h 05m 23s\n- tempest-full-py3 http://logs.openstack.org/61/629161/7/check/tempest-full-py3/1139b65/ : SUCCESS in 1h 42m 48s\n- openstack-tox-cover http://logs.openstack.org/61/629161/7/check/openstack-tox-cover/2322821/cover/ : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/7/check/openstack-tox-lower-constraints/67e747a/ : SUCCESS in 6m 08s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/7/check/openstack-tox-pep8/42f1b81/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/7/check/openstack-tox-py27/56877a8/ : SUCCESS in 4m 37s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/7/check/openstack-tox-py35/e85f20e/ : SUCCESS in 5m 22s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/7/check/openstack-tox-py36/da8d6db/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/61/629161/7/check/openstack-tox-docs/2f0a2c3/html/ : SUCCESS in 5m 03s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/7/check/pbr-installation-devstack/e45fb37/ : SUCCESS in 52m 58s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/7/check/pbr-installation-upstream-devstack/e7514ee/ : SUCCESS in 35m 43s","accounts_in_message":[],"_revision_number":7},{"id":"9c5f1e60a773f7aa5ccd04019c26be8829597e6c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-01-22 17:32:21.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"366023c8235617edee3e86dca1f75ff25ec41a68","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-01-22 17:32:25.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"2b7afc18f154e01527f73ea31c5e9ecd1da23c52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-22 19:29:46.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/61/629161/7/check/tempest-full/01b8c54/ : SUCCESS in 1h 48m 43s\n- tempest-full-py3 http://logs.openstack.org/61/629161/7/check/tempest-full-py3/59fbe23/ : SUCCESS in 1h 49m 18s\n- openstack-tox-cover http://logs.openstack.org/61/629161/7/check/openstack-tox-cover/3f081d1/cover/ : SUCCESS in 3m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/7/check/openstack-tox-lower-constraints/9e0d4d5/ : SUCCESS in 6m 41s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/7/check/openstack-tox-pep8/1c4e646/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/7/check/openstack-tox-py27/679c82d/ : SUCCESS in 4m 27s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/7/check/openstack-tox-py35/d6e9b19/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/7/check/openstack-tox-py36/891cbd5/ : SUCCESS in 6m 26s\n- openstack-tox-docs http://logs.openstack.org/61/629161/7/check/openstack-tox-docs/a9e9136/html/ : SUCCESS in 4m 45s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/7/check/pbr-installation-devstack/befa77b/ : SUCCESS in 51m 32s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/7/check/pbr-installation-upstream-devstack/663ef8a/ : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":7},{"id":"b85421ad0e26576161c50f474c2c4c28247665f3","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-18 15:55:41.000000000","message":"Patch Set 7:\n\nlgtm, but appears to have a merge conflict.","accounts_in_message":[],"_revision_number":7},{"id":"f418780b9269cca5f19b87fb6c79aa7f25cde951","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-25 10:22:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f87283bd04c54f600daa038d9514e9c858475e54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 12:34:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/61/629161/8/check/tempest-full/14085e6/ : SUCCESS in 1h 59m 45s\n- tempest-full-py3 http://logs.openstack.org/61/629161/8/check/tempest-full-py3/0a02309/ : SUCCESS in 1h 34m 58s\n- openstack-tox-cover http://logs.openstack.org/61/629161/8/check/openstack-tox-cover/c7ad89f/ : FAILURE in 7m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/8/check/openstack-tox-lower-constraints/d86e5d7/ : SUCCESS in 5m 47s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/8/check/openstack-tox-pep8/82012e5/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/8/check/openstack-tox-py27/ad08a0b/ : SUCCESS in 4m 52s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/8/check/openstack-tox-py35/955de97/ : FAILURE in 10m 28s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/8/check/openstack-tox-py36/2ae1cfc/ : FAILURE in 7m 18s\n- openstack-tox-docs http://logs.openstack.org/61/629161/8/check/openstack-tox-docs/ff8be97/html/ : SUCCESS in 4m 10s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-devstack/9cba299/ : SUCCESS in 33m 37s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-upstream-devstack/bb38278/ : SUCCESS in 33m 34s","accounts_in_message":[],"_revision_number":8},{"id":"744aac4817637c5cd2a20d1397dbd5289ccc5143","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-25 14:44:05.000000000","message":"Patch Set 8:\n\nrecheck\n\nLooks like ci was having some issues earlier, but things are passing again.","accounts_in_message":[],"_revision_number":8},{"id":"9ec7a07e1c530cba5cdd2f758d0b430a769cea95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 16:17:20.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/61/629161/8/check/tempest-full/37cb06a/ : SUCCESS in 1h 28m 33s\n- tempest-full-py3 http://logs.openstack.org/61/629161/8/check/tempest-full-py3/bc18a28/ : SUCCESS in 1h 24m 29s\n- openstack-tox-cover http://logs.openstack.org/61/629161/8/check/openstack-tox-cover/660468d/cover/ : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/8/check/openstack-tox-lower-constraints/7848330/ : FAILURE in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/8/check/openstack-tox-pep8/95882e0/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/8/check/openstack-tox-py27/8c30561/ : SUCCESS in 4m 43s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/8/check/openstack-tox-py35/df93bc4/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/8/check/openstack-tox-py36/4301d5b/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/61/629161/8/check/openstack-tox-docs/628b290/html/ : SUCCESS in 6m 15s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-devstack/45ebdb2/ : SUCCESS in 31m 36s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-upstream-devstack/0f0256e/ : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":8},{"id":"88deae2fccb2678c3b483a1b48d89439df90b30a","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-25 18:01:03.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"206020457c7491e682e4fd5a2b6f3e504334da7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 19:53:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/61/629161/8/check/tempest-full/450b0ec/ : SUCCESS in 1h 44m 24s\n- tempest-full-py3 http://logs.openstack.org/61/629161/8/check/tempest-full-py3/f6f0f9e/ : SUCCESS in 1h 46m 04s\n- openstack-tox-cover http://logs.openstack.org/61/629161/8/check/openstack-tox-cover/b86d0bb/cover/ : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/8/check/openstack-tox-lower-constraints/d24a9a8/ : SUCCESS in 7m 16s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/8/check/openstack-tox-pep8/2e03329/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/8/check/openstack-tox-py27/bbdfc22/ : SUCCESS in 4m 55s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/8/check/openstack-tox-py35/c0eca84/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/8/check/openstack-tox-py36/d9cab8c/ : SUCCESS in 4m 45s\n- openstack-tox-docs http://logs.openstack.org/61/629161/8/check/openstack-tox-docs/cde8614/html/ : SUCCESS in 4m 00s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-devstack/f6dd8e9/ : SUCCESS in 32m 35s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-upstream-devstack/20e6602/ : SUCCESS in 34m 28s","accounts_in_message":[],"_revision_number":8},{"id":"da48b1f32f44988bc21ddd71ebbe7a125b548545","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-25 20:18:56.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"b39c6479032c567efc5995f83d580a7e111b6f34","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-26 10:54:11.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"c8bfc773072a42e10a9acce243a05231c9636b3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 10:54:22.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"be956ef0f48cfcf354c3e0c2d2fbeb35d1942cbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 13:06:59.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/61/629161/8/gate/tempest-full/2b65f54/ : TIMED_OUT in 2h 10m 27s\n- tempest-full-py3 http://logs.openstack.org/61/629161/8/gate/tempest-full-py3/7188f9c/ : SUCCESS in 1h 37m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/8/gate/openstack-tox-lower-constraints/7c0f90d/ : SUCCESS in 6m 27s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/8/gate/openstack-tox-pep8/6ba1f81/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/8/gate/openstack-tox-py27/0ff66f7/ : SUCCESS in 4m 35s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/8/gate/openstack-tox-py35/f1f0127/ : SUCCESS in 4m 17s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/8/gate/openstack-tox-py36/b28befb/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/61/629161/8/gate/openstack-tox-docs/8640218/html/ : SUCCESS in 3m 45s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/8/gate/pbr-installation-devstack/12ad247/ : SUCCESS in 34m 06s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/8/gate/pbr-installation-upstream-devstack/4375385/ : SUCCESS in 32m 54s","accounts_in_message":[],"_revision_number":8},{"id":"dc2896a56f3f546b11e7cbdf414a8b970517b364","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-26 15:19:34.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"add24144123281c0127f3c997868e14304bb6dc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 18:07:25.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/61/629161/8/check/tempest-full/d28a929/ : SUCCESS in 1h 45m 53s\n- tempest-full-py3 http://logs.openstack.org/61/629161/8/check/tempest-full-py3/eccc8f8/ : SUCCESS in 1h 39m 43s\n- openstack-tox-cover http://logs.openstack.org/61/629161/8/check/openstack-tox-cover/7085d0d/cover/ : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/8/check/openstack-tox-lower-constraints/23c307f/ : SUCCESS in 5m 42s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/8/check/openstack-tox-pep8/83663a7/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/8/check/openstack-tox-py27/2d33498/ : SUCCESS in 5m 05s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/8/check/openstack-tox-py35/ac0d86b/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/8/check/openstack-tox-py36/bfabf52/ : FAILURE in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/61/629161/8/check/openstack-tox-docs/291817c/html/ : SUCCESS in 4m 23s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-devstack/f7fd35f/ : SUCCESS in 29m 46s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-upstream-devstack/6e64ab9/ : SUCCESS in 39m 41s","accounts_in_message":[],"_revision_number":8},{"id":"5df3fc413458f004aa4deed6db80e1372e17d4cb","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-06 16:39:54.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"4720ec6322173cde0f8c108878acc14f361ed37e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 18:28:00.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/61/629161/8/check/tempest-full/4b8411e/ : SUCCESS in 1h 41m 48s\n- tempest-full-py3 http://logs.openstack.org/61/629161/8/check/tempest-full-py3/50d6af6/ : SUCCESS in 1h 47m 05s\n- openstack-tox-cover http://logs.openstack.org/61/629161/8/check/openstack-tox-cover/62834f6/cover/ : SUCCESS in 6m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/8/check/openstack-tox-lower-constraints/c29ae74/ : FAILURE in 5m 44s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/8/check/openstack-tox-pep8/16dd7d8/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/8/check/openstack-tox-py27/de48761/ : SUCCESS in 4m 55s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/8/check/openstack-tox-py35/a14541b/ : FAILURE in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/8/check/openstack-tox-py36/9780307/ : FAILURE in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/61/629161/8/check/openstack-tox-docs/6339bbd/html/ : SUCCESS in 4m 21s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-devstack/8920ea0/ : SUCCESS in 32m 21s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-upstream-devstack/0343263/ : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":8},{"id":"bce35658a54bcfeed76d83af698d4198afc82ff4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-10 09:03:57.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"5f2f8eb47855fe8102e4a31f4fdfaec0fd873d09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 10:51:12.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/61/629161/8/check/tempest-full/4fe637a/ : SUCCESS in 1h 42m 03s\n- tempest-full-py3 http://logs.openstack.org/61/629161/8/check/tempest-full-py3/3e38ad5/ : SUCCESS in 1h 33m 47s\n- openstack-tox-cover http://logs.openstack.org/61/629161/8/check/openstack-tox-cover/ce34c1d/cover/ : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/8/check/openstack-tox-lower-constraints/50f04da/ : SUCCESS in 5m 36s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/8/check/openstack-tox-pep8/43c167e/ : SUCCESS in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/8/check/openstack-tox-py27/3d8cd8a/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/8/check/openstack-tox-py35/2c93ec1/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/8/check/openstack-tox-py36/436cb4a/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/61/629161/8/check/openstack-tox-docs/4a516d8/html/ : SUCCESS in 4m 08s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-devstack/995c434/ : FAILURE in 30m 24s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-upstream-devstack/f459a3f/ : FAILURE in 30m 16s","accounts_in_message":[],"_revision_number":8},{"id":"c06ac8b6b78d60462821b024c5508ebd7c065420","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-12 16:02:26.000000000","message":"Patch Set 8:\n\nI hope that pbr-installation-devstack fixes the devstack jobs...","accounts_in_message":[],"_revision_number":8},{"id":"4cad390d37d9520d461e564919859ca9d4923b4f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-23 12:55:24.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"26dbe061b18410ab6efe447931665b961dd66ab4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 15:09:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/61/629161/8/check/tempest-full/a343052/ : SUCCESS in 1h 47m 02s\n- tempest-full-py3 http://logs.openstack.org/61/629161/8/check/tempest-full-py3/bec7d93/ : SUCCESS in 1h 38m 15s\n- openstack-tox-cover http://logs.openstack.org/61/629161/8/check/openstack-tox-cover/7db712e/cover/ : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/8/check/openstack-tox-lower-constraints/0eb60fb/ : SUCCESS in 5m 39s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/8/check/openstack-tox-pep8/07a208e/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/8/check/openstack-tox-py27/31d91c2/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/8/check/openstack-tox-py35/ffe0ede/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/8/check/openstack-tox-py36/5c9df03/ : SUCCESS in 5m 50s\n- openstack-tox-docs http://logs.openstack.org/61/629161/8/check/openstack-tox-docs/2e215c8/html/ : SUCCESS in 3m 34s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-devstack/fe455e1/ : SUCCESS in 29m 40s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/8/check/pbr-installation-upstream-devstack/1652804/ : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":8},{"id":"08e22164be0b443b1e1eb07e63a0aa3bcce908e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 15:09:35.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"bf1204fb4a9354e1b6a16ee09eef021119d73245","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 16:56:41.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/61/629161/8/gate/tempest-full/d9e4dc5/ : SUCCESS in 1h 34m 30s\n- tempest-full-py3 http://logs.openstack.org/61/629161/8/gate/tempest-full-py3/9c44e21/ : SUCCESS in 1h 26m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/629161/8/gate/openstack-tox-lower-constraints/a8000a4/ : SUCCESS in 5m 51s\n- openstack-tox-pep8 http://logs.openstack.org/61/629161/8/gate/openstack-tox-pep8/9554656/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/61/629161/8/gate/openstack-tox-py27/08ce6eb/ : SUCCESS in 6m 23s\n- openstack-tox-py35 http://logs.openstack.org/61/629161/8/gate/openstack-tox-py35/4d242b3/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/61/629161/8/gate/openstack-tox-py36/ae5b020/ : SUCCESS in 4m 16s\n- openstack-tox-py37 http://logs.openstack.org/61/629161/8/gate/openstack-tox-py37/4d6f172/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/61/629161/8/gate/openstack-tox-docs/a775b8f/html/ : SUCCESS in 4m 14s\n- pbr-installation-devstack http://logs.openstack.org/61/629161/8/gate/pbr-installation-devstack/f683fe1/ : SUCCESS in 34m 32s\n- pbr-installation-upstream-devstack http://logs.openstack.org/61/629161/8/gate/pbr-installation-upstream-devstack/4adb568/ : SUCCESS in 35m 26s","accounts_in_message":[],"_revision_number":8},{"id":"92b332967df94364a0f31d90f0c7bfc47a167115","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 16:56:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"50356da8393abdb47cc6c64de4fa387d8499cbc4","revisions":{"844be9439134e452f3bba6fd19cca199597dc4db":{"kind":"REWORK","_number":1,"created":"2019-01-08 12:13:33.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/61/629161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/61/629161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/61/629161/1"}}},"commit":{"parents":[{"commit":"0f8aa9d0c8bf39c27bec9c7bed732237b27f7935","subject":"Do not globally replace path prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0f8aa9d0c8bf39c27bec9c7bed732237b27f7935"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-08 12:13:56.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-08 12:15:30.000000000","tz":0},"subject":"Fix white space handling in file names","message":"Fix white space handling in file names\n\nChange-Id: Id2cc32e4e40c1f834b19756e922118d8526358d3\nFixes-Bug: 1810934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/844be9439134e452f3bba6fd19cca199597dc4db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/844be9439134e452f3bba6fd19cca199597dc4db"}]},"branch":"refs/heads/master"},"cd32f4900ea375ccad6ba4ed3445aca423934e4b":{"kind":"REWORK","_number":2,"created":"2019-01-08 15:03:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/61/629161/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/61/629161/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/61/629161/2"}}},"commit":{"parents":[{"commit":"0f8aa9d0c8bf39c27bec9c7bed732237b27f7935","subject":"Do not globally replace path prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0f8aa9d0c8bf39c27bec9c7bed732237b27f7935"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-08 12:13:56.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-08 15:05:58.000000000","tz":0},"subject":"Fix white space handling in file names","message":"Fix white space handling in file names\n\nChange-Id: Id2cc32e4e40c1f834b19756e922118d8526358d3\nFixes-Bug: 1810934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cd32f4900ea375ccad6ba4ed3445aca423934e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cd32f4900ea375ccad6ba4ed3445aca423934e4b"}]},"branch":"refs/heads/master"},"fd159774d275bc79df1cc4bdf199c5c4ad425c82":{"kind":"REWORK","_number":3,"created":"2019-01-08 17:35:33.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/61/629161/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/61/629161/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/61/629161/3"}}},"commit":{"parents":[{"commit":"0f8aa9d0c8bf39c27bec9c7bed732237b27f7935","subject":"Do not globally replace path prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0f8aa9d0c8bf39c27bec9c7bed732237b27f7935"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-08 12:13:56.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-08 17:27:07.000000000","tz":0},"subject":"Fix white space handling in file names","message":"Fix white space handling in file names\n\nPreviously, when using data_files with a glob that matched a file with\nwhitespace in the name, pip would error with a message that the file\ndoes not exist, e.g:\n\n    error: can\u0027t copy \u0027ansible/roles/yatesr.timezone/templates/timezone-Arch\u0027: doesn\u0027t exist or not a regular file\n\nThe problem was that ansible/roles/yatesr.timezone/templates/timezone-Arch\nwas a truncated form of the actual filename:\n\n    ansible/roles/yatesr.timezone/templates/timezone-Arch Linux.j2\n\nNote the space in the filename and that it has been split on this space.\n\nThis change allows you to use a glob that matches files with whitespace\nin the name. It does this by quoting the path.\n\nChange-Id: Id2cc32e4e40c1f834b19756e922118d8526358d3\nFixes-Bug: 1810934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fd159774d275bc79df1cc4bdf199c5c4ad425c82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fd159774d275bc79df1cc4bdf199c5c4ad425c82"}]},"branch":"refs/heads/master"},"151d83d58f75a55d01f80c3dbb40a1a888361bc9":{"kind":"REWORK","_number":4,"created":"2019-01-09 11:03:02.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/61/629161/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/61/629161/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/61/629161/4"}}},"commit":{"parents":[{"commit":"0f8aa9d0c8bf39c27bec9c7bed732237b27f7935","subject":"Do not globally replace path prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0f8aa9d0c8bf39c27bec9c7bed732237b27f7935"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-08 12:13:56.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-09 11:04:20.000000000","tz":0},"subject":"Fix white space handling in file names","message":"Fix white space handling in file names\n\nPreviously, when using data_files with a glob that matched a file with\nwhitespace in the name, pip would error with a message that the file\ndoes not exist, e.g:\n\n    error: can\u0027t copy \u0027ansible/roles/yatesr.timezone/templates/timezone-Arch\u0027: doesn\u0027t exist or not a regular file\n\nThe problem was that ansible/roles/yatesr.timezone/templates/timezone-Arch\nwas a truncated form of the actual filename:\n\n    ansible/roles/yatesr.timezone/templates/timezone-Arch Linux.j2\n\nNote the space in the filename and that it has been split on this space.\n\nThis change allows you to use a glob that matches files with whitespace\nin the name. It does this by quoting the path.\n\nChange-Id: Id2cc32e4e40c1f834b19756e922118d8526358d3\nFixes-Bug: 1810934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/151d83d58f75a55d01f80c3dbb40a1a888361bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/151d83d58f75a55d01f80c3dbb40a1a888361bc9"}]},"branch":"refs/heads/master"},"afb83cf12f5dddf1a46bf559165c4c3acfc97c36":{"kind":"REWORK","_number":5,"created":"2019-01-09 11:47:34.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/61/629161/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/61/629161/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/61/629161/5"}}},"commit":{"parents":[{"commit":"c39d26dc28732d3f9c6a8682630d11bce3979474","subject":"Do not globally replace path prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c39d26dc28732d3f9c6a8682630d11bce3979474"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-08 12:13:56.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-09 11:49:34.000000000","tz":0},"subject":"Fix white space handling in file names","message":"Fix white space handling in file names\n\nPreviously, when using data_files with a glob that matched a file with\nwhitespace in the name, pip would error with a message that the file\ndoes not exist, e.g:\n\n    error: can\u0027t copy \u0027ansible/roles/yatesr.timezone/templates/timezone-Arch\u0027: doesn\u0027t exist or not a regular file\n\nThe problem was that ansible/roles/yatesr.timezone/templates/timezone-Arch\nwas a truncated form of the actual filename:\n\n    ansible/roles/yatesr.timezone/templates/timezone-Arch Linux.j2\n\nNote the space in the filename and that it has been split on this space.\n\nThis change allows you to use a glob that matches files with whitespace\nin the name. It does this by quoting the path.\n\nChange-Id: Id2cc32e4e40c1f834b19756e922118d8526358d3\nFixes-Bug: 1810934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/afb83cf12f5dddf1a46bf559165c4c3acfc97c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/afb83cf12f5dddf1a46bf559165c4c3acfc97c36"}]},"branch":"refs/heads/master"},"4a5f10e2d125f833d3682fe18afbb4d9205f4dbe":{"kind":"REWORK","_number":6,"created":"2019-01-11 14:31:30.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/61/629161/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/61/629161/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/61/629161/6"}}},"commit":{"parents":[{"commit":"a9b31113b76bde3debc696c1ce71aa7dd594faa3","subject":"Do not globally replace path prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a9b31113b76bde3debc696c1ce71aa7dd594faa3"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-08 12:13:56.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-11 14:32:32.000000000","tz":0},"subject":"Fix white space handling in file names","message":"Fix white space handling in file names\n\nPreviously, when using data_files with a glob that matched a file with\nwhitespace in the name, pip would error with a message that the file\ndoes not exist, e.g:\n\n    error: can\u0027t copy \u0027ansible/roles/yatesr.timezone/templates/timezone-Arch\u0027: doesn\u0027t exist or not a regular file\n\nThe problem was that ansible/roles/yatesr.timezone/templates/timezone-Arch\nwas a truncated form of the actual filename:\n\n    ansible/roles/yatesr.timezone/templates/timezone-Arch Linux.j2\n\nNote the space in the filename and that it has been split on this space.\n\nThis change allows you to use a glob that matches files with whitespace\nin the name. It does this by quoting the path.\n\nChange-Id: Id2cc32e4e40c1f834b19756e922118d8526358d3\nFixes-Bug: 1810934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4a5f10e2d125f833d3682fe18afbb4d9205f4dbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4a5f10e2d125f833d3682fe18afbb4d9205f4dbe"}]},"branch":"refs/heads/master"},"4cb15fcdce2f0e4ce01e8c366f6cab721d832474":{"kind":"REWORK","_number":7,"created":"2019-01-21 16:44:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/61/629161/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/61/629161/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/61/629161/7"}}},"commit":{"parents":[{"commit":"a9b31113b76bde3debc696c1ce71aa7dd594faa3","subject":"Do not globally replace path prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a9b31113b76bde3debc696c1ce71aa7dd594faa3"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-08 12:13:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-21 16:44:26.000000000","tz":0},"subject":"Fix white space handling in file names","message":"Fix white space handling in file names\n\nPreviously, when using data_files with a glob that matched a file with\nwhitespace in the name, pip would error with a message that the file\ndoes not exist, e.g:\n\n    error: can\u0027t copy \u0027ansible/roles/yatesr.timezone/templates/timezone-Arch\u0027: doesn\u0027t exist or not a regular file\n\nThe problem was that ansible/roles/yatesr.timezone/templates/timezone-Arch\nwas a truncated form of the actual filename:\n\n    ansible/roles/yatesr.timezone/templates/timezone-Arch Linux.j2\n\nNote the space in the filename and that it has been split on this space.\n\nThis change allows you to use a glob that matches files with whitespace\nin the name. It does this by quoting the path.\n\nChange-Id: Id2cc32e4e40c1f834b19756e922118d8526358d3\nFixes-Bug: 1810934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4cb15fcdce2f0e4ce01e8c366f6cab721d832474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4cb15fcdce2f0e4ce01e8c366f6cab721d832474"}]},"branch":"refs/heads/master"},"50356da8393abdb47cc6c64de4fa387d8499cbc4":{"kind":"REWORK","_number":8,"created":"2019-04-25 10:22:20.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/61/629161/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/61/629161/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/61/629161/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/61/629161/8"}}},"commit":{"parents":[{"commit":"55429ef85692298a6f2dacb38bfa27f2aac5477f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/55429ef85692298a6f2dacb38bfa27f2aac5477f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-01-08 12:13:56.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-25 10:25:56.000000000","tz":60},"subject":"Fix white space handling in file names","message":"Fix white space handling in file names\n\nPreviously, when using data_files with a glob that matched a file with\nwhitespace in the name, pip would error with a message that the file\ndoes not exist, e.g:\n\n    error: can\u0027t copy \u0027ansible/roles/yatesr.timezone/templates/timezone-Arch\u0027: doesn\u0027t exist or not a regular file\n\nThe problem was that ansible/roles/yatesr.timezone/templates/timezone-Arch\nwas a truncated form of the actual filename:\n\n    ansible/roles/yatesr.timezone/templates/timezone-Arch Linux.j2\n\nNote the space in the filename and that it has been split on this space.\n\nThis change allows you to use a glob that matches files with whitespace\nin the name. It does this by quoting the path.\n\nChange-Id: Id2cc32e4e40c1f834b19756e922118d8526358d3\nFixes-Bug: 1810934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/50356da8393abdb47cc6c64de4fa387d8499cbc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/50356da8393abdb47cc6c64de4fa387d8499cbc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
