)]}'
{"id":"openstack%2Fpbr~647492","triplet_id":"openstack%2Fpbr~master~I7982f9d40cb0979ddb89d7bc53964167f8e4b269","project":"openstack/pbr","branch":"master","topic":"subparser_required","hashtags":[],"change_id":"I7982f9d40cb0979ddb89d7bc53964167f8e4b269","subject":"Set subparser argument required","status":"MERGED","created":"2019-03-25 14:43:37.000000000","updated":"2019-05-23 17:02:50.000000000","submitted":"2019-05-23 17:02:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"647492-1558630970337-81eb7413","meta_rev_id":"65d516a0c72a749fb90015ada74facec583c0fd8","_number":647492,"virtual_id_number":647492,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-05-23 12:58:59.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-05-23 17:02:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-05-23 10:42:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-05-23 10:42:08.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":1,"date":"2019-05-23 10:42:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":2,"date":"2019-05-23 15:16:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-23 10:42:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":1,"date":"2019-05-23 15:16:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-25 18:49:31.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-03-27 12:52:03.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-04-02 15:23:24.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-04-02 15:28:30.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-05-20 15:39:13.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2019-05-23 15:16:29.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-23 17:02:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f69affa7057d60c78fa296904abf0074a586dcf3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-03-25 14:43:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2474f18c83576af27d1190107c8bf90de9c3e18","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-03-25 14:55:56.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"dacd4e11d3a5d4f8efabbaf17ab4f045e63459b3","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-03-25 15:23:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"521d9608c57e4b4e831589f53d8ace6fec82b141","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-03-25 15:43:07.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nYour way doesn\u0027t work with python2 and mine work fine the python2 and 3:\n- https://docs.python.org/2/library/argparse.html#sub-commands\n- http://www.howtobuildsoftware.com/index.php/how-do/cyFw/python-argparse-python-34-argparse-with-required-subparser\n\n\nPython 2 add_subparser method doesn\u0027t accept the `required` parameter.","accounts_in_message":[],"_revision_number":2},{"id":"e12716899e1cc3556c87684cc55207867ca0b54d","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-03-25 16:12:37.000000000","message":"Patch Set 2: Code-Review+1\n\nSorry, I have learnt something.","accounts_in_message":[],"_revision_number":2},{"id":"c2c9814bc052a7afe74b838635be9187c76eb7cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 17:09:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/92/647492/2/check/tempest-full/7831652/ : SUCCESS in 1h 51m 01s\n- tempest-full-py3 http://logs.openstack.org/92/647492/2/check/tempest-full-py3/e3c0c63/ : SUCCESS in 1h 39m 42s\n- openstack-tox-cover http://logs.openstack.org/92/647492/2/check/openstack-tox-cover/315c790/cover/ : SUCCESS in 8m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/647492/2/check/openstack-tox-lower-constraints/86a3597/ : SUCCESS in 7m 54s\n- openstack-tox-pep8 http://logs.openstack.org/92/647492/2/check/openstack-tox-pep8/7debe95/ : SUCCESS in 5m 07s\n- openstack-tox-py27 http://logs.openstack.org/92/647492/2/check/openstack-tox-py27/a0b1f45/ : SUCCESS in 5m 52s\n- openstack-tox-py35 http://logs.openstack.org/92/647492/2/check/openstack-tox-py35/603177e/ : SUCCESS in 5m 57s\n- openstack-tox-py36 http://logs.openstack.org/92/647492/2/check/openstack-tox-py36/b66c7d0/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/92/647492/2/check/openstack-tox-docs/ed42226/html/ : SUCCESS in 4m 53s\n- pbr-installation-devstack http://logs.openstack.org/92/647492/2/check/pbr-installation-devstack/28070f9/ : SUCCESS in 35m 05s\n- pbr-installation-upstream-devstack http://logs.openstack.org/92/647492/2/check/pbr-installation-upstream-devstack/a943329/ : SUCCESS in 37m 34s","accounts_in_message":[],"_revision_number":2},{"id":"fa9558b0ccf2354eb35bb20ae1ed22bf9ad648c9","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-03-25 18:49:31.000000000","message":"Patch Set 2: Code-Review-1\n\nIt seems that calling pbr without parameters triggers:\n\n$ pbr                                                                                                                                                [18:49:00]\nTraceback (most recent call last):\nFile \"/Users/ssbarnea/.pyenv/versions/3.6.7/bin/pbr\", line 10, in \u003cmodule\u003e\nsys.exit(main())\nFile \"/Users/ssbarnea/os/pbr/pbr/cmd/main.py\", line 105, in main\nargs \u003d parser.parse_args()\nFile \"/Users/ssbarnea/.pyenv/versions/3.6.7/lib/python3.6/argparse.py\", line 1734, in parse_args\nargs, argv \u003d self.parse_known_args(args, namespace)\nFile \"/Users/ssbarnea/.pyenv/versions/3.6.7/lib/python3.6/argparse.py\", line 1766, in parse_known_args\nnamespace, args \u003d self._parse_known_args(args, namespace)\nFile \"/Users/ssbarnea/.pyenv/versions/3.6.7/lib/python3.6/argparse.py\", line 2001, in _parse_known_args\n\u0027, \u0027.join(required_actions))\nTypeError: sequence item 0: expected str instance, NoneType found\nFAIL: 1","accounts_in_message":[],"_revision_number":2},{"id":"543531f899c97507043fd9f0026ffbe3312484a0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-03-26 08:51:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c06b65864f13f652b459cd53fedb0d84bd49cd17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 10:43:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/92/647492/3/check/tempest-full/89633cb/ : SUCCESS in 1h 49m 49s\n- tempest-full-py3 http://logs.openstack.org/92/647492/3/check/tempest-full-py3/c2c2634/ : SUCCESS in 1h 36m 20s\n- openstack-tox-cover http://logs.openstack.org/92/647492/3/check/openstack-tox-cover/cfa47eb/cover/ : SUCCESS in 7m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/647492/3/check/openstack-tox-lower-constraints/423f1ee/ : SUCCESS in 7m 43s\n- openstack-tox-pep8 http://logs.openstack.org/92/647492/3/check/openstack-tox-pep8/2286aee/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/92/647492/3/check/openstack-tox-py27/ee20708/ : SUCCESS in 4m 47s\n- openstack-tox-py35 http://logs.openstack.org/92/647492/3/check/openstack-tox-py35/06723c2/ : SUCCESS in 6m 13s\n- openstack-tox-py36 http://logs.openstack.org/92/647492/3/check/openstack-tox-py36/c5073cb/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/92/647492/3/check/openstack-tox-docs/0f6f989/html/ : SUCCESS in 4m 06s\n- pbr-installation-devstack http://logs.openstack.org/92/647492/3/check/pbr-installation-devstack/8a5a9c6/ : SUCCESS in 50m 35s\n- pbr-installation-upstream-devstack http://logs.openstack.org/92/647492/3/check/pbr-installation-upstream-devstack/bbf8a68/ : SUCCESS in 46m 57s","accounts_in_message":[],"_revision_number":3},{"id":"b97cdafc9c9ab7f894d5d4f171403edce957d1df","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-03-27 12:52:03.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"403a787bcb37a70366ff4641808c4d624fb034bd","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-02 15:23:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d4d47c872ea31d6ce7555957e967569b7f6ade28","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-04-02 15:28:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"baf99d2ec2631b58e2c4e794f67ef3358c905300","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-05-20 15:39:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"438e4fb67ee6cafbfd8909bf26567ee9d356c392","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-20 15:41:59.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e4b6ac6d77a21893225490d0567c90b6f45ca479","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 15:42:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0db2d93b87782791554a1873db5627786b437bf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 17:26:37.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/92/647492/3/gate/tempest-full/afb0357/ : SUCCESS in 1h 37m 49s\n- tempest-full-py3 http://logs.openstack.org/92/647492/3/gate/tempest-full-py3/688c88b/ : SUCCESS in 1h 34m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/647492/3/gate/openstack-tox-lower-constraints/62b3402/ : SUCCESS in 7m 13s\n- openstack-tox-pep8 http://logs.openstack.org/92/647492/3/gate/openstack-tox-pep8/0fc38b5/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/92/647492/3/gate/openstack-tox-py27/1b00092/ : FAILURE in 5m 19s\n- openstack-tox-py35 http://logs.openstack.org/92/647492/3/gate/openstack-tox-py35/267d97f/ : FAILURE in 6m 07s\n- openstack-tox-py36 http://logs.openstack.org/92/647492/3/gate/openstack-tox-py36/56cd548/ : FAILURE in 4m 52s\n- openstack-tox-docs http://logs.openstack.org/92/647492/3/gate/openstack-tox-docs/c8af6f6/html/ : SUCCESS in 4m 54s\n- pbr-installation-devstack http://logs.openstack.org/92/647492/3/gate/pbr-installation-devstack/0bcf7c9/ : FAILURE in 35m 57s\n- pbr-installation-upstream-devstack http://logs.openstack.org/92/647492/3/gate/pbr-installation-upstream-devstack/c6880b3/ : FAILURE in 30m 08s","accounts_in_message":[],"_revision_number":3},{"id":"911c0143aae3751f5b9d27fcb12834b9aa905c07","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-21 07:06:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7c4ad37e604442e385d948ac2d6a58552c76c6e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 09:06:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/92/647492/3/check/tempest-full/a22fbcb/ : SUCCESS in 1h 55m 26s\n- tempest-full-py3 http://logs.openstack.org/92/647492/3/check/tempest-full-py3/02a53bf/ : SUCCESS in 1h 34m 03s\n- openstack-tox-cover http://logs.openstack.org/92/647492/3/check/openstack-tox-cover/04631f8/cover/ : SUCCESS in 7m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/647492/3/check/openstack-tox-lower-constraints/e649251/ : SUCCESS in 5m 32s\n- openstack-tox-pep8 http://logs.openstack.org/92/647492/3/check/openstack-tox-pep8/d5914f6/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/92/647492/3/check/openstack-tox-py27/98e892f/ : SUCCESS in 5m 02s\n- openstack-tox-py35 http://logs.openstack.org/92/647492/3/check/openstack-tox-py35/78e3522/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/92/647492/3/check/openstack-tox-py36/5f3a527/ : SUCCESS in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/92/647492/3/check/openstack-tox-docs/fce2f7a/html/ : SUCCESS in 4m 10s\n- pbr-installation-devstack http://logs.openstack.org/92/647492/3/check/pbr-installation-devstack/26e488f/ : FAILURE in 34m 43s\n- pbr-installation-upstream-devstack http://logs.openstack.org/92/647492/3/check/pbr-installation-upstream-devstack/e936ecf/ : FAILURE in 39m 19s","accounts_in_message":[],"_revision_number":3},{"id":"3df35e7867ae2f1204ef812b778c46e932caeab9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-21 09:10:25.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1f1d1b700ac364ae9011c03ecb3f92f4cf759f67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 11:13:00.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/92/647492/3/check/tempest-full/8eefd46/ : SUCCESS in 1h 49m 36s\n- tempest-full-py3 http://logs.openstack.org/92/647492/3/check/tempest-full-py3/45b23da/ : SUCCESS in 1h 42m 16s\n- openstack-tox-cover http://logs.openstack.org/92/647492/3/check/openstack-tox-cover/8c89e6c/cover/ : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/647492/3/check/openstack-tox-lower-constraints/737a952/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/92/647492/3/check/openstack-tox-pep8/0e5585c/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/92/647492/3/check/openstack-tox-py27/e32f76d/ : SUCCESS in 5m 08s\n- openstack-tox-py35 http://logs.openstack.org/92/647492/3/check/openstack-tox-py35/442153f/ : SUCCESS in 4m 51s\n- openstack-tox-py36 http://logs.openstack.org/92/647492/3/check/openstack-tox-py36/2e6124d/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/92/647492/3/check/openstack-tox-docs/5b76c4a/html/ : SUCCESS in 4m 05s\n- pbr-installation-devstack http://logs.openstack.org/92/647492/3/check/pbr-installation-devstack/22a5d95/ : FAILURE in 34m 26s\n- pbr-installation-upstream-devstack http://logs.openstack.org/92/647492/3/check/pbr-installation-upstream-devstack/18f0727/ : FAILURE in 35m 38s","accounts_in_message":[],"_revision_number":3},{"id":"fb38f6be54608c4a73400a323cb7a7d18d3346cb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-21 12:19:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8f1e5d8d0a12ad1759ab5b276fde162649073a7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 14:08:25.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/92/647492/3/check/tempest-full/28797ab/ : SUCCESS in 1h 41m 08s\n- tempest-full-py3 http://logs.openstack.org/92/647492/3/check/tempest-full-py3/8a47579/ : SUCCESS in 1h 34m 15s\n- openstack-tox-cover http://logs.openstack.org/92/647492/3/check/openstack-tox-cover/9a68886/cover/ : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/647492/3/check/openstack-tox-lower-constraints/c8340d3/ : FAILURE in 6m 48s\n- openstack-tox-pep8 http://logs.openstack.org/92/647492/3/check/openstack-tox-pep8/b820c5a/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/92/647492/3/check/openstack-tox-py27/2b8bab9/ : FAILURE in 7m 52s\n- openstack-tox-py35 http://logs.openstack.org/92/647492/3/check/openstack-tox-py35/cadaa22/ : FAILURE in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/92/647492/3/check/openstack-tox-py36/b0130c8/ : FAILURE in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/92/647492/3/check/openstack-tox-docs/be3a2d0/html/ : SUCCESS in 3m 50s\n- pbr-installation-devstack http://logs.openstack.org/92/647492/3/check/pbr-installation-devstack/3979255/ : FAILURE in 33m 12s\n- pbr-installation-upstream-devstack http://logs.openstack.org/92/647492/3/check/pbr-installation-upstream-devstack/4267c3f/ : FAILURE in 34m 54s","accounts_in_message":[],"_revision_number":3},{"id":"395f8b511f5ca1f0f8f2465a50971b06aa912e6d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-22 15:27:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"06e54f1e57af99154d9b81a4a76b3c6844e76e8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 18:45:42.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full finger://ze01.openstack.org/cbe377dfbf29455c8ea031dc5ead7fa8 : RETRY_LIMIT in 36m 41s\n- tempest-full-py3 finger://ze02.openstack.org/07dea665e1784ab0ae7d32c0701ab497 : RETRY_LIMIT in 48m 31s\n- openstack-tox-cover http://logs.openstack.org/92/647492/3/check/openstack-tox-cover/6189a29/cover/ : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/647492/3/check/openstack-tox-lower-constraints/25687ee/ : SUCCESS in 5m 33s\n- openstack-tox-pep8 http://logs.openstack.org/92/647492/3/check/openstack-tox-pep8/069c1a2/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/92/647492/3/check/openstack-tox-py27/cc267c3/ : SUCCESS in 4m 51s\n- openstack-tox-py35 http://logs.openstack.org/92/647492/3/check/openstack-tox-py35/c4ee4a9/ : SUCCESS in 3m 55s\n- openstack-tox-py36 http://logs.openstack.org/92/647492/3/check/openstack-tox-py36/3556f87/ : FAILURE in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/92/647492/3/check/openstack-tox-docs/da9e86d/html/ : SUCCESS in 4m 32s\n- pbr-installation-devstack http://logs.openstack.org/92/647492/3/check/pbr-installation-devstack/bc472f4/ : SUCCESS in 35m 14s\n- pbr-installation-upstream-devstack http://logs.openstack.org/92/647492/3/check/pbr-installation-upstream-devstack/3bb4f69/ : SUCCESS in 35m 18s","accounts_in_message":[],"_revision_number":3},{"id":"1ba2bbdb35dc1a9c43f46742ef024a7fe6a33403","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-23 09:39:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5a306f8db7ee5470c1582eccb85b5ff9d77ebe9c","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-05-23 10:42:08.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"3a36e7f8927d8248c82392d0c9660767b0cdc7c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 12:53:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/92/647492/4/check/tempest-full/4c962c2/ : TIMED_OUT in 2h 06m 38s\n- tempest-full-py3 http://logs.openstack.org/92/647492/4/check/tempest-full-py3/8f8beb7/ : SUCCESS in 1h 26m 30s\n- openstack-tox-cover http://logs.openstack.org/92/647492/4/check/openstack-tox-cover/1cd8ccf/cover/ : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/647492/4/check/openstack-tox-lower-constraints/1652fde/ : SUCCESS in 5m 56s\n- openstack-tox-pep8 http://logs.openstack.org/92/647492/4/check/openstack-tox-pep8/460179d/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/92/647492/4/check/openstack-tox-py27/dc76fbf/ : SUCCESS in 5m 18s\n- openstack-tox-py35 http://logs.openstack.org/92/647492/4/check/openstack-tox-py35/bfa22fb/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/92/647492/4/check/openstack-tox-py36/c45e191/ : SUCCESS in 4m 45s\n- openstack-tox-docs http://logs.openstack.org/92/647492/4/check/openstack-tox-docs/2ae675a/html/ : SUCCESS in 4m 24s\n- pbr-installation-devstack http://logs.openstack.org/92/647492/4/check/pbr-installation-devstack/b30721d/ : SUCCESS in 32m 46s\n- pbr-installation-upstream-devstack http://logs.openstack.org/92/647492/4/check/pbr-installation-upstream-devstack/62fd93b/ : SUCCESS in 38m 40s","accounts_in_message":[],"_revision_number":4},{"id":"7574b1835c4205f7055a113bf52c5a4e03fcca29","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-23 12:58:59.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2201621e63a80d767b68aec09b472027bb2245c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 15:12:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/92/647492/4/check/tempest-full/6dbe89c/ : SUCCESS in 1h 48m 16s\n- tempest-full-py3 http://logs.openstack.org/92/647492/4/check/tempest-full-py3/b091960/ : SUCCESS in 1h 41m 10s\n- openstack-tox-cover http://logs.openstack.org/92/647492/4/check/openstack-tox-cover/5a320f7/cover/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/647492/4/check/openstack-tox-lower-constraints/b6433ed/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/92/647492/4/check/openstack-tox-pep8/97cdb14/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/92/647492/4/check/openstack-tox-py27/f9f538c/ : SUCCESS in 4m 19s\n- openstack-tox-py35 http://logs.openstack.org/92/647492/4/check/openstack-tox-py35/807bb1d/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/92/647492/4/check/openstack-tox-py36/7441337/ : SUCCESS in 4m 51s\n- openstack-tox-docs http://logs.openstack.org/92/647492/4/check/openstack-tox-docs/237ced6/html/ : SUCCESS in 3m 51s\n- pbr-installation-devstack http://logs.openstack.org/92/647492/4/check/pbr-installation-devstack/f521e27/ : SUCCESS in 35m 06s\n- pbr-installation-upstream-devstack http://logs.openstack.org/92/647492/4/check/pbr-installation-upstream-devstack/17be0f4/ : SUCCESS in 34m 40s","accounts_in_message":[],"_revision_number":4},{"id":"bb18185e8664742577aa5fa57d6f223a7f824524","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-23 15:16:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8e4d87c6ac983c47a4dcafb8fa53697dbea2b77f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 15:16:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3297051135a5ada5751513b7de62d0644c547b7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 17:02:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9be0cfbb3a0d218c3a061ecbf8f791f45f49f620","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 17:02:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/92/647492/4/gate/tempest-full/8eedfa1/ : SUCCESS in 1h 43m 28s\n- tempest-full-py3 http://logs.openstack.org/92/647492/4/gate/tempest-full-py3/838326f/ : SUCCESS in 1h 31m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/647492/4/gate/openstack-tox-lower-constraints/2f24c27/ : SUCCESS in 5m 46s\n- openstack-tox-pep8 http://logs.openstack.org/92/647492/4/gate/openstack-tox-pep8/fe175c9/ : SUCCESS in 5m 58s\n- openstack-tox-py27 http://logs.openstack.org/92/647492/4/gate/openstack-tox-py27/48022ae/ : SUCCESS in 4m 52s\n- openstack-tox-py35 http://logs.openstack.org/92/647492/4/gate/openstack-tox-py35/b34ca13/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/92/647492/4/gate/openstack-tox-py36/d5fe363/ : SUCCESS in 4m 52s\n- openstack-tox-py37 http://logs.openstack.org/92/647492/4/gate/openstack-tox-py37/7b3ec4d/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/92/647492/4/gate/openstack-tox-docs/8deab93/html/ : SUCCESS in 3m 42s\n- pbr-installation-devstack http://logs.openstack.org/92/647492/4/gate/pbr-installation-devstack/4fbcc11/ : SUCCESS in 32m 49s\n- pbr-installation-upstream-devstack http://logs.openstack.org/92/647492/4/gate/pbr-installation-upstream-devstack/e59b8dd/ : SUCCESS in 36m 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"713aff286b124aeaa6b2bea3b0d70488b43de4a5","revisions":{"a16919f02c6e6c0be3785bfa79c818fae7eebb44":{"kind":"REWORK","_number":1,"created":"2019-03-25 14:43:37.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/92/647492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/92/647492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/92/647492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/92/647492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/92/647492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/92/647492/1"}}},"commit":{"parents":[{"commit":"fd045e6e40aa6208132ee5a3e189701aa528f843","subject":"Merge \"Fix nits and typos on release note message.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fd045e6e40aa6208132ee5a3e189701aa528f843"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-03-25 14:34:56.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-03-25 14:42:09.000000000","tz":60},"subject":"Set subparser argument required","message":"Set subparser argument required\n\nWhen subparser argument is not provided by user\nargparse return an error message not really useful for user:\n\n    \u0027Namespace\u0027 object has no attribute \u0027func\u0027\n\nThis is due to the fact that when we launch the pbr in cli mode\nthe subparser argument is not mandatory (required) and directly\nwe try to execute a undefined function.\n\nSet the subparser required is more helpful for users due to the\nfact that argparse display the helping message with the available\nsub-commands that users can use\n\nChange-Id: I7982f9d40cb0979ddb89d7bc53964167f8e4b269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a16919f02c6e6c0be3785bfa79c818fae7eebb44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a16919f02c6e6c0be3785bfa79c818fae7eebb44"}]},"branch":"refs/heads/master"},"664bbe23e3d22056c18909b29d8276a14fd70426":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-03-25 14:55:56.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/92/647492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/92/647492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/92/647492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/92/647492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/92/647492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/92/647492/2"}}},"commit":{"parents":[{"commit":"fd045e6e40aa6208132ee5a3e189701aa528f843","subject":"Merge \"Fix nits and typos on release note message.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fd045e6e40aa6208132ee5a3e189701aa528f843"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-03-25 14:34:56.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-03-25 14:54:19.000000000","tz":60},"subject":"Set subparser argument required","message":"Set subparser argument required\n\nWhen subparser argument is not provided by user\nargparse return an error message not really useful for user:\n\n    \u0027Namespace\u0027 object has no attribute \u0027func\u0027\n\nThis is due to the fact that when we launch the pbr in cli mode\nthe subparser argument is not mandatory (required) and directly\nwe try to execute a undefined function.\n\nSet the subparser required is more helpful for users due to the\nfact that argparse display the helping message with the available\nsub-commands that users can use\n\nThese changes provides the following output if the argument is not\npassed:\n\nusage: pbr [-h] [-v] {sha,info,freeze} ...\nmain.py: error: too few arguments\n\nChange-Id: I7982f9d40cb0979ddb89d7bc53964167f8e4b269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/664bbe23e3d22056c18909b29d8276a14fd70426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/664bbe23e3d22056c18909b29d8276a14fd70426"}]},"branch":"refs/heads/master"},"185a871973cc72d91c8cd7a494542ae55cba38f3":{"kind":"REWORK","_number":3,"created":"2019-03-26 08:51:56.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/92/647492/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/92/647492/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/92/647492/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/92/647492/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/92/647492/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/92/647492/3"}}},"commit":{"parents":[{"commit":"fd045e6e40aa6208132ee5a3e189701aa528f843","subject":"Merge \"Fix nits and typos on release note message.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fd045e6e40aa6208132ee5a3e189701aa528f843"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-03-25 14:34:56.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-03-26 08:51:45.000000000","tz":60},"subject":"Set subparser argument required","message":"Set subparser argument required\n\nWhen subparser argument is not provided by user\nargparse return an error message not really useful for user:\n\n    \u0027Namespace\u0027 object has no attribute \u0027func\u0027\n\nThis is due to the fact that when we launch the pbr in cli mode\nthe subparser argument is not mandatory (required) and directly\nwe try to execute a undefined function.\n\nSet the subparser required is more helpful for users due to the\nfact that argparse display the helping message with the available\nsub-commands that users can use\n\nThese changes provides the following output if the argument is not\npassed:\n\nusage: pbr [-h] [-v] {sha,info,freeze} ...\nmain.py: error: too few arguments\n\nChange-Id: I7982f9d40cb0979ddb89d7bc53964167f8e4b269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/185a871973cc72d91c8cd7a494542ae55cba38f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/185a871973cc72d91c8cd7a494542ae55cba38f3"}]},"branch":"refs/heads/master"},"713aff286b124aeaa6b2bea3b0d70488b43de4a5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-23 10:42:08.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/92/647492/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/92/647492/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/92/647492/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/92/647492/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/92/647492/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/92/647492/4"}}},"commit":{"parents":[{"commit":"30926f6dbe05996009d407750acb5ea52a039e62","subject":"Update Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/30926f6dbe05996009d407750acb5ea52a039e62"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-03-25 14:34:56.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-05-23 10:42:08.000000000","tz":0},"subject":"Set subparser argument required","message":"Set subparser argument required\n\nWhen subparser argument is not provided by user\nargparse return an error message not really useful for user:\n\n    \u0027Namespace\u0027 object has no attribute \u0027func\u0027\n\nThis is due to the fact that when we launch the pbr in cli mode\nthe subparser argument is not mandatory (required) and directly\nwe try to execute a undefined function.\n\nSet the subparser required is more helpful for users due to the\nfact that argparse display the helping message with the available\nsub-commands that users can use\n\nThese changes provides the following output if the argument is not\npassed:\n\nusage: pbr [-h] [-v] {sha,info,freeze} ...\nmain.py: error: too few arguments\n\nChange-Id: I7982f9d40cb0979ddb89d7bc53964167f8e4b269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/713aff286b124aeaa6b2bea3b0d70488b43de4a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/713aff286b124aeaa6b2bea3b0d70488b43de4a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
